)]}'
{"id":"openstack%2Foctavia~846547","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79","project":"openstack/octavia","branch":"stable/yoga","topic":"3cf866-yoga","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-29 19:18:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79","subject":"Validate L7Rule value and cookie name","status":"MERGED","created":"2022-06-20 06:41:46.000000000","updated":"2022-06-29 19:19:35.000000000","submitted":"2022-06-29 19:18:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846547-3cf866-yoga","meta_rev_id":"88ce6984d0b23513e03148df0ff3357adb02386d","_number":846547,"virtual_id_number":846547,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 19:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-06-20 11:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-06-29 16:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-29 16:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 08:56:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 11:58:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-06-29 16:50:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"48e04a0362170ff2bf97fe69a348861c0a99adb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-20 06:41:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e696d53ebd4cce6dfa577f9140eb477c43194455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 08:56:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83051141e0d24356a3f5d35116c767fc : SUCCESS in 12m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e855f78ccd0d4622ae7a726acbb1b38f : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f0296dd6594c59bd43444e9a70ea1f : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc884565418247ec8a8850b52f8bad1d : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e29239e4ef1b4060b0082f1b4012a099 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d40aecc123420bbd7560ce74d77258 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca1e494e4214602b576b02b08fcfaaa : SUCCESS in 5m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/380aa61610d8480da3b51ad6444df5eb : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/76bb4251e21643079fc28edac9255c9b : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/317a4033653e42adabdb5125f6998a82 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ece83837fbb44b7b8845d35ddc26dec0 : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e41c6a2104c45aeabdf5d38eb13909d : SUCCESS in 1h 33m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c1fdd2d72304fd59b9f4ff340584a40 : SUCCESS in 1h 35m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4ec8d3b85c242a4b82e80971af9f1cc : SUCCESS in 35m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/825cda133a244b9e9aa0b615a4c1c7ec : SUCCESS in 51m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c2fed58b5ccd4745bccdc198465451ed : RETRY_LIMIT in 8m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/015eaecf96e84c4e892583ed0f273b0a : SUCCESS in 40m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/474b4ec3788e4e6c94d4623b688772ac : SUCCESS in 1h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62abffed081c4421b9f9d52a3342bba8 : SUCCESS in 2h 02m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10c4782dda93485389d10f9d90f9084c : SUCCESS in 1h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d1c868a345224a4f9361eaebdb9561c8 : SUCCESS in 1h 27m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1fa4bb6ccde54a94852e89050195f329 : SUCCESS in 1h 25m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/60c349e0c5c6452d9c46897a0894ed36 : FAILURE in 35m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b690aa0a3691daf7c2dc76cbd66ce0bada095d3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 11:58:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba6fb89707bdadad589ca727155b3f8c099913cd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-20 13:36:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2d1214e5ead689c6657a906391a2fd38f1347d4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:50:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c4c1ef87a77f8ba1a2f2db2a01000f5b11ce2d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 16:52:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d12cf7bb18d346f35883d75c720aa62555ea1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94d6eabf798a426f9f533232920d51fd : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac14de935f0740eca03ea67c50fe635d : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8fc30db92d4294b33724e0075a4058 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0918d6ec9a840059cff142dbe43a230 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc1484557f642de9699d30d4defce4c : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/746b84f0eb624da6bee88d3b7f80cac6 : SUCCESS in 5m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64e34a7039de47e887366dba17c173a1 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e605cc9e71934ee9b58d99324fe7f4f5 : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d28dc05f95d84713ac0527bbc438bcf2 : SUCCESS in 2h 09m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e80d1adf3a045b38aada8b8e3a848bc : SUCCESS in 1h 19m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3bac18a1b64142248bf58f9a86fa7ff4 : SUCCESS in 34m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a414905a97d48fe8ff12971287aed97 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":1},{"id":"157cdf7fa844f41465a3e603b01559018167b338","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88ce6984d0b23513e03148df0ff3357adb02386d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:19:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56da1a26ad0e4bfbb83ff68a0da39676 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38885015ade243fa95a5c6271aecf15e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"133ec4763db7889179ee09b897f503eb68ccd42e","revisions":{"133ec4763db7889179ee09b897f503eb68ccd42e":{"kind":"REWORK","_number":1,"created":"2022-06-20 06:41:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/47/846547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/846547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/846547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/846547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/846547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/846547/1"}}},"commit":{"parents":[{"commit":"399c2f17504fa156e65c664cc4bfeb1c98f7d9d0","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/399c2f17504fa156e65c664cc4bfeb1c98f7d9d0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-25 13:49:38.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-20 06:41:24.000000000","tz":120},"subject":"Validate L7Rule value and cookie name","message":"Validate L7Rule value and cookie name\n\nAdds validations in L7 rule and session cookie APIs in order to prevent\nauthenticated and authorized users to inject code into HAProxy\nconfiguration. CR and LF (\\r and \\n) are no longer allowed in L7\nrule keys and values. The session persistence cookie names must follow\nthe rules described in\nhttps://developer.mozilla.org/en-US/docs/Web/HTTP/Headers/Set-Cookie.\n\nStory: 2008994\nTask: 44859\nChange-Id: Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79\n(cherry picked from commit 3cf866dbc096ff2b6136d1e06da3f668f6ffeaaa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/133ec4763db7889179ee09b897f503eb68ccd42e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/133ec4763db7889179ee09b897f503eb68ccd42e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
