)]}'
{"id":"openstack%2Foctavia~846570","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79","project":"openstack/octavia","branch":"stable/victoria","topic":"3cf866-victoria","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-29 19:17:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79","subject":"Validate L7Rule value and cookie name","status":"MERGED","created":"2022-06-20 06:42:59.000000000","updated":"2022-06-29 19:19:57.000000000","submitted":"2022-06-29 19:17:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846570-3cf866-victoria","meta_rev_id":"29b1cb1a67f5e5daa8a826018347e6b745c755fe","_number":846570,"virtual_id_number":846570,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 19:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-06-20 12:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-06-29 16:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-06-29 16:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 09:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 12:00:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-06-29 16:50:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a30ef6c05a2e1b8e101e151ccb4980142bdab87f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-20 06:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2441bc9989fdbf80efe99f2f17ffdd44a60ada40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:10:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfcd4bbf7cb84000b7f1396bd200522e : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d828d44e33a44df3a95debe67f5833c0 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd2256db7174f99aadbcbd6510178a4 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c959249090554a1e893c8338b510b3e6 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f77ae8eabd4624a37447acf3d3aa49 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42321406929a4dfcb706dab3d769cb32 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/800eff5ff3fb48f1959100b124783a80 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/16f1e390ac78431787380e22c6360bb2 : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf9ed143c89a4b4d8f4a1341e6e5c964 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/433305b22c7f4a2ebb6dd5689d4d6885 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bec9cc51dbfe460c8852f988b0428fc7 : SUCCESS in 1h 38m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/411a6d077aa8456e950aca462345e618 : SUCCESS in 1h 39m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82da3cc5ed2a45f284923feeaae686f9 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39a3b290b9574263b6186854de0bbdd0 : SUCCESS in 29m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/729ba0a3870442cb9e28dd33142f2f40 : SUCCESS in 33m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/48b34152968d458e88bd73184a36419d : SUCCESS in 2h 08m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc22709684294eba9dc96533fffa714a : FAILURE in 30m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e432d49bb51f423584ac7679c32b1e84 : SUCCESS in 1h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6cc58aba9e3842bab2da742861fad48f : SUCCESS in 2h 15m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e693f0af6e5e41af8ebb23b01bc7009a : SUCCESS in 1h 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ae12e59605864a2185dd04a22ef7ddbe : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54ed73c3ec824605a1b03730a1722d8c606e115b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 12:00:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86a66870ff1b96c976362079a45391cd59f54ee8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:50:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e97dae91fda8d43892858fc2a07b765171bfe235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 16:52:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d00a3c28d67f92c44ad6a30f24eefdd070f3fc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:17:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9876985f34274b75a6abba80c8868eba : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cd83f8e26294eb5b2ca7faaab706d3b : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4584fbc46cd84f8cb06ff04a72c01f25 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d14625773847d0b0925b0dccc74135 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae398ecd37f4efc991fbccbfe59e954 : SUCCESS in 5m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7a45e66284940c799ffc5b7a293e3ef : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29de897623e7434d9bf5fff4bfd026be : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00e7e0884b1b47b9b83a380e423c038e : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e682cb3000e464bbdeccd33b6aafd3e : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a22d9b01d7494bf18a5f7c5455e25a0d : SUCCESS in 36m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/81bb3847bd1c4c1095eaa6918d268dba : SUCCESS in 33m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08836d0ea610487ba0561a440ec213a3 : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8528ee65d63bdc9ea47d1a6b53fff4e8eeae98eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:17:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"29b1cb1a67f5e5daa8a826018347e6b745c755fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:19:57.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a223ddd8376948dfbc437f2c6a4cf29d : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53af114c6abb46ca88430f2d35026bbf : FAILURE in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9736bb8f4a9011aec1780aea902a46ec83b9a9d0","revisions":{"9736bb8f4a9011aec1780aea902a46ec83b9a9d0":{"kind":"REWORK","_number":1,"created":"2022-06-20 06:42:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/846570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/846570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/846570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/846570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/846570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/846570/1"}}},"commit":{"parents":[{"commit":"f224fa3294844abd3605270e0ee7debf4a41215d","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f224fa3294844abd3605270e0ee7debf4a41215d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-25 13:49:38.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-20 06:42:41.000000000","tz":120},"subject":"Validate L7Rule value and cookie name","message":"Validate L7Rule value and cookie name\n\nAdds validations in L7 rule and session cookie APIs in order to prevent\nauthenticated and authorized users to inject code into HAProxy\nconfiguration. CR and LF (\\r and \\n) are no longer allowed in L7\nrule keys and values. The session persistence cookie names must follow\nthe rules described in\nhttps://developer.mozilla.org/en-US/docs/Web/HTTP/Headers/Set-Cookie.\n\nStory: 2008994\nTask: 44859\nChange-Id: Ic370e9edc3fb5548e9cf0d66b85df66e01c41e79\n(cherry picked from commit 3cf866dbc096ff2b6136d1e06da3f668f6ffeaaa)\n(cherry picked from commit 133ec4763db7889179ee09b897f503eb68ccd42e)\n(cherry picked from commit c30aa9df228548f30ff7609ee8f8f3cb36b5f89d)\n(cherry picked from commit 6c52fd0e46c1a1d2b7f631af1c1e393c19a69a06)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9736bb8f4a9011aec1780aea902a46ec83b9a9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9736bb8f4a9011aec1780aea902a46ec83b9a9d0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
