)]}'
{"id":"openstack%2Foctavia~846584","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I7d92b011455d95a2df7782984b8f4648d4b00941","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-17 15:11:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d92b011455d95a2df7782984b8f4648d4b00941","subject":"Fix disabled UDP pools","status":"MERGED","created":"2022-06-20 08:35:20.000000000","updated":"2022-11-17 15:13:07.000000000","submitted":"2022-11-17 15:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846584","meta_rev_id":"a4daa51b35e232b5e9560fdfc90f77462c78d7dd","_number":846584,"virtual_id_number":846584,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 15:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:01:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-17 13:18:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-17 13:18:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 10:35:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:49:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:01:35.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-11-17 13:18:27.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"4808f7fbe8bec50acc47df902e39e661302bc697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 08:35:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be54d11f22529b73d79838ae2fdb86794607df94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2073499e981e4b67b8cff8235e6a8bed : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6824bdbb1d497ba7dd41a6de3d8ed8 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d59d4cddfa6e47a79e2cedafe143bb89 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17b31df378174df39c7ecc98cc424633 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d65ce2e73e462ba65124f3b40eecf1 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0df3c6486bca48a88c86f7ce5476da9c : SUCCESS in 6m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/295a829288614c54b0282ce93d697114 : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c5efdc97a4c454a97d49156a59bf563 : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d222511f1ba84a678639f3a986c89133 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2242742bfffa4cfdb7d09b068c9ea6a9 : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a6da52911d74cf3a04401f754db53b8 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0009defc54c54ed698f315a3bf87851f : SUCCESS in 1h 42m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f97d59f1566d4cf0ac2c2679e6be204c : SUCCESS in 34m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/18168fe6b4c84b768bcdbb69e84e3274 : SUCCESS in 33m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c6d4929940b417fabe222639a7e97e4 : SUCCESS in 32m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64671d62e0e148c38548571f16769fe6 : SUCCESS in 1h 31m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96aba6e6dafa4e6fac9f4973a31d650d : FAILURE in 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb4d657030214d54b92e1ed30c3a1208 : POST_FAILURE in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a8cdf135bcca45e9b47232098b6e3f2b : FAILURE in 1h 29m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8fa0c446796c4c2a9c72bdab330160ea : FAILURE in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cef50366bbc541f9a19b08fece4fec60 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8103bc31e3226a7873e8a6ea5ae55101f52e71d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:49:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84f3ee545940cffcffd41f6ac5705be02dd56d77","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:01:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8c97a41c0d8f7d521b2e645126731add52778392","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-17 13:18:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abc8b5f5fc41637e5b999177df98b962032bf5f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:19:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40d49fa5de9a7a6979520467574dd81d02fa89bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:11:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d25493523d4f9597ee018dcf79323f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84264efaf90342bdbb880cc37a3b229a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d433edd0bd452cb00868e7d247c6b8 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e040173dac0b493284e82a2b8cd8ec2f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e2cb02a7674971ac0b15b10b52cd5d : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6697de6dc75459e9ed7a07c05575edb : SUCCESS in 5m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8aa93fd5fe7447bc8b3a12de48d86b2e : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b764c69daa144ed283fd5dad9a3ea182 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/479481ca77d445d1b3a170eacc0352a2 : SUCCESS in 54m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d574a72f17241ab85d7976511879863 : SUCCESS in 1h 49m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d8c83bd90484973b7c44ba5557166e9 : SUCCESS in 35m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ff262fcacaa64414abf4500e19c78c76 : SUCCESS in 35m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f11e32cae7ae4e44bc7ec51ff1a11c91 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e57bbbaf407ba8ada01975818098778e5f0e6891","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:11:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4daa51b35e232b5e9560fdfc90f77462c78d7dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:13:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1db8b8686684d68abc8db14f4dba206\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c521aab046af4f99beeeff8c9595a9a7 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c800298ee52c4e76b3d79ae750a51919 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8fb6a37bec74bd55e4b7de1299a3d3d0953386a","revisions":{"a8fb6a37bec74bd55e4b7de1299a3d3d0953386a":{"kind":"REWORK","_number":1,"created":"2022-06-20 08:35:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/84/846584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/846584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/846584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/846584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/846584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/846584/1"}}},"commit":{"parents":[{"commit":"f224fa3294844abd3605270e0ee7debf4a41215d","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f224fa3294844abd3605270e0ee7debf4a41215d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-26 07:23:07.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-20 08:32:15.000000000","tz":120},"subject":"Fix disabled UDP pools","message":"Fix disabled UDP pools\n\nDisabling a UDP pool didn\u0027t disable it correctly, requests were still\nforwarded to the members.\nNow, when disabling a UDP pool, the members are removed from the\nconfiguration file.\n\nStory 2010011\nTask 45164\n\nConflicts:\n\toctavia/common/jinja/lvs/templates/macros.j2\n\nChange-Id: I7d92b011455d95a2df7782984b8f4648d4b00941\n(cherry picked from commit f4b32ac313d751f81b90438cbedb7fd941322c61)\n(cherry picked from commit e397a118cc2d12e0ae586eb508bd28564fe89df7)\n(cherry picked from commit 839fa4581e2f2095b20bad940495c767187e91c0)\n(cherry picked from commit 5d45e56b47625095203e3f6dddfe3fb9f7019751)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8fb6a37bec74bd55e4b7de1299a3d3d0953386a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8fb6a37bec74bd55e4b7de1299a3d3d0953386a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
