)]}'
{"id":"openstack%2Foctavia~846586","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I7d92b011455d95a2df7782984b8f4648d4b00941","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-06-29 19:18:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d92b011455d95a2df7782984b8f4648d4b00941","subject":"Fix disabled UDP pools","status":"MERGED","created":"2022-06-20 08:36:21.000000000","updated":"2022-06-29 19:21:06.000000000","submitted":"2022-06-29 19:18:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846586","meta_rev_id":"5ab22523c24fb97736385482aaa88f9525b1c86b","_number":846586,"virtual_id_number":846586,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 19:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-27 06:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-29 16:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-06-29 16:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 09:47:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 06:51:30.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-29 16:59:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e09fc70ea01123ba32e52af9fbb6a26a05a23fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-20 08:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf403f845e297dcc4088d7ba937e65ec78499439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:47:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56722d03f2ad467181752053cb7ddcb8 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428cbd1d2ebd40e6a6b845770271ccc6 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44ad11990f6940e7b186c70e55632319 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b49dfa24f0f341cba91fd8717679a0a8 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a5868ab9d8a405fa7446346ca408fe8 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741c0a044af044e58bed3f74cde9d5a5 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f8a8c1e6a04bc3b6adfe7788fea7c7 : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce8063c8ed6e41658800fe59c17fe4eb : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d79c757b6cc448428de0801f3790b156 : SUCCESS in 8m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e30cc545f654ac6b8cacdfe12697256 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59648cdbae184bfd8d05b6506588ee33 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4309833fb1804d9ab117ccbb234ace23 : SUCCESS in 1h 04m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a190df845589414890cc350493707c5b : SUCCESS in 1h 09m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32db0ef87951482aa0b318baf5749a16 : FAILURE in 19m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d07ee6b262ae4d1b852c3804ca37dd61 : FAILURE in 20m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e5d49c1403e64c91af86abc0b6506b3e : FAILURE in 12m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5a73d396b564ebea35d95ce8f751d79 : RETRY_LIMIT in 6m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e3dcc6e252d47f2b6eaf3c67a22de36 : RETRY_LIMIT in 8m 33s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56670c6b5aa542f3a483c31f25cbaef7 : FAILURE in 21m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39ed2f4152d44e5da905727dcdfcbe49 : FAILURE in 20m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1212b545bea4e9baa9d312af37ae7a8 : FAILURE in 20m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/53d9f6d656b646c3a81113e37e4e6398 : FAILURE in 18m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/61fb63d5ccfd4e66afd964b49c20899c : FAILURE in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ac018af46eef8cd14d959a7d38e8a811882307b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-24 15:15:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf79038a0d6d02cdde600a695c63c03ac85594d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 16:53:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/676730772c1644e3a00a2feb52a2c430 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171f799806224927ad4134fdea96bff3 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/905a173d7fa64c038a656b26b049d4e8 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83b1932fe884dc0b93225cba6862bc6 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2714e53598524c0a9078e5c77f839074 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553b5965687944678dcca6c384e591ee : SUCCESS in 17m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e47e69ecc07439eb095ad1f904b1bf7 : SUCCESS in 5m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2e7f3a23442249ffaae6d65efa238842 : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/45c725bc5b934dbdb5e696004d812d28 : SUCCESS in 9m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f8b406b275f44a9fbb4f7782c7324fc5 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/541d97c14e644c1186efa306b90afbaa : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2324794701c54d1f8dd8a22ab1d5e582 : SUCCESS in 1h 01m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8b3487dbfd324f7c9c679647b8d1f82a : SUCCESS in 1h 08m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a092985649a4450cb02f877a18cf024e : SUCCESS in 1h 26m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0787aaa9e9ce4fa58a9a3afc371f2565 : SUCCESS in 1h 34m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8bb242eae7bb4071bc411637bdf5e9b5 : FAILURE in 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4bae96e336934c9dbbebb96fbb9f2b57 : RETRY_LIMIT in 9m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c88c1f044cf45b0b5a94da88121a14b : RETRY_LIMIT in 6m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b7981bf8d5a4de3be9e1ab9e4d5a4c3 : SUCCESS in 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a242661c70ff4cfc914f7ed171c101d0 : SUCCESS in 1h 25m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e42e0efd547c42e5bf73313d43389a59 : FAILURE in 29m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a273f1f5a2534336a8274843a86b2e49 : FAILURE in 31m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9ff604762cfd43e1ada0862904ab6414 : FAILURE in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb9beb3e4c5cbefe8e8866060501e2b6647a73f7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-27 06:51:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c42178fa55408d97c56ba33646b3299b960a3cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:59:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3d87d1be0e9921a0d3cc2ee2eca8f8913e8cfe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:00:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4abddc1c0f49f0079b60c4416532952f82c19642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5273f82e3aa64f39b84d3253a9e55518 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/579c5e6ae5c14c9ebcc41ea4910d65ad : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72c268679fd44aa19ae81a76ce463562 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c1eb49e0d9b4e0bb382e72d5891e09d : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f595ccd76db4763acb54957aba760de : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/555187b96b25414fad5581f1aa13a0b9 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b414908f1fe84e76975cce9a1c0b3f1a : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d712020decb24b9b9e6aded6fe726b3e : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74899222811b49a3b1e52f86cd590a1a : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5953478eda3b4e408218cb2692a04af3 : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91ee38bb4bc94dbe9045c06597313955 : SUCCESS in 1h 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1653fbf3848c5ba1b4891cedf51216c7e2992196","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ab22523c24fb97736385482aaa88f9525b1c86b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:21:06.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6904f6a9d2844e608a8375f103faa0df : FAILURE in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d263dd11b024162958a2663a9f90eb8 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1467ec3eff6cd139603bb98e7cbef2050db5fedb","revisions":{"1467ec3eff6cd139603bb98e7cbef2050db5fedb":{"kind":"REWORK","_number":1,"created":"2022-06-20 08:36:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/846586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/846586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/846586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/846586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/846586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/846586/1"}}},"commit":{"parents":[{"commit":"38d79f37204bb671216184ca9621c7050ae497eb","subject":"Reject invalid whitespace in HM url_path value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38d79f37204bb671216184ca9621c7050ae497eb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-26 07:23:07.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-06-20 08:36:16.000000000","tz":120},"subject":"Fix disabled UDP pools","message":"Fix disabled UDP pools\n\nDisabling a UDP pool didn\u0027t disable it correctly, requests were still\nforwarded to the members.\nNow, when disabling a UDP pool, the members are removed from the\nconfiguration file.\n\nStory 2010011\nTask 45164\n\nConflicts:\n\toctavia/common/jinja/lvs/templates/macros.j2\n\nChange-Id: I7d92b011455d95a2df7782984b8f4648d4b00941\n(cherry picked from commit f4b32ac313d751f81b90438cbedb7fd941322c61)\n(cherry picked from commit e397a118cc2d12e0ae586eb508bd28564fe89df7)\n(cherry picked from commit 839fa4581e2f2095b20bad940495c767187e91c0)\n(cherry picked from commit 5d45e56b47625095203e3f6dddfe3fb9f7019751)\n(cherry picked from commit a8fb6a37bec74bd55e4b7de1299a3d3d0953386a)\n(cherry picked from commit 94a6c2e8f9c5adf925ffc2d78441fb314bd11ca0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1467ec3eff6cd139603bb98e7cbef2050db5fedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1467ec3eff6cd139603bb98e7cbef2050db5fedb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
