)]}'
{"id":"openstack%2Foctavia~846939","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I66ed18b9cd89f49a3ddb08ac4fbc1afff92ccac9","project":"openstack/octavia","branch":"stable/yoga","topic":"story/2009113-stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-12 14:10:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66ed18b9cd89f49a3ddb08ac4fbc1afff92ccac9","subject":"Exclude invalid TLS version","status":"MERGED","created":"2022-06-21 17:20:56.000000000","updated":"2022-09-12 14:11:38.000000000","submitted":"2022-09-12 14:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846939-story/2009113-stable/yoga","cherry_pick_of_change":804270,"cherry_pick_of_patch_set":5,"meta_rev_id":"f4693c088519aaf81fe7f1a0257cd35887e6eca6","_number":846939,"virtual_id_number":846939,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 14:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-06-29 16:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 11:30:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 11:30:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 11:32:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 19:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 06:32:13.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-29 16:54:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a9e74ec1754445accaa5d1379fe4b874985a4658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 17:20:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c41a8ff150362c9162077a2fb55faef277ec2277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 19:34:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e56666e21e8e4e2c80f2dfa5539e6d21 : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7328e1d8438a4724aadc7fc963708646 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c32fa904825496bb188bf0d48cdab37 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3b4222fd8f64769968562dd11afb8f9 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2df236f1471c4094b214b80a9cecc82a : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc9c4eabf82459f97cb44aba9656a6a : SUCCESS in 11m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aa226654721b4ab18a2a3a01e63dbfec : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/29c96f76c4204e55bc42b90bc51828a3 : SUCCESS in 10m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac3b7bb910a74ca09ba1ba04538dc016 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/052cc56e68f24e8db92b3c16cd317163 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac8e9a5b50934ad9890614987ad04081 : SUCCESS in 2h 08m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a7ede088a984bdc816356c9bac56cce : SUCCESS in 1h 17m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd6812953ce847f2a6a307b75fa93776 : SUCCESS in 36m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a410f74ff67b44bbb89eeaa5edaed645 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ffed4376a7974627a06212d3bbe3b159 : RETRY_LIMIT in 8m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f680a2845d24c4d8bff2edc0fdf5c98 : SUCCESS in 37m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ebf1b62ccec84620a213e95d1d58f45e : SUCCESS in 1h 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce9af73f3bcc43708bcfc3e0fd9c06f4 : SUCCESS in 1h 55m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f245932bd304fb4b99a97663ecb9864 : SUCCESS in 1h 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f2df212bfc24defb991a0a46e2ece46 : SUCCESS in 1h 34m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/974678245bb4468a977e4c22da00cf18 : FAILURE in 17m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a20dcffc97946ffb744057a5123e591 : FAILURE in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2abe45b0524c93d03ceb07dc0615201583f5058a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-22 06:32:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"453cd7fbc6dcd072497bbed18dc73119706bf6e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:54:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fc0d266e0e937ad01396c6d776e83584e9fc1c5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 11:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a482946a34cae17619ebe6245e6ec17f6ddaa9c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 11:31:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6224405bd1392ab6e072e9715249fe1c2c083c0e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 11:32:18.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"8992afe43032501695c8c19ae3cf943d699d685a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:10:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94b428e47104e379697361c18befef8\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bcb1ec23abd474284c468ac39743aba : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e41d179c9241f7a4327cf5b9394634 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75e2ebd0797465c93fd697dc434471d : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13b176430b5e4b77b928b618d08077bf : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c2e0972f61430997785ed599d1edf2 : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee621eefdf0740ad916fa9634339b9bf : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/baa56f2b67c24802b23e2c2083ac5d86 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c2840a364d34fdea7bbaeefdf3ed751 : SUCCESS in 1h 37m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6602b4d3fb64b57a2af43d3c2c81442 : SUCCESS in 2h 11m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/530e31b028784c5b92e9c1bdae2659bb : SUCCESS in 42m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ebabfc7b1db492f8e27ce251477cc5d : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9c26bcb382a8062f13d9c2a58295759df58fe838","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:10:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f4693c088519aaf81fe7f1a0257cd35887e6eca6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e9d32bc0d574a03851b8d4223dc75b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafd52360f4943c9b5954569fdaf7a03 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c3f2a57ef23661fbd9390b3cf54569eeb0fea72","revisions":{"7c3f2a57ef23661fbd9390b3cf54569eeb0fea72":{"kind":"REWORK","_number":1,"created":"2022-06-21 17:20:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/846939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/39/846939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/846939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/846939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/846939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/39/846939/1"}}},"commit":{"parents":[{"commit":"a237182303d6a5559ea1148310a6d8fb18c6cbd3","subject":"Fix compile_amphora_details when using UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a237182303d6a5559ea1148310a6d8fb18c6cbd3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 15:43:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 17:20:56.000000000","tz":0},"subject":"Exclude invalid TLS version","message":"Exclude invalid TLS version\n\nThis change ensures the agent_tls_protocol parameter accepts only\nvalid protocol names.\n\nStory: 2009113\nTask: 42994\nChange-Id: I66ed18b9cd89f49a3ddb08ac4fbc1afff92ccac9\n(cherry picked from commit 0059f05aeb8b7bd1fb83d45148a9926d9dafaae0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c3f2a57ef23661fbd9390b3cf54569eeb0fea72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c3f2a57ef23661fbd9390b3cf54569eeb0fea72"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
