)]}'
{"id":"openstack%2Foctavia~847540","triplet_id":"openstack%2Foctavia~stable%2Fxena~Id1dd029995d41e6913a07eeb120c889d71d583eb","project":"openstack/octavia","branch":"stable/xena","topic":"3d3e38-xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-11 13:36:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1dd029995d41e6913a07eeb120c889d71d583eb","subject":"Set sensible nf_conntrack_max value in amphora","status":"MERGED","created":"2022-06-24 08:26:07.000000000","updated":"2022-07-11 13:38:23.000000000","submitted":"2022-07-11 13:36:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847540-3d3e38-xena","meta_rev_id":"83daaf7c5ca21ae50cbb19e55c330bf55082f62c","_number":847540,"virtual_id_number":847540,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 13:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-11 07:41:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 11:19:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:56:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:09:53.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-11 07:41:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"4c15d347551a58e5fc681af8d26d8c37cba3b4cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 08:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25e6cfee8ffd3b99f4ce7b84db30ee7fc94bbd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aad8626ed40413c849ef327843aa460 : SUCCESS in 12m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa25c26199a54db7a7af55b4432e1117 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4237a9c342a74047ad7c78ba7795f223 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5ac1be1a7cf474e90012b390dc72ff5 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b2c2bc564ab4294b7184521434e3f2f : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f5272fa0d0248839b5815298e99cd11 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad1cb340d054ba0802875a7d23080a6 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/833b72dcbea24897b86f7b2755dee5f5 : SUCCESS in 5m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/372f7658d45e47c8b1b50db726bb8067 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9b8c71de0b1a4b15818fb3794245ac70 : SUCCESS in 11m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/981880ab312043e0ba5821f0d323db6c : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17600b4b6a2349cc9ce0cd079c46b098 : SUCCESS in 9m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63c43946fb70489c970b884eb323f8ef : SUCCESS in 2h 12m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ac7957561cf40e39e376f56e1425a18 : SUCCESS in 1h 58m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7280215e85f94d71aea10f4dd06aa2e7 : SUCCESS in 48m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7d5abdddbb44afe97250e29ee4c1cf2 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8bbcb75cb7ad4c3c8c1762984acb82b3 : RETRY_LIMIT in 12m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fb7792d15484076b0e57b830fc4c692 : SUCCESS in 35m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4382831f23524238aa35018a9402903c : SUCCESS in 1h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1ed15578c144c5c911ad5dc5a35913c : FAILURE in 2h 42m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d222c41ba6b40d082773fa6ecbc61d4 : FAILURE in 1h 57m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e4ce1195ce4a46a7beed3ca912799f1e : SUCCESS in 2h 17m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fe29f4b58db64016a16acadec30d20c4 : SUCCESS in 1h 58m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/41f67f470f694c5e883c4ac2d1042405 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eeb37b6969cb1998a0540df27b3035ab78a159b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:56:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66655388a525e563cbf0e5df2afb70a4e743d877","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:09:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ec374740fa02451e3e76f43d2faa19029d73cff","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:41:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"695e7f70aa528064f8529390af48787baa6b6abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:43:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d87f47b31d09480c40bbd61d4661f6b4af5ca8c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:36:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259335cc14054eec969f6afe04bf480a\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d3e94ac1dca424f8b8c04db7f557eb2 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9531ad8b8a4c64804df74250f75af6 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50646297eb8b42928109dc158c28d16c : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25eac5a586924d6ca4d78f8500df8007 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78a487f4ea249388ed9c95edbe8d47b : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0bd25aa182a49168c66bafc4015c53f : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/db0ce741dec44f87ac1a0da265a0bf26 : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58255038c76b4f46ad8b178b93bfdb75 : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b62bea5fd9c425e9222d14dc3ecd380 : SUCCESS in 2h 24m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b91760438c9a4ba7ae3e1450dbc4d188 : SUCCESS in 2h 10m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05da9e6b1b024df7bb7ff03a65c82d4b : SUCCESS in 58m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb3bb6f99d244055ae3d3d9eb6c91a77 : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5c00263ede166c9638aa4fa179cd0e69ad213654","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:36:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83daaf7c5ca21ae50cbb19e55c330bf55082f62c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:38:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee90a95075f4480b7b49128a1957b48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddc38efd2a343b78573789af647474a : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6d7a35d5cc04f9d9db5bc509de8caa4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33eaff55fc59c711e9a74dd94f55d84a54f09f56","revisions":{"33eaff55fc59c711e9a74dd94f55d84a54f09f56":{"kind":"REWORK","_number":1,"created":"2022-06-24 08:26:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/40/847540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/847540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/847540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/847540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/847540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/847540/1"}}},"commit":{"parents":[{"commit":"d668e1f2055e0124cce4a4b7abae06ade340e24a","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668e1f2055e0124cce4a4b7abae06ade340e24a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-24 12:43:26.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 08:25:55.000000000","tz":120},"subject":"Set sensible nf_conntrack_max value in amphora","message":"Set sensible nf_conntrack_max value in amphora\n\nThis sets the values of nf_conntrack_max and nf_conntrack_expect_max to\nsensible values. Previously, the default values set by the kernel were\nused, which limited net.netfilter.nf_conntrack_max to 32768. On the\nother hand, Octavia sets nf_conntrack_buckets to 125000. A value of\nnf_conntrack_max lower than half of that makes no sense.\n\nThe new values are calculated using the formula for the default value\nin the kernel [1]:\n\nnf_conntrack_max \u003d nf_conntrack_buckets\nnf_conntrack_expect_max \u003d nf_conntrack_buckets / 256\n\n[1]: https://www.kernel.org/doc/Documentation/networking/nf_conntrack-sysctl.rst\n\nStory: 2010050\nTask: 45460\nChange-Id: Id1dd029995d41e6913a07eeb120c889d71d583eb\n(cherry picked from commit 3d3e38afe9f0186f45e9800090afecc3aa434846)\n(cherry picked from commit 78603681cce95048b76e2fcfabcdb0b5baf58c51)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33eaff55fc59c711e9a74dd94f55d84a54f09f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33eaff55fc59c711e9a74dd94f55d84a54f09f56"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
