)]}'
{"id":"openstack%2Foctavia~847543","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Id1dd029995d41e6913a07eeb120c889d71d583eb","project":"openstack/octavia","branch":"stable/ussuri","topic":"3d3e38-ussuri","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-12 08:11:36.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Id1dd029995d41e6913a07eeb120c889d71d583eb","subject":"Set sensible nf_conntrack_max value in amphora","status":"MERGED","created":"2022-06-24 08:27:02.000000000","updated":"2022-07-12 13:36:54.000000000","submitted":"2022-07-12 13:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847543-3d3e38-ussuri","meta_rev_id":"203614220e296b655f034a87a3204d527e70be17","_number":847543,"virtual_id_number":847543,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 13:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:08:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-11 07:42:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 10:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:55:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:08:46.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-11 07:42:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"2de482e7da45c24a28587af1241c504f86e39e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 08:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5070b3dcf4b45e6a4cdc94da3b1bc1253cf2b874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:08:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d5c3ae9709432c9efc86fc60b51741 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bcf6606d664cb28a02346d725c7e62 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0332f866354342628d2c2597186df0eb : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f57f7a08740d430583e02f477417ebf9 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259c9607b77e49d4ad426ca2babb7537 : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74cb335d4801405190661acba19705c0 : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/48c2debc297a48a6a3271d00db5b073a : SUCCESS in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/443d9e3472f14ec6bc4e0c84f0dedd3c : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/723e4468b91e4c909b2ae60b18ce9f9f : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa0ab0816bf246818671df9a76e386bb : SUCCESS in 1h 20m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/232031451910405fb73f693f58752a8d : SUCCESS in 1h 33m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be3e412b4141481191090486a0b11c2d : SUCCESS in 35m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c0f5ff46001841f481abbcd39326d571 : SUCCESS in 33m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c38b378f8b9f414b94aafd2b2847e6ba : SUCCESS in 33m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ebe42eebf5443308fe972861bce41a5 : SUCCESS in 1h 32m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d06b11ef50d546d1bf48b6a178e9dda6 : RETRY_LIMIT in 7m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a931f5917cb407baf860feea9560785 : SUCCESS in 1h 38m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78f9e1f568f82314a1eaab5c9a72e3d696ee0076","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:55:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a80fdd34d7221e44dbf5b1f7a684410946bdb91b","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:08:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38d3761e003fdb8dc5b1ea262b0e7010f975973b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:42:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1b576b0f4f3d3655baf91137a7e06009cd484f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:43:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2e4efa9ad1917459d41ff308f68406a5903227a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:37:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/810f3e9680f9455987e6e0f3c966bd56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf5cb6dac3b40de8854affa87af5490 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/518b78f9d4e148d69c06350d6376cd79 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c732d951a83431e880c543c1cd37c73 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d918f854c9a4185ac99364016a0aef0 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c2f7f49e9a44670a7fb383e4d19255a : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3682556f5094abc9cb2d7657b257976 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d64b6c3f7614dc1b73ebb65fe360585 : CANCELED in 44m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/135488dbdda6482ab1aa7a7c51a43f0c : CANCELED in 44m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ecec68151044d87a0abc1e8cea9c7ff : FAILURE in 43m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ffa7dbcc723848d8a30537c79bc3d298 : SUCCESS in 38m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ff9fa5d94e2492ca3fbb7f57d8c3690 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5fb8305acb9a1a8da175f6bb0fccb9f112a00e3e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-12 08:11:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff5f001c8d7feb6ae7899ca8a49ae90863557754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:15:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ffee798f2b4bb6b7cc6b510c5b6492\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b8f5e621484474ebc7df0c68dea88a0 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7278f073cbaf4b63b1ca3f51b9281c59 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5624b7d9769b479988d9f3d2d28af46c : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c365447eaebd44cdabbd8e09b8494c82 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36f5731e8dd4797b3291061d11e77a4 : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2899e687047946658a7e5dd7e94b09a9 : SUCCESS in 5m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb7697d2a80a4fde9730f46f4ca64def : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d9bdaae593ce4c74b757482432df01f3 : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/307079b2f86342b1a2663fd1fd99eba0 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ba0d1316e8e4fada2d3294f432cc10f : SUCCESS in 1h 37m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f45fbe1fc0524bdf893e77dc9ec4da21 : SUCCESS in 1h 36m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbd5e66f15654a89bcd92bfbffcae358 : SUCCESS in 40m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a3cbb5d9bf0f4e51a7a0900358ea1478 : SUCCESS in 43m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a4dfcf9191344b689c52651f7c255e7 : SUCCESS in 44m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6be25e960fd141b3ba5d21805f59a5a2 : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/18a226c85eff4a3ca0c0f4876155ad5d : RETRY_LIMIT in 8m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c164948ad85a4beb85225f660fbcfe13 : SUCCESS in 2h 01m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c181a9e03ef57ff9fa70f85e9f85454bda5f33ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:16:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"212474174e4a07593d68765960f39d39510d8fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:35:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497fa09817f34c25abbda4270d6d7ecb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc50621cf7d45559dd9379e1ac42c4c : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a51b37d1b77b4f0b856fd6429448030f : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0f5abc298dd441b9abb970abc16b2c6 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3000a21b7a084afb962e56fe8701a34a : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63b21a3ab149484a9fbcd2b7932562f5 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e0fa1a6c58c4342a288ab14ae60ef7c : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ba6871d03594585908778445aa50258 : SUCCESS in 1h 09m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fac53e073254f619d6ee7ab0361c102 : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05149c7259eb4e2fb9081a6df8be5874 : SUCCESS in 46m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf08fc4922c244079d77c931d47a21eb : SUCCESS in 40m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f97c5e14e8c479394f3993b78d227a0 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a7096e8de754b190c3f1b9eb414d4e068ca6eeac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"203614220e296b655f034a87a3204d527e70be17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:36:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f3166ed2b74338b9d862e28db6b042\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1902c0fe384efd9054bf1a45636c74 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/680e825aec7d43849f150089a9133356 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e367cd09478811113d225b6f59a11df5f31c827a","revisions":{"e367cd09478811113d225b6f59a11df5f31c827a":{"kind":"REWORK","_number":1,"created":"2022-06-24 08:27:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/43/847543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/847543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/847543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/847543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/847543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/847543/1"}}},"commit":{"parents":[{"commit":"efeb4da03e31770246a8e715133ce9153561c8a5","subject":"Fix amphora build on stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efeb4da03e31770246a8e715133ce9153561c8a5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-24 12:43:26.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 08:26:50.000000000","tz":120},"subject":"Set sensible nf_conntrack_max value in amphora","message":"Set sensible nf_conntrack_max value in amphora\n\nThis sets the values of nf_conntrack_max and nf_conntrack_expect_max to\nsensible values. Previously, the default values set by the kernel were\nused, which limited net.netfilter.nf_conntrack_max to 32768. On the\nother hand, Octavia sets nf_conntrack_buckets to 125000. A value of\nnf_conntrack_max lower than half of that makes no sense.\n\nThe new values are calculated using the formula for the default value\nin the kernel [1]:\n\nnf_conntrack_max \u003d nf_conntrack_buckets\nnf_conntrack_expect_max \u003d nf_conntrack_buckets / 256\n\n[1]: https://www.kernel.org/doc/Documentation/networking/nf_conntrack-sysctl.rst\n\nStory: 2010050\nTask: 45460\nChange-Id: Id1dd029995d41e6913a07eeb120c889d71d583eb\n(cherry picked from commit 3d3e38afe9f0186f45e9800090afecc3aa434846)\n(cherry picked from commit 78603681cce95048b76e2fcfabcdb0b5baf58c51)\n(cherry picked from commit 33eaff55fc59c711e9a74dd94f55d84a54f09f56)\n(cherry picked from commit d0adca9a34b5d2f36b821478d4e505f6592f222a)\n(cherry picked from commit 4ea8953a698205764fcbc9617f441fd32f6a5ce2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e367cd09478811113d225b6f59a11df5f31c827a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e367cd09478811113d225b6f59a11df5f31c827a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
