)]}'
{"id":"openstack%2Foctavia~847544","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Id1dd029995d41e6913a07eeb120c889d71d583eb","project":"openstack/octavia","branch":"stable/train","topic":"3d3e38-train","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-11 14:14:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1dd029995d41e6913a07eeb120c889d71d583eb","subject":"Set sensible nf_conntrack_max value in amphora","status":"MERGED","created":"2022-06-24 08:27:18.000000000","updated":"2022-07-11 14:16:21.000000000","submitted":"2022-07-11 14:14:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847544-3d3e38-train","meta_rev_id":"b3d513d7e42a3773f60f7eb617e566a7febf8ee1","_number":847544,"virtual_id_number":847544,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 14:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-11 07:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 10:11:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:56:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:08:58.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-11 07:42:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"7925e3465890e0c6665ad49c0c9b8e0a506dbb6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 08:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21d30f15e3cb8d59d46752009fbe6bfe4bfa2b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:11:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58bf3180cc65449092312657d3226aed : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713f7da60a2f449fb3f9896bf17077ff : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e65a67626bb04f64bcf268e559b5e2eb : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8082f3d8063a4eeeb45ea83bd9624163 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9089cc75f0db4a57945f2d645abfcdc4 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487024b5e402460b82e519493494c5f1 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da274c50f208469ca478ad0926bec52b : SUCCESS in 7m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/24769d2bf2d947a1a34f3f4b060857d5 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/868a07164aee4edcb216481b3a976407 : SUCCESS in 8m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4e84c54f3be54a2badbc1e7230886ea4 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b161d30e2744b21956c8f15586e958b : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f20763d591ff4d9c94ecc3bea4dca879 : SUCCESS in 1h 10m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3f1c8b1c91954f7a9d794207f34adb45 : SUCCESS in 1h 08m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3408371fd70a4d2b8c7e193d7aa8fd5e : SUCCESS in 1h 26m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/63f2717ef35c407088b29238fedffc11 : SUCCESS in 1h 30m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9991cffc019f4a2287f720cb2debc164 : FAILURE in 17m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ecd82e54c06a47928843aab0f78dd16d : RETRY_LIMIT in 6m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d3eca980d514ccd943c793a525b5a9f : RETRY_LIMIT in 7m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6318a11f25ad41eb93b2b6a132fd4084 : SUCCESS in 38m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f971549b0c3c4fb29fc92a9f4d4fd39d : SUCCESS in 1h 40m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46bcbb204d0642acb70e57a004f02da7 : FAILURE in 31m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7a45e5662f514448bc2236944da29987 : FAILURE in 30m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b674e206ae8640b4b7c14a68672e6e0f : FAILURE in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"333065121d142036db3d31e85d3ceca4033769c3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:56:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bc20cb18a28dd4a2b0247df7adacf8f422a4b26","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:08:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a00705f10c0335e461c152152b21576a19bbb12b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:42:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88c1448331ae133e9363a0a3b746385a4369bd02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:43:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3530b5e7b3f3279217b8225f75d0225b70187b81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:14:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d085552ea2444562a5d112a6864c2b17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d927392d63a4b20b08d264b18fd85ca : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e5308b8d54849c9a4ebf526e2911311 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1fca5614b3643d586b064da78ce0d9c : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87af49099a3f4bea8b16ff3b16702ee1 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a83cc223df64e72929f5cc1800e0529 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f41d41ff8c274e3bae630e2b3de25cfe : SUCCESS in 4m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b76de3c37bef41e7ac39b92db8a9f647 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd14c7b39f944b4fb9f10331e74bdf0a : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4da3068f73034b6ba77e9af9ebfaa2f6 : SUCCESS in 1h 10m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a318a4bfbe8d46ef905b1cc7eba6c1bd : SUCCESS in 1h 17m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ede82fdc13e24a87a0f076890144cba3 : SUCCESS in 1h 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1c529844bd48ed6fa4c58ca5985608ddd625d17e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:14:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3d513d7e42a3773f60f7eb617e566a7febf8ee1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:16:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3526b33edcac421b86a7ddde6eea4416\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef94b504aa6421e9a7be16a6e34294c : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a4ed6c9a6be4f47ae6d2e32c2b31ffe : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f12875365c90bd50d337afac0fd69202cd6c0eb","revisions":{"5f12875365c90bd50d337afac0fd69202cd6c0eb":{"kind":"REWORK","_number":1,"created":"2022-06-24 08:27:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/44/847544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/847544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/847544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/847544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/847544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/847544/1"}}},"commit":{"parents":[{"commit":"5be32ed7401aff6eab8c076f70317eaa8380ca69","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5be32ed7401aff6eab8c076f70317eaa8380ca69"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-24 12:43:26.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 08:27:07.000000000","tz":120},"subject":"Set sensible nf_conntrack_max value in amphora","message":"Set sensible nf_conntrack_max value in amphora\n\nThis sets the values of nf_conntrack_max and nf_conntrack_expect_max to\nsensible values. Previously, the default values set by the kernel were\nused, which limited net.netfilter.nf_conntrack_max to 32768. On the\nother hand, Octavia sets nf_conntrack_buckets to 125000. A value of\nnf_conntrack_max lower than half of that makes no sense.\n\nThe new values are calculated using the formula for the default value\nin the kernel [1]:\n\nnf_conntrack_max \u003d nf_conntrack_buckets\nnf_conntrack_expect_max \u003d nf_conntrack_buckets / 256\n\n[1]: https://www.kernel.org/doc/Documentation/networking/nf_conntrack-sysctl.rst\n\nStory: 2010050\nTask: 45460\nChange-Id: Id1dd029995d41e6913a07eeb120c889d71d583eb\n(cherry picked from commit 3d3e38afe9f0186f45e9800090afecc3aa434846)\n(cherry picked from commit 78603681cce95048b76e2fcfabcdb0b5baf58c51)\n(cherry picked from commit 33eaff55fc59c711e9a74dd94f55d84a54f09f56)\n(cherry picked from commit d0adca9a34b5d2f36b821478d4e505f6592f222a)\n(cherry picked from commit 4ea8953a698205764fcbc9617f441fd32f6a5ce2)\n(cherry picked from commit e367cd09478811113d225b6f59a11df5f31c827a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f12875365c90bd50d337afac0fd69202cd6c0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f12875365c90bd50d337afac0fd69202cd6c0eb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
