)]}'
{"id":"openstack%2Foctavia~847555","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","project":"openstack/octavia","branch":"stable/yoga","topic":"f166bb-yoga","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-29 19:18:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","subject":"Fix duplicate object error messages","status":"MERGED","created":"2022-06-24 09:36:18.000000000","updated":"2022-06-29 19:20:46.000000000","submitted":"2022-06-29 19:18:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847555-f166bb-yoga","meta_rev_id":"80476689ab65a91dc5cbeb27c29edeac1bf40dd1","_number":847555,"virtual_id_number":847555,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 19:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-06-27 06:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-06-29 17:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-06-29 17:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 09:36:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-06-24 12:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 06:52:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-06-29 17:00:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b8e39ef06470252e1075559d3cfd73bc2478a7a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 09:36:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f29f8ece5b4cded919cf2f3f92bbf91d33cf774c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 12:07:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76e3aeb531924d74960007680f2251d3 : SUCCESS in 12m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c97d18211044409fe82ffceef2ab45 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647465379c9c4fd9a53d1233fda2bf5c : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ef7b42ac975479da50b30f61b8e8dac : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a15c4bcd131f47ce810533b77b57ab53 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3778dd597694ea8953184d32559abfb : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c483e8e56bd54cfd8d9a585055e21f69 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/747098a1b3454b919043a161aeea4a3d : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25cf3b1d20984cdb8be8befc495fd1e8 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e92658842dff4dd7b3bac8f49bbbabfb : SUCCESS in 9m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/291513040b1d4e6a912c336f5790b0fe : SUCCESS in 2h 23m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8331be19bfe419d92bf46a993f4d934 : SUCCESS in 1h 22m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/246f2ac0b58e488981606f1b599c6834 : SUCCESS in 38m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd22b47773ab45fa9d42b79669c2a58c : SUCCESS in 47m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/45c62549f36d4358bff9611190177cbb : RETRY_LIMIT in 12m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d124563afa74da6a9db301f41a290cc : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0fe2dbdeecbf4b4185ebc46d2be5edaa : SUCCESS in 1h 40m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53d78004945f45b79f57bf95c433fa40 : SUCCESS in 1h 45m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/940fac8931cf41e6a02a22f1f330d6af : SUCCESS in 1h 38m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11d81bbdf2ed4b2cb1c2470faed69fa1 : SUCCESS in 1h 30m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/32a61404ffb14896a4e05c5acd7de6a9 : SUCCESS in 1h 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/059a1080587f4bbb8675eef36b0cef25 : FAILURE in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1da074cc1808e2028740c274e2ebc41d7d7628d8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-06-27 06:52:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01d6f0bbd76e83e18bb32ea3c00c1e9b57fdebeb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 17:00:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15f5e9e046eeb8bd6411c1539d7d89a93bf43d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:01:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bee9fae29411b058c79a223f62c837acff3fb7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04d5e33697e649228ce3f526918aff2e : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1910c9e9264eaf80379ac7e83ba1b0 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6070f57fa3f144eaaaf99f54f720a4ba : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3faf9ec271af482f94b1218de798c15d : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a883bb4a8a82401d912e55e62f8ae250 : SUCCESS in 13m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54097176c03c4cb2a160125fb30806dc : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba259c8ebb5b4f7c9c0a49e419a80f82 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55c0b682b6214a0aac7f389d66665019 : SUCCESS in 2h 01m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb54d1772e404077b8fb45196b040874 : SUCCESS in 1h 25m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9b4e28d56ea24e8284801d600195224d : SUCCESS in 34m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b4edeba6613e403fb0af31e2648db17f : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a6fd05b22f36f588b75da1946c9cfbdf80b65cfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:18:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80476689ab65a91dc5cbeb27c29edeac1bf40dd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:20:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5af10e87054c50b0ea839e913166e5 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e69899824ce09b8eb72c294562fdae5ad7ad5c13","revisions":{"e69899824ce09b8eb72c294562fdae5ad7ad5c13":{"kind":"REWORK","_number":1,"created":"2022-06-24 09:36:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/55/847555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/847555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/847555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/847555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/847555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/847555/1"}}},"commit":{"parents":[{"commit":"a237182303d6a5559ea1148310a6d8fb18c6cbd3","subject":"Fix compile_amphora_details when using UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a237182303d6a5559ea1148310a6d8fb18c6cbd3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-12 21:08:24.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 09:36:06.000000000","tz":120},"subject":"Fix duplicate object error messages","message":"Fix duplicate object error messages\n\nFixed an issue where creation of duplicate listeners or members may have\nnot returned the correct error message to users.\nThe API would traceback with AttributeError: \u0027NoneType\u0027 object has no attribute \u0027to_dict\u0027\n\nStory: 2009128\nTask: 43045\nChange-Id: Ibbccda5859c8c6129d43e0742b28b319b49ddd1f\n(cherry picked from commit f166bb1481fb58acd7e7329ec394f43249001718)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e69899824ce09b8eb72c294562fdae5ad7ad5c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e69899824ce09b8eb72c294562fdae5ad7ad5c13"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
