)]}'
{"id":"openstack%2Foctavia~847556","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","project":"openstack/octavia","branch":"stable/xena","topic":"f166bb-xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-11 09:56:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","subject":"Fix duplicate object error messages","status":"MERGED","created":"2022-06-24 09:36:31.000000000","updated":"2022-07-11 09:57:23.000000000","submitted":"2022-07-11 09:56:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847556-f166bb-xena","meta_rev_id":"dfb6b4105d7c741677a2362689d9123942e46368","_number":847556,"virtual_id_number":847556,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 09:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-11 07:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 09:36:31.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-06-24 11:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:57:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:10:47.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-11 07:37:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"057ae901e6df129c85e7052304d2b05c8f5fdb22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 09:36:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"200b4f9b00336695cf78feb4cb3e5cd6a9519e66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf0a5789a8b430b985c5906ef00ba9a : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17b4f8a0da694b55a20caff88a218c9d : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3780fc3ff04e369551dbf5da129d68 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9446e8a62c844029be11561b0f1329e3 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7361a17f7410496e85d7a4c06414b909 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e1c7e35a0a742d0b24a1e3c586072f8 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3295e209ae4b459c8e1b234ae99bf0bd : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/63b8d0f981534d2f9eef778400d2bf0d : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6b9279173594f01912bd7fe84d70dac : SUCCESS in 11m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7d1b6146863a471e8f361b8512b33f85 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39996e89af99480d92e6e5eaeb531ca6 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d38afb0714c48c3be938d793e5dbb4c : SUCCESS in 1h 55m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0a47bc94f5248fd96cc1504296f17c0 : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9388b7c54bc45e6a8f1687c58f20425 : SUCCESS in 36m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/93647d63a96742eea1f46dafe87ce4b8 : SUCCESS in 52m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/463cc0c341ca45d48cd2d070eaa2af18 : RETRY_LIMIT in 13m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc59de44cf554ea9b95416f2e485f6dd : SUCCESS in 30m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/93bc4daaea47434087d33f82851aa76d : SUCCESS in 1h 39m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9312ce44593247cebc1601674ffa632f : SUCCESS in 1h 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb5aee8a908a421a947d937aea3dbecc : SUCCESS in 1h 27m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32781cf3b35244cf967d0170961bb7b0 : SUCCESS in 1h 25m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7f86b443a7a84f85943aa7167a04ad56 : SUCCESS in 1h 24m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21da5fac72bc4f9a80896392f3af35b3 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7960626f53374d0c5a3c00893ad2ecc79b02b69","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:57:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9eb543ae59b4b55a4fcfbb03880cfd0aa0d9fd4","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:10:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c013d5d5d94ab2ae0562264ccc2310189800666","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:37:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e39d1e5bfca45d6f8f65fb78e352e67e747093fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:38:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c50328a7eb0b0a59a4676c4d3f113921171ba58e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:56:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f45472e90c84fa2b89840a4a82cd9ba\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc9d2d366f5e4112a2ecb9e2a3fc8185 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1441690df3bc40428fc92ebdd108e090 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc43caced8d45038d65c904dcd8fc95 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec3f8d5fe0f44f54a2fb0f01ed8cb2bd : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474d53a08edf4aa89128d2096aa71bd8 : SUCCESS in 12m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7d2f5c4b12ba49518c441eaa8cd686ad : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c28f45c317e147d19e6c5e6d41ce7dc4 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03c7dbe50cb741c59f28264ea7eccaf2 : SUCCESS in 2h 12m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08189931b7da4cb8842664fef9f8157c : SUCCESS in 1h 42m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ae20e3a790b42a087c3677dac2c9d53 : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b00648f55c2c4176aa421036c8f3c805 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9f86aa22df21947269e5817c5bf963ee9678ad11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:56:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfb6b4105d7c741677a2362689d9123942e46368","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:57:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea50089e84f427ab48df41171df826c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf3957c281146a38f512c84062b0a62 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5647db94c941f582bb8292599379a3693b470cbe","revisions":{"5647db94c941f582bb8292599379a3693b470cbe":{"kind":"REWORK","_number":1,"created":"2022-06-24 09:36:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/56/847556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/847556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/847556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/847556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/847556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/847556/1"}}},"commit":{"parents":[{"commit":"d668e1f2055e0124cce4a4b7abae06ade340e24a","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d668e1f2055e0124cce4a4b7abae06ade340e24a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-12 21:08:24.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 09:36:23.000000000","tz":120},"subject":"Fix duplicate object error messages","message":"Fix duplicate object error messages\n\nFixed an issue where creation of duplicate listeners or members may have\nnot returned the correct error message to users.\nThe API would traceback with AttributeError: \u0027NoneType\u0027 object has no attribute \u0027to_dict\u0027\n\nStory: 2009128\nTask: 43045\nChange-Id: Ibbccda5859c8c6129d43e0742b28b319b49ddd1f\n(cherry picked from commit f166bb1481fb58acd7e7329ec394f43249001718)\n(cherry picked from commit e69899824ce09b8eb72c294562fdae5ad7ad5c13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5647db94c941f582bb8292599379a3693b470cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5647db94c941f582bb8292599379a3693b470cbe"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
