)]}'
{"id":"openstack%2Foctavia~847557","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","project":"openstack/octavia","branch":"stable/wallaby","topic":"f166bb-wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-11 09:56:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","subject":"Fix duplicate object error messages","status":"MERGED","created":"2022-06-24 09:36:44.000000000","updated":"2022-07-11 09:57:29.000000000","submitted":"2022-07-11 09:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847557-f166bb-wallaby","meta_rev_id":"9baa727e2c5f717e196a964227f97a8cd030c204","_number":847557,"virtual_id_number":847557,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 09:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-30 04:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-11 07:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 09:36:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-06-24 11:34:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:57:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:10:34.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-11 07:38:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"2122e8f726f2137176dcb38e8d3d11629f434a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 09:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02887e505c402207f59aab813147a8f73154b6be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:34:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fcbe91d06af4009b9542970baa42ccb : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b7ebefcbb3f4d1e84e6a2910ac6a18e : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ce1a7466e64b118c0ead9e6d86e5f5 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37d0e40354ef4e7883bde4c77cc2296d : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/235b0204e7f3446d876f5e600e98fa71 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0dffa337ef74099b3d6fb6aefbb7ab5 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023dfdeb8dea40999c38d3ad7d1ca89f : SUCCESS in 17m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1ff177dcac734c6985b03ef95875fa41 : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8a6bdaf09074442b067b2fca08c22c4 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3acbdb7d08040768bd7ae5a932b0cc3 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5fd6e383e9a4ef99b1658c920d501cb : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5a69922fffd47e4990ab33fe0501648 : SUCCESS in 1h 19m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba202b18d0784939a2620120b226d117 : SUCCESS in 1h 14m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2312cbd23914dc1993baa19baf2f2af : SUCCESS in 36m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/33018f35530d49bdb79d2f2552097866 : SUCCESS in 32m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4dac81f3af5448198a7cb5b6d5363e54 : SUCCESS in 31m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0e24e124fde54606ba6843cb7305df80 : RETRY_LIMIT in 9m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca5aa34cd8f846249b5d1f1d1bb4bea5 : SUCCESS in 1h 24m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23c0ff984ac5410aa679b236991ca58d : FAILURE in 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/924207de27694074a94c9e77a4a0dce4 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73ca9372ca2e4f9594f18cb6e87b73d3 : SUCCESS in 1h 53m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b6d760a152224206a0453103f313bed3 : SUCCESS in 1h 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7f8849fee03e4511afe79c5d2326eee0 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efe29bd31327c6dc77dc70773e2817abc8f5da06","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:57:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"285ae8a9413d15542a21bc05187e201543cdcdf6","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:10:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"624bf24e3012c598ec209d7a70885677624857d3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:38:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fbaedf9c6ffdf49baeec64ff35933781ea9c885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:38:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a0ce77b8a2ef35ded0f66a9cad05bdad49d087a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa72e32870f041368b67ec1248daf643\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/982052fe6af647458976272f9c8359eb : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fc439468e0406ab5aad0fdca8b88b9 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f4124bdb234eb09bc0fc01703e0f44 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d787591368e94d4bbe853ec96ace7fa6 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a7dd104b664e21a252f3566242a145 : SUCCESS in 12m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bffbbf4511654aa78908ea34ca14400f : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbb3425436fa48179b494b5189fedbb2 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d3e34f656b447a59b2df363f1374ab0 : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a1743341a2f44dab06e0f4b17b218f8 : SUCCESS in 1h 36m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/155aaa319ad342f181b47b00829a14e3 : SUCCESS in 38m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e83e018d873f4ce8b03455620129c05b : SUCCESS in 43m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cdd7643ed5146db937f4c1add92b8a5 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2285c13a6e8986c5a2a7d4e0c59d6a12cf2c8b59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:56:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9baa727e2c5f717e196a964227f97a8cd030c204","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:57:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ed2ceeb9c54a0d843ca0fa9ce88228\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de6dd7a856f43ecb00943b9e1e26f18 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca66aac3750d7e2ff005111ade5212a2092e25a8","revisions":{"ca66aac3750d7e2ff005111ade5212a2092e25a8":{"kind":"REWORK","_number":1,"created":"2022-06-24 09:36:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/57/847557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/847557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/847557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/847557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/847557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/847557/1"}}},"commit":{"parents":[{"commit":"a870525df503605b6e7c1ff932487ac61234dbb6","subject":"Fix new pylint issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a870525df503605b6e7c1ff932487ac61234dbb6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-12 21:08:24.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 09:36:35.000000000","tz":120},"subject":"Fix duplicate object error messages","message":"Fix duplicate object error messages\n\nFixed an issue where creation of duplicate listeners or members may have\nnot returned the correct error message to users.\nThe API would traceback with AttributeError: \u0027NoneType\u0027 object has no attribute \u0027to_dict\u0027\n\nStory: 2009128\nTask: 43045\nChange-Id: Ibbccda5859c8c6129d43e0742b28b319b49ddd1f\n(cherry picked from commit f166bb1481fb58acd7e7329ec394f43249001718)\n(cherry picked from commit e69899824ce09b8eb72c294562fdae5ad7ad5c13)\n(cherry picked from commit 5647db94c941f582bb8292599379a3693b470cbe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca66aac3750d7e2ff005111ade5212a2092e25a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca66aac3750d7e2ff005111ade5212a2092e25a8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
