)]}'
{"id":"openstack%2Foctavia~847563","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","project":"openstack/octavia","branch":"stable/train","topic":"f166bb-train","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-12 10:24:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbccda5859c8c6129d43e0742b28b319b49ddd1f","subject":"Fix duplicate object error messages","status":"MERGED","created":"2022-06-24 09:51:40.000000000","updated":"2022-07-12 10:25:51.000000000","submitted":"2022-07-12 10:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847563-f166bb-train","meta_rev_id":"35d16df4803d09e3608089b920343ad5d69c509c","_number":847563,"virtual_id_number":847563,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 10:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-07-09 12:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":1,"date":"2022-06-30 04:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-06-29 16:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-11 07:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-12 08:30:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 09:51:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-06-24 11:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 16:58:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-06-30 04:12:26.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-09 12:04:45.000000000","updated_by":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"reviewer":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"state":"REVIEWER"},{"updated":"2022-07-11 07:39:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"bd300e88323881c1a8023012c1d21575267bacb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 09:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9225146d82b67ad70c482e77b669952ba828b34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9fab8cd6345455288063bc936e97d99 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b560b919afe6443488ded1bfddd0613d : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a89b0b5222ca4880a878273012c9ca62 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/829c4d76a1e44d6c8fbf368995ac22ba : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26ffad4df9ce472b863fd7098aaf6947 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80772ad74cc745b3a76d43fa29c39396 : SUCCESS in 10m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/02d68d7aa91a472eb21f9cd3a5d7d2d0 : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3409b31b5db943e6b58de92c4b6dcded : SUCCESS in 7m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/861d0f11e6c24a7fa3e322bf9e0782f9 : FAILURE in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b0977a97c0243dbbbb87912e6ab8415 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e7557e1251ae4a04998138624336ec70 : SUCCESS in 1h 15m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4056973f5bda4c5a9ff69f2dde3fdefb : FAILURE in 22m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7a82cbf772446c78e296d91ab83f963 : SUCCESS in 1h 38m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b3f42cee28b9430aad8fbc1464f12796 : FAILURE in 26m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/deb0a3e37a2a42fb9b0674e8331d2696 : FAILURE in 16m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0023f05e33644603b595103cb978077e : RETRY_LIMIT in 7m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff0359cb04184900965d28d03ea91d17 : RETRY_LIMIT in 6m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4efdf33c9a194786ac6ff9274e3211a0 : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb37769e1b19459d93aa7cdc077c10ec : SUCCESS in 1h 48m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e625b51df6f49e8a51290467af7bd7b : FAILURE in 36m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c730792927774d12ab66c78c1c576d38 : FAILURE in 33m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6346497b2fab4b45a5c414c2adeaab2e : FAILURE in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77a68610a86e788aa0ad6461def7e579e16fe5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-24 11:56:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c71f7e530e9fb005d6193f4b20d0e2ff96d0866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 14:28:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72c7db00cf5e4aac83af00aa018e3d48 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fddef554e0494752abf42488a6997134 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd4e22bec7c64275abe6f4a23496120a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11ae7b81af84a218eb3f2d575b63200 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90c8015d7d7e41ee927985de8188e830 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542431542c45412b8c9d57de7bc8fec1 : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/775b688ecf134f898686de0ccf5a14a6 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cbb9855e019241cea33e14aea56deb2e : SUCCESS in 8m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/21e2678111cf4c0cb592b3f76937cae0 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/614b726b923347a6b1da694c7e7f2230 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2724f21f93e40339cafe9fce4c94d98 : SUCCESS in 1h 14m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6b1348bbbc094e7c83d21ec0696aeed2 : SUCCESS in 1h 39m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/140a120a71b349a383aa15b417a165d5 : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/328d24bfa1954eb88c36d70f14607d2c : SUCCESS in 2h 25m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d65dcac163614d19b56e01faeab973e7 : FAILURE in 15m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3104b9685fa14da8998659804f868f6f : RETRY_LIMIT in 7m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/03cebea7e4e945f6ad52bd97bb8c8cbf : RETRY_LIMIT in 6m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd607628ddd549ca9ec47f3e23983956 : SUCCESS in 47m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2811dbdddd1f40d6965c13fdf5ea1bc0 : SUCCESS in 1h 51m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7a7ded0a9cb4b1b82eae39a0a2d5bb8 : FAILURE in 41m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/029f543180a8447a90ca652e2e854794 : FAILURE in 31m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a53ae5ad91574398b7132630eb7c57ca : FAILURE in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3e2c215e8999a21b144bd04a9adf29caa006ba0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 16:58:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75d711a8d01fcc8a6745ae6f38f3ec0c35c4b001","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-06-30 04:12:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d95e7fe9f6ebaf4697e24d520d3a1d6a28d73f97","author":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"date":"2022-07-09 12:04:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"25008f973cf5449fbe74cfcd484835e4f76adc90","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 07:39:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9365de99931c44db1856f5b9024e13e7c0ad3089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:40:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4ef5c9cbad2faf6d93f38eb29fa4a93da26f89b","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 11:08:46.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"3038ae0ef4748460f59d0e8b7c470b2c0dde3e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 11:09:05.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"4409131bd5603b7d063ff32a283750fe6fbb35db","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 11:09:23.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a718dd28908f8086334b664325851e18918ffa42","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 11:11:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55cde3ddb628e514d76a437c90c34eb3dad34f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:23:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea1ac6a3ff6407abaad91772415da65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5645093f76a44d73a8e17b8edbe08f80 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859e73ea2a1245b691ba121d5d06624d : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b510236b8a245ac908a8c4bd522f0df : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c33bf190d14cf88e60273aabfe38ca : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb20bb912d9c47d1a25297e0f28aa2df : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5d48802bef4dbeb7ff02044e129c78 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/def55bd34f74463ca3436926fbdb9a0f : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b39fb9d86bf94420ad5780f82d2f16db : SUCCESS in 8m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c472fdd1fe624e1f85f97ba9bb9f3e06 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/233434e208b64795b87ebe344feba62a : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5871984d77284840b6236a08b37fade5 : SUCCESS in 1h 10m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/26f788dd47ce4eccaa9b12e43ef4edf9 : SUCCESS in 1h 17m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8748bf91452140669802c9624aa0163c : SUCCESS in 2h 05m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aa23f11066354c7abfab7d75878f2864 : SUCCESS in 1h 48m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/99a46e2e6fee48ba99ca7b5c9e486947 : FAILURE in 21m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f3b8b073c454b53a2e9317d2e99dcf4 : RETRY_LIMIT in 17m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2aa0bff9dff48c690d6dcfc73c2d83c : RETRY_LIMIT in 10m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c21edd6c02d480ea92f0c4dff576838 : FAILURE in 44m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4fb669ffa0674d61be7a621031301c09 : SUCCESS in 1h 59m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/593b2c521862494aa3ecf0b371f7c454 : FAILURE in 38m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f5da619dcaa14e119a9912c89e67e213 : FAILURE in 38m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c2bf1c2b536f4d93b6e9e31bd23b2179 : FAILURE in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"370d154b6818c44b5b477c1181eb1d14370ebed9","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-12 08:29:47.000000000","message":"Removed Workflow-1 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"dd56c478e9334a711d75549f8e170377e8c8a97c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-12 08:30:03.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d1048233c58c05a28a808377df8d1f6ec87d45a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 08:30:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abceb0216d4e0e0155b10a37e93149713b3863fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:24:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5426b8050a4945639e51fa4b8a55f238\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965e24771f744ca4a40d251ee9aa7600 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdbafc09deff4c19872090472acbec73 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b093be57e7e450687dd0ec3a676b1b7 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f351dc21ea5441bdbfc3b6dd9d3a5950 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0860fa7202d04aa8895ec14eda2de53c : SUCCESS in 9m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/42511c5655c94a20b5eb5bd78493a37f : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c434d85654640cda24a718fdd0755dd : SUCCESS in 10m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4d05872ea9d426296f922f4b20a9ab1 : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/56f0acdb7aac45c5add4280ace0d1773 : SUCCESS in 1h 47m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5621750a83d04c848de081b85ce183aa : SUCCESS in 1h 44m 07s","accounts_in_message":[],"_revision_number":2},{"id":"16e4c6123a281574160cf93ce1ac621e913704f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35d16df4803d09e3608089b920343ad5d69c509c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e142f472d14d4bb1a828c798c0127ea9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70ad5fc1f884c0ba979c0ee672b91fd : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d139e157e8e67e86c8378241feaf5e12df6e682","revisions":{"e984c241cbc40e40ca2839edeaab34cfc0c3649c":{"kind":"REWORK","_number":1,"created":"2022-06-24 09:51:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/63/847563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/847563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/847563/1"}}},"commit":{"parents":[{"commit":"5be32ed7401aff6eab8c076f70317eaa8380ca69","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5be32ed7401aff6eab8c076f70317eaa8380ca69"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-12 21:08:24.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 09:51:18.000000000","tz":120},"subject":"Fix duplicate object error messages","message":"Fix duplicate object error messages\n\nFixed an issue where creation of duplicate listeners or members may have\nnot returned the correct error message to users.\nThe API would traceback with AttributeError: \u0027NoneType\u0027 object has no attribute \u0027to_dict\u0027\n\n Conflicts:\n\toctavia/api/v2/controllers/member.py\n\toctavia/api/v2/controllers/listener.py\n\nStory: 2009128\nTask: 43045\nChange-Id: Ibbccda5859c8c6129d43e0742b28b319b49ddd1f\n(cherry picked from commit f166bb1481fb58acd7e7329ec394f43249001718)\n(cherry picked from commit e69899824ce09b8eb72c294562fdae5ad7ad5c13)\n(cherry picked from commit 5647db94c941f582bb8292599379a3693b470cbe)\n(cherry picked from commit ca66aac3750d7e2ff005111ade5212a2092e25a8)\n(cherry picked from commit a1e22d89a4852a28a3e60b31b4b293199daac0ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e984c241cbc40e40ca2839edeaab34cfc0c3649c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e984c241cbc40e40ca2839edeaab34cfc0c3649c"}]},"branch":"refs/heads/stable/train"},"0d139e157e8e67e86c8378241feaf5e12df6e682":{"kind":"REWORK","_number":2,"created":"2022-06-24 11:56:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/63/847563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/847563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/847563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/847563/2"}}},"commit":{"parents":[{"commit":"5be32ed7401aff6eab8c076f70317eaa8380ca69","subject":"Merge \"Deny the creation of L7Policies for HTTPS/TCP/UDP listeners\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5be32ed7401aff6eab8c076f70317eaa8380ca69"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-04-12 21:08:24.000000000","tz":0},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-24 11:55:31.000000000","tz":120},"subject":"Fix duplicate object error messages","message":"Fix duplicate object error messages\n\nFixed an issue where creation of duplicate listeners or members may have\nnot returned the correct error message to users.\nThe API would traceback with AttributeError: \u0027NoneType\u0027 object has no attribute \u0027to_dict\u0027\n\nMade adjustments to patch for Python 2 compatibility in Train.\n\n Conflicts:\n\toctavia/api/v2/controllers/member.py\n\toctavia/api/v2/controllers/listener.py\n\nStory: 2009128\nTask: 43045\nChange-Id: Ibbccda5859c8c6129d43e0742b28b319b49ddd1f\n(cherry picked from commit f166bb1481fb58acd7e7329ec394f43249001718)\n(cherry picked from commit e69899824ce09b8eb72c294562fdae5ad7ad5c13)\n(cherry picked from commit 5647db94c941f582bb8292599379a3693b470cbe)\n(cherry picked from commit ca66aac3750d7e2ff005111ade5212a2092e25a8)\n(cherry picked from commit a1e22d89a4852a28a3e60b31b4b293199daac0ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d139e157e8e67e86c8378241feaf5e12df6e682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d139e157e8e67e86c8378241feaf5e12df6e682"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
