)]}'
{"id":"openstack%2Foctavia~847726","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I4c218793948b0ed7ce0c143d6c794516fdeb224f","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I4c218793948b0ed7ce0c143d6c794516fdeb224f","subject":"Improve string representation of DB models","status":"ABANDONED","created":"2022-06-27 08:37:54.000000000","updated":"2022-07-12 15:28:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93f7aa00d03e205f389e9471928f47877ea139ad","_number":847726,"virtual_id_number":847726,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-27 10:22:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 10:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75e4d9e430dbbca412f681be8c4f844e2d515e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-27 08:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"844217af2af0138068a56acec40774176777f7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 10:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/704c920c23814e568d430eb6c0b0b46f : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f198665ae08b47989c4a285bd10ffb83 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5f31615f645453aac02551006f138f2 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98042de0a6554fb4996b81c2a7f25a6a : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8732968ed638477a9a8c62dbb783da24 : SUCCESS in 10m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc618fa1ff3648dcb8d4b03ef109e96a : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/672f6f3289154a72b9c19c1877ef6aef : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f8d0e950d704d099c22a46625f0c305 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e776d8a739e4570b722613bb8b28f00 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/412ca7ff125a40d2bd16c92e7bba17e1 : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce0614b212354958af091d0b4d051099 : SUCCESS in 1h 38m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a7788abc63f4e4fb540efa26f571f67 : SUCCESS in 35m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/248541940c074d25978d9770a7aa1368 : SUCCESS in 32m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ae3aa2a49bd481f898cc59ed0b84b4c : SUCCESS in 32m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba0cc3a9d3c449b990b391825dd642a1 : SUCCESS in 1h 39m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8261dc96b33424da526d1a21029b0cb : FAILURE in 29m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bb1a98b29044411a997dcb3de6bd9023 : SUCCESS in 1h 42m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc655a0100944a82834290a67880797f : FAILURE in 1h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8ef84ecacbf640babfbb7de03776dec7 : FAILURE in 1h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/48af53edf2ad482894830aff0a69cdfa : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93f7aa00d03e205f389e9471928f47877ea139ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-12 15:28:19.000000000","message":"Abandoned\n\nbackporting not to victoria","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a57c0221b4ef29aa30be816503335949f7461dcf","revisions":{"a57c0221b4ef29aa30be816503335949f7461dcf":{"kind":"REWORK","_number":1,"created":"2022-06-27 08:37:54.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/26/847726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/847726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/847726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/847726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/847726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/847726/1"}}},"commit":{"parents":[{"commit":"f224fa3294844abd3605270e0ee7debf4a41215d","subject":"Merge \"Fix compile_amphora_details when using UDP listeners\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f224fa3294844abd3605270e0ee7debf4a41215d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-04-06 13:00:54.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-27 08:37:44.000000000","tz":120},"subject":"Improve string representation of DB models","message":"Improve string representation of DB models\n\nThis replaces the default \"\u003cXYClass object at 0xdeadbeeff00\u003e\"\nrepr() for DB objects, which is pretty useless.\nThis change makes it easier to identify DB objects in debuggers\n(and logs, if objects get logged).\n\n__repr__() and __str__() have slightly different purposes.\n\nThe purpose of __repr__() is to return a string that uniquely identifies\na single object. __str__(), on the other hand, should produce a human\nreadable string. Another thing that is worth to note is that str() will\ncall __repr__() if __str__() is not implemented, repr() will always call\n__repr__(). More details can be found under [1].\n\nBy implementing a generic __repr__() in the base class\nthere should be no need to override it in the models classes.\nHowever, one might not be interested in seeing all\nthe object\u0027s attributes in a string representation\n(which is what the generic __repr__() would do) or\ncertain sensitive information should be hidden from users.\nIn those cases it makes sense to implement __str__() for certain\nclasses specifically.\n\n[1]: https://stackoverflow.com/a/2626364/6265131\n\nChange-Id: I4c218793948b0ed7ce0c143d6c794516fdeb224f\n(cherry picked from commit ff6d0c5280fbdbe8f5361469fb88448fa650f365)\n(cherry picked from commit bc95e34819bed67444c752e59c01a90c0c6df89b)\n(cherry picked from commit 77c86a2409a7e244c713a6d680d2d38f9f6e7bbd)\n(cherry picked from commit b55247d1feaa99b7b383a0a41e1685b57a547678)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a57c0221b4ef29aa30be816503335949f7461dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a57c0221b4ef29aa30be816503335949f7461dcf"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
