)]}'
{"id":"openstack%2Foctavia~848105","triplet_id":"openstack%2Foctavia~master~Icaefa292d9a4f081aeea2934f20b0304e139a9b7","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-22 06:47:21.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-06-29 17:16:48.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Icaefa292d9a4f081aeea2934f20b0304e139a9b7","subject":"Proposal for Amphora vertical scaling","status":"ABANDONED","created":"2022-06-29 13:07:43.000000000","updated":"2022-07-22 06:47:21.000000000","total_comment_count":10,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"3aa0ad2cd9038d47b2869cd7faefe5d1d00baf81","_number":848105,"virtual_id_number":848105,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"reviewer_updates":[{"updated":"2022-06-29 13:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 14:11:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"aa88f6ce3db97c3961fa4263d6476180c5378c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-29 13:07:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e1d708356e762849e134a636d1fa3f5895b0417","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-29 13:08:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b1b88587f2c3f859e772b83b48393a79cbe07215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 13:26:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd68701581248b3833ff2087b42a10a : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e402ef43484ca38f834c41e621d81a : FAILURE in 6m 48s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":1},{"id":"6b34d4454f2d44c329691ee68b0b0b02bab82e9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-29 14:11:08.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4e87acdf0e1af985b30a8f42992de2318378c5d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-29 17:13:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"34077a956337bf64f8f862998f444aa8984b7c8d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-29 17:16:48.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e726e2db912302c011a509514b1e638491427301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 17:22:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27467f35b7941fab251836d94436849 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5e0161a1cf419a982dcfb2db1f712f : FAILURE in 6m 59s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":2},{"id":"14aad34b4d7bd781682fd296b5bb2c8c78324e6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-04 13:24:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0674df2b7c447ca03c7022421f59b224e6458d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:33:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/370cb21b9dc24b5cbbcfc71a5a9ceb15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba764aa0c0124f60a6d113fe70824b1f : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1d95036150429b95cf8589399f72aa : FAILURE in 5m 43s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":3},{"id":"d59a54fe5461264666b464c7b501ea3c62e0ebc8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-05 14:00:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e572f260a80934d34bb8f0933b8e42e71db37e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:14:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66669ec14d09475b98df3dec7cd86208\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54519da1c0f54a3988c3e4d947284027 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7747f3d4a08438483277514a5d63998 : FAILURE in 6m 54s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":4},{"id":"ceacce0d34490930cffaaf4a2f13b1ea8afd18cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 15:52:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f91a1688e2a47231e50adcbe3c9ae26cb3d29bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 16:04:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0ca5a5f8033495a8ee0c054d241ec4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eeba2b287e04a20b2307d817c6beaa5 : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e789e4cdb5044db19096ff467914288b : FAILURE in 6m 46s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":5},{"id":"ee310aa9b00b7574631caa9730b2d2eeb8360f33","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-19 16:45:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"82d452a529ed89b979a457d5c4711fa165fa9860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 16:53:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e947b0f317f4e15a9e11c243de09bda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab81cb2065b14e1f80da4440697cedcb : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a6a7d59d62428babff82ada761d7b9 : FAILURE in 7m 18s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":6},{"id":"f271ee345eaae367bbcf7a82c076f1a133969f2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 09:25:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7f0b44f640db73f964f699a4f1eec439b4c5dafd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:33:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/910f0c91056a4aac8846c6c45c948765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4044ba9f0fa74f439baa691ff47bda2c : FAILURE in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a41117322ea4eeabd3fd297c50fe7a3 : FAILURE in 7m 18s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":7},{"id":"55abf3d2e166888b04b006ca652128df3fa724ae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 12:09:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ac36ea956fa68f01d1f0e789626859d516ebbcc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:17:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bba2dd9715f489daf38890bccf0a4c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f52470af0c54f7796a5f98433abb945 : FAILURE in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708ad947c3104c5f9f9961bd77a62b46 : FAILURE in 7m 06s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":8},{"id":"a5b6e948480a0add095ddcf25097881cd7e4386a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:24:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3aa0ad2cd9038d47b2869cd7faefe5d1d00baf81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:47:21.000000000","message":"Abandoned\n\nunnecessary","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"60b8dff8e80581fe125df967bd7328bd3e7237d5","revisions":{"408a841e31168c9e7f352b3d96e6cf00317e8edf":{"kind":"REWORK","_number":1,"created":"2022-06-29 13:07:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/1"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-29 13:06:49.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-29 13:06:49.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/408a841e31168c9e7f352b3d96e6cf00317e8edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/408a841e31168c9e7f352b3d96e6cf00317e8edf"}]},"branch":"refs/heads/master"},"c82cd828ce6fca3d964c3ea0c221eff3f1bc8b68":{"kind":"REWORK","_number":2,"created":"2022-06-29 17:13:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/2"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-29 13:06:49.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-06-29 17:13:24.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c82cd828ce6fca3d964c3ea0c221eff3f1bc8b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c82cd828ce6fca3d964c3ea0c221eff3f1bc8b68"}]},"branch":"refs/heads/master"},"2753035faa4c296345c15fb129690cae33c66248":{"kind":"REWORK","_number":3,"created":"2022-07-04 13:24:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/3"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2753035faa4c296345c15fb129690cae33c66248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2753035faa4c296345c15fb129690cae33c66248"}]},"branch":"refs/heads/master"},"92ec04d65159e4fbcf13d60c41c0275437683d72":{"kind":"REWORK","_number":4,"created":"2022-07-05 14:00:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/4"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-05 13:59:40.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92ec04d65159e4fbcf13d60c41c0275437683d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92ec04d65159e4fbcf13d60c41c0275437683d72"}]},"branch":"refs/heads/master"},"f4fecec076aa6ee688174a5ed702e058ef1cba71":{"kind":"REWORK","_number":5,"created":"2022-07-08 15:52:45.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/5"}}},"commit":{"parents":[{"commit":"b7730f51eab04ceddcdd8c3b6462a3558c7396f0","subject":"Merge \"Add missing [cinder] parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7730f51eab04ceddcdd8c3b6462a3558c7396f0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:52:08.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4fecec076aa6ee688174a5ed702e058ef1cba71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4fecec076aa6ee688174a5ed702e058ef1cba71"}]},"branch":"refs/heads/master"},"d7a6c1fa16a0505e0d780348f9614dbd6bdf6254":{"kind":"REWORK","_number":6,"created":"2022-07-19 16:45:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/6"}}},"commit":{"parents":[{"commit":"b7730f51eab04ceddcdd8c3b6462a3558c7396f0","subject":"Merge \"Add missing [cinder] parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7730f51eab04ceddcdd8c3b6462a3558c7396f0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-19 16:19:27.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a6c1fa16a0505e0d780348f9614dbd6bdf6254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a6c1fa16a0505e0d780348f9614dbd6bdf6254"}]},"branch":"refs/heads/master"},"0f86784d6156fc2438cc29b11f5b41096ded652c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-07-21 09:25:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/7"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 09:25:00.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f86784d6156fc2438cc29b11f5b41096ded652c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f86784d6156fc2438cc29b11f5b41096ded652c"}]},"branch":"refs/heads/master"},"c1cdd61cc812a326873ebaa907324d366d3e89c4":{"kind":"REWORK","_number":8,"created":"2022-07-21 12:09:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/8"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 12:08:17.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1cdd61cc812a326873ebaa907324d366d3e89c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1cdd61cc812a326873ebaa907324d366d3e89c4"}]},"branch":"refs/heads/master"},"60b8dff8e80581fe125df967bd7328bd3e7237d5":{"kind":"REWORK","_number":9,"created":"2022-07-22 06:24:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/05/848105/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/848105/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/848105/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/848105/9"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-04 13:22:50.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 14:57:55.000000000","tz":120},"subject":"Proposal for Amphora vertical scaling","message":"Proposal for Amphora vertical scaling\n\nChange-Id: Icaefa292d9a4f081aeea2934f20b0304e139a9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60b8dff8e80581fe125df967bd7328bd3e7237d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60b8dff8e80581fe125df967bd7328bd3e7237d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
