)]}'
{"id":"openstack%2Foctavia~848816","triplet_id":"openstack%2Foctavia~master~I4a0985fc93aa071e68491fe60478bee903de39ed","project":"openstack/octavia","branch":"master","topic":"fix-missing-webtest-test-dep","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-06 12:43:21.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-06 16:43:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a0985fc93aa071e68491fe60478bee903de39ed","subject":"Add WebTest as an indirect test dependency","status":"MERGED","created":"2022-07-06 08:22:08.000000000","updated":"2022-07-06 16:44:33.000000000","submitted":"2022-07-06 16:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848816-fix-missing-webtest-test-dep","meta_rev_id":"3b83509b99c0c5deec923f48a7d47f73bbb05d4c","_number":848816,"virtual_id_number":848816,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-06 16:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-07-06 12:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-07-06 14:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-07-06 14:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 08:42:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-07-06 12:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 14:47:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"bdaadfa0e6e1346544723073c16fc0d9657b3a82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 08:22:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd89efea13e1bb1b5806121895edeff41cf7b415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 08:31:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"208c5ed7e5f11ad7a8a59bc397aec3fc50a2e9de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 08:50:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9fb29d25b365253f7ae40812b02bf201bf7f77e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 09:02:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d9b6f52bac034c79e7165b9e85f059696e8cb93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 09:07:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"81704c0bfdb08d1872599f9f8e5ef0cdfbeadd86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 09:29:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e52f52683ed8d8b2698eaf280d66ce5115ce395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 09:56:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8cea2fd2acbeda414fc80c4c8eb13b9d776d0fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 12:33:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d83643fe80e4e17bd56d2dcd5f55d27\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a126a3b298e346b981455a2289d60a41 : SUCCESS in 5m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d49a0ea432c4cf99e2589bdbd0bfb9c : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdfbe5f45dcf4cdc8f1465031c4f3f5f : SUCCESS in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d02ea50d9444808b9ea23bb82c7d904 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ff8cd757904e468ccc664ab00fbbda : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33e873aa61984defaa7f1daaacb42648 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/785c779f02ea4fd89091a7ccff83722f : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb83d08ff3644516af9993e626250d3b : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48338cb964ed4cfb8be6ca8d1de32360 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81b472ef3b7741a29433d9c358c3f24e : SUCCESS in 4m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21bfa253a8a04ccf88f2b9776df161b6 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/211aa885d57d4237bbc1bf1076787a78 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dcbe602bf6b14d648fe259f02c22dbad : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b11784fe54964e799bb5e2c6fb743564 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0896a2c7d1ca4a128c13c23ba1cd5440 : SUCCESS in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9c09b19bcae4df5815ba9d1201dffb5 : SUCCESS in 6m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc53216d316f4122ae6b190b7b2d6bc7 : SUCCESS in 1h 54m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d62688ce958b4bcfa7c1a9a6d51bc110 : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f4a60f0e182414fb22e465f03161bb7 : SUCCESS in 38m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7c7e9e42d644e519876d2ce385b590d : SUCCESS in 1h 05m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/332f9773652b4b319fa342eb720a0be8 : RETRY_LIMIT in 11m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e023a1660abd45b189579140d84a2dae : SUCCESS in 36m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3464585b9f854fd08343a14aefe765bf : SUCCESS in 1h 51m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f0800bf8d984f95bbe23117442f503a : SUCCESS in 2h 31m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/64a7a63f27e6410e91bd1aa4e18a0d91 : SUCCESS in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d74bb69073324b64a9098d47dfbdc5b1 : SUCCESS in 2h 08m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/76007a94c42e4b6f99d4845e91f7b7c5 : SUCCESS in 1h 56m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1572a58eddc34bc08124442cf44846fe : FAILURE in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fa429b1b2774ecaead0b9b754017dc50492839ca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-06 12:43:21.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f0d9c846641e460405f622d759427f1952229d96","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-06 13:06:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c177fe2c3e59b18e9629ca667101524ea281dbf2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-06 14:47:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b72a8dcc28a37d34d80a25d42bc382f706f6bfc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 14:48:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ff116cbc491f817ecbfec957a78421c51a984db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:43:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a061e7d7638429f98cc83986ebf3388\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/14fe4db1992548ff94144c6f621b6a6f : SUCCESS in 6m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6771476deaec47af98ef0909e27d16d5 : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/488ec1d26ef847d5befc882876c7e8bd : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390660619c3c40fc8ba42d808bb204cc : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94796673a9444948b57bd766eef102cb : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9a4ed6a821d4519b796a753f98e70ec : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752139aa9ff541d9a1ff955eb9ff2fec : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de52b62212e44fc7ba2c2cf98b060dfa : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9aa48772d3bc42968fe38ca0e06e23d4 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cbf1cdc05464e149d4084f1bff9dfed : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c552069357f541e187717eb6f05711d9 : SUCCESS in 1h 53m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/beae7552b6b846979aa11499a4a7379d : SUCCESS in 1h 39m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca1c0adf273243d281c90d7a81b5ba12 : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/928bda0fee074aa3946487b5ddbe9b39 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":7},{"id":"9514439f8c27284c0c6c482ee7cd0c3f5b0cf136","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:43:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3b83509b99c0c5deec923f48a7d47f73bbb05d4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 16:44:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75b490eb49a4693bb13adcf61cf1ac6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/301dc443f61a4598962b3f63ca60f0ce : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3947aaada14578b8ad43fd44eb23ea : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d86609e1d8c4b9db766bdaead57d3e5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f45388643f0423199c9ac86a21a163e535cb0007","revisions":{"33ad716a675eb12fffaed70e69fca3eede2ceb4b":{"kind":"REWORK","_number":1,"created":"2022-07-06 08:22:08.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/1"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:22:01.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only. Since it\nis still used for testing we need to add it to our\ntest-requirements.txt.\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33ad716a675eb12fffaed70e69fca3eede2ceb4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33ad716a675eb12fffaed70e69fca3eede2ceb4b"}]},"branch":"refs/heads/master"},"95b31226a0ade1cf1b98d2cd77f8a3374c233c34":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-06 08:31:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/2"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:30:59.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b31226a0ade1cf1b98d2cd77f8a3374c233c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95b31226a0ade1cf1b98d2cd77f8a3374c233c34"}]},"branch":"refs/heads/master"},"1640f21881ef7bfdee0433dae2d76ee42389b48e":{"kind":"REWORK","_number":3,"created":"2022-07-06 08:50:23.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/3"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:50:13.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1640f21881ef7bfdee0433dae2d76ee42389b48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1640f21881ef7bfdee0433dae2d76ee42389b48e"}]},"branch":"refs/heads/master"},"fa5fe8d5e3d0c86152006ca44304e38fe820cf72":{"kind":"REWORK","_number":4,"created":"2022-07-06 09:02:47.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/4"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 09:02:37.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa5fe8d5e3d0c86152006ca44304e38fe820cf72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa5fe8d5e3d0c86152006ca44304e38fe820cf72"}]},"branch":"refs/heads/master"},"94a7c2cafa6a3628abc5af4f275fd4f1411d8fe6":{"kind":"REWORK","_number":5,"created":"2022-07-06 09:07:08.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/5"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 09:06:55.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94a7c2cafa6a3628abc5af4f275fd4f1411d8fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94a7c2cafa6a3628abc5af4f275fd4f1411d8fe6"}]},"branch":"refs/heads/master"},"3d21c6e8a7e439336f9febf65a2624cdf6c20851":{"kind":"REWORK","_number":6,"created":"2022-07-06 09:29:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/6"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 09:29:08.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d21c6e8a7e439336f9febf65a2624cdf6c20851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d21c6e8a7e439336f9febf65a2624cdf6c20851"}]},"branch":"refs/heads/master"},"f45388643f0423199c9ac86a21a163e535cb0007":{"kind":"REWORK","_number":7,"created":"2022-07-06 09:56:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/16/848816/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/848816/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/848816/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/848816/7"}}},"commit":{"parents":[{"commit":"7557c8f1783a3d66a69d4d45280a0cc993172221","subject":"Merge \"Add explicitly vim-minimal in CentOS-based images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7557c8f1783a3d66a69d4d45280a0cc993172221"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 08:19:19.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-06 09:54:53.000000000","tz":120},"subject":"Add WebTest as an indirect test dependency","message":"Add WebTest as an indirect test dependency\n\nPecan has made webtest an optional dependency for testing only [1].\nSince it is still used for testing we need to add it to our\ntest-requirements.txt.\n\n[1]: https://github.com/pecan/pecan/pull/140\n\nChange-Id: I4a0985fc93aa071e68491fe60478bee903de39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f45388643f0423199c9ac86a21a163e535cb0007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f45388643f0423199c9ac86a21a163e535cb0007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
