)]}'
{"id":"openstack%2Foctavia~849129","triplet_id":"openstack%2Foctavia~master~I1a0591de79be867483a044705e866b2368b2a567","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-11-28 16:28:27.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-12 08:07:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-12-08 15:59:28.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-12 10:01:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a0591de79be867483a044705e866b2368b2a567","subject":"Add cpu-pinning element","status":"MERGED","created":"2022-07-08 16:02:27.000000000","updated":"2023-01-12 10:03:05.000000000","submitted":"2023-01-12 10:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"849129-amp-cpu-pinning","meta_rev_id":"a05ed02b9094970c4ec73a6cb9f0451db67169ab","_number":849129,"virtual_id_number":849129,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 10:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-12 08:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-12-08 15:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-12 08:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-01-12 08:07:45.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 18:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 10:38:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-07-25 10:38:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-08-17 15:28:00.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2022-08-18 14:12:08.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-08-25 10:27:20.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-08-25 13:27:01.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2022-10-26 11:35:18.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-12-07 20:16:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e69ce97ccca04156a1ffe89676fa0bce58c5c69f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 16:02:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27019e0e44ea240057d45d8dfb25fa5b3e77d315","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 16:12:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"74face0550e6faa8496a5fd4f3d5d831cf7db559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 18:13:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2169ec261f5242a894e6c770ec0b668e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b87a30da2ac549a88d2c6f5f739ccc46 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9dd35d35be4b3680377f0a25fda213 : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0551e4578354109a2d7af8f785d4d83 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61887c28ce0948aea140d9666632b99f : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb2086ded3724804bcff29b9bd5cf2d2 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cee31d208ab40e69c5425cf3955eb97 : FAILURE in 7m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/37522e8dec3c4a9b945cc3ab873d8745 : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ad39d3e367b4499b83d4a85f5dfc3edf : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b4c039b77f74fbfab4b7ef5e1f955e8 : SUCCESS in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1bd5f3c7bd494062a2c89ae5bcc9096e : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b520bb3aaf3b40b680e5590a36cd4f7c : SUCCESS in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/801d7a18889b4d998f20d375af09ea43 : SUCCESS in 7m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ef24c1c67274e0a8715e3695b644de0 : SUCCESS in 1h 59m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0b64251ba8c44ce881d39e1f29ea634 : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/216769e5066e4c9781ff35c4af297fb3 : SUCCESS in 34m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd2bf09d5af243278f6c9e6a821f7c0f : SUCCESS in 52m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1bc473f482a042bda57e0d3586741b38 : RETRY_LIMIT in 8m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9cf9ae5cd1049f78a0db46d941fd486 : SUCCESS in 34m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73b08fbc95694a60b73960dee7a9ef45 : SUCCESS in 1h 33m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/633477ce99384d72ac05a2f496c1dae6 : SUCCESS in 1h 54m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cc61f3a6d7f46b0ba43fef8731e4baa : SUCCESS in 1h 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec7a0d3d05d04ea2b35197dfdc0bfce5 : SUCCESS in 1h 40m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb52005d294148b7a72f3db2fbe368c4 : SUCCESS in 1h 54m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/714f165d22f34f0eb47a5fd62c2d88e3 : SUCCESS in 2h 05m 19s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":1},{"id":"a2d08c02dd5c5a59cfef16bbfd5567c0f8eaa5f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-12 07:31:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33741d46466f4a89194cdca0fee597eeaf58d69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:07:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ca3b78664e640b6a3ebf63502f2b673\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d0efb76abf74393b5d7fc0982272020 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2773799be4444c7695a6dc0edc2d7c5c : FAILURE in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79272437af1c47ed9187dc5b9c307e4c : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc44b1c6f7794b52bbd16f30587db00e : SUCCESS in 8m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dce7646f954f4f8b8e3e107b1f853ead : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e8fd5fb5d4440fb1c8637ddc3185a0 : FAILURE in 9m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ce3eb2fb5b084d3295386fd0aa484ffb : SUCCESS in 8m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5f0b9546a84c423e816aca693d37e5b0 : SUCCESS in 10m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf06c15af2014727b2ec1d885448150a : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5464d90b9cdc4b0785901d0c19c4d0bb : SUCCESS in 9m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f1c46e73e9e4206a6e58438fb45b070 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e38ea279b8824c9582f813bc29294e9a : SUCCESS in 9m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fad3bf9ca96047dab1e229ba6a9f0e7d : SUCCESS in 2h 20m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/879ced00e43f44a1bdeb4cf07f20bb58 : SUCCESS in 2h 15m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4dd2fa13975240049ab56d81be83e7ac : SUCCESS in 38m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4adbd05b98ce48e4b872f363cf5c65ed : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cfc65b6e4e4944f28690a40654715fc7 : RETRY_LIMIT in 17m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e560f5ecd4e6442fa033fa133f89ee5d : SUCCESS in 53m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/02048a300a564c78b31800351c3afcbb : SUCCESS in 1h 37m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b741fa5bb6414393b25ff687e1bf23f2 : SUCCESS in 2h 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97d2c4e0a4484e1abc9bba0431fab4ac : SUCCESS in 1h 52m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/561acfcc9fac48f3b34c514da5571823 : SUCCESS in 1h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/365b8ae8f9574284ab1182325d6986b3 : SUCCESS in 1h 57m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a6ad9fe000f46aeadfd45c2477478b3 : SUCCESS in 1h 52m 35s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":2},{"id":"3bb87ddbc45ba1190e1d6691c66655b16d978b2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-19 16:45:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd8ab47565e3637761b3507d5bf7b8877ab1e0ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:03:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/becbf143732442ad9384aed69702e508\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd3a034d7444f68be5f46aa3ba7ccfb : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8276db61314bed8e5846301281e55d : FAILURE in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e2be90ac697431ab0b06b90e3bdde42 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e831993c75a1418e8b305b24b79c9be8 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6252fc6d056c4666a7b9e4e171e2bb13 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd55e6c477341eca4ef1211500ef125 : FAILURE in 7m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/db9db9a2b5db487ba1786f9678a7a9ac : SUCCESS in 6m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3b23a0c02395420ea2993fed709748f3 : SUCCESS in 9m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/05a33d217a6e48e2ab8c6a71253d0baa : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c773c08ff5a74374ae3cd8e9cf5f4c6e : SUCCESS in 8m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8dea7e2c6042430b828771244fb3c355 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec26ec8af0a64319a791e7926575c6f1 : SUCCESS in 8m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd128101e9bb4e4ea2ade531916c2fc2 : SUCCESS in 2h 15m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a8ddf9ea1eb4f6fa9f1c3c35a586c5f : FAILURE in 28m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/564ae6e1c23648b9ba39643503ed5bda : FAILURE in 36m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef49c509463c4362843098b9050c3b8a : SUCCESS in 1h 03m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2aba41c6edf944389dd7cea552ec2103 : RETRY_LIMIT in 13m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d49dd88ccc440ab9bfc6bc3edffd790 : FAILURE in 33m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fcedd75063f401e9b03ab70429cc41d : FAILURE in 40m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8c7eb96b66c4f87abd5d5e01aeb979a : FAILURE in 50m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/223c1cbfd11447dca1ebeeada1293d94 : FAILURE in 25m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb56fc8d7a5f4d3ba936a18fd89242b7 : FAILURE in 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16a45eeedbfd4fc8a08f6c83da529ac1 : FAILURE in 25m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd8cb789cbf9430fb267ca44e8f8a851 : SUCCESS in 2h 06m 16s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":3},{"id":"28f6cec79568fa1801916199a4fd43c62675114a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-20 14:47:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"128a118557d49afbba16f88db8abd909bb76549d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 17:35:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19a0a9686e3c4d64bdb32429b12dea39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97a253c821304b2a86ac808218e6d6d6 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d93332abb04f5d8b55fcc99b1b8b32 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e0e89af798c427b933e69392659d19e : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f7dbe33780b47b89513ec3fe1b1662b : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0dfd81f81fd45dca78d850982df9b68 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05dffebf2b4940439e32bd116deab282 : FAILURE in 7m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bd1f6022f4f54159b863a4c9bb1e6a64 : SUCCESS in 6m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/db96eac4e65f4134bd26da1b00853b7e : SUCCESS in 10m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcee06d552294306877dcade3dadb54b : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84783835b55b4d45bfd64cae0ad8c9ed : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61261a55a5d042fcb3797bc55b4700ed : SUCCESS in 13m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d90a0e6f55c44fd963bc2c3899e28a8 : SUCCESS in 8m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e08a3f9cc014342ac896f32f187a0c0 : SUCCESS in 2h 38m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f89b8475686d4d5cb977799981b0b14c : FAILURE in 40m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8193df8eed4a4cd284bc75ca425a6c38 : FAILURE in 41m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78f6a57cd3314f27aa5f9593938c758b : SUCCESS in 1h 22m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c00e1b2ae2a424198543deb6f8accd7 : RETRY_LIMIT in 24m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7ca8435493f4fd7a36c3f6b33615312 : FAILURE in 40m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a6820f187e149c99cd20cb6d99bb9ca : FAILURE in 39m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f84c468219e24896bb6a6ac3c2ed3a17 : FAILURE in 50m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37dd3e1124c644f18162286dd5d8713b : FAILURE in 39m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f07e334a60246ebbcc4b455fdbde865 : FAILURE in 39m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0ed10418c305444c86e8612701b13510 : FAILURE in 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbf704a7895740ad9fadd0ae3485cdb0 : SUCCESS in 2h 07m 13s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":4},{"id":"ad12dd39432c1042212458641cef52a9a222e282","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 09:25:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c62636abe9be1fd64f56c7796a40f130f1c317e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 11:39:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ebb2673d3ea4ba2a597ab8c53cf1d53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f5c32263c164b7883d5bf7561793c19 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e7be1e50cfb45f4903384fbdd09746f : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d91170488794551b00bb7b99625bd84 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82087f9d2ab146c69c9d5ddc14e500e3 : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7583d02605348fba847869cfcdc1a6b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9b1a82395b4e1a8331e66dd12b5232 : FAILURE in 7m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/73dc6e42b2d84fe2900263c1298fb3c7 : SUCCESS in 6m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/eb63e92986854d9aa18f4fecc5aea7de : SUCCESS in 9m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/171bc5eb73a949b4b7b7d1094db5e13f : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dea30ec45a0b4f259095bda51769ba17 : SUCCESS in 9m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ec6a6884cee4134955d0da3cd3b0644 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8db15403a1194105add7820b9294a004 : SUCCESS in 8m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28215839f8614c229e197c530aa87a31 : SUCCESS in 2h 12m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22510c3c2fb948b0addc24f6624d356a : FAILURE in 25m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98d72200c6b6419bafde685772035df9 : FAILURE in 26m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18cdc5a134374653b3566bb2ce2e61ab : SUCCESS in 1h 10m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/18805f4d043d4e10ac9df3cb62966e33 : RETRY_LIMIT in 13m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8bb1cf22d2345449c67cfbdb7c7774e : FAILURE in 25m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/585b866d18094ce5880fef41596eab0b : FAILURE in 26m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f33173934e8444de8886451a93a40f80 : FAILURE in 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07d2c6b1954c4025b6b8c9336a2e17a5 : FAILURE in 25m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6929f6d317354eccafd2eceeea8ba998 : FAILURE in 24m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3e8330601d8a4145994178fbf7e5e82b : FAILURE in 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c19ecb79dcde4369ba8aaf4ee30c721b : SUCCESS in 2h 12m 39s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":5},{"id":"1715c06187877ffc0dad01c0b03e1d7ea305c5ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 12:09:24.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f792c6779269250df3b43dcbdc7847cb923cb20b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:04:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcf06f1f1e4740deb610d74787dbbf85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c32e427f8a314c5ab5fcbfaf1325f300 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29253b27c01c4096b50a1ce6b1a14a06 : FAILURE in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ff92b6264d540babb42eb11a88b55a8 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46944656515b472db4fcb6072f3df33d : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c69c444b5f34126bda740a2a55f7e75 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad62f27f99a486c912ab2097dcbda75 : FAILURE in 6m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/24c20a08be7246a0831ed187de79f573 : SUCCESS in 6m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f2f01f8d2980466682c62fb28c4a8c3a : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce3a984dee2b43db8aa67e2c2a4a2acd : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e4ab1ca689f747eaa71049075e51694b : SUCCESS in 8m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/799acf05ad07470c9187a3b896d8de7e : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/604f160fa5dc4216a2602c20112c2a69 : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efae80d942324e52b495021df05ddf92 : SUCCESS in 2h 14m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6c21efb08a84002a060626093bad205 : FAILURE in 37m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39e996ec5d454577af14be6d6be87b0c : FAILURE in 42m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b21fb271f1241ea874921fe542a802c : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/326c901b65ac4b4187ef4bf43bec019e : RETRY_LIMIT in 25m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2ad5cefed1e453994a182f40b5a59c9 : FAILURE in 40m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a18fc8bee164466da1cbfd6748c71c64 : FAILURE in 39m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b19498b6578b4ea498e65d8824e92a5a : FAILURE in 51m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b17dbbadd55849728b9746214f849acf : FAILURE in 39m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e25fa5f89b714224b64cd6e83e3ba317 : FAILURE in 38m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b9809a3f9eaa4a08b91b8c4cb26897b3 : FAILURE in 39m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5053bcfe75064379ac2282d03a735f25 : SUCCESS in 2h 48m 17s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":6},{"id":"6525526118b00674bd34e417f6ecfee28eb5961f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:24:24.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"46ea8a74dbdeba04d0b8b2ec19d7c75a6faa60ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:47:59.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"000be272f29ad75bb6991ae0cb7d548685860a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:26:21.000000000","message":"Patch Set 8: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/610bc3d27a17464fa6ffcf044fca90fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c639dec5eb6e4e899b7d01cf5918badd : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3b94d07d294dc2be323a28f17006b6 : FAILURE in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/992dc7b8d25b44bbb9387ed40785e0be : FAILURE in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff182a81efb4c30a3f6114d7c2675f5 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e8751159b784ee6aa0658ccd274fc8e : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299755c7aec140aebdcf8438fb4b53f4 : SUCCESS in 12m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a38bda39ed5c4e058dd58e0ff8ccccd6 : SUCCESS in 6m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4c7171b22e3949f8bddda14e40d872dc : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/415b934530824a8f8da55ae28b555442 : SUCCESS in 7m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3e9045885134beeb2d2c14a9fd876fe : SUCCESS in 9m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/824f369968854fb78f8b2f19901320ff : SUCCESS in 10m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6edc21d76a34172b1940765b621ca29 : SUCCESS in 9m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f915537645e425eba5e838f51867905 : SUCCESS in 2h 35m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58cb0265824d47a58ead3e66d63a93d8 : FAILURE in 38m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9aa729c61e1a4d2d86fe3e63d2d18839 : FAILURE in 36m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55e1b5053fbf4498941a7f9b9962dc78 : SUCCESS in 1h 20m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/93491d6f50cc4f87b278fcbdbb6f383f : RETRY_LIMIT in 13m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d39efc3bda74ebf836071cb9836f8fd : FAILURE in 38m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8f5307e20fdc4b18ac2709d975c83acf : FAILURE in 39m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26c8d6c657ed4203ba0b9958b1357cda : FAILURE in 38m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2afbb0e5414413ab809b3a53e4e2786 : FAILURE in 24m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e4b54daac684b3cb5fff41bbf382c9d : FAILURE in 29m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/38d6d10b8ab64c43a67c0b1b81f260b0 : FAILURE in 27m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d864605c1c63450cb81f1789aba30fa8 : SUCCESS in 2h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dbb86223e9225721782edfb8dcf8428155eaf437","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 10:31:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ea4efcee9c04f1ef096c894ba18a7518d2356660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:45:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c5a164009f41dfa80e502d652f0b5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a3a5db7c7942c493969dce84913c87 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec075156f9104484af0bc365aab425c6 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f8a9cf302c74ffc81349dd55240ee00 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a5e856232df41d7901768d1f941d9d6 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f393b97dfc74f00b95e153fe9247102 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c238cb7524495db8e26f99c1cd3083 : SUCCESS in 12m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/571fefccf2a9418d97c1b17853f0fa54 : SUCCESS in 8m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/60779187d40b41d180e55151e37095a7 : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c02abb869a3d46db9860c14e5f26f068 : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8efda382db74075b21da6fc53f166de : SUCCESS in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12669bb3451045278d5ee7e713da140b : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/967d6321d53b41b595ce825259157907 : SUCCESS in 7m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bbb3569004c4a15847b26b3c3fd859b : SUCCESS in 2h 00m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0603f34ac56a4033a70b98da50aa386b : SUCCESS in 1h 49m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/587f65b8f9ad46fd9ce9f6c7eef76670 : SUCCESS in 50m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a06a123051a847ab944f72c3c1c0fa40 : SUCCESS in 1h 05m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c83a22d4fd5d4387a07b54db0b7395d5 : RETRY_LIMIT in 24m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72aaa44510b74a098208fa35af23903e : SUCCESS in 41m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45473478e821459c98632a2195027741 : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7da4fb451ff44e98001f23ff72cf47e : SUCCESS in 1h 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f9dc3619ffb84ca8a8204b62dfb04f46 : SUCCESS in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/17511e6866e94cb081bfd0cb16a31daa : SUCCESS in 1h 35m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/50afb063976f4c4cb5decdadb9ab352f : SUCCESS in 1h 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0972fc682f404ade9c01fac9ff15b02c : SUCCESS in 2h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5934ade35f76735cc33a3d96bb16c17fde7297a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 10:35:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"566d5c1547bc879a1addb78135fc6619ff2d3f4f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 10:38:06.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1d9475f096c8efb3ceb81b4737a8973949019d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 13:27:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/763db25f37554b14a090c9fbd13ecdd5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d82d106e0d384baf96bfb7b5013d2bca : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d53c799be24e7db3193b38c7f3ddc6 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e209fa975e4242df91362386310d4c32 : FAILURE in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5ac97712f2146238c070b968c7ac399 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25d05eee046b453fa7b56aa1f4cd718a : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9282d7c263747b2ad28e98397575842 : SUCCESS in 11m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ae4f3bebf9ae4d5688ff528df5655113 : SUCCESS in 6m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c005ec58b0984f279267954d2888b3a9 : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c579bcd1c943437593fbbdb635083eee : SUCCESS in 5m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cd8bdace9f247c889acec42961b783e : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/adb57cdeaa564eae8aaba749ff4cfc1d : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ecf422d5ea344aca34a2ade141e6011 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24cd0fdc1369457d9be5d804f3e0bb9b : SUCCESS in 2h 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24e4a6ade26a43a8b30f85497268d498 : SUCCESS in 1h 40m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98854bb0eaf34222836ba339cfdd571b : SUCCESS in 58m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9589beb857b4b27a8446ac5a547877d : SUCCESS in 1h 04m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b931c849bf1f4b38a70b30c9b7b63a8c : SUCCESS in 1h 29m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7c9e0708a9243479dfa85adeb7611e9 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c53df5139b514cddb34edb34a05aa36b : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c2a431b83524b8e8dc177d06abf34fb : SUCCESS in 2h 02m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a79148b5244162bb602c2e491509c7 : SUCCESS in 1h 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aea2c81378044e0983188d32acb3f11f : SUCCESS in 1h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd18cda19cd1421fa025473818b91b9e : SUCCESS in 1h 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/194a12e9e9f941e0858e8ad6901ca621 : SUCCESS in 2h 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"811056a28fc5c1f5fba45652dd52b96b77605756","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 14:53:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"da4c0cb3432c7b7491ced0c5e6e4c38f56d76955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 17:03:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb3fb4d3cce144ffa11d5a0ac1388186\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b42eb3cc1734616b8bd2bdeddfdc42f : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3744d2744b449e9171ea5b9a54cbe6 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/073cf17230b949b58e872ba3c24546ea : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85a9bc8c06cc4b159f80be60c632d70e : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2a20fee221447dc81b251914912ca68 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537211bde9364d89beccad2696d48994 : SUCCESS in 14m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9e4956dc186a48c892d610924541ced7 : SUCCESS in 6m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/018d025831e14d118c5d64dd8e51c4eb : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/114fa7ba360546aeac91e98c2668eba0 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/882af5f2c0f24278b1e807cbc4c9eef8 : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fd121f11ab4242e48645ee768a647009 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c67800945da499c9c3b6589ce69573d : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0eea4785e934ee7bf5eaae082edf530 : SUCCESS in 2h 08m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2da262e334040719b83ae963c18f0d5 : SUCCESS in 1h 40m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b21539925c46427b863f93624c7306ac : SUCCESS in 49m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7cbbcb84fe1445bb3979f9c6edc0a8e : SUCCESS in 1h 01m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/257ef7ea7aac471482d41d36b070b5ec : SUCCESS in 1h 10m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d73191ceee645ffabb783eb42910503 : SUCCESS in 43m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d2e47110d5f446f78c0ff9d7179d8859 : SUCCESS in 1h 44m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d460544d029346158ddf02e1d1888c5d : SUCCESS in 2h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2cfdf1c809d346829690377ef254af14 : SUCCESS in 1h 49m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b473d9f9a34f4951974d5258b023331b : SUCCESS in 1h 49m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c451a1e923374cc292b70c25fc57e0dc : SUCCESS in 1h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f0deb2ff6c9438ebf8502931d1f5be5 : SUCCESS in 1h 54m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d9ce79c16ec8ee0592ad24ca8eed978f4cfb5e17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-01 16:08:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ae457a50aa3aec44b0d0d047f8faf9e63cd2189d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 18:39:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a48798b505ae46458a90c4753295a77c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/281bb4443d264267844613fc8f52c6ea : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8c736526ac480cb37bcb9b6a3ed755 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad41a3444da64d4f96705428a2774469 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf5f5b1389f24232b783a0c20d03f95d : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08b8ad34b7374376a0236aadce605fba : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f14592dab14828a088eceff9e7672a : SUCCESS in 11m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cc7d939570a64745b64a4e888ca19e9a : SUCCESS in 6m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e7e3d8f2f1014a9981ca00d0a6e8ae80 : SUCCESS in 10m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f40cb5e229ea4aab9a379cb2dac7340f : SUCCESS in 5m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/338cc06d39914080ae0b05746392d8a2 : SUCCESS in 8m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/212d9ed511a14b7f95c7b5d01cd91363 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98f1b3ac29ef46e5a4b4f5dcfda40e70 : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18abc94f81494e838ffe057101ba1518 : SUCCESS in 2h 19m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/903dcb8779b741faa7bcee15696c9b7a : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70ca8980076b4d27ba10c0e4e4e97572 : SUCCESS in 41m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bfd082518ba14e56ab3b8267ba65eae5 : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bbc0d036db514d70b4493a197d1f46d9 : SUCCESS in 1h 33m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca9f6158ce5e43b5b23221b602b9a851 : SUCCESS in 41m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/37bf25040524434b8cedb8df9833219e : SUCCESS in 1h 35m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1d841d672f545f191ff2a73d28c1b70 : SUCCESS in 2h 28m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f94022f3ca054a3c92213fc136593587 : SUCCESS in 1h 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ecd3fa6343944f6c86a3e55cbbb65783 : SUCCESS in 1h 55m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/545c6eca5aca427e8b0cfaf269e09c7f : SUCCESS in 1h 31m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a706dab81c64dd8befb900e4fa5b577 : SUCCESS in 2h 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d3953cd3f23fa9dc6928e551ba555b633b1480b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 08:07:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a23f172748918976e2321c6ad2ef0b6df04eef19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:35:08.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b812efabe574c479fa7f38e105c6d9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3587828a6ee464ba8a2c532388ad840 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd259300dc94b948980d6d9707dfa6b : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61a99260a1e846749c1aedeb423b47c0 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ac19054d154d6ca7bfcf74cc640639 : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/676960efd06c49bca186a0cb4e7f8f65 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18107099c41f432d9445f0744a619d27 : SUCCESS in 12m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/da87936cd9a8427e9d21a8ff5900ad87 : SUCCESS in 8m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8bc0a88175664f5c9d62331e9841e9ef : SUCCESS in 10m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4fb7400409541a984eaade09bbaf35a : SUCCESS in 7m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e92b9fddf8646e491efc1c436bc407d : SUCCESS in 10m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c99c07269bac4f2cbf43f073f748bc3f : SUCCESS in 10m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c736b7be84f489795f9aad49a157af4 : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73a26ba42db947919b4526923f7fd5b6 : SUCCESS in 1h 59m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0217e4f88d2f4302bea666efa400899e : SUCCESS in 1h 36m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb67cef415c94dee944f0ad41e5a458a : SUCCESS in 43m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ab92952d36147faab48acebf54bb68a : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c196b002fdde4932b7ab4066fd42b3d4 : SUCCESS in 1h 13m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30a67b8715b34fd7a90f8cd0fe5ee0cb : SUCCESS in 55m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/127b1c9f62ee45fe8e1f9d902bee4089 : SUCCESS in 2h 10m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e49e05eeca494c6092398eee4137b3f5 : SUCCESS in 1h 54m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/620526f50a454678ad41b9f97dc1765b : SUCCESS in 2h 12m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/654b2680cf5145dcb09886bfe97dbcc7 : SUCCESS in 2h 09m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/351139e283dd4afe846f56c89419cbfb : SUCCESS in 1h 59m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa448fa55d614f87a3d7bff688c55abb : SUCCESS in 2h 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7d9e39085d0c579690fb061f7209ae76e9876ef1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 10:59:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":12},{"id":"763f23378174878d925e99860207c8c7461145f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 11:28:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d4e1d18004bc2f4c036505f2cf1afddd95aa5b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 13:10:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c91b88ae5dc426b9b612ec5f8cf4cb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a217e29e44e4bdeb90c0726dac34858 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784b5de56dee4b8e833aeb97307dfe61 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5954312e834d46bda061be734d8350cf : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aecce1c0db844f2fa0316b6c7f81faa4 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29ae3988ee034c089127ca10a32f861d : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96ddaa1df6940459ed96b7de490f8f5 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de234c87e45448ab9a51f335a7948fa0 : SUCCESS in 5m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fa82bf9dcbe2436d8d8fcd6f351f720c : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f21c6f92eb5d46659708e2bcf50607a7 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2328d15a075a47d2bc7f28ae5ec076e4 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e58a64f8c98649e09c9e92b23fcbff7e : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/84deba4de5a744c28d1d479f5a53bf9a : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/afdafbce84e34160b69d1d784f9deaf5 : SUCCESS in 8m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb1dba982ca140f79e12191c117a9fdd : SUCCESS in 1h 36m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf26c40de27e4ef68821d079051fcd15 : FAILURE in 24m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d93006a9995a456d90c6401897a0adea : FAILURE in 26m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55c57e1a60424eb09eb0e94cb8d97a0a : SUCCESS in 1h 03m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/647abbc23d04468291e2a19d9daa4a14 : FAILURE in 39m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f36422ff394a42abac99532408dcedf3 : FAILURE in 23m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfd77b3aef364a49bcb4fe3d3737ba13 : FAILURE in 26m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f333d50592f444e6bc0eecdf93386170 : FAILURE in 28m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc41d7f6d3f943c8af7ae4a63a638a1b : FAILURE in 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d219e1a28f25454ea7b0ec6010b18b97 : FAILURE in 25m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/45ec2ae846114b7ba231d571bb90eb6b : FAILURE in 23m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdc0af8357354835b5ef9c1c931377cc : FAILURE in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c74b24656eb49560eb2a8839b31c2ef90bfd396b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 13:43:50.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"53eb9e1a5c7298ad4e4eb11cae66750c1d1e8a2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:12:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e367b36de3764e2eb040b650fff2be89\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83604e1e83e2491995ec8c297ef9ceb5 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2cf17d89bb745799187bd9860121d5b : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8e35e0c36e40dda7f322c01e1918a4 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/504ca21a9ba44de3937356a8a5b002dc : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a46aa5fceec450a9efc5aa9b9e2a8f5 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9d5c490296468bb63dc2d846416ba6 : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/779599394bac42b0974ea097050fb18a : SUCCESS in 5m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ecc7352022094e14987c6ccafe9a9674 : SUCCESS in 6m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b8af661b795a4080903dea2a5c98b0df : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a0f5cac2521e44f49ad707ab3577ea09 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/370cb443fdf8438e9f63203955d098a6 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4de0973a554e4f479ef91e219239e817 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84fbb39259094f419d49f76747186d0b : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5cd41469c65749868261c174ae8b4575 : SUCCESS in 2h 09m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03fe1728a5c747d6b03f831c2cc30d22 : SUCCESS in 1h 26m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3acacd27224549ff82b54f98876d706e : SUCCESS in 42m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0def2bddaed94cc68854958178fcd8cc : SUCCESS in 55m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af7ffef69dbc45c1853ccf46b3e07517 : SUCCESS in 1h 26m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c27593d5fa7b4e8b914c844912d5d0a3 : SUCCESS in 41m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce6c7c8e3f06400a94c57a5811bfdb97 : SUCCESS in 1h 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55727171945047219f2fbac3fb117de3 : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/075243ded5f948ff955949759412796c : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c30234ff1fb4e6da6d4a89649454bc5 : SUCCESS in 1h 31m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cf0225e84a6e4d9cbdbfb0c134f6059b : SUCCESS in 1h 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7aaf27a004a412f8841b0df5ad091a9 : SUCCESS in 2h 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"73a623c56aac9a616a5ebddd60009584184ace65","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-10 16:21:56.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ae1adc1fbdc90e289a00b47a55a49aa2fa53ec13","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-08-17 15:28:00.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6a9173ebf71b0789bd3a0ddc500f087d15d4b4cb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-18 14:12:08.000000000","message":"Patch Set 14: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"988b56a49416aa76e63bcd637d04de58bd06cb1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:44:17.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"0f66597513db395d8fbfcae90cbf3153f79d40a7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:48:06.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"743f0acf2fe6f4acb4954adcf2e4cc62fa58721c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:37:35.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"02e48a7955bb083cb107693df2d882e7af21e4ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:47:51.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"4806c7698773344fdebf2be497888e792f9f65af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 13:03:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53508c1eb01449c9b63dd9e00dbb404\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c60abcb77b4d7497b5e96dbc87f33e : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e2574767544694a679c32dc989a285 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6988c831fbd472dbca04f25261ef5ad : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8114bb795e4940afafdcea45ff2ba465 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1980b248c64f40328cf9a4f849961c53 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3c9fd4b0bd424899cd94acc5102f1e : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f153c8370ef4f428ee105780b383d9c : SUCCESS in 5m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/06a7786b6553416bbf2db49ca6bc2eb4 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/31c19a0352b24f5cbba5550a890de838 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/221642f0107944148be36bfca6942e1d : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b12716eb15ca4aa59773682edb4d5cf9 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddb969a6eb744c2eace5399ee39e5aa4 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc98d8e7f6ba476f969206a7afcf2963 : SUCCESS in 8m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df3f435a83534c86a79c29b7615bede4 : SUCCESS in 2h 04m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/466982994d7c46d2a449410da4cad343 : SUCCESS in 1h 28m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a05a65cf0c54348b9508dc7c71bf785 : SUCCESS in 44m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65b568806601481fb1b91eff10f0b9cd : SUCCESS in 58m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/229f52343053493896f76749fcac9224 : FAILURE in 1h 08m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6af773b2d7d14f4e9fb056ab246c6965 : SUCCESS in 40m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/603fdbf623534362988f689f29fde8af : SUCCESS in 2h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73309505233d4b81b8c44f699aa52479 : SUCCESS in 2h 13m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9cab14ab92854e558e966b5f5b179659 : SUCCESS in 1h 49m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/abdd28d67b4b41769891f457634830f8 : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/25a3fddcef974b4bb642de326996a9d4 : SUCCESS in 1h 48m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee36740dfb54ad09cbcd08f8239e0a1 : SUCCESS in 1h 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"431023a255e2212be75351cdbd43001f448d554f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 13:22:19.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"401b9aaa1409dc1032d9d094d8448eb534884d3b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 13:22:23.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1897527f47c09be467662b13def5ddfb0b9fa8ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 14:46:32.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"bf9883569193a09c2bc4c7e5cdcf0ec58f06afba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:31:59.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2008114d574dbd986d0684d2f0cfe0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b05246853334917b6f0c68087af7174 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e691a0ef5e49ca87e727a52135b970 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb44cc2a66f442bbbb321cd230699475 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5b0714acedf4f39ba8a8d9a4c3138d0 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9daccb250c8241e2b56bb93f15b53f87 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a182e066068444958c8bf16cb2a4efc5 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f681a256c0e44649bc3035799ef661c : SUCCESS in 5m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/576789a29d344688b3372aaad2fc4fc9 : SUCCESS in 6m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9a2b81013ad244d8b9d11f62d4cc23ba : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/beb0ac292f3b4c6db06a7ced25c3aa11 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9aee5e7a119f4cf7881cd5c85b9d14db : SUCCESS in 8m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b7ece8ef11740d8ba8ed97423e2b34a : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c7a412aa72b42698fc52134af032eba : SUCCESS in 9m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2abdd67e412e42e19dc55ff22c4ccd38 : SUCCESS in 2h 16m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8db2196663ef48b59dc6d045bb6a989a : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e9bf703ba6c4beb9e459ad014512408 : SUCCESS in 42m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4df8dfe284914a4694c65bcce12d3161 : SUCCESS in 58m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5864d8772c2c4a6eb0711aaea655339a : SUCCESS in 1h 28m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/296ceaa7b2a84a1e90661dffe8d9862f : SUCCESS in 46m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e46606ace834cdd91acdd6ccef64c95 : SUCCESS in 1h 34m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/525226b73ce8481493826de13d14ddeb : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d465ec6d6bf84be4b61dbb491f387ae5 : SUCCESS in 1h 48m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f08811bf2b1845ed9cba47f6b95a22b6 : SUCCESS in 1h 48m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7f600e2dde9843a29821d582364333fa : SUCCESS in 1h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d263abfd171417fa01d7f39ed2b834e : SUCCESS in 2h 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ade6a76259c8a66220ec42728316a8aa70c6ac61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:00:58.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"21ac5cc15dc1c4e40b70d90b88b8458385a0f5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 17:31:55.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818fdad5f24642abbbc03a759b84df2b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82708458d5a8436cb468ad138239bfca : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7b5d72ecaf4e478bfd94db32a0f0ba : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85993efe657542cc952116e25c37cc9e : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def4323a47274c529348d51769733b05 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cec01e0eac64e32a85f1b38c74fb1ec : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfdc14c5b1944e2983f9120ff19bbe2 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2c76a803d664e969cf683923c2cbfd0 : SUCCESS in 5m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b9a524faac2448c8ad2bcec277a5e81e : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2a60d2b614714dfb993e4277aad2fd05 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b7d64870ae64640bb9b2022ea116576 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d50faa9129b4db39a8b59dacbe46d8f : SUCCESS in 10m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e69a0b088484038ad00a455f307f568 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15447045487047dbb371496f7b660ec9 : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbf3ade7747942208dac2638294400cc : SUCCESS in 2h 23m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d7d173f972d419c998187655670a92d : SUCCESS in 1h 37m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/570ceefdd1a640a0b2d0f572bfb1c2e3 : SUCCESS in 43m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/366245f256c1436f916ad03d0dd3d8ef : SUCCESS in 1h 05m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1e5982a10c704869b3eb3c26c1dc80ec : SUCCESS in 1h 32m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e896bb539b1f43daa34f782954228d44 : SUCCESS in 42m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f67bb1cefb2a4f25944cd5011d3c9b87 : SUCCESS in 1h 40m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35f45f1e32fa404abe0b3e4f323e1714 : SUCCESS in 2h 02m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/736602821f6147f6a63dca789afe15cd : SUCCESS in 1h 36m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a118dd73c2e44e4db441f719a221098e : SUCCESS in 1h 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/de87c5b93036477d9e475bed42246275 : SUCCESS in 1h 44m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/47a12e3418d74a98b15f46d4b5d4814a : SUCCESS in 2h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6bda57dbafac5e4dff3fe5bf43aba8f90a224936","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-22 08:02:18.000000000","message":"Patch Set 20: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"48c2af54597787469bfb0b46a2b11964f6523736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 13:52:53.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"dc56d53c3b41704c94074fa170212dc12a993039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:58:13.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ab4c5da5608449ca25c8372b3ae0373\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab81ab6317d84f6a9fbee03718ab6493 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13259fe885804d81a0b739e77951ba74 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ab578f55e0842368a708d5a997f03f3 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4b09c6638e04e39a84f80f7a9494952 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64995f4a62f14fda86708e72c8a0eadb : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e94050dc9944fdb1c45e65d65fd5c0 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57d6d786aab4c0a87f197bdba44c617 : SUCCESS in 5m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9aac9e23fe33432faebe0388767f611f : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/27d3803da7834f3399beba91e21f34d1 : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3464264bc84847bcab8fba010308d6fb : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7a5e5ed1fcf4e68ab8035491363da6b : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/467a31b5448746d7b8567fc2b93f7127 : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/489cd9cbb0c04f23ae64d85bfa501d65 : SUCCESS in 10m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4f9803645b749479aaf6b01506289b3 : SUCCESS in 2h 03m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6149d6842f14e758a232d29872d0bfb : SUCCESS in 2h 16m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c94f30842ffb491faf1a843d141512fd : SUCCESS in 57m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e36f543e38984a2eb941d58b1746852a : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0aea4919cee24d809157b32b9dc0a261 : SUCCESS in 1h 35m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89b529db39944ac19ae5d894af4f9163 : SUCCESS in 56m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd85231d061544a3b013cf51f1d1ed5a : SUCCESS in 2h 25m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41fea31a57ec4823ad7e521326f99004 : SUCCESS in 3h 02m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf1bf30d52a4ca695a3769ce63c0193 : SUCCESS in 2h 08m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a3a62032577845d386d88045322436de : SUCCESS in 2h 12m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8accb43e0836423d8de7e97c9c641a4d : TIMED_OUT in 2h 42m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d19cb7f5b50459fb2b6af11bffb4061 : SUCCESS in 2h 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"616d88fe21a40715d2b9a92f7564be0f10246a3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 10:07:03.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"56948a73b973650bd371643ff5493b4981aa1cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:39:38.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0ae21e39c5a46d9a8427f95ff135cbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b3ca5d8592d4b3b956315d0782a31e7 : FAILURE in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d4907818ea48068282a1b9e5ea8e49 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c1b72cf4cb4ed299a9557ba1455b11 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f0a2a1002de496f8a96432bfac00e8d : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66631e97657d466c91b029c0d3a25fbf : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c965af6e90f42a6b15a85bd7c9bac3f : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08e1c080f47f41ca8a43e89539f6b646 : SUCCESS in 5m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/60f4e44606914e148ada84fb051cdbf8 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/468d56b0029b4dbdbc75ff634de4da86 : FAILURE in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f1590dde42147ba97a162f380e38f66 : SUCCESS in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e23c6562da44c5385a9e92d9195ab90 : FAILURE in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32cbd1b92e5e49948e2b970b1f438e34 : FAILURE in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85fda442c0f94daf8ff1e5c8d7d27564 : FAILURE in 6m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/740327624a394fd69bb5d2c85df01c2a : SUCCESS in 1h 53m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf088b9e04ac48d08c8b17b8eb7343fb : SUCCESS in 1h 50m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f2c6884d7cd4e3dad52fe5204893c39 : SUCCESS in 41m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68a0e1572d72466baada1e425cdad81f : SUCCESS in 1h 27m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/841bbc77bdcc4b5ca9788102d55ac84d : SUCCESS in 1h 12m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7923abe30a1a45efbc5e889a97515b64 : SUCCESS in 41m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e50f1496d9c49c5896fd992b30bbb9d : SUCCESS in 2h 09m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9221d7d1f30a4756a070bfc534e872ab : SUCCESS in 2h 22m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/263babc7aa1a4f0ab7f8a7abb70dab12 : SUCCESS in 2h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f15c102ea794446bd17e562329b5c60 : SUCCESS in 1h 50m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/30660b49f51048829c1f451649a1cea5 : SUCCESS in 1h 54m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fbd7a8ac2754c28a2e76c1b787cefef : SUCCESS in 2h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6b045f23d4721df8208c4c1abccbff286c553e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 12:45:59.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"c9cf1bf547225cadcfcb0a2300c3637fbd24cd0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 15:35:10.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a38d1653f254a26ab06cf26c28bc98e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49473ac9acd2427caa80239d90332b4c : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd2978e3cc249cdaa08e638aa243871 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d05e264f1b4e4e53ba56919132ba0664 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab7e4a4a81b042c6879310d6346a272a : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f88a97a26e584c33a4fb710b395d3a80 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cdde634ce94f5585ec65371dcd5b19 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b49aae4f290947f285fe4b838b9e1cfa : SUCCESS in 5m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cc888d24ab574022beeee26509f85504 : SUCCESS in 6m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/48a5ef6227f149679f6f78e3f284048e : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bab3186378ff4c039dfc2ca62fccdac1 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c245cb382c30405798b2990411d929a7 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/607574fdb19a4481900b192fb1837ab4 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ee418d5dd65403abfe415be699f8d09 : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df275e8b7d99446ab640be6411fb754c : SUCCESS in 2h 42m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce4a0c2aec164f148dabd7843ab9b4fb : SUCCESS in 1h 50m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71a4c0a20b004e0192854e75a57b554e : SUCCESS in 41m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ddce3b3230e45ce85d3900d2c6fbca4 : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5da8ce1577e349c2b68b359d9a8bd39d : SUCCESS in 1h 13m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7c4eaf4894e44e2ae573ebc3090b49a : SUCCESS in 41m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54f3f6f3c8bf4374a8534378c6fd59d1 : SUCCESS in 2h 05m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a730390e388b4b56aa853c878f4ba884 : SUCCESS in 2h 14m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed47ec176f194c8fba2e8b4e603587b0 : SUCCESS in 2h 00m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e43972a941794e8abc44b8aa9adde283 : SUCCESS in 1h 59m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e579265c9284447d872c4c4ff4969610 : SUCCESS in 2h 01m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5ddb24203e544649d65320dcdfd9000 : SUCCESS in 2h 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"eca4c3630a3219b8c2b74bccac2f6b57dd2419a1","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-08-25 10:27:20.000000000","message":"Patch Set 23: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2df65b956c3c1f6b96f91ce12dddf7de1fd6a3ed","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-08-25 13:27:01.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"174bd107418fa03a4231ea3e1e2ca79ac2850d76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:21:15.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"903185a1d40a4d2d06e9c51ae03fce46dcc60f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:35:14.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"bbb631242e5b93f5ec9c9ddbf1f412a44a3ee21e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 14:04:55.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f27943f7edb455f8212d4ec42e23813\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8396d1c52c394555a2b90ae914d1e520 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce62bbc2f6fa4047b7b7552f5b4225dd : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c061df2289294ca1b5535a7c34734308 : FAILURE in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47db02e1e84f44f69aeead40f0546da9 : FAILURE in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4416bdaac0d34ad486e6a8fbc32d59e5 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604ad0f6ebdb4f49b813ee8bd87e255a : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eff4230bc1242d28717ce7013091975 : SUCCESS in 5m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f7f2c344092e48c4b574ee74c933f47a : FAILURE in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/742c5a0451bb430f968198548129efc6 : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0952f92d5ecf4a5784e251ed6853c035 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/174f443ff7ec4185ba0e432ce641c7b0 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fdf1ee42482a453fb7c621203901e1eb : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33f10b72fbc747ac929ce009996653a5 : SUCCESS in 8m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7cae1e18b234624bee48c3a99d33251 : SUCCESS in 2h 19m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef2ba72e741d4dcca1dc1387bed3c4b3 : SUCCESS in 1h 27m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/477fc01e490d466cb7a4ee4444799be6 : SUCCESS in 50m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca39e379282d445c8fd64228b6f06fdc : SUCCESS in 1h 05m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d51513a913644dcda455aad519c8ecf9 : SUCCESS in 1h 10m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf5f85586dc74edba9fc77658ba9d540 : SUCCESS in 46m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8d09e09cf901428fa599a0f01c60b2d5 : SUCCESS in 1h 43m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/909054078c794582a3df4b8577db8f2b : SUCCESS in 2h 00m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10a561bb2580402e89f4c5b375c12992 : SUCCESS in 1h 52m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/284bdd26e91a40a8aafab67ea1a6f3d0 : SUCCESS in 1h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/42811ff476fa4b5b8f98ccfd329072a8 : SUCCESS in 1h 39m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c83f1c5fad7a4ce8aa47eba684f4dc4f : SUCCESS in 2h 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b757ef6153354f1918836c780e177b1145d93806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:22:33.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"c801d89675fe2e4818e8e8fd7d07fb3dcda61e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:53:12.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"ec23924903e8d7b340f376db8dfe98fbca486adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:20:29.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f934e9becf242a0a02a6478937a9160\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6734221ebcaf4427b0677e78b8ebfe4b : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cddffed766834c62829ccc7e53ff9a32 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c007466c99464cebaf1faa6c40d63720 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daa27f8c97bc4f309a9da3fb1a6290ef : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e677964368eb4d26a36759e94166826e : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf16de6de06c4f02bc3438705a941a37 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22ace1bf683c427ab70b61d56b5b8493 : SUCCESS in 5m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7c5f9275f4e943479f119bfccd18e680 : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/362b669f71334df1872d1c474ef1aa75 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3d4badff434495e912494a5ae4061f4 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0ef1c8dce1645a598b5229ded319b16 : SUCCESS in 8m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e5759ba69154e3bb8e973fb2d5094f5 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4806862a93494d1e82fbe472cd0bc4ea : SUCCESS in 9m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c521155861994aa3861f8ab33f045b4f : SUCCESS in 2h 11m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/969999dc53544fdd8a1b3928a50e1203 : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b20e186ec78497590348bc9827f0cf3 : SUCCESS in 45m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a1d6ca17054649acb496feeeeb3a4fd6 : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5f45e7a28e6348f281f67d3bdc2f7f73 : SUCCESS in 1h 19m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cce3027260e44325b34782bae80fba1d : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3b737c80c3b4f599c59f746765a13d4 : SUCCESS in 1h 55m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66756c6885a344c09efd143675d1d613 : SUCCESS in 2h 09m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07ae4da7ffe84f42826a8a553001d793 : SUCCESS in 1h 55m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/321ba0dbfaf54bde958398e766cdbbb5 : SUCCESS in 1h 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/74e73a99e129408d8a5588f7189ca1bb : SUCCESS in 1h 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c908eb5e906446784462a4206b7a3da : SUCCESS in 2h 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"2d549418c8b299f9abc3bd3efb054338e1dead65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 07:29:59.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"67ad6fc562f79038f44022fa7d098c26d09dcd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:09:52.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"391e2343ff90112343a2e5a8efa7ce2f77a9bbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:06:05.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50e14f04e77e4b779b7ce7de1fac4f8b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3021ac3bdd694ce8aa2fe2a30a6bbbd9 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4861523b8099454ba5efea8a61c21474 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9152391243e948ba99ab54f968e4d7a4 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd8c42026b8474285a22bf23b80042e : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4268b0fd813a4965986b7c26e90f90dd : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58472e58730438cb310a72a3e209484 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1e180c92c8447da2c2ac67d1b2ef49 : SUCCESS in 4m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/de71e1521f3640e789755db946d86de6 : SUCCESS in 4m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/510d141573864efd8a51434057f5ed4f : SUCCESS in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7068dd89ab5a444cae4fb445ad432678 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a22b911966b443d8a580fee21592f823 : SUCCESS in 6m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2c032a26aee4266aecdaf79fb4cc3ab : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64afb18c4143482c8841801b29bbaa57 : SUCCESS in 7m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c35d62938444bdeaeac2e68c018c005 : SUCCESS in 2h 20m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab563a7ce3584a44a0b641ef30c60a80 : SUCCESS in 1h 59m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e16d0e56397640629ba0f789fe46f783 : SUCCESS in 39m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b9155898e9f43979cd8140be3bb45e7 : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e08936ad85a44987b3d2cd980339246b : SUCCESS in 1h 33m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb027830bcc54910949c23621075fd1a : SUCCESS in 57m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16650374dd12442daa3d95a48e190ed0 : SUCCESS in 1h 49m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59ad5706b2414964a951a142241eed9e : SUCCESS in 2h 49m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/329d1b0ed2f04e7aa1d5d6c083b32a16 : SUCCESS in 2h 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/50c272fe37ca4a689f60568ca6c88e49 : SUCCESS in 2h 25m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f56bdcd552b342b58864e4a7f642995b : SUCCESS in 1h 30m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6fd88ad48e84114a2a3acd0bf6253fe : SUCCESS in 2h 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"31f266e69afe12c6e583ffd0de7b1f91b490adf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-01 08:56:27.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"c731bf27073f8c4cc7e8fa32c22f3c9906323e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 12:07:19.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11d5db8ab2c5457691f8cd206f15ff31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e570efe0ac46c9b8fa8c0fa5d11bbd : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5489d1f81c54ad0b4907ac81f3fb41b : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/370e2e49a67341c49c8aaca68e3a09d8 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80b1f5e859840bcbdd33b5c05acd462 : SUCCESS in 9m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c363d86141504a92ba71e00992ef167e : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2629c108a06a4509abd399d9c156559d : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f5253aa9a4b46509cf4ca140716b284 : SUCCESS in 8m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1d2000dd78bd4c908a6fba0b791fe33a : SUCCESS in 6m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/78fd9f22adc84a3ea93a07336716ab50 : SUCCESS in 10m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/114bb0cf06e04741acf05055fdbf11ef : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b36215cc005149ffb6fac6862a5413e3 : SUCCESS in 10m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f61dfda60b544bdd8b539db3d78be38b : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79f02201c94348f28faa7c530aba29e3 : SUCCESS in 9m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/327195d8fd364c8092f9f14c27c02d90 : SUCCESS in 2h 37m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef065e4d31ec468ab4d69d1ac57238d2 : SUCCESS in 1h 35m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8f57ada87f34bfba3792885887f67f7 : SUCCESS in 39m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6528bc32dcc64b9d9636e62657f1fa22 : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c04b38eb9ca2450e925e59790863f5be : SUCCESS in 1h 35m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0316009e6655459881a124549a0cccb8 : SUCCESS in 41m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d5be4c4c6f104a2daa9035c76ba32c49 : SUCCESS in 2h 39m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c63ec8ba718f47458193148c13904bc0 : FAILURE in 3h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e3220a46b8ff461c855e684c3997d250 : SUCCESS in 2h 19m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/061316bf49ff476cbbb481e207878dad : SUCCESS in 2h 03m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/18dad18fabf2469abef8fcbaf8760f13 : SUCCESS in 1h 53m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/75eb37930f274508bae7891470a029ef : TIMED_OUT in 2h 46m 32s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"aa6840d4c6502b216dddd519cd32155af0c6130f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 09:57:43.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"fe439846dd974c1ceeb5b37ede3d2c4d94cdc751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:03:44.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"24ea93702212103a3a6501d7e01de832338c21ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:08:17.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.","accounts_in_message":[],"_revision_number":33},{"id":"b2ceccb94b907ffe4021f1c5b82caefffa9a8dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 12:28:16.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.","accounts_in_message":[],"_revision_number":34},{"id":"0f9149143f477b48e5fba95747665343489e2c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:02:08.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbaf0fa42287449ea99026c4fc262538\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1819fb368d8944f4abf7040c9692cfe7 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b9f7179a0d469999ebcd1ca22ed3d1 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1299c73be144e799df26efc282f092 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be33f563348d48d7939da827477ba193 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b4933790c324e45be3dee62ef5d42dc : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a481b666316498bb778b0ed4ee5ed77 : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdd14059c1684dafa7ab06346f6a084b : SUCCESS in 6m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/05a2588b00ba4b13920a4817dcd6d6a9 : SUCCESS in 6m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/331de6e296394c739fda759d02d9db3c : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/51c14132e47d4cc089085929256a3f56 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8419a8c65b8a4458a3e74cc070e55181 : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f7c3ea2522c9418da026d7584a792239 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0109d864b2fd4d84aee30506e941db40 : SUCCESS in 7m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7bba3b21a78441a39f091c6b10288ed3 : SUCCESS in 2h 30m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc7f1e336ea94edaa4c551ec0b167218 : SUCCESS in 1h 45m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/761e2f7ec8ee4585a3921a7a7508aa3f : SUCCESS in 40m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3869cb50e57f4a8fa64d4c93bc9dcfda : SUCCESS in 1h 07m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/20672d2ceaa24509a859d17c164935e8 : FAILURE in 1h 12m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc9484476528465ab1f53a097add3ff3 : SUCCESS in 53m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a5a4a9885e7f4f349c36dac2b27f5a24 : SUCCESS in 1h 47m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a79c5699c26c4767bf31bc9f2a07e24e : SUCCESS in 2h 22m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2215816bd1bd4144abf06b7e27069e88 : SUCCESS in 2h 31m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51467c563a884d85bda88c870668ce16 : SUCCESS in 2h 25m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7fa54492d8d440b4b527a1f4cc238ebf : SUCCESS in 2h 07m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bc7938af04a421881ab420262e355e1 : SUCCESS in 2h 09m 20s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"c720d864e8f5172966805489967e4f28e48abce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:09:36.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"73eae1bb4aadc80c9786d029f69769548130a20b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:11:01.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"9a3ddeb10f243e56dea43000e403e0db20474c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:49:03.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74757bca029e4b94a32cfcdd015a3ebd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a233e4126b4f26a990548f377c0b69 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9723679e4a3c4cadadf9adae4145467c : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81458b7763d34082b82e2476669c8f98 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0b44b60510442929a21fd1f5b19060e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4cc8ad1067a47cf96047253cdbb0095 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8b9c3e8a04740bc9108191a994c89cf : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/966e177bc4c84b48a8689a4109606fdf : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/286ffe33e9454645a3b0522f1f907963 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fde274452c774bf5b72a6a69a3c2f8a0 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae930231a8be4fcdbfa119f539d13d72 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/99cf2897ffc84e85a784fd8ddfeb475e : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fcee9768468b4037a13a0aaf52742076 : SUCCESS in 8m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8f2cd1f4a874406a9d2d999869eba38 : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10a098e60eb54a489ceeb5fb9577d812 : SUCCESS in 1h 30m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d35de8bb85984da0a878ab293f06ea59 : SUCCESS in 38m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a98c11bf5dea45d1b16ed724cf79fdb0 : SUCCESS in 50m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/149542fce89e461ab9922ba97f03dc2b : SUCCESS in 1h 16m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e89a9183c79441da20f45fdfc18e1e1 : SUCCESS in 32m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f75f2fa37967464cb48ed33c463b71ec : SUCCESS in 1h 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ce199a5cbfb4092aa2c09e8d82557cd : SUCCESS in 1h 59m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d5b303268f2444b927eafb7bdecf4c6 : SUCCESS in 1h 48m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/42ea7c9da39c4f70887584aaceb62ac7 : SUCCESS in 1h 40m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cb55c1d1f88947f48bf8a1014fe66eea : SUCCESS in 1h 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9154886b8936460db6f35154d2944608 : SUCCESS in 2h 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"f3901cfc3c8a01e3be39cbaf146d6540b49a39ef","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:35:18.000000000","message":"Patch Set 36: Code-Review+1","accounts_in_message":[],"_revision_number":36},{"id":"f80c065df85f1c005c1a098ec832c4eed035788f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-15 09:56:47.000000000","message":"Patch Set 36: Backport-Candidate-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"5bdfb0998f5f723967ef0335cc14c1a02477e52d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-16 10:46:01.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":36},{"id":"919faa66b7282233f17e544c0f230e2a2412403f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-23 15:29:39.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"1447c522b0b9257a3f070b3d1f519615bb52b30a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:06:53.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5533a939bad4d8e9bf252d3ba3345a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4a377d7c4447b1bdffedeb68990740 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4683faa6c641eca911dd6c32e87f07 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca8f5e4bafaf4b6bbf50b8f81faf48ba : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a7fee50b8034b009c6ecfe3ba769be2 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b6d6e013484e5a9bcf29b6748a5e7d : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/702f901d90c048ada443c00e91378395 : SUCCESS in 4m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/06381a9c986a4e30b2e94f13928a0aa7 : SUCCESS in 6m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e94b496881904cc0b725383376031175 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9fb6a4767a5f4763aaad275e86ec8ac0 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f713c6beb02d4d8496b6e7c279adbebd : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08b07ededbc049739ccfa7d1b4faf983 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84bd7d49086b4b6fae62f9d051f753fa : SUCCESS in 7m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d34cc3893bda472fbbc9324768d48a7b : SUCCESS in 2h 02m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e442410c44f4b7d9e9aa045854b5989 : SUCCESS in 2h 05m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e32419188684333a9e15f0cd9b330d3 : SUCCESS in 38m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7c822fc4d104242b5e2c95daf2f027a : SUCCESS in 1h 19m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/13626e2dd3d14e289fa3edb764dac1aa : SUCCESS in 1h 16m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0dc5a067350a4e1e8b9df3a241bbe91a : SUCCESS in 38m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0bd9bf85477b4bdc9c8223b83a2fa911 : SUCCESS in 2h 01m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00d40ebc0cb443c5888d506fba82541e : SUCCESS in 2h 26m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b944065b51742c0becc2b77591fcaf4 : SUCCESS in 1h 52m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/661b91834b80479f964efda7e5704ba2 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b47d0c690046487cbff57318572a1005 : SUCCESS in 2h 14m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfe7f5ea35234e61954463fd880f1700 : TIMED_OUT in 2h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"2b2c89a39332c9320739d4f6564c347dc12b7590","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-24 17:04:27.000000000","message":"Patch Set 37: Code-Review+1","accounts_in_message":[],"_revision_number":37},{"id":"c14d15abd94e25374752c2dbb7e7efcf30949f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 11:00:01.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"6be81c816aabd3aa8744297a06ec996fa0f117df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:28:02.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d67e8bd96f4fa19e8c50f1d1c9317a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d6a03b0b46480da816708bafeec05a : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac44efedde545ea8e91e4849b2fb3ef : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7900cf5ae0d14ff1b5451f25f4f057ab : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97a62d45bac34351bef08bd104152b3c : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e61f53916c04959a433e79c7ae82fac : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0af3ae44e4e049d0b200e20af5629dd0 : SUCCESS in 6m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e26f9788e0b648529833d5bf78a01592 : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa19a676f5db43ddb50410acadcf2b02 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d41f13b967149f49e5a5caf37c58f91 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfe2237a403b4b7b8f49930838f89060 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/713747778cb149dba6d66a090843908f : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f9a0552ea38f46bca4ce4bd3d8c1cbd0 : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acfcc26884ac48e483378741b2edee54 : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec5f165e46ca486c8c6679701b10bcb0 : SUCCESS in 1h 32m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a9a9cc4cc52c4f9da0d7cbf2f03b0cf4 : SUCCESS in 32m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/df591d5dfbef4a8ab4627a083942afad : SUCCESS in 59m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe7a330871db419bab72023fb2999a5f : SUCCESS in 1h 10m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a52692ef1f1d415dadcfa2c4673a871e : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e382d85a04f4a7b9253d7c5c7d65a88 : SUCCESS in 1h 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71138aab58a24e5d911e00dcdc2c8a91 : SUCCESS in 2h 25m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dfe644ebdf6943b79b7554941d8c105c : SUCCESS in 1h 47m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df63ae9af22f4be082a8897303e92243 : SUCCESS in 1h 45m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/baf1b9768c2a46ccb67dbe49b0991f79 : SUCCESS in 1h 46m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed6a003acc5a46af8a21322107925289 : SUCCESS in 2h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"2e615b4413bd9850ca3c74a2e81d8af2a109352e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-28 15:10:03.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"be4fe43f35346417377cd9218e2160d8e1624d86","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-28 15:11:03.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"bd3722a49808a4d03f5a2760b5f8240e92e83b81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 15:46:59.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"d40ec785ad2c892b0fefe2b1295518101f76b40b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 15:47:46.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"f7e8eec262e2aed5651c3cfce6bccea72bc1312d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-28 16:28:27.000000000","message":"Patch Set 39: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"78ed12dd9b31493c69f2e0ec1c7c9a6e1ea001e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:22:49.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa5e1e293fe47a7be3159a53b304d8b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd53c0699c7f46d7ba50323c8e29efff : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5334fc8fd04212bc1b7103dd87341a : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5e8b77a0724e7fa60ebd5cef0c7481 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e8475ec97af429ab0bd1a5b42131b40 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5570d014e7d4043a91558f1916ee790 : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e717ef16a3442f7a015b54eabbabafe : SUCCESS in 4m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21bc06d4fc2e48e38078dc55dca63276 : SUCCESS in 5m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e36407c590a548059d822bc7265b85ec : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/27cc419f75ea47049059609818ed59a0 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/06d71934382e4a62bd4710d507e6ca86 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d892f0c1a0f4b24a2ea0cc7497bcc83 : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8550fcc603a641feb05077dc2ca7bb4a : SUCCESS in 7m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3d73d5161f8418da372a06f71fdc0c8 : SUCCESS in 1h 13m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19d43fe5b6f0465f8cc750295f94b717 : SUCCESS in 1h 52m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb33446b634e40a1b289668579c98e16 : SUCCESS in 35m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c96faae1fe248c584fb9062b36d51fa : SUCCESS in 48m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b9c19aece3dd4f2fb87d3c206f861a45 : SUCCESS in 1h 12m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0de7a691907e42d0896cb7b1054cc486 : SUCCESS in 32m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77288cf539f34cda9e805f2cdf6910d3 : SUCCESS in 1h 36m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f043bd9a236b4edf9d40c25335071094 : SUCCESS in 1h 59m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1322621111945ba830c3ce10344be02 : SUCCESS in 1h 49m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/254343b9cca945f89fccd4ec0e130dca : SUCCESS in 1h 50m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/33f2dee870df4045b9b7d259966e3564 : SUCCESS in 1h 37m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6a1b26de7c6493db60d74310c6e3599 : SUCCESS in 2h 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"e221f0493ca33758b029b580e76d1472ba6cab14","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-07 12:06:21.000000000","message":"Patch Set 39: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":39},{"id":"772464047d913808929925692e418316cfc2c9a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:18:13.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"140e88b845eeaa2f3b27ca0ae74b31663f102a19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:21:35.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"312fab7963489d9e2613676a57017478913e7d00","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:26:29.000000000","message":"Patch Set 41:\n\n(3 comments)","accounts_in_message":[],"_revision_number":41},{"id":"0b5c3dc5737b9b8c89293acc3e32ab19561a8f8e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-07 14:25:48.000000000","message":"Patch Set 41: Code-Review+2","accounts_in_message":[],"_revision_number":41},{"id":"dd2b420f254afa93a5ae3c843bd29d11716ad0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:52:12.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad54291b00e5499898076a47d2d0f71d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5b39d3db2284d1c8688eb5ba448783c : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f32247570f4e6a88acb5ff8302904c : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca658f501562411b9e64c5cad7082e41 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f032f222cc9a47ee934f60a2d1ba4b70 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eaa90a1c3ad4761b363ca1e15b7c750 : SUCCESS in 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/644db61366b84ea8b28703e5d79ee65e : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e059fc9feba04361903348d84ee6fee5 : SUCCESS in 5m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/416e3c507bd64dc1bddf451954d05b1e : SUCCESS in 8m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3b86e94268c4ab08ce10dbf2abbd83a : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15159085b248497eb8d53d34ee7696a0 : SUCCESS in 8m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e9af6750fbe4dacb40b6930f12e19ba : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f47e454520ff4bd0a042d2bed3f5795e : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f3b967819574f58b7fbbec2b082ff70 : SUCCESS in 1h 45m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8444f869f25846879684f4477595b03b : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/16e89f0f22b14c55af10e9385afc995d : SUCCESS in 32m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0d9f35a68ca4b77af2694aac7219c28 : SUCCESS in 53m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/63f9416a63b147db86b7b8de6e799c6f : FAILURE in 1h 21m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7bcb291b82c43ac9e9a957652e9e6f2 : SUCCESS in 31m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7694042bb0dc412e9ea5a3fac35a0e10 : SUCCESS in 1h 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b68b702ff3a4f94b932c81dbc65e280 : SUCCESS in 2h 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a83b9c05096844b280a2515099155691 : SUCCESS in 1h 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3cdc47d8a26d4b1baefb89b2aacb042b : SUCCESS in 1h 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/85a5a75bea2043c5be16f6a449efe248 : SUCCESS in 2h 10m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d32183e459c44b4a18091a8190c466c : FAILURE in 1h 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"49ecfd9705095e234e0f1feb79dfee4f05dccd92","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 20:16:58.000000000","message":"Patch Set 41: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":41},{"id":"9696c01e3920d934c11674687e0692ce80b488e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 09:27:30.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"335790426d4c949d185509c3f6f43262a2c36747","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:39:28.000000000","message":"Patch Set 42:\n\n(5 comments)","accounts_in_message":[],"_revision_number":42},{"id":"73fdd38ae07c267f1429073d13c9cd8fb357ae16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:54:33.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"63b070819d2f5cdddcd910c5bc8a0f49d70b5819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:59:41.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"71415148f5e26b6dbce11bbf6dc2bd86fa5880b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:18:06.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e53d6276fee40a089c08d4ec315ea48\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b32e92b596742e380753ba2711d1f22 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f198f7cd8244e1b0541a2260ed99ad : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c460f19770444583cf7d4abe088584 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a446c58a896240a1855ae3bc2828ce81 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ca69523ba6498282fbaa5b08cda47d : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab30311b59949e3a331a71716c67892 : SUCCESS in 4m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/106f5fd1668840f793bc0d30a3c6bb1e : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/193a7a40c1c14eadaf870b901e226091 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ab5b95b54484d14879dcadaf23633a6 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64f1a4398acc4ed2921536f78bb19319 : SUCCESS in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6f64e5be870403aa50ae0835435794b : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa516c87e4534639bfbf0381a8e06c21 : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f706ed6a44e54f14a179abe996ef4011 : SUCCESS in 1h 25m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3c6bc498e754244ac8f7a4f500148d9 : FAILURE in 26m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3dfb2a2a1e34951a282bfcf6c669cd2 : FAILURE in 30m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54b0181c6b7d4674ba89ebd0c5997ab5 : FAILURE in 29m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/df0d7d4154194d0289e64f00605c7572 : FAILURE in 1h 19m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05032a743ea64e9ab90ecb51ab503c0f : FAILURE in 28m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8437bf7f54a40fb8daf2062bc16b93b : FAILURE in 20m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6f725a8f4e39495395b8ef540ad00456 : FAILURE in 30m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/64723c81a1aa42db8984c7404d8a4ccc : FAILURE in 24m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d17d963db68e45f5a335addeb63784bc : FAILURE in 26m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e17a896ac72e497bb3fc7ff31f864d7e : FAILURE in 26m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2e50e4c71b644f3814725269d9a26fb : SUCCESS in 2h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"f6370e7d5bfcb7ee1416a72404efd767e890be46","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-08 15:59:28.000000000","message":"Patch Set 44: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"c13067c2d2fb00058c06b9c634a1fb348f935a0c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-09 17:54:23.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"0a2aa88c75eb09f697fe5d4d0db7c3d3b0b805c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:38:32.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e125166d89e7418aa7f33475337e0aa1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8056551358a4db992f75814cc192774 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8b087a9c924449ab9a285dd34c6eb4 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888b7def8d764c0a9d1df05d3f397a8b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a85f5e995fec46a5a612265a8e5537e1 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b807f7c4514760845d385a0e3b73bc : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f1c668e712f459498ba8d661f2e7899 : SUCCESS in 4m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/77a7f61bdb7049319b93463df76d29ea : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d14a2feb032a4b87a392baa631547d8b : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/234f7178b4f3437bb3d4ba34bcddf568 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3dadbaf4bf24417a8d67f8b29a572014 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7eb9260a3481468caef16de3d990d553 : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba76451e15ee4866a4990106ee3ccdfa : SUCCESS in 8m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c33ea10efa54c9c8d5d35f2ba4c4252 : SUCCESS in 1h 38m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/285ee1ebf96a4c5f988553e6ccdfe3e4 : SUCCESS in 1h 38m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b78e907f9a9f4fb08535bc2ca2cdd927 : SUCCESS in 33m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7ee76293f1540c5965e629ff6037c88 : SUCCESS in 48m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e963efe43c434d81a1bd3c35dddeabd6 : FAILURE in 1h 21m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6f4090eeb034ce6a375ff06ca8ac693 : SUCCESS in 41m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54310e4318e04a41a96cb8cc1b9b67d2 : TIMED_OUT in 2h 33m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8e5fbe306fd415d975a62ebe642c83d : SUCCESS in 2h 38m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3058dc745ff24a97ab9de51b131b0cfb : SUCCESS in 2h 30m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/437407f652c14410b92e849308256179 : TIMED_OUT in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6da79661e06d48de8b759cb314807ff1 : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/78d2ec048b544cbd98d8b4304f18c229 : TIMED_OUT in 2h 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"691ead2b11431e130d5e148ca6618e9ce0e5bd5c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 08:07:45.000000000","message":"Patch Set 44: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":44},{"id":"b967a6c26b11aa3e217145ade6818a8fafdbecf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:08:26.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"3a6c04b8bae1f40035c9972f230ea25e26e96e01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:01:49.000000000","message":"Patch Set 44: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77836ab363cb43629b8cb7d5706df49d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10dc001047df4e239df0ff679ee9c173 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4825a52d83954dd5a9aacefd3eba3686 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bdc0fdf5e0d465ab09aafb92a9a9b5a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc4eeab35ed4c25a4731a4dd9dd3c87 : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f2cd0491bf4d448da3594e332a4a32 : SUCCESS in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/406d49ad8f9045fe893e62997e3e9e5f : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e2d1cea27174e01be14aae860c3140e : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bbdf30e41a784a95a51dd8932b6a68d5 : SUCCESS in 1h 34m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1562be3aef1047fdade82642dc055ebe : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53cf760bd0aa46dab7d48015dfcaa9d0 : SUCCESS in 32m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/627c22dbbb454ad79e9d90fa3bf98972 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":44},{"id":"1875e8205c7548ac35b2c8b839c6cf6a7d19ad60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:01:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":44},{"id":"a05ed02b9094970c4ec73a6cb9f0451db67169ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:03:05.000000000","message":"Patch Set 44:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0715ee17e04a4c34a9203453bc94dbd2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7206e2704368403aa5754b90b6630804 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6a4b7d34eb4302868e9bb518d60114 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3","revisions":{"bb4f26a833cf3186e77a9f787c73d6fd46d6c537":{"kind":"REWORK","_number":1,"created":"2022-07-08 16:02:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/1"}}},"commit":{"parents":[{"commit":"f4fecec076aa6ee688174a5ed702e058ef1cba71","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4fecec076aa6ee688174a5ed702e058ef1cba71"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:59:10.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f26a833cf3186e77a9f787c73d6fd46d6c537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f26a833cf3186e77a9f787c73d6fd46d6c537"}]},"branch":"refs/heads/master"},"aa434f4d6cb1eac12751825cb4a94ffe9f277284":{"kind":"REWORK","_number":2,"created":"2022-07-12 07:31:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/2"}}},"commit":{"parents":[{"commit":"f4fecec076aa6ee688174a5ed702e058ef1cba71","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4fecec076aa6ee688174a5ed702e058ef1cba71"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-11 16:21:35.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa434f4d6cb1eac12751825cb4a94ffe9f277284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa434f4d6cb1eac12751825cb4a94ffe9f277284"}]},"branch":"refs/heads/master"},"fa21afc5e103c7a19a761e2b6bea9861a5d98704":{"kind":"REWORK","_number":3,"created":"2022-07-19 16:45:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/3"}}},"commit":{"parents":[{"commit":"d7a6c1fa16a0505e0d780348f9614dbd6bdf6254","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a6c1fa16a0505e0d780348f9614dbd6bdf6254"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-19 16:44:02.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa21afc5e103c7a19a761e2b6bea9861a5d98704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa21afc5e103c7a19a761e2b6bea9861a5d98704"}]},"branch":"refs/heads/master"},"20780fcb3050d6ed19e2f571300d61d1ed5845cd":{"kind":"REWORK","_number":4,"created":"2022-07-20 14:47:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/4"}}},"commit":{"parents":[{"commit":"d7a6c1fa16a0505e0d780348f9614dbd6bdf6254","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a6c1fa16a0505e0d780348f9614dbd6bdf6254"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-20 14:45:43.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20780fcb3050d6ed19e2f571300d61d1ed5845cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20780fcb3050d6ed19e2f571300d61d1ed5845cd"}]},"branch":"refs/heads/master"},"2bd3c555b14c5b93443fd3ccdcd62a8c12cbd975":{"kind":"REWORK","_number":5,"created":"2022-07-21 09:25:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/5"}}},"commit":{"parents":[{"commit":"0f86784d6156fc2438cc29b11f5b41096ded652c","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f86784d6156fc2438cc29b11f5b41096ded652c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 09:25:00.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bd3c555b14c5b93443fd3ccdcd62a8c12cbd975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bd3c555b14c5b93443fd3ccdcd62a8c12cbd975"}]},"branch":"refs/heads/master"},"705feaec74278ca267f27e4357c07e2eef61c8e1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-21 12:09:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/6"}}},"commit":{"parents":[{"commit":"c1cdd61cc812a326873ebaa907324d366d3e89c4","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1cdd61cc812a326873ebaa907324d366d3e89c4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 12:08:17.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/705feaec74278ca267f27e4357c07e2eef61c8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/705feaec74278ca267f27e4357c07e2eef61c8e1"}]},"branch":"refs/heads/master"},"77974b6e79876819455fbac3075339c0001ab050":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-07-22 06:24:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/7"}}},"commit":{"parents":[{"commit":"60b8dff8e80581fe125df967bd7328bd3e7237d5","subject":"Proposal for Amphora vertical scaling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60b8dff8e80581fe125df967bd7328bd3e7237d5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 14:57:55.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77974b6e79876819455fbac3075339c0001ab050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77974b6e79876819455fbac3075339c0001ab050"}]},"branch":"refs/heads/master"},"14ac901b92666311ecb654589b68109296811838":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-22 06:47:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/8"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-22 06:47:43.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14ac901b92666311ecb654589b68109296811838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14ac901b92666311ecb654589b68109296811838"}]},"branch":"refs/heads/master"},"8774a2feebccad9e5ca03ef8397544e8c6dd5268":{"kind":"REWORK","_number":9,"created":"2022-07-22 10:31:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/9"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-22 10:30:48.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8774a2feebccad9e5ca03ef8397544e8c6dd5268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8774a2feebccad9e5ca03ef8397544e8c6dd5268"}]},"branch":"refs/heads/master"},"ca5355809164586d173901a11f99d031d4baef00":{"kind":"REWORK","_number":10,"created":"2022-07-25 10:35:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/10"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 10:35:22.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca5355809164586d173901a11f99d031d4baef00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca5355809164586d173901a11f99d031d4baef00"}]},"branch":"refs/heads/master"},"89b417224fe264851cebb5bf9e658cb2efbdb370":{"kind":"REWORK","_number":11,"created":"2022-08-01 16:08:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/11"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-01 16:08:42.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89b417224fe264851cebb5bf9e658cb2efbdb370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89b417224fe264851cebb5bf9e658cb2efbdb370"}]},"branch":"refs/heads/master"},"00c0fcdc8e9f3e084e5095917fc791ed73afb1d5":{"kind":"REWORK","_number":12,"created":"2022-08-08 08:07:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/12"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 08:06:39.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00c0fcdc8e9f3e084e5095917fc791ed73afb1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00c0fcdc8e9f3e084e5095917fc791ed73afb1d5"}]},"branch":"refs/heads/master"},"d278598ad9f9429d447e6b1c59e0febcec76ca1a":{"kind":"REWORK","_number":13,"created":"2022-08-08 11:28:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/13"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 11:19:33.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d278598ad9f9429d447e6b1c59e0febcec76ca1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d278598ad9f9429d447e6b1c59e0febcec76ca1a"}]},"branch":"refs/heads/master"},"326b7f19b39d8b16f85cbbdd17d088812cfd0972":{"kind":"REWORK","_number":14,"created":"2022-08-08 13:43:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/14"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 13:43:39.000000000","tz":120},"subject":"Add element for TuneD and Tuna","message":"Add element for TuneD and Tuna\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/326b7f19b39d8b16f85cbbdd17d088812cfd0972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/326b7f19b39d8b16f85cbbdd17d088812cfd0972"}]},"branch":"refs/heads/master"},"a6bd897fd35b5af9cf6dfca6615106a5bffe5c2c":{"kind":"REWORK","_number":15,"created":"2022-08-19 09:44:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/15"}}},"commit":{"parents":[{"commit":"726baf2af76fa8467e4bc6065a96d7236e7caecf","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/726baf2af76fa8467e4bc6065a96d7236e7caecf"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:43:47.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6bd897fd35b5af9cf6dfca6615106a5bffe5c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6bd897fd35b5af9cf6dfca6615106a5bffe5c2c"}]},"branch":"refs/heads/master"},"71422d3536886b575560662bf55f6ac8b3679ee0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":16,"created":"2022-08-19 10:37:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/16"}}},"commit":{"parents":[{"commit":"e35009ebd2fabd483b99e6ee10a78a2a148a533a","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e35009ebd2fabd483b99e6ee10a78a2a148a533a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:35:21.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71422d3536886b575560662bf55f6ac8b3679ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71422d3536886b575560662bf55f6ac8b3679ee0"}]},"branch":"refs/heads/master"},"2aa91bdea18e9b5263ca3412963c521b149a17f1":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-08-19 10:47:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/17"}}},"commit":{"parents":[{"commit":"b60c47a6110a03baa594116f5a36de439ad8d2e1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60c47a6110a03baa594116f5a36de439ad8d2e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:47:40.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa91bdea18e9b5263ca3412963c521b149a17f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa91bdea18e9b5263ca3412963c521b149a17f1"}]},"branch":"refs/heads/master"},"bcf07e0af3d5347dd5cb9000c3f0af96c6869416":{"kind":"REWORK","_number":18,"created":"2022-08-19 13:22:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/18"}}},"commit":{"parents":[{"commit":"b60c47a6110a03baa594116f5a36de439ad8d2e1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60c47a6110a03baa594116f5a36de439ad8d2e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 13:22:10.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcf07e0af3d5347dd5cb9000c3f0af96c6869416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcf07e0af3d5347dd5cb9000c3f0af96c6869416"}]},"branch":"refs/heads/master"},"60c5917cc40b01e5772ab35018b9f49757d3520a":{"kind":"REWORK","_number":19,"created":"2022-08-19 14:46:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/19"}}},"commit":{"parents":[{"commit":"b60c47a6110a03baa594116f5a36de439ad8d2e1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60c47a6110a03baa594116f5a36de439ad8d2e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 14:46:17.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60c5917cc40b01e5772ab35018b9f49757d3520a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60c5917cc40b01e5772ab35018b9f49757d3520a"}]},"branch":"refs/heads/master"},"24ea898125a01b7f23f0f8a166e0a0cf522443c7":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-08-21 15:00:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/20"}}},"commit":{"parents":[{"commit":"3518c7c2af9695514148d093129d98c6e0a71782","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3518c7c2af9695514148d093129d98c6e0a71782"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:00:45.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ea898125a01b7f23f0f8a166e0a0cf522443c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ea898125a01b7f23f0f8a166e0a0cf522443c7"}]},"branch":"refs/heads/master"},"2dc7679b76d2b0a36bd8c601e4d462de08837e39":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-08-22 13:52:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/21"}}},"commit":{"parents":[{"commit":"e2167dcfd405b2aef95a5dfbb5007fe170caacfe","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2167dcfd405b2aef95a5dfbb5007fe170caacfe"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 13:52:45.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dc7679b76d2b0a36bd8c601e4d462de08837e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dc7679b76d2b0a36bd8c601e4d462de08837e39"}]},"branch":"refs/heads/master"},"dcc945117ede18b498d362d9648107f2b2e3aedd":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-08-23 10:07:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/22"}}},"commit":{"parents":[{"commit":"bc1579954f32e9bcb43c6ebfc6aa0d566879b7b6","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc1579954f32e9bcb43c6ebfc6aa0d566879b7b6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 10:06:51.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcc945117ede18b498d362d9648107f2b2e3aedd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcc945117ede18b498d362d9648107f2b2e3aedd"}]},"branch":"refs/heads/master"},"9514f5b9db9de4cc9dc267a9de4891358c859013":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-08-23 12:45:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/23"}}},"commit":{"parents":[{"commit":"9ebba86664248b02826d2175fc8494ff5c1b0a7f","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ebba86664248b02826d2175fc8494ff5c1b0a7f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 12:45:50.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9514f5b9db9de4cc9dc267a9de4891358c859013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9514f5b9db9de4cc9dc267a9de4891358c859013"}]},"branch":"refs/heads/master"},"1c9092d1944afbbf49b05097eee26e93235ddbdc":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-08-30 11:21:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/24"}}},"commit":{"parents":[{"commit":"2a33dfb7ff006a75c11176d4211e030d430a0fe1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a33dfb7ff006a75c11176d4211e030d430a0fe1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:16:37.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c9092d1944afbbf49b05097eee26e93235ddbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c9092d1944afbbf49b05097eee26e93235ddbdc"}]},"branch":"refs/heads/master"},"d234f7e83dd5c2999a61088a5c5f4475def722a2":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-08-30 11:35:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/25"}}},"commit":{"parents":[{"commit":"bf980acc70760542cd42270272392fabdbe4f3b3","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf980acc70760542cd42270272392fabdbe4f3b3"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:35:03.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d234f7e83dd5c2999a61088a5c5f4475def722a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d234f7e83dd5c2999a61088a5c5f4475def722a2"}]},"branch":"refs/heads/master"},"7f5d541c0b1bb966c581d07576c074cb5f19e920":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2022-08-30 14:22:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/26"}}},"commit":{"parents":[{"commit":"920efa82345f24ac2261985e53bdbf7a451c9314","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/920efa82345f24ac2261985e53bdbf7a451c9314"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:22:20.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f5d541c0b1bb966c581d07576c074cb5f19e920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f5d541c0b1bb966c581d07576c074cb5f19e920"}]},"branch":"refs/heads/master"},"4048346a3121224f66da777202b1db24a862ba10":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2022-08-30 14:53:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/27"}}},"commit":{"parents":[{"commit":"e6c8f4bc9e6e4927bc4544690737e8d1f49e92bc","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6c8f4bc9e6e4927bc4544690737e8d1f49e92bc"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:29:31.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4048346a3121224f66da777202b1db24a862ba10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4048346a3121224f66da777202b1db24a862ba10"}]},"branch":"refs/heads/master"},"799c16ed0219af080f57334d446778ce59d11506":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-08-31 07:29:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/28"}}},"commit":{"parents":[{"commit":"796c0458889a212a34b9fdf6ae89e8813507cac7","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/796c0458889a212a34b9fdf6ae89e8813507cac7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 07:29:43.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/799c16ed0219af080f57334d446778ce59d11506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/799c16ed0219af080f57334d446778ce59d11506"}]},"branch":"refs/heads/master"},"bf6d4022f29636fa7f920e67b856a710e110b88b":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2022-08-31 09:09:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/29"}}},"commit":{"parents":[{"commit":"b723feab3a49e0e114b2fca093995bcdf6c474aa","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b723feab3a49e0e114b2fca093995bcdf6c474aa"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 09:09:40.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf6d4022f29636fa7f920e67b856a710e110b88b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf6d4022f29636fa7f920e67b856a710e110b88b"}]},"branch":"refs/heads/master"},"c8be509fbb5a7c7886a0b1c4b975122641ba2a98":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2022-09-01 08:56:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/30"}}},"commit":{"parents":[{"commit":"ef2d9b17a3c254ffdd29f8f7764ec6d88f846cf2","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef2d9b17a3c254ffdd29f8f7764ec6d88f846cf2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-01 08:55:12.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8be509fbb5a7c7886a0b1c4b975122641ba2a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8be509fbb5a7c7886a0b1c4b975122641ba2a98"}]},"branch":"refs/heads/master"},"91e7f0d1fa2d7220972c246e7eafb36f6d95b965":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2022-09-12 09:57:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/31"}}},"commit":{"parents":[{"commit":"72d452221fcc92d0e5bdb542a138c327c9e4426d","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72d452221fcc92d0e5bdb542a138c327c9e4426d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 09:53:34.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91e7f0d1fa2d7220972c246e7eafb36f6d95b965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91e7f0d1fa2d7220972c246e7eafb36f6d95b965"}]},"branch":"refs/heads/master"},"47acdcad26e52d352e9a6d685942fe40b05b3606":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2022-09-12 10:03:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/32"}}},"commit":{"parents":[{"commit":"8f803f39dd74efb9b387c015c24f39e79e3a4c3f","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f803f39dd74efb9b387c015c24f39e79e3a4c3f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:01:55.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47acdcad26e52d352e9a6d685942fe40b05b3606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47acdcad26e52d352e9a6d685942fe40b05b3606"}]},"branch":"refs/heads/master"},"01888dcedb35e85b9a4a6fbfc71272f15e5bbf92":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2022-09-12 10:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/33"}}},"commit":{"parents":[{"commit":"1a23670f15249b75699f58edbbdfd21d0330544a","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a23670f15249b75699f58edbbdfd21d0330544a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:08:05.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01888dcedb35e85b9a4a6fbfc71272f15e5bbf92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01888dcedb35e85b9a4a6fbfc71272f15e5bbf92"}]},"branch":"refs/heads/master"},"de7ea6e40a09a166829c88e540ae8c12c9728df3":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2022-09-12 12:28:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/34"}}},"commit":{"parents":[{"commit":"3752dbc0b0b5c6f3c8e817db0cadc89e97698ca7","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3752dbc0b0b5c6f3c8e817db0cadc89e97698ca7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 12:27:45.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de7ea6e40a09a166829c88e540ae8c12c9728df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de7ea6e40a09a166829c88e540ae8c12c9728df3"}]},"branch":"refs/heads/master"},"deeda821420da1ffb8c5fca52cfed173019f30ff":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2022-10-19 09:09:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/35"}}},"commit":{"parents":[{"commit":"fd3fa2a745486c02800aa51acb2ce7ec4c8d0bc1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd3fa2a745486c02800aa51acb2ce7ec4c8d0bc1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:09:22.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deeda821420da1ffb8c5fca52cfed173019f30ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deeda821420da1ffb8c5fca52cfed173019f30ff"}]},"branch":"refs/heads/master"},"a299f7eb7c8a8fa2759a910512abfe740e306a9b":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2022-10-19 09:11:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/36"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:10:48.000000000","tz":120},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]},"branch":"refs/heads/master"},"66cc89ab21f400f115052706e10cb886b91539c8":{"kind":"REWORK","_number":37,"created":"2022-11-23 15:29:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/37"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-23 15:28:54.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66cc89ab21f400f115052706e10cb886b91539c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66cc89ab21f400f115052706e10cb886b91539c8"}]},"branch":"refs/heads/master"},"76743e238650b11eb54b146c429c7b925978b807":{"kind":"REWORK","_number":38,"created":"2022-11-28 11:00:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/38"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 10:59:34.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76743e238650b11eb54b146c429c7b925978b807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76743e238650b11eb54b146c429c7b925978b807"}]},"branch":"refs/heads/master"},"74b49e5aa7d139fce56c789a29fadf446ba68dcc":{"kind":"REWORK","_number":39,"created":"2022-11-28 15:46:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/39"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 15:46:47.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74b49e5aa7d139fce56c789a29fadf446ba68dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74b49e5aa7d139fce56c789a29fadf446ba68dcc"}]},"branch":"refs/heads/master"},"ce3796639c9bb3b82fa6894229ad705969f79d99":{"kind":"REWORK","_number":40,"created":"2022-12-07 12:18:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/40"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:18:01.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3796639c9bb3b82fa6894229ad705969f79d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3796639c9bb3b82fa6894229ad705969f79d99"}]},"branch":"refs/heads/master"},"97a2fbdeb764714b88a345c2892b84a4dd6ff263":{"kind":"REWORK","_number":41,"created":"2022-12-07 12:21:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/41"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:21:23.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97a2fbdeb764714b88a345c2892b84a4dd6ff263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97a2fbdeb764714b88a345c2892b84a4dd6ff263"}]},"branch":"refs/heads/master"},"00b655c89e4878dfae34d0142205fbeb1aec0654":{"kind":"REWORK","_number":42,"created":"2022-12-08 09:27:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/42"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 09:27:17.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00b655c89e4878dfae34d0142205fbeb1aec0654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00b655c89e4878dfae34d0142205fbeb1aec0654"}]},"branch":"refs/heads/master"},"dd0964b4c0b5cc8a8d4d5fc1019076757da1070f":{"kind":"REWORK","_number":43,"created":"2022-12-08 10:54:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/43"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:51:49.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd0964b4c0b5cc8a8d4d5fc1019076757da1070f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd0964b4c0b5cc8a8d4d5fc1019076757da1070f"}]},"branch":"refs/heads/master"},"c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3":{"kind":"REWORK","_number":44,"created":"2022-12-08 10:59:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/29/849129/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/849129/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/849129/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/849129/44"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 15:57:59.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:59:32.000000000","tz":60},"subject":"Add cpu-pinning element","message":"Add cpu-pinning element\n\nThe new \"cpu-pinning\" element optimizes the amphora image for better\nvertical scaling. When an amphora flavor with multiple vCPUs is\nconfigured it will configure the kernel to isolate (isolcpus)\nall vCPUs except the first one.\nFurthermore, it uninstalls irqbalance and sets the IRQ affinity to the\nfirst CPU. That way the other CPUs are free to be used by HAProxy\nexclusively. A new customized TuneD profile applies some more tweaks\nfor improving network latency.\n\nThis new feature is disabled by default in diskimage-create.sh.\n\nStory: 2010236\nTask: 46042\nChange-Id: I1a0591de79be867483a044705e866b2368b2a567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
