)]}'
{"id":"openstack%2Foctavia~849130","triplet_id":"openstack%2Foctavia~master~Ifbbe714c66117e57f96614534e6f20a9634c26eb","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-11-16 15:37:45.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-12 08:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-12 10:47:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbbe714c66117e57f96614534e6f20a9634c26eb","subject":"Amphora vertical scaling optimization in Octavia","status":"MERGED","created":"2022-07-08 16:02:27.000000000","updated":"2023-02-09 14:59:23.000000000","submitted":"2023-01-12 10:47:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"849130-amp-cpu-pinning","meta_rev_id":"160366deff5f4b30d7ce012df4ccc5e565242aa1","_number":849130,"virtual_id_number":849130,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 10:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-12-08 10:59:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-12 08:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-12-08 10:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-12 08:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-01-12 08:15:31.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 18:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 07:38:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-08-19 21:39:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-10-14 15:06:05.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-07 20:23:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1dc8223f05685978dc0f8af8b23544a2fac1590a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 16:02:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf090d635cd5f4de0fb57b593a0dda7e81924b9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 16:03:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"aadb17add8298e949393f84f8517696728d913b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 18:11:36.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e22ebc266ca447fb21d070ea577a920\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d23a69c2bcda4351b3976679317299fd : FAILURE in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d78bc3b7734817903346008590c885 : FAILURE in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8865347af7104aceb812f965746a9a8d : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bbf9c7f7fd44566abb40b76672e8a6f : FAILURE in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c222f02bd0384a209e3a4c9a99e26948 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ef5985fe464a90a6b8f53b5e508cbd : FAILURE in 5m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cc879c835df84f21b90bb8abfea58ff3 : FAILURE in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dc635617fe354cc08a103f764ce26739 : FAILURE in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5166a98f9740429fa1e699f9803d85fb : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5364715b0df74a24b67a8d7cf02b251c : FAILURE in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d64482a342a45fb9eb8976cde759b7a : FAILURE in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0075bc844bd942c4827d16e026fc0f18 : FAILURE in 6m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f47288757924f95825ae261a9aec1ae : SUCCESS in 2h 02m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d21884e15cc148c1ad4ceae99e91cf9a : FAILURE in 1h 13m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47a65089384b486b8fb4a43670eecf0a : FAILURE in 1h 07m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c714c111222248ff85f99f4a851ccb45 : FAILURE in 58m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9efd9ede96c445e5a1da73108ea475fa : RETRY_LIMIT in 8m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58041bdf9adb40ce88feb8e94c9f8766 : FAILURE in 32m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7497d040f0b442c98b7bcc92b1abb17b : FAILURE in 1h 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2358e772d574481bbacebe47ea6161ae : FAILURE in 1h 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/683dfee873e04eddb6272096187419b8 : FAILURE in 1h 17m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcdf056a3d754270a36d0e47b1e22f0f : FAILURE in 1h 00m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/414adbe72733418192c28c3fa27fbf97 : FAILURE in 1h 10m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/31a01c38427a418a89bc809639b496c8 : FAILURE in 1h 31m 56s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":1},{"id":"676b0ecd1067110ac4c05453e8950f98fd44b976","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-11 10:47:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc1c294fd9d6fe1fa0a812f8611daf1bf578002f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 12:50:34.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98fc7c54423449cb8e8139f27add591d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c39608c86eb48aa8cd1ee7e2ed47498 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352778ca6339416a8d79b81da1722a58 : FAILURE in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8faddb9fb85e4c38a2cc6edd47bc1ea9 : FAILURE in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ee2537ab0d64bc89cf6c7de803fb877 : FAILURE in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c01c7ae165e64c21ab3e2548c32012d2 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3cd80d135b4353b2e33fc546557601 : FAILURE in 6m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ad58936c0756479cb4e6364f77a7954e : FAILURE in 5m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/15c613b0d31340b991c16656449c82de : FAILURE in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c40efa16c9245599716bd0f5491a210 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/857aad933b3d44b1b9d0cba3556eb853 : FAILURE in 8m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/528d7623ac394e55b26f1e5b1cf18e77 : FAILURE in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae79f68106f74944bdd46c37f061028e : FAILURE in 8m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edd055edc7d840978b396b84855da086 : SUCCESS in 1h 55m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/156f39bcb8864314804014cd0bf8b16c : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9042e04fb0f488a8235d85bb43375c6 : SUCCESS in 43m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d62e7885436c4d35abc43402f67b890f : SUCCESS in 58m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c1cef74bb75429a915db434d5325ef5 : RETRY_LIMIT in 13m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a58bea4aa75c4c7db6f7b4eca85986f3 : SUCCESS in 39m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10291348a710412f809afb92af937fee : SUCCESS in 1h 36m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/068a9d3c708b455180815ed7582a0ee4 : SUCCESS in 2h 01m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dece41a4c59743db8bd0de5f2648282c : SUCCESS in 1h 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c6b5c35525b44ae89539dfcc4ba6d74 : SUCCESS in 1h 39m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/85c2048e4f494cba9278c11f75c06c14 : SUCCESS in 1h 46m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d94fa647257843aa9a8bc4211970aa9a : SUCCESS in 1h 49m 11s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":2},{"id":"d95cce17b97a0e39d65d69ada5da4e68f76ae4dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-12 07:31:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d61490624a66008ee987b60afe25fac7669488f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:58:01.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dacba353a024ac59b9075e1857fb31b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/896900288e6c4492b23e9ac1ccce6b85 : FAILURE in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e695d939b0d14ed48faa1189e0391f9b : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ac2b08c29a4207aa5af576ba926d73 : FAILURE in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18093d00c1134368b4fc194662ba5fae : FAILURE in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e88feb6ab73a435abb9a999fb3c99848 : FAILURE in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae35d8e64274d2a9fcab0ae3640ffd5 : FAILURE in 7m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/caaf6838a86a41a5adf3387e27a240d4 : FAILURE in 5m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ebc1b166131a45e9a0cde41eca393206 : FAILURE in 10m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee9bc2b732314685b7c41ca7da67c744 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59558a9ef122462ca0bdb53a48f462ab : FAILURE in 8m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/417679bb2ec041578773dff091ac98ba : FAILURE in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/674f582572ea477b874a69170dad3277 : FAILURE in 11m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af02d59df46e4bc99c9225ade464a7ac : SUCCESS in 2h 15m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fa0f9e7154a4f7e8c0ede8b07eaaccc : SUCCESS in 1h 50m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6180c06aa994d62b31c30e30e513748 : SUCCESS in 41m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/586ae39c96ad4182860ac3b454d36b7c : SUCCESS in 1h 10m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/daae7c987eab4bb2b99328357e51340d : RETRY_LIMIT in 14m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfb63d84648f4ee6ba6f694167b5dc5e : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e44511dab394df9a76d6e395d0b897e : SUCCESS in 1h 55m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa4b9d25e4484706906c58ad518d1ada : SUCCESS in 1h 57m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e34d6fe0359f4febbfa36574e6d28d29 : SUCCESS in 1h 59m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eddfdc43dbc64f24a70c087ed6650687 : SUCCESS in 1h 46m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3722ff68df8545959b5395ce74a0ee15 : SUCCESS in 1h 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7e1cd3459044a03b88de9c6ee868dd1 : SUCCESS in 1h 56m 37s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":3},{"id":"07d5d0222b989d839dc2429b9ad61f80b8114ed7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-19 16:45:10.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"053f6b598340e0a8a36ce9cff00607f907ad6658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:11:16.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f302701c7b4546958e297a2083240bf5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d08beea1b124d9fb10f4e32590c1e4f : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1454275b230d476d86867bbd312914df : FAILURE in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eb428d3cc074f21ab2329bf2f15b055 : FAILURE in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83a532f56893423588077c19f97f5397 : FAILURE in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed6ac1e77c9641f2acf53ece5d4b81d6 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b150d37874934f33b9bdf680c5de92f5 : FAILURE in 9m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/494c2a02d23e47aaa59ccffef9b25ea1 : FAILURE in 6m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/afe17ba7b8f3494f821442efb7bf9c8c : FAILURE in 10m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99ad306f076448848a94bb4ca02d254d : SUCCESS in 7m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/359021911337494a84b21237b76c779d : FAILURE in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/889c173d7a0b46d590c75de48672d6e0 : FAILURE in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a005cc77e48a4502864836802b6f0d71 : FAILURE in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b5fbd9e4d534f078656de7d22ec72ab : SUCCESS in 2h 15m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/862d41baa15543e48996955be872f273 : FAILURE in 24m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2dd277f16134cd8a9c68168f992fdc4 : FAILURE in 29m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c29ea40ce8ca426dbfabb9c267971654 : SUCCESS in 56m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6a1c12de9b21407f85f563be31814e55 : RETRY_LIMIT in 13m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d527171b8771421dbdef3254ef5f57b1 : FAILURE in 39m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/78c6cd2c5f884a6c9de7d7f0daffd6de : FAILURE in 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42d6613b6d434d3293ced250ed95fc81 : FAILURE in 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4475bfed50494a798787aff2f350bc27 : FAILURE in 25m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d228a8a062644555b7a92683ac9d2984 : FAILURE in 40m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/795e7f20d47345f7800f8a30c7074057 : FAILURE in 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e172ee433ee4dfdafdacd5e6a5246ce : SUCCESS in 2h 23m 57s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":4},{"id":"d0e71a7a135695436ff308cbf09fdcb52dc8384b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-20 14:47:24.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0c71e4174b7f0ae9ca07322ad515093ba1dcddb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 17:24:22.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/452eade0dabe4a47bb0d192820ae1638\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5bb4c0ca6b54977ae7d505412eb4540 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbec9870d840427fba6c5f85ddfb2596 : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e1640bb56d640c9b9b072cf107194ae : FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72315b5e847a4eb4a810cd18e380418d : FAILURE in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9138f65bce97478fb4568053668c2666 : FAILURE in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4739d6ce77a24dd6873006fcce376343 : FAILURE in 6m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3da32c35205a439486568bf546c92c7b : FAILURE in 6m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/60f1e82430774378bd1151ababfc1755 : FAILURE in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4a173a83e6d84bdfbfecfbc495bae8bf : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b808ecc9f24045f4a9b7f788d186ebcb : FAILURE in 8m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c280c56500ac4597ab76c85cd59bf0dc : FAILURE in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d3df431f55014c74ad05dcbb31b8153b : FAILURE in 7m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc55690fbb1649349b987bae5b0d828a : SUCCESS in 2h 10m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae8f05eb145943b8a29408e6fcd771ed : FAILURE in 39m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/87a188da2242499691ae7e5610447e67 : FAILURE in 27m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cb7f8e169be431882b0d6dc1ad810df : SUCCESS in 1h 10m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/20cd9a17947444b28e7c3f57eec8c2b7 : RETRY_LIMIT in 13m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/baa02392a120428b9d17b7d2488df803 : FAILURE in 44m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86d02eca885c47cb9b94a2a637760709 : FAILURE in 27m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0a1f3d9184d47649a8f2832c93806a9 : FAILURE in 49m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f4f9c982299a4b4598a495a58d52c838 : FAILURE in 25m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f44c11b37f244f9e86ed8a5ce2c0c3c0 : FAILURE in 25m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/185c6e77a61241f6ae0e85bd79f7651b : FAILURE in 25m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/67eb1dc5178c4dd597f377991711eed6 : SUCCESS in 2h 33m 24s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":5},{"id":"e287ddc9faf384f3aa8f56b2f2d95c78719e1865","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 09:25:27.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"03918551a956e98e7962356212289f13480411f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 11:59:02.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95fce95891674b58ab56c50cb4efab20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/617a6a1a8649446895ae76314299a9b6 : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64fe78fb4d24644a464926acab34ccb : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a72372b36ce48cca179cbfac33b2661 : FAILURE in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff54f636279d44f29d0fb58cc326acfe : FAILURE in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f479da0f1924f1ab920f53e831451b4 : FAILURE in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23f7699e7b94071a728b4075c838e00 : FAILURE in 7m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8c6b0432111444e29352084285d40314 : FAILURE in 6m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5b02b8ce584c4a66af49360ded73bbc1 : FAILURE in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cce5bc2279d4039b864fe5276e05a51 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08f3dcf761394ab480f6b0aab7c38d4c : FAILURE in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46734a0c02de4a01a1ac4cd95156c2b4 : FAILURE in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66b264b5e50f4e4cbfe80d1cb84b3f41 : FAILURE in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d34b311c70c24acbba121b12bc73c246 : SUCCESS in 2h 09m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a286d1ed77354ec28e771a84081ebd0f : FAILURE in 30m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c3ea4b676504d3f8de8c4d01a888a9b : FAILURE in 42m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9655d27c3294216bbd0143abe0231c7 : SUCCESS in 1h 09m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5891395e7d8343a39396e33f109e69c6 : RETRY_LIMIT in 25m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d72ff8e14d34e2a9b64598a35fd1b90 : FAILURE in 25m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55ecbd7b7a204a949872b4855688beeb : FAILURE in 41m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4fe20ccbd79c404c849abe94e06c302e : FAILURE in 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f58a9a01af44730a1d2891b9e8f39b3 : FAILURE in 40m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df5225775db943458e5a12cfff749f6f : FAILURE in 25m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6cd2d0efa8d844eeba9ba6c128cf41d9 : FAILURE in 40m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e42065c9a458498692c8b3309b5d2796 : SUCCESS in 2h 24m 08s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":6},{"id":"62b7c1cf2eb66bd4175359213eed8171eb305e21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 12:09:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1b76a8fa5f78ca6c3dc6d801ddecb3d82fe5af07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:28:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d99b90c5bf4943ada462cd52600195\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921ed1b8ff254629bc9c1cb2313cb59b : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a179e608e0c247d8b8782630faaae7e3 : FAILURE in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03aafb3cd48b44aa88505792a74fd1cf : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf27dd72442444dd8a181d7e8ae14f38 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7e5973d6f9a4f59ac0984d713d37809 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e68a321da44f7daeb22f7de92d9487 : FAILURE in 7m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1d81198d3e934f159025e0e47a990171 : SUCCESS in 6m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0aad0ef0e75a475882cfb642756a177e : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64b76ba4a6754545b34ccf16360d2c7c : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bcdaebe423f471fa61f60652c8e230c : SUCCESS in 8m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8317cd9084b4b93bd4d2261c10b6498 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/611f08e1c21140adbf5fe3eeede5cf80 : SUCCESS in 7m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c94d4f95d576403db521abc70bc51276 : SUCCESS in 2h 11m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae924fe2e1ab4215953ff4689b701495 : FAILURE in 25m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6fd47842b89b420189646e8520fea39f : FAILURE in 26m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0510bdb6afa146668d1ccb0050301cf8 : SUCCESS in 58m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c5c27bea73849aab61c2aa2d69f9526 : RETRY_LIMIT in 25m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5acbf34839e9422ba4e61dc7dc20b87e : FAILURE in 24m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6480635dcdd4466d80688a2fd1488091 : FAILURE in 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/338e086b765143e2baf89429784152d9 : FAILURE in 39m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8e067cb2b5144174b1973919a4c6fc1b : FAILURE in 30m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e70e50aef6664f9394a465673da49b24 : FAILURE in 31m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/876077c2098a40aabdf0f91f0fa672c6 : FAILURE in 24m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/db7b491dec49402f982791560952cfd9 : SUCCESS in 2h 17m 06s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.2/amphora-vertical-scaling.rst","accounts_in_message":[],"_revision_number":7},{"id":"9f5b762c40a4bb4963aebdef93b6b20a07dd06a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:24:24.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f6f270656c9e76f77562f8d1d84581606803896b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:47:59.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"3fc25b88f11f33c5bc2a25cd7245fdd7667795cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:10:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07dbd9961747474398dff1b7bed4cada\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e63705289742138dedc91179f970f7 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5eadc3dbc943a9bd3ff018787cc3d0 : FAILURE in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bcf649aad344ce7ba7862ef21cd3d74 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47b62a8ceb4546b2ac79bfd85f7c2daf : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb9fe70c26b74ab8802dd1cd0cb9233d : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdc6794eee94d79b57f1e9a75b9870b : SUCCESS in 13m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6c54ccc89b8544ab8604957b34e5f287 : SUCCESS in 7m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5a3b881673c5422aae853a7bc61a657d : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/78f308927ec3498fba3472834a25af01 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc35dd3cd82a47f9876e3f4af1a4c537 : SUCCESS in 10m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9bd61ab86abb43c494d52cad713725ea : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a03c64128724b7982c74319d4e47191 : SUCCESS in 9m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23983c80cbda449d9580df6b7c216566 : SUCCESS in 1h 56m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ea9dceaf8ab4db980c1452d37a35ac1 : FAILURE in 24m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2b2f21cf0f14f919af1efa6fc4a5bed : FAILURE in 26m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0be9a3faedb4dcb953f11b54ca09ac1 : SUCCESS in 1h 09m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1daf6e22c4d44bf8915f2860cb20b97f : RETRY_LIMIT in 13m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34fdd53b438f4db6873471d7b122de14 : FAILURE in 25m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d7369ae30ecc46df9f6abeefabad6e21 : FAILURE in 28m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c66a56dd574d431795a5fe51c1a04d90 : FAILURE in 36m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/334e8164ae3141ffa688792923269a16 : FAILURE in 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/942194bad4104fc1b8a93f2e3e7a62a2 : FAILURE in 27m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/22336690f3fd46e2a4b21b1029cbcfcb : FAILURE in 25m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/00faeeed41be4c91bbbf5905e37b7852 : SUCCESS in 2h 18m 34s (non-voting)\n\nWarning:\n  Comments left for invalid file elements/tuned-tuna/static/etc/tuned/amphora/00-tuned-pre-udev.sh\n  Comments left for invalid file elements/tuned-tuna/static/etc/tuned/amphora/script.sh","accounts_in_message":[],"_revision_number":9},{"id":"e199b15e365e97b131d68165ccb31cd41c180ae4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 10:31:02.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"27ed9cef71f651d4523eec17bd9f7f36a541d541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:59:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfef6f7b0cba40c7888ee362d31a2875\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d32aebdaa84bd4b14099b88a113385 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1129eb8c5ef147e1849929b557a4d9ec : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921724fc81bb418e868d98aa793dc0f5 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b46b74a953e4b6889d198b2bae817f2 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/496676e26ce84be08055d67f58a09cc1 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4e6474d561403ca93bdb542a26c127 : SUCCESS in 14m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/61be4f77816b44ba812eae7527cbb188 : SUCCESS in 6m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c92daa846c164a22a7c3f7441cf9b5ac : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6a383fec60745cbbce6af3e36afdbe3 : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aeb77cef986a4300ad10c02db4cf40d1 : SUCCESS in 9m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/011c62ea5f494a068312b75747db6429 : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec6fb069d50b4f63940cafd3bcf44fc3 : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a4b943abd2a440fa6673ac3484e7997 : SUCCESS in 2h 23m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ef7208eec8d416385cf9abfc03c0b04 : SUCCESS in 1h 40m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78bec80bce9442fc831bf26bc1f4719c : SUCCESS in 40m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/111c8d7e32cd4df59d5f27de5024b3b9 : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/92d149903042415a9e459987b64ed6ca : RETRY_LIMIT in 25m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef4610ec0cf14d34b352304ccbd7d517 : SUCCESS in 40m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1bbfa1b21bd74b379b7472bd1bbade26 : SUCCESS in 1h 44m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d23117979b545f7b04721bb54dac88b : SUCCESS in 2h 02m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f70a163425247ed9b5247f44171f6b4 : SUCCESS in 1h 51m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1d4a847ecdb46bab102ef742f72b2b2 : SUCCESS in 1h 44m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b78abaaa905c4a428aa4bb71c26106de : SUCCESS in 1h 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8100615bb67d4132b073d92eff4150e3 : SUCCESS in 2h 05m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ae44c5dffec0e298803b5b5a7389d6a0d58fc539","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 10:35:40.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"47384287029e2e9a6d81ed8caca1fa34840e52df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:52:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48105066cc3f47ea8c92a9d64a496257\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ab7bc35ba54cfdb8c7d5be967854d1 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637f75d9ca6b47edb9e97c291f9a9e33 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93ca715ed78f4a719052c36dc3fbbd86 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2fdd1b3affd44efbb92756396c6d532 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8444c45f78cf4bb99ee03d56f4c49ab0 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4083a3caa48471694ce906e4fc140f5 : SUCCESS in 13m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6d970538dc1f45e28232bc803f6e034d : SUCCESS in 5m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/28493de8c6924ee4b10174754eefb2c6 : SUCCESS in 10m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75f4271144994c76b343290227a3e54b : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35d459a363d04eb797762de383a5ff49 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0963554106854247befa708bb274e7fa : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc8e8ce2eef84ee2bd2e2a5378b4ffda : SUCCESS in 7m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05a2d60e4661403f980579de370bdaaa : SUCCESS in 1h 52m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27f0b625956d40bda400933a2c9a7c17 : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1691a1ad1ad480d99093bc56af3f56e : SUCCESS in 37m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3a4f40a55874790a41af38dce39779a : SUCCESS in 59m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2d302d7eeb7f4322a9d734362abd94a4 : SUCCESS in 1h 09m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/885544b0085448638fc0a25b98a544d0 : SUCCESS in 35m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0d6cd305794a4eae8a42c6ad0979f540 : SUCCESS in 1h 48m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23a8eb314fef47ce909ee8fca6988e0f : SUCCESS in 2h 13m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e5e7028c3e75468c9c3db357f6561304 : SUCCESS in 2h 01m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/391f6a3618be46748a1ce58fcd75cfe9 : SUCCESS in 1h 44m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a28af535eea14fedb6e008cc6517178c : SUCCESS in 1h 44m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e92da40ce5df429986e68d3da84d7e96 : SUCCESS in 1h 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"353c9ecaef2c04b9b2041ddf87b4cfe69fec34a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-01 16:08:53.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"a2e20bcf32a9e200e50766bc66a91fab0cdd8ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 19:06:55.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a3ad666ab84ba2b93b69728a1abe2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cca8f4a2034c4a0f9e7432d6bb3b14b1 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6353158968cd44ffa7f88052fb37c95b : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa01516d0b0e4c38aaed3375a88a40c4 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/735d419284844dbcb55afe8d0a4ae74c : FAILURE in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c65fe7300864d098074a2833d7ccd8e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402ef342490d44568fc8d4da30abb222 : SUCCESS in 13m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2f7abb52b55f4ec39d74d1ea516bd4cf : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1e1b42addd4e4348bbe86be16e576099 : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6633f0f820cf43f49a34b8cd89cf103b : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/502d55601397420ab784b69f562dd3a8 : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70be6753d0d44ea9868fc5b0ff4bc1d1 : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62869110527945d0bab51263920ad2bc : SUCCESS in 8m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43d711f780504216b7ba494ebdf6fa85 : SUCCESS in 2h 55m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2998d024fd94b7aa011bb270a24153c : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dfdfdd74e374c67be1c8b9b6176e313 : SUCCESS in 57m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ca813945fd54ccab509b5b498ba0436 : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d7c0fe724a5f42aa959193e023841302 : SUCCESS in 1h 30m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a40974d37394bc1b12db2167e3d5917 : SUCCESS in 47m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ee588044ada4427c8ea9a9d5518a0394 : SUCCESS in 2h 06m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65669547019e41d4a473e8b428a432e9 : SUCCESS in 1h 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4c3315f3764b07a6ca9c1dc9616aa2 : SUCCESS in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8d6ffa4bf5c64ae3ab3a3c06c49e4a58 : SUCCESS in 2h 06m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ca1ebec83a334d26ab6f4106d2663315 : SUCCESS in 2h 11m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/569039ea39584cc99ccf76ff8bbc22ce : SUCCESS in 1h 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"13c317da792aae0bb4fb84fbe99cf6c074b002d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 08:07:19.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"54cb7e60989e5f14720ee6f1a328c0c991a62c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:54:56.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591cd604857e44eba7ef065ef7a2cb4b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df44c30877246249ee9f3169807ddcd : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c1ea35285a4c3ca68f07daa878440c : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d752cd67d9401ca56a19873d9e01d2 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72557236c1944ce48e8848c5d4395483 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e7f1226a91544108eca3854e4c4485d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f5c38bd8a84d8391d75dd265fd4014 : SUCCESS in 11m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/99b25e560e13489ab6a501cedadf13bf : SUCCESS in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fa9a91032da94dba975f986448334557 : SUCCESS in 9m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/deb3963e8cce4a448ec68914e7099265 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5972a6029e714e13bf6c159e666f5d97 : SUCCESS in 7m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/04c363fb2d4f4327802ea7081a773372 : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75c1672a6f54448fba0e7184a41780fb : SUCCESS in 9m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/387900d394f04f5f960490faef6a53bd : SUCCESS in 2h 13m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a70aea63406440ed9a9716dcf2a103d6 : SUCCESS in 1h 33m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d2c00cdfa054f69a2aa20149959811d : SUCCESS in 41m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0c05346e6744de586c720fb2cc2b5cc : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c11dc6ca9f3748db80f798e7a2ad60f8 : SUCCESS in 1h 12m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c244c1e9d40844cc8da998274cb5cafa : SUCCESS in 41m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/81d17d8e3e4b49149a75c3b8eb1602e7 : SUCCESS in 1h 50m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aba7c1c2b2574dfe88119ac82029a40a : SUCCESS in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae9144637b0b4aafbf787a677a66a77f : SUCCESS in 1h 58m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4019d52bbdbf424b898c6ae6c5a61016 : SUCCESS in 2h 08m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f0ad83e365a4345b6784b26b204d824 : SUCCESS in 2h 18m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb189fe3ec13460e85f57c696601137f : SUCCESS in 2h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4baa898661028d22174f5a9bf75e7225f5cdaeab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 10:58:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"30e16313f0b5ee65fb5eba09b1f0ed4ccdc77e20","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 10:59:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"933fdaf41294dddb1e778406528c53ef87550d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 11:28:31.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"22d1fe05e5c22a5c47ba3c69ce096042b0dd1f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 11:30:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2f5279292ed705a5b73f7eaef8e90aede6d39218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 13:35:29.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7703cccc542740a4b147eb14b1d1efcc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f3f108b3a704c33b727d677ee0a7a99 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b217ef7ad64ce690b71cb2931463c4 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c6a8e3f188f4f09baad734e7168f180 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/298c18b8f5aa4ddea1719d2ff272616c : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94300cd5c8374850a413c9f282231fe5 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e612fc745aca4806b5e2a9b18e6469ce : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3ecf446cf794327bbe5bdaf58755dc1 : SUCCESS in 6m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d27e215095304a608463522a1fa3e50d : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ada540a8dec34bc1bd637333d41e1e77 : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b337ed66be7e4d4ba7867a486c5c0acf : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/438186e9de3e45098e25ef534301bfe1 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55b236d4f151491e87a1bfa320a50fcc : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8797e3d6c275467085b6c348d3f56a04 : SUCCESS in 9m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b966f37b8c0840b48b30e5780a9127f8 : SUCCESS in 2h 03m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8937dd846904cbbb7eee887d589f29c : FAILURE in 23m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c74a97395abc4037a918f5eb8aecabd6 : FAILURE in 25m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e96faf06fe204a6aa9182a45ded24528 : SUCCESS in 56m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0d9c600812ec4f279a1ef9c0cb4c0026 : FAILURE in 57m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d31226161fb94695b49034aa57c6fe2c : FAILURE in 30m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6be1885b7d7d4d46812ffaf0514bfd3f : FAILURE in 28m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/157771155ea94e018b0b1bb089426889 : FAILURE in 37m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7016e01075874e8e9fdea0f46cf6b97c : FAILURE in 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02cbd2709e834ed0800f29af3e0a3c25 : FAILURE in 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/74071b84cdc2498aaf119a514b56197b : FAILURE in 24m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8965f35c40f64cb1a3fdd066a39f3327 : FAILURE in 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a90251ec36fff1db9a6139b76d530a2f9cee46fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-08 13:43:50.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"fcb84823e347060ffeef5682a33edf30b7b9be6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:10:24.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfb86f29ab140bf987120b03be4cd39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1214164edb4888b62212e016a25892 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec57d12b84946f2b93f60572b460c87 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6dd30edf0874a37872f990e0bc1ca89 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33bca63ab2c417c8c0b01d70f28c3ac : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90e7716464f344febae5d83723bc84a8 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0def05b3e03419ab7b5bfc8567aff18 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f0a475b23f4435abae8cae3e572716a : SUCCESS in 5m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/534ef4a6adce4e3fa4d9ad6303708ced : SUCCESS in 5m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5d304fc4a91a46f3915ce050651c7f79 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8acac880756a4e12aeff13560c897d14 : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1c86c6504884e2789660e2ac31b050f : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b576fc297a93460d8adc55acd66aeebe : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c48217b1e494641a6c522186453c154 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4deb844374494f22b9df12d4ad141204 : SUCCESS in 1h 51m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bb821e9e5fb4f3b8134f5322164cda4 : SUCCESS in 1h 43m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/939e0b5e8d7e4a4792f409a5d86c950e : SUCCESS in 40m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08bf0c4348a04bd9a13034084f6272b0 : SUCCESS in 1h 14m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7af4a4d8d83a4009991f3886f378d86a : SUCCESS in 1h 13m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c2024ce57343409a37d150a5537351 : SUCCESS in 40m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ae0618c2d0d4467a775b21add24f866 : SUCCESS in 1h 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d431689ea914dbd9055b2fa937c443a : SUCCESS in 2h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/44d01056450b47c7bd3d074ada474ca5 : SUCCESS in 1h 55m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a4af4bf6cad4bafa67df46c4544bec4 : SUCCESS in 1h 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2a476417310f4d9fa81bd87dc4c4505b : SUCCESS in 2h 05m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/03adefc991cb49e0aad06c27eced0f21 : SUCCESS in 2h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8047f56c109d07632318cfa838aa1edc2168bf40","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-17 07:38:58.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b04c6960b3795ed36c51620f614eff99d0ee9c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-18 09:10:44.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"4e9c16febf21f0e3f0789fe74c99a5d32819b037","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-18 09:11:22.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5f1b7e684d109360b45319bf7e1b41f94dcfa1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:39:41.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cf17edeb2d64369834895ef5f67ba31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/affe7d80ce2e4c368037acc157213465 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7230e80e893d47deb2f8f3995c37912a : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/083b31eaf26741dca52da370254b6dc3 : FAILURE in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c4f1e586eff42c1b815754813c4368b : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d61a3a6384a4806936183a6240a3c03 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea50b32b0214f3eb1b0b040fa5ff58d : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cc231042dce4604802e2172797cdd82 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dee8483a9caa45128ac624e8893a8cf8 : SUCCESS in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/15af3abab7704362a471d15897bd1c04 : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f23dbd5d9e9b4686a586ebe31c6f355a : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2bc77dc83f3440e4b6177c9de936d053 : SUCCESS in 9m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a78ad158ccb5444da4778fffb2e42139 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c80d3412a29c4592b35dbda894b96f43 : SUCCESS in 9m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aea00577edb5402b83e48164ede984b6 : SUCCESS in 2h 00m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/129d143341e747088d2e098a920bae03 : SUCCESS in 2h 22m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/299fd3889d4d400bba3bd662eeddf0d2 : SUCCESS in 45m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9a85956c86447a9a34218c3a31d697c : SUCCESS in 1h 27m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ea1f4addb2e44ded88b24ce02dbac216 : SUCCESS in 1h 16m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75df256f923a4e7b944003212b665e5a : SUCCESS in 57m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d738a787e0ee4353964e568b6b82caa1 : SUCCESS in 1h 53m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c0fc173c17748f188b8709ca191216e : SUCCESS in 2h 25m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf520cb9e1764ab8a19cc1fbbd53c515 : SUCCESS in 2h 07m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8091b7e1dba24434bc16d788de4cd3c0 : SUCCESS in 1h 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a9fa8dcf16bf427a8297d68255ed39c0 : SUCCESS in 2h 24m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e52b9988b78646ef80a36f4c295f7275 : SUCCESS in 2h 05m 22s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6c4886e3bf2061e22612701794c5de4bc59ab392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:44:17.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"f90bc648d4a8eab83ad571420feaa94965162de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:37:35.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a7a8081c8d883d03970642f516cfcb414b2b3426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:47:51.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"eabc600cda66bf30c728d9c40a30f24655bcbfb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 13:01:03.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f204df4d9fbf4836bab240a7240cfa50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b160ebf46cf7407e9d34eb976fb33709 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7bbddbe078f46dfa5bdefabaaf709c4 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6b17b5324045649cee49bc51f0f20b : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c347490ab34e3a82dd1a4fe4b7f3b1 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4d71cc518c248cabd475b25323e0eb9 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d5f8f94ea144ab8b1cabe3ff9ad3ce : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75847489da724ffbaa8cef2c9c36795d : SUCCESS in 6m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/73f5dc56ad6e4f8da64df691076dd456 : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b6ff5cb6c6a54f4c88986ed9a875e97e : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/121e95bdb2454177adad3f2a36e3cb57 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9649b4d80e8c4e2283fe91ac5eafffca : SUCCESS in 8m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1acd2e6163c4fc6a2c077c9ab862f76 : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/875df26395a54367939661102d65f0a0 : SUCCESS in 8m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23b84f7fcabc4a04a1c18278b01e37c0 : SUCCESS in 2h 11m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dfeefa25f5a41be827a5714d58c1e76 : SUCCESS in 1h 43m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47634367aff64880b0201032fc235de4 : SUCCESS in 45m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0f66437e1e14bc8a363c6f765119530 : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/73bd156b4d7147038376d76c68ad4658 : SUCCESS in 1h 02m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bde054500aa4ae7adb8e5d98a2c3e18 : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ca9ca575b13473cb94d07e4f8cee4f1 : SUCCESS in 1h 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59725c511d9c4503ab86db53e9e0a292 : SUCCESS in 1h 57m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/015e6ec9fbd54b548a20eed5a700330d : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce1555eb96704d9996042e6bca7d7af1 : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/267bcaa6897d4efe9d6931952a456a9e : SUCCESS in 1h 28m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4069323e040e4faab590d84fbaa5b214 : SUCCESS in 1h 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"507c30206817d76db65e6814babab127c488b767","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-19 21:39:05.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"fa25819f751cdc491c30df042f7c84fa9e8d74a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:00:58.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"6b91e06080195137c04f6a724b09315af4f3b937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 17:34:37.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff65721b748a4d1888f598ec9f193512\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f30045268818484e840edd90e6d53f51 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f7def2466d4e2c9ab5fd474653aa1d : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56642178b3a74deda4ad27eee84d967b : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af2d29e43044dc7b4a000a22e798d2b : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97b9faad6f4a4bf58fcc0379519a112a : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60dac4dcf02f4fb892c3826e7713fe1b : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd07260c3e9a46348fc3146d8dbc2eb9 : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f16ce38f4044831a1c190ba98e1a920 : SUCCESS in 5m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/386ab2d246bb442cb22534a958e82190 : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e082530f81544ca197a0bcdf18cac810 : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/954c48587acf4dba9dd9f57475f83fed : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/674459c4e3fc4414bc336ee5c2f84a1f : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/385fc45dd38c48708b176b4388c01075 : SUCCESS in 7m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a468803c00f49c28b7efcfb49fe1448 : SUCCESS in 1h 39m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99439a944e01431283732c05838bc5fa : SUCCESS in 1h 37m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32e462b60db44fea8e4394028e45e42e : SUCCESS in 42m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d5f2e9a9b4d49daa4c05420d4e7ec0a : SUCCESS in 1h 12m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c436ff6b2a884d0a83884663be2b9f60 : SUCCESS in 1h 11m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1a6ac8b5b764f508404f2f57a3e1774 : SUCCESS in 38m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c697b2d63eb24c56afd81638302c3063 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e39984bf04ee4b52bb4498e67a2efa73 : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ff4b7ac9a68425c9b191bbcfbbbfb2f : SUCCESS in 2h 29m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d3fc9529faa464987afc8cca1875baf : SUCCESS in 2h 05m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6989b3e6358345e5bbaf9df613a2aba2 : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f395ffa8a7ef48279367274473acaef9 : SUCCESS in 2h 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"af32af81248db08619e763cb03c1ead4990e4df5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 08:59:08.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f0980f32ef8dd6231a3fb4b50be5fe16bb38b587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 13:52:53.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"f0cc0963ac03d77efde982a58985d84e1eada157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:49:45.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4665c04804f4740a696b7cf4cc87350\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3a5b8b958d4f5389be0eedb861339e : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e3f2cb91ef4d0dbeb4de74038a15df : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3009e6d1996d4bcb8398d7f55ff6d777 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f4e62313adb4e2bb333f2a5aa0b46e7 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e202a96b79b4cefbd7102eeb06fb097 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f579fbee5ecc4e05ade119d29bab79ec : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a995c817e2ae4fe2b10aa296341e350b : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e25c19845b334b948f3383f2a2b06411 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/05fe3ed184e14a91807e2f79d90383e5 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e34f646f5fe74956a9aec5c10d46128e : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14c4f361329e4d8586c990de335883a0 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f3302e4af404cd88429c915dbdfa08e : SUCCESS in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e6ab618c1e9456b95a8fc5416a2432c : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50504a9db8db4a7bb6a853b4db123834 : SUCCESS in 2h 03m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c3c1d4e637f40aa8ab69dbef019d87e : SUCCESS in 2h 25m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61298ca3b32147c998be1702fa4bbc48 : FAILURE in 39m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23e45ab69c3847128213b01d9aeafa66 : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67b8375fa1fa48478ae7472c3f9d78fc : FAILURE in 1h 39m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/625993bdde37419198951cc2abd90036 : SUCCESS in 55m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/48db3ee3e4da4ae7a803aa1508950dca : SUCCESS in 2h 05m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/09cb6582957b45d7a55792ed7eb31f88 : SUCCESS in 2h 54m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e4f3f55d4bc1461b8dbec36283c94e40 : SUCCESS in 2h 21m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/204113c2ba0f441f8be04f5f24190ed0 : SUCCESS in 2h 35m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a1166e968404592ac7927509af6ed26 : SUCCESS in 1h 56m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/42ded970631e4f2a9c221392adfd467f : SUCCESS in 2h 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"9cbd2743c8aa0048e0ba6cdfb0f2c2b9896b06ce","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 07:03:27.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c5b094c777d6dc8345285f09fe4e782ee4aef766","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 08:12:16.000000000","message":"Patch Set 23: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"56564d73f451bdd68fc56ea71c843cc495656b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:38:13.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5906decf036a4b03ad627647a0eac39f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da0d242fba6c4d97992aaeb13cbaca70 : SUCCESS in 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf9a605890849d39e0f09eb527c1002 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b6732141c74218aeb746279e028aa7 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2693c3dd038744c8b145daa537ae2742 : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e51467de2c74993bdd5581f66ee4a19 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e1643da0e040ed835f904b3ab0e3f2 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/208c8eae50ea497f8b095ac92ca68b24 : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3988e771ada0447e9a117ff3fe4e8563 : SUCCESS in 8m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8fa307161ca142dfbd87df513c8fe5b1 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5739a603afcf4182b97acb4e565a4ca9 : SUCCESS in 7m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c939df15fe704fec967474cfa0b73c63 : SUCCESS in 9m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1471cba40ff44385a20a3f47b47c58d3 : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d14d36d76182400797c37815d392d89c : SUCCESS in 13m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f98a1a819b1426dbe68a989b62eb276 : SUCCESS in 2h 15m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de910fbecfd048339d2aca244a035e2e : SUCCESS in 2h 05m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a286972b53784c9d85964f474c3a320e : SUCCESS in 57m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ae516a67002416ba0ee055521fe0785 : SUCCESS in 59m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/81d9b4c5fc134cbba21ff628c354e627 : SUCCESS in 1h 18m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e2b00f8f50845b98c05f75cac03d7cf : SUCCESS in 47m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c9abd2283e4d40128d3114f54ceef248 : SUCCESS in 2h 01m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/225ae2faae33481bb323d688f8ee8f2a : SUCCESS in 1h 56m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37c430cb1d5543b58880b4de1635c8bc : SUCCESS in 2h 24m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aad11394f9d1443a980ebac4c21d5b8c : SUCCESS in 1h 46m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8e80fbc6f6364427aaba258231357785 : SUCCESS in 1h 57m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f98cf2550e4548cd97b62c20260fc1c2 : SUCCESS in 1h 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bec970e529f4ce1a41f8ea5a2f5daaad9585f68d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 10:07:03.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a18d08ecf826996370dcfb0caf2c60bf82ac1df9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 10:08:36.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"c34669d27ad7d0cc043162bec2cfd7a6fcb474ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:42:12.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65d4828a5b3b4facaf0f81aad4f48df9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4567a6da26be4a9990cebde857bbdf8b : FAILURE in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5553fda2226147e5a9dd6b8074bc4e36 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0828dc1633ea4736af3009d3c7edac36 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6e9603ab2941518c756d0955258229 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a11707da55e149a19a10b85f7c971fea : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7ab2ccf01b44c7903e1d387ef1f100 : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75105aa27932407a86e9e0b8365be030 : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fc636b06363c4ddd908173a48649d637 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/878e9dedc25a485ea40cd7c2544932ee : FAILURE in 11m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa7e822a1bdf4049baee37536d5ec1eb : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2fcfdf718fd4750889824a7398fab79 : FAILURE in 7m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fccf90a49ea439190208d0a5a1c3a45 : FAILURE in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/532561fd9ba74e49b95b9d6290f48447 : FAILURE in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0f5b1965e6a48c1a3d5420e19953746 : SUCCESS in 2h 28m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/968bcb99abf44c26acaaf4d23340a3e7 : SUCCESS in 2h 03m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11e6f497af10463084471cbe40bc052a : SUCCESS in 45m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95e9fddec8a74dcb84baed668340e3ca : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e240070108394c258c7ca563eba1d84c : SUCCESS in 1h 13m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afbad9c7b51c47d5b375cea29ea28451 : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fdd7fc6dffd6444d98ca142532d8b363 : SUCCESS in 2h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/630ef7ae25cc4b4c976a46bbb9f0556a : SUCCESS in 2h 01m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f18b3eafb3a24c659c86b75c3ec81f6d : SUCCESS in 2h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c3cc4192c2f4c9596f666807bb46c4f : SUCCESS in 1h 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fa33009a7d824188b918bd577d4d94c7 : SUCCESS in 1h 58m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/69d1d41716a24424baf6ab71b334aa0c : SUCCESS in 2h 04m 01s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"201c60e7be424d9ee083ba2b1de9e3a943214d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 12:45:59.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"f81944584dfea6b3181fe990d09354567c58fb45","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 12:48:03.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"4233381a309a201ffe72bd063852e2b261e9faa0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 13:51:24.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"164f1654f641d810b8be0b7c7732a951f4ef5e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 15:06:14.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f501f8262444ff8a8d6744eb59cc047\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/defdf3abf9e94294adc2e9bda2572767 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096ef0cedcd84c5fb4e241b107bf4c53 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b944ca984f9a4df6b2ee3ee296f853b9 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d37347d3976f4cdd84017e8f5b65d2c2 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3da6b377fc9d4a9486065aaeef4c987f : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f370bea7088a49c7b4e76f21f2152839 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c806b5a1e849418fe0f7d6d2dca565 : SUCCESS in 6m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8a386169937b400d9202530d9fe9d6bb : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a7a48ad7dbe84528abce388ef2fba59c : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/17c5f6fc9b96478da46483e1380d4059 : SUCCESS in 5m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/720db64903e04dca997b6f6a597548de : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5c6adb322bb47a9adc82cf5341ae413 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6ff3d9ee4684cbaa92b055ef4403d65 : SUCCESS in 6m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6de420e36d64377accc9bab91edbe75 : FAILURE in 39m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d32b593d53794ce2bda35a9ce3bb3866 : SUCCESS in 1h 35m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5bed8ea38d94099b1ef19bfd278df71 : SUCCESS in 49m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc8982db559745dcadf811fc30e6aee6 : SUCCESS in 1h 01m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dfa71434a839441dbc33b02d83b3915a : SUCCESS in 1h 12m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/255ad20594064ceda7074f530e5a8ef4 : SUCCESS in 41m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/04493d21083740e38cf6b9291befa312 : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5ce650c78e24cd3b6dbf1cbeb82493e : SUCCESS in 2h 17m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/43bc80a7b5994daa9351ba0c8dbd8850 : SUCCESS in 1h 49m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/219b58b4299a419cbe092df4fce2d3eb : SUCCESS in 2h 10m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/296b8032f1c64ad085966f200522de7b : SUCCESS in 1h 48m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/369908ab568e4e31917b587316f0dacc : SUCCESS in 2h 08m 44s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"59262a04c30801767b21c5fb798ae5862225a5c4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 15:30:11.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"945e890b2225c8a95665d57dad543c81b821a16c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 17:53:19.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd938520e10455a870daa5bd084d027\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b076cfc11549452d8029026760f1c2ec : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e189136157f4a6e9ea42c9ffee6f6a4 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05e1696518e249729f68cce43f61a2f9 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43e9336be6a84b1593712ee22bb846cd : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/034eb82b2937467f92be61c6b7bf6c8e : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b56dc5d42174cdf9bea6c4eae2b7495 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728865495dca4131a4c4b0ade4679736 : SUCCESS in 5m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9a4dd0ee7f584f25a0f9d13a628b56d1 : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3947c8242b1848128e4245260e64b9b5 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d94664d2f10847bf96685793adf6c2f8 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e843f14ec77480dad6f9d33fa9a6735 : SUCCESS in 10m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8de825c76a7465d83bce482b1019ebb : SUCCESS in 10m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/55f990f34d934df28aecba2e66161d83 : SUCCESS in 8m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b1effa19dae4a02bab5fe6ea15ac1c1 : SUCCESS in 2h 21m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41b8105a1e8e40b598608229dfd83c47 : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1101ef460dd0407d9fc38d6c6175c2d3 : SUCCESS in 44m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c142e72ceaaa401aade2d26e576cd964 : SUCCESS in 56m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c764ad9e3a7140c8b0e4af47b42af7cc : SUCCESS in 1h 16m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23444e6f93f84fb5a0a4e591c4a19edf : SUCCESS in 39m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73088c1606724191bfa33afdf3d8aec6 : SUCCESS in 1h 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d95e6c9b64140fe81a05842bec21710 : SUCCESS in 2h 00m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/570b7f47433d494786ec983d9269e13b : SUCCESS in 1h 43m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b10eface52534cbab2aaeac84510e13f : SUCCESS in 1h 57m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/71d6a87184154d899cb79902c40d750b : SUCCESS in 1h 39m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b3c7708b670420889c36d95d57af160 : SUCCESS in 1h 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"600c09afa30a590d99999c552d78145821547166","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-30 06:58:29.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"e10be6276f4eddf7107f029818c5e194b7675883","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 07:51:00.000000000","message":"Patch Set 25: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"da6d9402ffd5754f15f8b581d63a9af865e5033d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 08:00:07.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"e1784fdc8593a4ed66640e0131fb1deba6c843b1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 09:26:02.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"63ae964db56c6dd2acc9bf9e759a514e25d54d78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:21:15.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"69df945a289456fdf132ca70f49a81612c33774c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:27:23.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e67317b1ed202ac0b95150345ad9a59cd767e342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:35:14.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"f61b1d5ca2553bb1f7d18605da92593a94767ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 14:08:17.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a37c5f3387843a68418ec60fc148b30\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b43abb1fa74241c2b5b96dd9cff15b24 : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587e4b3306d9483b976c3dc277ee7c7d : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17720197fbfb46898e337718f88fd3a1 : FAILURE in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8ce3036eb24705b8a114a0fe5a7b7b : FAILURE in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/164411dc6ed64a6e9268c5a7fcbb0bbb : FAILURE in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83460e4d4cfb445fa526ea5a6351acc7 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28d0971fa6fc4510958344575335269e : SUCCESS in 5m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/aea9ac2f43e44452858ca229252f757f : FAILURE in 5m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e7798cd2530c453c9271a31270593a79 : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cb407d419a9c4ddb9a41a8bb92be10ea : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b91486418f64492b923f6784b297dd05 : SUCCESS in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e36ab00891d4330a4710096b6a099a7 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c8959e7cce04e6ca364eaa9da8dc0ba : SUCCESS in 9m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b95c5e92a7a4f35b3c9b3d870705299 : SUCCESS in 2h 25m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ff385b3aab946f2bd82252a2029c58d : SUCCESS in 1h 33m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ad6644490324188979cf14bad57aca9 : SUCCESS in 46m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/310695cc9a8147b486c2e6dd9e80d619 : SUCCESS in 1h 07m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4cfd35e0812f48849bf85d29c6696b40 : SUCCESS in 1h 14m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74ad2317f147429ab3a90381bf50162c : SUCCESS in 48m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/941e5ce7034c44ce8e27d861dc6096bb : SUCCESS in 1h 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2819c049731f4246bc92ef9f34a3b906 : SUCCESS in 1h 47m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a21c8304791f4838bcb8a9b6da3d684b : SUCCESS in 1h 44m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdf6a9d17adc4d4392831653f551881f : SUCCESS in 1h 54m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/48a184367c9244d3aa02864f33d4119a : SUCCESS in 1h 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fd3552721a0430bbcf186b4d9dd4a3a : SUCCESS in 2h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a7e65cd094e40d3cb94825f6de48b55c366c313c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:22:33.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"002b96f0cbcf97c2a254d3cd3ccf04736372f54f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:53:12.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"4c6a28ccdf67b1be4d6aea1be20dbf5744ca9b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:54:27.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca670d93bf4414889e9d8a7425bcfac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dfdf912053a45f797fc8c001d5faef5 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f565a282750430f93ed00ed09b947fa : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ff900c169d45f28bfb11f3de4be98d : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e865afe1ef14d60a25443c8dd132d16 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a07462771cf949f8a387b2d994002bda : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bcb09525684b45a87bb916247cf7ec : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1db6cdb20a834500b8ef852bbb0ca621 : SUCCESS in 5m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3c8dff5bf981419c9a6624e914d63c02 : SUCCESS in 6m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b93af48bfddc44d48c312276b8d406d5 : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dbc07dbbed6e45898a3c0cdc688a1660 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b00f2d732e424e10a3dc21f3c45d3a1f : SUCCESS in 8m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35189027639e4928abb8e078e058dbbe : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7804bb64ad94ab58879fdff4773a002 : SUCCESS in 9m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc80ea4c42334cfa9b51ae40dc4da796 : SUCCESS in 2h 10m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99d5d45927454872b1d7c00e0d78f71b : SUCCESS in 2h 09m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d85407df5164475b8bc68b54ec09ca2 : SUCCESS in 56m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ccdf4378cfb407c923d4554a7cf5210 : SUCCESS in 1h 26m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d437320e91444087ad5ce7dbfd13ebaf : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a7a74a4957446dca7bb1e92723acb6d : SUCCESS in 53m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f4a8a5f879f4d95aa48d9ae6b5a3409 : SUCCESS in 2h 23m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae759c7da2424b12b393b7dbbdeb866c : SUCCESS in 2h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3eb884fc3f4b35bfe4944cbd1389c5 : SUCCESS in 1h 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c3f832595014b1199c0fc72c1fcf74d : SUCCESS in 2h 00m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bb1db05c302b40099e1195d153fd78c8 : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea4170b6ee484527bb571192016e0d49 : SUCCESS in 2h 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"bc1174e42dd2f550bc68183234d64bb6c8937e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 07:29:59.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"b484b07e4eb1303f255e769d60f6dda22b353540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:09:52.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"c0d631adbaf8c7b9ea9615305967dab505319df1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:14:45.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"dce5e6ea8ed352bfce9fcb00720bde9dbdb102cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:12:31.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb3e364ffc1419587204f41c3d4b46c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1356f70fd3884bfa9c1f86d2df2b8ba5 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c4430215a544917a97fe8567f25283f : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea576cbd5fcd42a9816d5e77a959203d : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da7d974db644eafa38e4a665cdee7a0 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd6afcdd49034492bbfefd2b24a3ec74 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d772a97558b4c429a0bbefdc1a2f10c : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90099a3014044109abcb953450eb3f4f : SUCCESS in 5m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/44df283718af4971a61665f4d19e8377 : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8fcc7cd889934b3db93a334f611346f0 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7b38a49361e842c687dd7cbc24d4112c : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c36413ba77ce44dc8cbf0a1f5dfbfbdb : SUCCESS in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cab63213b0494177b6f97f3fdd09082d : SUCCESS in 10m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa540f1974894780abc41c8457a09480 : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f28932279a440a6927964759787ab8a : SUCCESS in 2h 17m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10ac2e64855e4cbc81ef175241035d07 : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6dd1047d6064bc19ad0d3cd20de368b : SUCCESS in 46m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c341fd61bc654aaaa3ea539b472c0bd1 : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf1079ded2bd40ea86a813fd532bd4e1 : SUCCESS in 1h 34m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9153c73b64e4407a791df56cf8ddc48 : SUCCESS in 46m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/affc9739ca15401fbad710dbbd1084da : SUCCESS in 1h 40m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7f8b9d42ea64e4b898d81d63c8136cb : SUCCESS in 2h 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/989f6c3503074867a1e044101fb944fd : SUCCESS in 1h 46m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/059556988da5476f8848316763876d1b : SUCCESS in 2h 19m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/151e9ab676534428af304ece0ac33eb1 : SUCCESS in 1h 43m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/155cd2e62cf24b84bd788b92d22ec7a8 : SUCCESS in 2h 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"9964032fbbdf06ff2eb8205735d1d709f7945970","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-31 19:22:30.000000000","message":"Patch Set 31: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"a73861b08985455d00fb252197c1420dd95b2f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-01 08:56:27.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"651eecfe320477a114fd2eaee91c470f81558e25","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-01 09:23:47.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"e93180a57751a7714054b0cbe603a27f7fbde76a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-01 11:12:39.000000000","message":"Patch Set 32: Code-Review+2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"a8d94171fa1a664e8a28a835d1bdd45655f72b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 11:37:54.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09840ea675b64418a53bafba6c3332ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f953c520697c4fa08260e75113a81508 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d16fd930bb74feeb10036441aa01d4e : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec4b7777eff4844a8f6a07a762cf173 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7dbf0b86a94a1f936412a9bb26df41 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6c47e72bdeb4226b73514d809c955e1 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f681ec59244af79737ae3917acd895 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff7fc67ae3614c5ab64df8db44cf2b3a : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6f366da2bb53498bb1aef07225988c10 : SUCCESS in 6m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/04070fa59f7141a0986ea4556e10d842 : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ffb9e4d3b26a478d8608d84c55a3811f : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/369935eed37648fcb464294665e03c7f : SUCCESS in 8m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e720ffce954644cf85200c774ff98e25 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c87399cb822490cb49ab53d3c0d7ed7 : SUCCESS in 7m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d01584980dd47748d390372bdb0a9bb : SUCCESS in 2h 23m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04fba20f95e5479d8813ebdcbb0020ea : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6777982b610845db9d8f9dec1dd964ae : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/275843ccd3444e848fa5d58ebe958aa2 : SUCCESS in 57m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/945ff0d67d364710bde78ccd6ce970a6 : SUCCESS in 1h 05m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/347b45c916664fa192cb6396f59a4f4d : SUCCESS in 41m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8262645e42ed49828f46be532a87494a : SUCCESS in 1h 39m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cc7c59fa8324fca9e335500071d06bc : SUCCESS in 2h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/287a711c1eed4f81b79202844da31b3d : SUCCESS in 1h 52m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e408632fe9fc4f418b37d8c420ea70d1 : SUCCESS in 1h 46m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7b2579af12dc4b1ba96880ad4e4408b2 : SUCCESS in 1h 54m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/81692c0176414b53814884e5f8e0252d : SUCCESS in 2h 38m 37s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"9b6401c5d0eefe66537b5a746dbae8cb953f563b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 09:57:43.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"5594921d56477553675166efd21a378a6330ea95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:03:44.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"aefdd3ca4fcadfda60b75da81d7020bbabdedd0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:08:17.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"80b6826107a9c16b4836578f0476b35b4706c12f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 12:28:16.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"26bd04bbc20053e4b8e842e9eb8494316a02f55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:02:49.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20819f41f240490eb897fadd51a3189a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f025d03fbdb64b4ba016524443b65361 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3601298478f84d3da493eaf8e84b8ace : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74bb9d57ed584686837b0755a2b9225e : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29f74bdab06d4894a5ac0b70593d7ac7 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e64145c563b24b658a4b5fd528b09323 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b40e45be6fc46c7a46b6a6452a3e906 : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b1955df5e864f53812a46186c6d2b0e : SUCCESS in 5m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ed09fd11d2a34fc5a3ef7b027cb2ab39 : SUCCESS in 6m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f37b147901f14318ba350bf02e535497 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89aafc4265c44325881fe42b62a8efcd : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff7a9dc2f2cf40948009630ab130890d : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8664c033e53544c483c1c5422ea1b01f : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2d8d29e8ca349ffa0c937cae8f7a845 : SUCCESS in 8m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad78e703126e4375ab9891ed92e42655 : SUCCESS in 2h 22m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1b7b5f84aeb4b73967bd17e026eee3f : SUCCESS in 2h 03m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/308f93d4a79c48c497f0386ad162b217 : SUCCESS in 46m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a70bae5154542888cdd9328297aeb29 : SUCCESS in 1h 04m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/163d03113f214d5787b5158a4fa75cad : SUCCESS in 1h 29m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3970381ba4064144848f5bddcb77a01d : SUCCESS in 40m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0e2cb9b06b3492ca012138a2b8c673d : SUCCESS in 1h 44m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9696bf836733445eb3d7a3318afc9d02 : SUCCESS in 2h 12m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c48731a4c7e44f9381f632dee343efc5 : SUCCESS in 2h 07m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23d3c0786ea84d3388bf6b4bfad44cf8 : SUCCESS in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c167df106a104755944a98c5d6922902 : SUCCESS in 1h 46m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/187281654aa24bb2a292a0b4ec15f393 : SUCCESS in 2h 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"a104ce257daad664ba7de5d6fd36bc140cb62f3a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-14 15:06:05.000000000","message":"Patch Set 36: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"8cc691429ce8a08ae5d9982c727859cddeae8c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:09:36.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"f3c091a1f96e10ff2bf7424b6254522c3f017f10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:11:01.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"7510ae33706536577627867957c85e3682b9074d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:04:45.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f300c304f69423798ecf44d11f766a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cb3b76cce0f4401ad429b8fda86c47e : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d441e7c25abd4c71b079e396b6a630af : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81a59033144c45c8ae57ad4cd49740ec : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3062d8da646049c5b334f13f269ec89f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07acc85cf30f4f9da4a7fec2fb68063e : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03cc960df46040b8ab114710612dc56c : SUCCESS in 6m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/258b2bdabcee4c7aab0eb671bdc621f2 : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/13cbaa81f42142dbbdd03fdf8bdf7f37 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7becde8d81ab4785a3ef7af910c01519 : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56ff49e9de51449a8c0d3c72747f98d2 : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c3e27f14196469690ace8f530f866da : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa1c76f1f9ed48fcbe271e855a7c2d24 : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be0b27293b1b46b599f0d50ca038f10c : SUCCESS in 1h 23m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ace9e817f35749ea9c0ffd4b269854bd : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61c8e58a07bf4829b3c322e1df6cc2fb : SUCCESS in 37m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0aa94f24f5e241e68a9a15f86781bd0f : SUCCESS in 1h 02m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8fc44b7966774cb3ab492741b4ed45d7 : SUCCESS in 1h 15m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78b611beb9644c6c93cc5ab40ef04eaf : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de8e776f7954465989e7f7090e225423 : SUCCESS in 1h 46m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5bc75e8a9a14e989c24354a106271fe : SUCCESS in 2h 50m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9de0c834198e487ca701fb67cdc2f2a8 : SUCCESS in 1h 41m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/925ec0a1595f42c186898f28d325278d : SUCCESS in 2h 16m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1bc99a56f6b94f0aa247da20229797b3 : SUCCESS in 1h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4259ff2ac53344aa95bf1fc4f80a0ed5 : SUCCESS in 2h 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"442b66a6087312fcb2bc56c2770b3616539442ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 11:30:48.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"6d407016baee2cad313218c761a3ea979f23df26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:22:40.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"98ea38f3e65f79b54438fd7101d5d5f2d37f8b30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:04:30.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"25370cb4a5034c3bb21bc5eaf52680e90a89bb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:08:24.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"81f674a614773263ac7882fedbe5a795a9f57b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:15:46.000000000","message":"Patch Set 43: Commit message was updated.","accounts_in_message":[],"_revision_number":43},{"id":"c26560e2d7b149d18fc398598f1e1fd0b2095442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 14:41:34.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"c107f02c5d1325916b40fdfb1bfd0a6c4e5e4f6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 16:19:29.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"d4d862ce3d21f2b569c7c38da912c44b7e07696a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 18:36:58.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b31b3ce2c2499b847c5c653ea05ab3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35ee0a7bf7a2476e843504cbcbbfdeb7 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9905b0649ed74d25bfffed9ca321e70c : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca1d5b1b7a274e16acf38ef96cb614df : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82475370d0154c48accaf094553e3cae : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556ca5f0429e4cd7a648500bf56392ea : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5ac2e5adbde49e8aba325732461860d : SUCCESS in 6m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/05ec6af9f2d4404aab9efdeb4b08571a : SUCCESS in 6m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0dc7db3dd29c4e12865ba4291eb322f2 : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c816ff528cc4bdf96b3c6efd656c7cd : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/051b913ab662417f894b8c24d36bb61c : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8feaf0a8c0024f3ab4cacdfc0e91d7da : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c3ae27077ac4c0d92b9c918780a6960 : SUCCESS in 6m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbcb20e5f14f46979f81e4953eaf7dbb : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da965ce20f7841aca079eadf23212c24 : SUCCESS in 1h 55m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bae4030a78e84c60aa438a1d95f1caf6 : SUCCESS in 33m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d4db441f3e11452fb6864b8dca50993b : SUCCESS in 48m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/963e8869647449469ac78d9615536727 : SUCCESS in 1h 11m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/463c9d0f1220439996a0a584ab04c6a8 : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ee54878c14540c7a8a6ffae2961add0 : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/463ab3868ccb466db2de0bbddcf68dbd : SUCCESS in 1h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad0e14e4fdcc4477bcd919832447202a : SUCCESS in 1h 47m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f58317f36d3a4fcf848ab39fda4d9d81 : SUCCESS in 1h 46m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7ae1633a60044358b7210eac9ecd91ec : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/85b3cc294a9143d2b58921e8fecfb100 : SUCCESS in 2h 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"a51dca343068a4bdc8a66a90f97975059aa0cb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-28 09:35:53.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"c782cadd59972d5279693b7909288f0fa85a3ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:04:56.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/480385f33c9f445694eda039c791c250\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c614a8fd63ff4b3f84b34e783a32b02a : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be8864f8447849a7bb58f041cfb05569 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c4990458d254d34975bd51a69c9adb6 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28acf5158b844ff5a9459068097b0528 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b904657c56874a94a967719ce7c4f327 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720ebee71b1f458eaced503163d55f71 : SUCCESS in 6m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4aecf1c8627f4e1184f84822fbb2d566 : SUCCESS in 4m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee35f6c1d6de4f37bf6d8667ac404682 : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/703905db1cdd42e39eddce55bfa8b970 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea1977c535b8418faad3d3004961b649 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9055610f0efd4fdc86f9cffa631a0493 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f41b000bdc1b4a5f996c9a8cb3a4d4a1 : SUCCESS in 8m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/180c52d4574c4b0bbe601de62f777e36 : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67b2b5cda78b47f192b879052ff85787 : SUCCESS in 2h 06m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/899c900397c848f38acaf606fa769265 : SUCCESS in 44m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3f5fa411a994405852b0d44e84a4720 : SUCCESS in 50m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0720f3ecb096432c9b91c371923ea4d8 : SUCCESS in 1h 24m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfe6e278373e4bf1978f9bce7fd3ffbe : SUCCESS in 35m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ee45012f2854089b761323a7b46c285 : SUCCESS in 1h 34m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87c2c75361814f0fb4632a1144edf9b8 : SUCCESS in 1h 48m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eae628782bd2410cb287d3170e409279 : SUCCESS in 1h 34m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16876643a87441b2aa334e7baa76bdaa : SUCCESS in 1h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f8d407b5b01497e956b2158a1e760bd : SUCCESS in 1h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e81039ebdc1a49f09fb6768c4c0db185 : SUCCESS in 2h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"58c01c4ad85eb98f9917e25de73f7b22433e4d64","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-02 11:13:50.000000000","message":"Patch Set 46: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"067020e3675a3a8034f7ddec97125997806cf106","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-02 11:58:22.000000000","message":"Patch Set 46: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"41344a0fbbc1f539385d162ea8a9dd7a55cfe8fc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-14 15:50:10.000000000","message":"Patch Set 46: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":46},{"id":"a82371eb3a87b42cdcf832ec002d73bbae19f207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 16:24:16.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"16b9a7dcd5f433bb890025b08d6963c9a0f733a0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 16:24:30.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"6c4dbdd3b61b3f940f6a012056a586965a3c74f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 19:02:06.000000000","message":"Patch Set 47: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a9b3e6c61c942a5804df9f8a69add63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/637b8677954242e3a7e1676a57c52eea : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd1af151ceb447490a19cc834bcb4c3 : FAILURE in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf018bd5bc134e33bc5b61d1a773f255 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36f1345fc71545058af3645ffad954d2 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43233cccaa44cd59384cabd2dfb5a06 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3862328a7ace4c8696cf33b7c33048b0 : SUCCESS in 4m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a7f670f29e8846a9818c333c9ec851ff : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c13a20985e014a0bbd2d665e0e686428 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/974dea3244794698bb24056fac339cb3 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff9b03f86513415c9837de74a01ee27c : SUCCESS in 7m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f55000251b784106961592cfa165cc7f : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/731aedab7d044dec9646776b0478359a : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92943c287df6405f95323784c34d3d65 : SUCCESS in 1h 37m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f527bcc66e6b4b5ebbef45876538d1b0 : SUCCESS in 1h 43m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11366e2084cc46b686636b323b71d833 : SUCCESS in 38m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/edcf9ca24ee44b2ea5f3e4f1bf6738da : SUCCESS in 58m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2f783e274073428f91b8d15bc5e5e095 : SUCCESS in 1h 00m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/245d7ad5ce0e4a63acacf4ffd5af6306 : SUCCESS in 42m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c9c1b24f57d45b99c6e175ba79edc44 : SUCCESS in 1h 49m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84047277a9f04b0182627d0757cf0018 : SUCCESS in 2h 24m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/817b0d584b8949e788090b5510fa0bb4 : SUCCESS in 1h 57m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/886230b2947e455c9d1432d7a3c8b447 : SUCCESS in 2h 06m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ec8fc0542314e2dba46460092be2cef : SUCCESS in 1h 43m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6b4bd61f56e4238bf6f01e11eb16cad : TIMED_OUT in 2h 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"f7fc6f6df2df32a4bd57aa2828e27fb04e7b6805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 19:03:55.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"a6a84ad90de6b0b9c837f22e915b92d6c5bfcc17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 21:31:35.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdce6eb655b742139aa5bf08000e2665\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44721b5c715a4a7d98562a6c9f9f6366 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b907014a1840349749bf29a78b6cf6 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac071b6b6c89446c81cf31207fecde80 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8a4bcbf9cfd4917a8a76d38f5416d00 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be93026023de475499f2764cd515145e : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a1cf24cb35a4024a581f9e294257b38 : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ac4dc6cfeb9a4ac7a458b9cbdef226b2 : SUCCESS in 5m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a5eab973bacb40438a93dc2ca9aa06f9 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/93e3948f929f47ebb71708aaa5ecb3c4 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27c46b1a26f1410cbbf689b49fc8e0dd : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e836214d948347cab9fe02c996bb8ad9 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36e45433405f4c91bd0fb00ac95be865 : SUCCESS in 7m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7be69116859c4c8889dd15fcc9696f46 : SUCCESS in 1h 35m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e0f20362c5e4a548cca67c58e92b09b : SUCCESS in 2h 21m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24af476874984f43a313d1d54d755819 : SUCCESS in 46m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/705d8bd9c5b44f679ad96d8f1a7b9349 : SUCCESS in 1h 06m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/23c4b997dd114c848eb204d0d94f5ad8 : SUCCESS in 1h 24m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe42efd89df14eb2a3b142a128136e88 : SUCCESS in 39m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3052a402844f4593bec1d9240ddb7680 : SUCCESS in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/040d5f6ec7a0406493dac39e6b377f9b : SUCCESS in 1h 57m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cc4694499d0b4841aaabb8e00a85916d : SUCCESS in 2h 01m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c32c2d1794a4620aa86f391c99ec98a : SUCCESS in 1h 47m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3a41f2c08ae844e8b5a748c5d7e6198f : SUCCESS in 2h 09m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ce7bc95ee384a59a6d19af71dc668a5 : SUCCESS in 2h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"9c8403bd668a06fd21345f8d55b1164850d91f18","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-16 13:38:15.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"7092d0526cf5738a2e0e2573450041824799e8e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-16 14:45:18.000000000","message":"Uploaded patch set 49: Commit message was updated.","accounts_in_message":[],"_revision_number":49},{"id":"cdb14ee52e83a7b3c591721edc985631c4967460","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-16 14:45:45.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"08cf294a2d82d8cc465c1d80774d7a3a16622bf4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-11-16 15:37:45.000000000","message":"Patch Set 49: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"57110431f082cbbc21c3e0831d0b31dc18efa72d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:42:31.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80401384690348eeb85f432fe39d133a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a94d4f4ed1346afadee98356051cdc4 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a479370128cc4851aab7a9ffdb841f36 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9491b34ddc4404292d0697997505eb2 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2254f9cc195477a88fbcc0b0fdadaef : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b3f42d6995499fa94f602b4b5bf425 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171e5a55d2dc49fa9757876cfae56254 : SUCCESS in 15m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7c29f38943af46498f08af979394a736 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ac9477da2f8547978562037cd48ba523 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5938f7d565a746498ab8371fab7984b3 : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf3d5165fc324cb490484fa78ef7f2ab : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbcd25215c00481ba58a59e3d05a2ebe : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5961d1d74866447ca02dadf9c33c5e3d : SUCCESS in 6m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11230d5608714953baf4519dde2a5d5d : SUCCESS in 1h 58m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5320b7b6c94e44e0b818fcad183f60b5 : SUCCESS in 1h 40m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90d679850afd4d56ae3fa3be19faf87a : SUCCESS in 40m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b14b2265e23460a94f8a4aee52cd54d : SUCCESS in 1h 27m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/412d0a454673404eb6c3510bcf1be30f : SUCCESS in 1h 03m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3a0f0a637a14aa1ae2ec5f805cb02d6 : SUCCESS in 40m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e640c90788b24bcf98d0cdee84f355d0 : SUCCESS in 2h 01m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7329fc1ddae04c03b06905afa2f01cc3 : SUCCESS in 2h 54m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60735ee1d674494fb60a4a6c40a924f0 : SUCCESS in 1h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7f0f40e11f9d46df913925327cd85690 : SUCCESS in 2h 06m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/121d6b481332472fae82aecaa9ae5ca8 : SUCCESS in 1h 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eea62ddc5da4910a1f6f7acc37029f8 : SUCCESS in 2h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"45a028e4b5a76a76959e7d4765e663e3490a6dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-23 15:29:39.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.","accounts_in_message":[],"_revision_number":50},{"id":"9bba7e8f96e8ab43cc734fe346c438c30c7faeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:26:31.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f3a9504fbd473a8b52ce1e871f5403\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ea8a352fec40bf94f26815f90b9626 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ddb1871064e4f00a6688130b933b1c0 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8076189c4834bfd82e348064b488592 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c2e43d9132c4a0098fe51ca967659da : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb4d1d37ff04e6cbfd303370c77aa09 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3098bea5c941419036001df33f7f26 : SUCCESS in 4m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/54038db035664dd19fb633edf2df26b3 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ca4ace4919414e2f92866c34bd3d57b0 : SUCCESS in 6m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c104355a2e948528cc6f404dc001bc0 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67e5eddb20f947ea88542ff1ac6f480a : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a47887eb44cb495c855bb3b101922078 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/582a335aee5746828fa7906b072108f3 : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/791a21e327944bcc995c7e53df86971d : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5939bfc53f3419daae38c5353fa9b34 : SUCCESS in 2h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78d65eff25a1495eb959cdc3886712ce : SUCCESS in 46m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/458a2f7e8e9345628bc7ba3ab523e6a8 : SUCCESS in 1h 09m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aa54beebdceb442697a9ab7c61de46d8 : SUCCESS in 1h 24m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61f9823c93a84591ab0b9e39e9fdf399 : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a21fb37cf284234a4c2c9ad9b1f4369 : SUCCESS in 2h 16m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f88f5e021f594923b6b20b8c22203e3c : SUCCESS in 2h 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a38a64d9e9a4f3692479c9deafed586 : SUCCESS in 2h 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f43ee925e37445868430a8c642259161 : SUCCESS in 2h 09m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/34e53320a7994ad7b6f0c6d075b50ef9 : SUCCESS in 2h 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e310b1923755417585f8e46d2019339c : TIMED_OUT in 2h 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"0f0e782bdbf68f33897fa076f8f84fccf9c05c12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 11:00:01.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.","accounts_in_message":[],"_revision_number":51},{"id":"12958f1a201f6d56e8c0a9e40c6e086e6d5820e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:44:53.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf4c14356bc0472f8baaca4a932e3f1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/062dea402dfc46359092abb9fbe3f39e : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac21bec200f24f638fa3432989fafef4 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5915c877d8d540f19cf3a711c3551537 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5c56bc86e984fccb73cf4e7d5b9b2e6 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa68440bae4a47d0a2ccba94dfd32d2f : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6df4cfb8847541949147e4df4f2cd349 : SUCCESS in 4m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7dc87ecf65da46ef9c55ca873665d9b2 : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3cdf08f80d8c4c4285289eb5ea8d6066 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/051a3ce4e69b45028fb4c4fc94c86117 : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6199ac1a40904db2ac39d2487e76aa02 : SUCCESS in 9m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ebb62c874254f29a2854b29114a782e : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45efb82ba4044c5587867780a6383a1a : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee4ccc0b9d7b4c929d542ff8eea4a058 : SUCCESS in 1h 24m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edaaf084fd7c45f7b81dc319a88aac3e : SUCCESS in 1h 39m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a23635fc245427586524644be36e29a : SUCCESS in 33m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f456354dc4d44c03a290cf1f2aa5d4da : SUCCESS in 47m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0d88a23498e94b2683650d79f7bd79d0 : SUCCESS in 1h 19m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c0ad33f6ee842aba115148f77f6317a : SUCCESS in 28m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e33dbe2bd92542c2b5a9b5cbbbd5336e : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d9e46a1cd2ec44d1802029e940d2c715 : SUCCESS in 2h 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0e4b4fcb23e4cbb8a479ec1115da3e7 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a6a2f7277fd43fb9231fa872c34d4fe : SUCCESS in 2h 00m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5377afc680b54069a7bd6a1efcf23315 : SUCCESS in 1h 52m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f5d42ffcfc84f08b47e93ac5be8c82c : TIMED_OUT in 2h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"254a55870519edfc939fede3d6d4ff26781c20f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 15:46:59.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.","accounts_in_message":[],"_revision_number":52},{"id":"ca21453e4f19f5e9ec3516e99898d9e32d435b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:10:15.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a471592153f45e3a7cec75d7bdfcd3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98bfcaba01984c85bb60eb9ab87c81d7 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b9daafbe924f52b4fc09d2fd6d4a4c : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a31ea4493970465fa2191ce7d74133e0 : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83a1bbc8239b4892ab294d72e058ea03 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a3e775d24b460ea684bdc6525fc308 : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/585331e63916467ead9d7c7b5f53d794 : SUCCESS in 9m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/23ce470ce88643b2a58077a08bf58736 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d25db454d3eb4653b8789b2c5f89ef5d : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a56ffc0cb3d14ae7a5d1d7b6a1bccef9 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3d586ff5489414bb9c268405cd6c96a : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9b1d19a36b948c9a264d9d9c30bbda7 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09d627c0694e44d6adab6b490ce2909f : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41ba843654c246469ba1d12f77ea58bd : SUCCESS in 1h 21m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52b8e9e5ef3240c8a46f3e42a10af913 : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48cf6949acc24a0d95668d9f089ffa8d : SUCCESS in 33m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5601dc14650c4a28ab16112d9a1b02a1 : SUCCESS in 48m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0e444bf8b15144a08ef3a703cc9d67e7 : SUCCESS in 1h 12m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a819dce78b24e39b9012de2f18da47e : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2e988bd6c8604ed187a411c8c56c8ce9 : SUCCESS in 1h 47m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce2e2408f3184bd68307e29a386a8fa0 : SUCCESS in 2h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/666865debcd1418fa3c5cdc18c0e872f : SUCCESS in 1h 58m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8889dec0e4e546b2bae7b46ee6e361ea : SUCCESS in 1h 41m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b92fedef471e46f9befa755671e42bc0 : SUCCESS in 1h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/29bbf39dc80644a687c4c209e97ff2e4 : SUCCESS in 2h 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"35fd8d39b6c46c1aa152b648235a5c2926abca1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:18:13.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.","accounts_in_message":[],"_revision_number":53},{"id":"6d4318b00da05ba80b73481df7684de399805198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:21:35.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.","accounts_in_message":[],"_revision_number":54},{"id":"8978cfc5b23d164479da6182deaccfa1b9977689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 15:00:05.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c57848f32715410caf3f40ef6467c22e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef0487989cb1421e909ccc6e3e84335c : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef748a2eb924a54bee474a26ddc8b4c : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7c7d5b241343ce88a98bab2e085a35 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15b64db7275f4eb89aef2820d795c23c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a15033faecf427cb00ef922f01137a2 : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed59c5f9c754e6c80349894347fad7d : SUCCESS in 6m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/68aff2fd0e594cca84f9df79eff97755 : SUCCESS in 6m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8ac0bbbd2218456e965a71e5627f47c8 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0537ebb73c04857811523d11cbe4349 : SUCCESS in 4m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bfbe60f8b1340c3977923fb863da59d : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30839bf968bd4153b30c1ed42b0678d2 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f4c9aaf641f4c399a5b5500e2fd399b : FAILURE in 9m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbbe246e3153474eac42e134afe53cd4 : SUCCESS in 1h 16m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e80b0760497a453bb6ace341a59518d3 : SUCCESS in 2h 10m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0df759a603e94bf08c35727274172af9 : SUCCESS in 40m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/88ce7279b82e4846bdb6f09bb8d32554 : SUCCESS in 1h 10m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1a04631d32b04d81a0fc0edfc8d5654c : FAILURE in 1h 20m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3271ea03ab4540509e8bced594c13894 : SUCCESS in 32m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4326b13c20144298a960bf728420bab : SUCCESS in 2h 10m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f2378871f754321abf1a067938f4008 : SUCCESS in 2h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7aec8902f6b44c6baff667ef97dce6bf : TIMED_OUT in 2h 33m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/76cbf82dd98c40f28db5e080a5717ceb : SUCCESS in 2h 30m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2e7049de4ccf450586a2dae0402fab58 : SUCCESS in 2h 09m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3b38f3442c14992a61df52f4fcaabfa : SUCCESS in 2h 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"a4937668c271b1905466a1fd9c1f6618138c1f78","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 20:23:22.000000000","message":"Patch Set 54: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"9536ad9fd7eaaec9d4aeec1a805d47a3d6ad8853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 09:27:30.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.","accounts_in_message":[],"_revision_number":55},{"id":"851ac8318984e2afaf9a872a7f91862733b5edda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:54:33.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.","accounts_in_message":[],"_revision_number":56},{"id":"71cd716a349e46dd542990c94287f872a1c80b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:59:41.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.","accounts_in_message":[],"_revision_number":57},{"id":"72130880b2cb9c5fa1ced58c8561c87d0011c4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:46:54.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f09abf8b6d0a4c6b8ee5a5cca78e03ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a05e74dfd72947ebb548b730d34a8306 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa72e50ec8d4b4da8368f81bcd33bd5 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/defd0b72a2e049bdb68abbff6c3caf5b : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27419135297e42a2825f7f0614a13b84 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fec116d2ab44a0bea964fcdb6f28cc : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0af69dd827ad46c2b596f4ad8b3393ce : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cb8587093a4b49c389727e7cfa5ef0c9 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8eeda412957244d296a5d12312c556e0 : SUCCESS in 8m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68b815cc18c149eb8a4a8757ab94d318 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f80ff7d92a4540fe827176e30890f3b5 : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9227b5e052a64e69af35d3e8172aea44 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/139c04bae18c422b8fcb53de6fa560ce : SUCCESS in 7m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7500ef8658843159aef3e3579ad240d : SUCCESS in 1h 39m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c065f1f5a47d4bc489fd19015c0e5a54 : FAILURE in 27m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ed7897bc00849e8bc21dcceb63c7c9a : FAILURE in 34m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6fa1c3d3ac454daaa1da85dee19a58e0 : FAILURE in 27m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f9c9880ec5bb433fabc954f8930724f8 : FAILURE in 1h 21m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e71949000dbd46bb876ed3df1d0309a9 : FAILURE in 25m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e1e59d79c8a4405d8d036eb02f14fd0e : FAILURE in 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/619b4ce129284449a413d0d7e1444aa6 : FAILURE in 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/561610e700b7491d91aeb7cadfa86334 : FAILURE in 25m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57b54eb6498a478fbcc9f8324f3aeb5e : FAILURE in 26m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d328fd56bc7746238be6af95ca9c7e42 : FAILURE in 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3debdd8a00cf4bee98e9348086f550d4 : SUCCESS in 2h 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"a3c3929442c6c009bcb787df33d09385502da73f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-09 16:56:48.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"0c557d6ccd8b7f298befe21890fa74ba70110f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:57:01.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcca9573287340538a77638f9897fce7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d15231a17f954fdd99452c899af28c5b : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0ae718dffd499b809c1d6333fc7195 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12b196af90614681b8ba19a676b3dc72 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70d1649afc9b49f5968a725ed3b0deab : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3d4e0841054fb5a3894ad3eec4e8b6 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7da224207f04824b54686fe8d714a6c : SUCCESS in 4m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3c1ea2c74bf74e54b256aca1c8297554 : SUCCESS in 6m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bf059f6c95064ffba3f4ab6f0fffdd64 : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe1d8014aec54d3abc5e8dd5ae465e0a : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/991b5b5c59364cce9999746528ce7ff2 : SUCCESS in 8m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3db33949c76f49e980fcf17122a0a482 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f66b458462144572bd68447e877c422e : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/247dbffe89e84acc9fd3095101722a31 : SUCCESS in 1h 16m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e67b32d9ace459abb26518f12ecef58 : SUCCESS in 2h 31m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a86456d8a4c44fc8bcbe4e885b6d6a04 : SUCCESS in 38m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/516f4e9a6ee349e39dd66f06f767323d : SUCCESS in 1h 06m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3bd38fa4042d4896ae26616b7799db70 : FAILURE in 1h 26m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51596013efc74aa5bad56a7bb1f527cf : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7ae28dcabc074a7e9c793ab0820dc9fc : SUCCESS in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fbd60bedfe046849b47a3a4d2dda3fc : SUCCESS in 2h 47m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8363896760ad4ac88e72015a256e0307 : SUCCESS in 2h 19m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9d23751fb1e4670910589d93ac348c5 : SUCCESS in 2h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8df095b56b29442eb938ae53b6a8db8f : SUCCESS in 1h 51m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc6c41c3013456ba9b36ef3e1554f4b : SUCCESS in 2h 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"2a4a1462c8e57a69517e328ca615a3f1f7f50495","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 08:15:31.000000000","message":"Patch Set 57: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":57},{"id":"f5d6b00125078f70e118000ff92182a350d60098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:16:12.000000000","message":"Patch Set 57: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":57},{"id":"ce8c205ed8f2cb17819b884880a9c01d575e7d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:47:15.000000000","message":"Patch Set 57: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c36e8b12a9e4220920e0ae0f6e9d9d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790480998daf4725a56baf2947e8ed4d : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa91724570394444b2039059541f0461 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd32929fd5fe4b30a44c0345f470c2ce : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e41650bafa4b508be80c0234510b9d : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d38d981dcda34bce92869beb144dc4fc : SUCCESS in 5m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81320c8b13104fcb9a81c8f2d5f8a199 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1d4f8aad3314b46a5e417ab80823087 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf1a511cd3b14275a488ae417d1da4f1 : SUCCESS in 1h 21m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77f272025cb34763851e3d773801fa2c : SUCCESS in 2h 29m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7031be1629fa4b3eb20d13a0ca05d905 : SUCCESS in 45m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1ee8d43dd40454baf45f2efec2eb236 : SUCCESS in 1h 10m 07s","accounts_in_message":[],"_revision_number":57},{"id":"00e5050282f43c5d40f7fa72203856603279fd02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:47:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":57},{"id":"396ee647f67a30d79d5eabbff6724372499773e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:48:31.000000000","message":"Patch Set 57:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29328f0948a8490eb752a1380236ca46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a431606409fa45ecae481e04c15303a0 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa94981a8f324941be055490190f260a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":57},{"id":"160366deff5f4b30d7ce012df4ccc5e565242aa1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-09 14:59:23.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57}],"current_revision_number":57,"current_revision":"0b1b6c58f55cbe02bb9473c2a4e961d3f658f489","revisions":{"39f35fd0794130a3192525e26a3025efeaf8c71f":{"kind":"REWORK","_number":1,"created":"2022-07-08 16:02:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/1"}}},"commit":{"parents":[{"commit":"bb4f26a833cf3186e77a9f787c73d6fd46d6c537","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f26a833cf3186e77a9f787c73d6fd46d6c537"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39f35fd0794130a3192525e26a3025efeaf8c71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39f35fd0794130a3192525e26a3025efeaf8c71f"}]},"branch":"refs/heads/master"},"01d113253f22e0d30ca75bf36e73c349d2ffe75c":{"kind":"REWORK","_number":2,"created":"2022-07-11 10:47:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/2"}}},"commit":{"parents":[{"commit":"bb4f26a833cf3186e77a9f787c73d6fd46d6c537","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb4f26a833cf3186e77a9f787c73d6fd46d6c537"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-11 10:42:39.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01d113253f22e0d30ca75bf36e73c349d2ffe75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01d113253f22e0d30ca75bf36e73c349d2ffe75c"}]},"branch":"refs/heads/master"},"a98f3716dbf0221c5caba8d0eba71cfd0ba35d09":{"kind":"REWORK","_number":3,"created":"2022-07-12 07:31:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/3"}}},"commit":{"parents":[{"commit":"aa434f4d6cb1eac12751825cb4a94ffe9f277284","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa434f4d6cb1eac12751825cb4a94ffe9f277284"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-11 16:21:54.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a98f3716dbf0221c5caba8d0eba71cfd0ba35d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a98f3716dbf0221c5caba8d0eba71cfd0ba35d09"}]},"branch":"refs/heads/master"},"acea4264c1a0187592bc8f117ab5c9b45f5d345b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-19 16:45:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/4"}}},"commit":{"parents":[{"commit":"fa21afc5e103c7a19a761e2b6bea9861a5d98704","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa21afc5e103c7a19a761e2b6bea9861a5d98704"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-19 16:44:19.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acea4264c1a0187592bc8f117ab5c9b45f5d345b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acea4264c1a0187592bc8f117ab5c9b45f5d345b"}]},"branch":"refs/heads/master"},"25e26f4fdb9a4380b4f7847f5f6e0a1eb23254b9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-20 14:47:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/5"}}},"commit":{"parents":[{"commit":"20780fcb3050d6ed19e2f571300d61d1ed5845cd","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20780fcb3050d6ed19e2f571300d61d1ed5845cd"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-20 14:45:43.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25e26f4fdb9a4380b4f7847f5f6e0a1eb23254b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25e26f4fdb9a4380b4f7847f5f6e0a1eb23254b9"}]},"branch":"refs/heads/master"},"4a267ba6ca23f97c727434dd31429695f79472e0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-21 09:25:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/6"}}},"commit":{"parents":[{"commit":"2bd3c555b14c5b93443fd3ccdcd62a8c12cbd975","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bd3c555b14c5b93443fd3ccdcd62a8c12cbd975"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 09:25:00.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a267ba6ca23f97c727434dd31429695f79472e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a267ba6ca23f97c727434dd31429695f79472e0"}]},"branch":"refs/heads/master"},"a312581ad6f276204ef6bfb30c695358f9be64f8":{"kind":"REWORK","_number":7,"created":"2022-07-21 12:09:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/7"}}},"commit":{"parents":[{"commit":"705feaec74278ca267f27e4357c07e2eef61c8e1","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/705feaec74278ca267f27e4357c07e2eef61c8e1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 12:08:17.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a312581ad6f276204ef6bfb30c695358f9be64f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a312581ad6f276204ef6bfb30c695358f9be64f8"}]},"branch":"refs/heads/master"},"106429fb8360b27a3fe76f1bdc13ddc81ae000ec":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-22 06:24:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/8"}}},"commit":{"parents":[{"commit":"77974b6e79876819455fbac3075339c0001ab050","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77974b6e79876819455fbac3075339c0001ab050"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-21 14:57:55.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/106429fb8360b27a3fe76f1bdc13ddc81ae000ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/106429fb8360b27a3fe76f1bdc13ddc81ae000ec"}]},"branch":"refs/heads/master"},"782762f9dc8487e8d0ea019d2782524b48a72140":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-22 06:47:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/9"}}},"commit":{"parents":[{"commit":"14ac901b92666311ecb654589b68109296811838","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14ac901b92666311ecb654589b68109296811838"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-22 06:47:43.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/782762f9dc8487e8d0ea019d2782524b48a72140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/782762f9dc8487e8d0ea019d2782524b48a72140"}]},"branch":"refs/heads/master"},"0580d695e9b93e90aba463caaff71caa6d659a43":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":10,"created":"2022-07-22 10:31:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/10"}}},"commit":{"parents":[{"commit":"8774a2feebccad9e5ca03ef8397544e8c6dd5268","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8774a2feebccad9e5ca03ef8397544e8c6dd5268"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-22 10:30:48.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nTODO: Once merged, remove \"cpu pinning\" from the roadmap\nin the wiki\nhttps://wiki.openstack.org/wiki/Octavia/Roadmap#Considerations_for_Octavia_8.0.2B\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0580d695e9b93e90aba463caaff71caa6d659a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0580d695e9b93e90aba463caaff71caa6d659a43"}]},"branch":"refs/heads/master"},"913727cd056f0eb137c0797a611793fa54d4ed8d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-07-25 10:35:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/11"}}},"commit":{"parents":[{"commit":"ca5355809164586d173901a11f99d031d4baef00","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ca5355809164586d173901a11f99d031d4baef00"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 10:35:22.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nTODO: Once merged, remove \"cpu pinning\" from the roadmap\nin the wiki\nhttps://wiki.openstack.org/wiki/Octavia/Roadmap#Considerations_for_Octavia_8.0.2B\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/913727cd056f0eb137c0797a611793fa54d4ed8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/913727cd056f0eb137c0797a611793fa54d4ed8d"}]},"branch":"refs/heads/master"},"7d2fc167991350eb2f87359a355081a9c799edb5":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-08-01 16:08:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/12"}}},"commit":{"parents":[{"commit":"89b417224fe264851cebb5bf9e658cb2efbdb370","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89b417224fe264851cebb5bf9e658cb2efbdb370"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-01 16:08:42.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nTODO: Once merged, remove \"cpu pinning\" from the roadmap\nin the wiki\nhttps://wiki.openstack.org/wiki/Octavia/Roadmap#Considerations_for_Octavia_8.0.2B\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d2fc167991350eb2f87359a355081a9c799edb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d2fc167991350eb2f87359a355081a9c799edb5"}]},"branch":"refs/heads/master"},"807908d79ef1998de1312675ffdff267f221c8e6":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-08-08 08:07:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/13"}}},"commit":{"parents":[{"commit":"00c0fcdc8e9f3e084e5095917fc791ed73afb1d5","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00c0fcdc8e9f3e084e5095917fc791ed73afb1d5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 08:06:39.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nTODO: Once merged, remove \"cpu pinning\" from the roadmap\nin the wiki\nhttps://wiki.openstack.org/wiki/Octavia/Roadmap#Considerations_for_Octavia_8.0.2B\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/807908d79ef1998de1312675ffdff267f221c8e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/807908d79ef1998de1312675ffdff267f221c8e6"}]},"branch":"refs/heads/master"},"c3c102ad46a1924019f207898c92bd3e003de325":{"kind":"REWORK","_number":14,"created":"2022-08-08 10:58:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/14"}}},"commit":{"parents":[{"commit":"00c0fcdc8e9f3e084e5095917fc791ed73afb1d5","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00c0fcdc8e9f3e084e5095917fc791ed73afb1d5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 10:51:05.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\n* Set cpu_map setting so that HAProxy pins each of its worker threads to\n  one specific CPU (except CPU0, which remains reserved for other tasks)\n* If a HTTPS termination listener exists, it sets the tune.ssl.cachesize\n  setting to 500000 which allows for more resumed SSL sessions and\n  therefore less computationally expensive SSL handshakes.\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3c102ad46a1924019f207898c92bd3e003de325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3c102ad46a1924019f207898c92bd3e003de325"}]},"branch":"refs/heads/master"},"cd61a811c3dc4bfcc12827f962d90ab9b3a39966":{"kind":"REWORK","_number":15,"created":"2022-08-08 11:28:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/15"}}},"commit":{"parents":[{"commit":"d278598ad9f9429d447e6b1c59e0febcec76ca1a","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d278598ad9f9429d447e6b1c59e0febcec76ca1a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 11:28:21.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\n* Set cpu_map setting so that HAProxy pins each of its worker threads to\n  one specific CPU (except CPU0, which remains reserved for other tasks)\n* If a HTTPS termination listener exists, it sets the tune.ssl.cachesize\n  setting to 500000 which allows for more resumed SSL sessions and\n  therefore less computationally expensive SSL handshakes.\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd61a811c3dc4bfcc12827f962d90ab9b3a39966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd61a811c3dc4bfcc12827f962d90ab9b3a39966"}]},"branch":"refs/heads/master"},"0dc4c571e5cf8e5462bf78e7917e753e7b444411":{"kind":"REWORK","_number":16,"created":"2022-08-08 11:30:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/16"}}},"commit":{"parents":[{"commit":"d278598ad9f9429d447e6b1c59e0febcec76ca1a","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d278598ad9f9429d447e6b1c59e0febcec76ca1a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 11:30:17.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\n* Set cpu_map setting so that HAProxy pins each of its worker threads to\n  one specific CPU (except CPU0, which remains reserved for other tasks)\n* If a HTTPS termination listener exists, it sets the tune.ssl.cachesize\n  setting to 500000 which allows for more resumed SSL sessions and\n  therefore less computationally expensive SSL handshakes.\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dc4c571e5cf8e5462bf78e7917e753e7b444411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dc4c571e5cf8e5462bf78e7917e753e7b444411"}]},"branch":"refs/heads/master"},"5e96a7ecb1e95068580962d841689a86d029b3f0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-08-08 13:43:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/17"}}},"commit":{"parents":[{"commit":"326b7f19b39d8b16f85cbbdd17d088812cfd0972","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/326b7f19b39d8b16f85cbbdd17d088812cfd0972"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-08 13:43:39.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\n* Set cpu_map setting so that HAProxy pins each of its worker threads to\n  one specific CPU (except CPU0, which remains reserved for other tasks)\n* If a HTTPS termination listener exists, it sets the tune.ssl.cachesize\n  setting to 500000 which allows for more resumed SSL sessions and\n  therefore less computationally expensive SSL handshakes.\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e96a7ecb1e95068580962d841689a86d029b3f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e96a7ecb1e95068580962d841689a86d029b3f0"}]},"branch":"refs/heads/master"},"5a4abb615fa9ead5dc7bfbcfc793a31612f50581":{"kind":"REWORK","_number":18,"created":"2022-08-18 09:10:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/18"}}},"commit":{"parents":[{"commit":"326b7f19b39d8b16f85cbbdd17d088812cfd0972","subject":"Add element for TuneD and Tuna","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/326b7f19b39d8b16f85cbbdd17d088812cfd0972"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 09:09:07.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4abb615fa9ead5dc7bfbcfc793a31612f50581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4abb615fa9ead5dc7bfbcfc793a31612f50581"}]},"branch":"refs/heads/master"},"726baf2af76fa8467e4bc6065a96d7236e7caecf":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-08-19 09:44:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/19"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:43:47.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/726baf2af76fa8467e4bc6065a96d7236e7caecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/726baf2af76fa8467e4bc6065a96d7236e7caecf"}]},"branch":"refs/heads/master"},"e35009ebd2fabd483b99e6ee10a78a2a148a533a":{"kind":"REWORK","_number":20,"created":"2022-08-19 10:37:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/20"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:34:09.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e35009ebd2fabd483b99e6ee10a78a2a148a533a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e35009ebd2fabd483b99e6ee10a78a2a148a533a"}]},"branch":"refs/heads/master"},"b60c47a6110a03baa594116f5a36de439ad8d2e1":{"kind":"REWORK","_number":21,"created":"2022-08-19 10:47:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/21"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:47:40.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60c47a6110a03baa594116f5a36de439ad8d2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60c47a6110a03baa594116f5a36de439ad8d2e1"}]},"branch":"refs/heads/master"},"3518c7c2af9695514148d093129d98c6e0a71782":{"kind":"REWORK","_number":22,"created":"2022-08-21 15:00:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/22"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:00:45.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3518c7c2af9695514148d093129d98c6e0a71782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3518c7c2af9695514148d093129d98c6e0a71782"}]},"branch":"refs/heads/master"},"e2167dcfd405b2aef95a5dfbb5007fe170caacfe":{"kind":"REWORK","_number":23,"created":"2022-08-22 13:52:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/23"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 13:52:45.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2167dcfd405b2aef95a5dfbb5007fe170caacfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2167dcfd405b2aef95a5dfbb5007fe170caacfe"}]},"branch":"refs/heads/master"},"bc1579954f32e9bcb43c6ebfc6aa0d566879b7b6":{"kind":"REWORK","_number":24,"created":"2022-08-23 10:07:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/24"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 10:06:46.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc1579954f32e9bcb43c6ebfc6aa0d566879b7b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc1579954f32e9bcb43c6ebfc6aa0d566879b7b6"}]},"branch":"refs/heads/master"},"9ebba86664248b02826d2175fc8494ff5c1b0a7f":{"kind":"REWORK","_number":25,"created":"2022-08-23 12:45:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/25"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 12:45:50.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks)\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ebba86664248b02826d2175fc8494ff5c1b0a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ebba86664248b02826d2175fc8494ff5c1b0a7f"}]},"branch":"refs/heads/master"},"2a33dfb7ff006a75c11176d4211e030d430a0fe1":{"kind":"REWORK","_number":26,"created":"2022-08-30 11:21:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/26"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:16:34.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a33dfb7ff006a75c11176d4211e030d430a0fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a33dfb7ff006a75c11176d4211e030d430a0fe1"}]},"branch":"refs/heads/master"},"bf980acc70760542cd42270272392fabdbe4f3b3":{"kind":"REWORK","_number":27,"created":"2022-08-30 11:35:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/27"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:35:00.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf980acc70760542cd42270272392fabdbe4f3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf980acc70760542cd42270272392fabdbe4f3b3"}]},"branch":"refs/heads/master"},"920efa82345f24ac2261985e53bdbf7a451c9314":{"kind":"REWORK","_number":28,"created":"2022-08-30 14:22:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/28"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:22:20.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/920efa82345f24ac2261985e53bdbf7a451c9314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/920efa82345f24ac2261985e53bdbf7a451c9314"}]},"branch":"refs/heads/master"},"e6c8f4bc9e6e4927bc4544690737e8d1f49e92bc":{"kind":"REWORK","_number":29,"created":"2022-08-30 14:53:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/29"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:29:31.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6c8f4bc9e6e4927bc4544690737e8d1f49e92bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6c8f4bc9e6e4927bc4544690737e8d1f49e92bc"}]},"branch":"refs/heads/master"},"796c0458889a212a34b9fdf6ae89e8813507cac7":{"kind":"REWORK","_number":30,"created":"2022-08-31 07:29:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/30"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 07:29:43.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/796c0458889a212a34b9fdf6ae89e8813507cac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/796c0458889a212a34b9fdf6ae89e8813507cac7"}]},"branch":"refs/heads/master"},"b723feab3a49e0e114b2fca093995bcdf6c474aa":{"kind":"REWORK","_number":31,"created":"2022-08-31 09:09:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/31"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 09:09:40.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b723feab3a49e0e114b2fca093995bcdf6c474aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b723feab3a49e0e114b2fca093995bcdf6c474aa"}]},"branch":"refs/heads/master"},"ef2d9b17a3c254ffdd29f8f7764ec6d88f846cf2":{"kind":"REWORK","_number":32,"created":"2022-09-01 08:56:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/32"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-01 08:55:12.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef2d9b17a3c254ffdd29f8f7764ec6d88f846cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef2d9b17a3c254ffdd29f8f7764ec6d88f846cf2"}]},"branch":"refs/heads/master"},"72d452221fcc92d0e5bdb542a138c327c9e4426d":{"kind":"REWORK","_number":33,"created":"2022-09-12 09:57:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/33"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 09:53:30.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72d452221fcc92d0e5bdb542a138c327c9e4426d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72d452221fcc92d0e5bdb542a138c327c9e4426d"}]},"branch":"refs/heads/master"},"8f803f39dd74efb9b387c015c24f39e79e3a4c3f":{"kind":"REWORK","_number":34,"created":"2022-09-12 10:03:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/34"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:01:42.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f803f39dd74efb9b387c015c24f39e79e3a4c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f803f39dd74efb9b387c015c24f39e79e3a4c3f"}]},"branch":"refs/heads/master"},"1a23670f15249b75699f58edbbdfd21d0330544a":{"kind":"REWORK","_number":35,"created":"2022-09-12 10:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/35"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:07:58.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a23670f15249b75699f58edbbdfd21d0330544a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a23670f15249b75699f58edbbdfd21d0330544a"}]},"branch":"refs/heads/master"},"3752dbc0b0b5c6f3c8e817db0cadc89e97698ca7":{"kind":"REWORK","_number":36,"created":"2022-09-12 12:28:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/36"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 12:27:45.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3752dbc0b0b5c6f3c8e817db0cadc89e97698ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3752dbc0b0b5c6f3c8e817db0cadc89e97698ca7"}]},"branch":"refs/heads/master"},"fd3fa2a745486c02800aa51acb2ce7ec4c8d0bc1":{"kind":"REWORK","_number":37,"created":"2022-10-19 09:09:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/37"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:09:17.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd3fa2a745486c02800aa51acb2ce7ec4c8d0bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd3fa2a745486c02800aa51acb2ce7ec4c8d0bc1"}]},"branch":"refs/heads/master"},"0929dc83680eb6d587ff4cd5addc15268dd760ed":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2022-10-19 09:11:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/38"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:10:48.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0929dc83680eb6d587ff4cd5addc15268dd760ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0929dc83680eb6d587ff4cd5addc15268dd760ed"}]},"branch":"refs/heads/master"},"431a94fffffce58f78d44cf87c2def79df132cac":{"kind":"REWORK","_number":39,"created":"2022-10-27 11:30:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/39"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 11:23:20.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431a94fffffce58f78d44cf87c2def79df132cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431a94fffffce58f78d44cf87c2def79df132cac"}]},"branch":"refs/heads/master"},"3c3ac18e3cb4d66ba0ee6b415dd91eace316641c":{"kind":"REWORK","_number":40,"created":"2022-10-27 12:22:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/40"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 12:22:29.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c3ac18e3cb4d66ba0ee6b415dd91eace316641c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c3ac18e3cb4d66ba0ee6b415dd91eace316641c"}]},"branch":"refs/heads/master"},"0666c752cf132524b8585d5a84ce67fe86e26498":{"kind":"REWORK","_number":41,"created":"2022-10-27 13:04:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/41"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:03:44.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0666c752cf132524b8585d5a84ce67fe86e26498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0666c752cf132524b8585d5a84ce67fe86e26498"}]},"branch":"refs/heads/master"},"4a548aba8bd112624867369cbe5208fac4982d62":{"kind":"REWORK","_number":42,"created":"2022-10-27 13:08:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/42"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:08:15.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a548aba8bd112624867369cbe5208fac4982d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a548aba8bd112624867369cbe5208fac4982d62"}]},"branch":"refs/heads/master"},"e82f217a580d70aaec08a4ea8d6f4659b8679e93":{"kind":"NO_CODE_CHANGE","_number":43,"created":"2022-10-27 13:15:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/43"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:15:46.000000000","tz":0},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling optimization\nenabled (the new  default).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e82f217a580d70aaec08a4ea8d6f4659b8679e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e82f217a580d70aaec08a4ea8d6f4659b8679e93"}]},"branch":"refs/heads/master","description":"Edit commit message"},"db1b365a0356130c7439279776defcdf816e8409":{"kind":"REWORK","_number":44,"created":"2022-10-27 14:41:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/44"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 14:41:24.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db1b365a0356130c7439279776defcdf816e8409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db1b365a0356130c7439279776defcdf816e8409"}]},"branch":"refs/heads/master"},"abbc75d04c6b56803942fcc84b985b205e19bbe5":{"kind":"REWORK","_number":45,"created":"2022-10-27 16:19:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/45"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 16:19:16.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abbc75d04c6b56803942fcc84b985b205e19bbe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abbc75d04c6b56803942fcc84b985b205e19bbe5"}]},"branch":"refs/heads/master"},"a85b78e5d78d1ebb01a94528edf98ee05c992cf6":{"kind":"REWORK","_number":46,"created":"2022-10-28 09:35:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/46"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-28 09:35:36.000000000","tz":120},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a85b78e5d78d1ebb01a94528edf98ee05c992cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a85b78e5d78d1ebb01a94528edf98ee05c992cf6"}]},"branch":"refs/heads/master"},"8ffefcbb7d0ecbe4ca1578cc75a9a334cb46b6a1":{"kind":"REWORK","_number":47,"created":"2022-11-14 16:24:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/47"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 16:24:02.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ffefcbb7d0ecbe4ca1578cc75a9a334cb46b6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ffefcbb7d0ecbe4ca1578cc75a9a334cb46b6a1"}]},"branch":"refs/heads/master"},"19c6774c1b488803659dbf41b875d8de7ff00e9a":{"kind":"REWORK","_number":48,"created":"2022-11-14 19:03:55.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/48"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 19:03:40.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nTo remain backwards compatibility with old HAProxy versions this\nfeature will be disabled with HAProxy 1.4 or lower. Also, in order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19c6774c1b488803659dbf41b875d8de7ff00e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19c6774c1b488803659dbf41b875d8de7ff00e9a"}]},"branch":"refs/heads/master"},"e2cfec242e1a136773f9710ee5d0754b23a0e103":{"kind":"NO_CODE_CHANGE","_number":49,"created":"2022-11-16 14:45:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/49"}}},"commit":{"parents":[{"commit":"a299f7eb7c8a8fa2759a910512abfe740e306a9b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a299f7eb7c8a8fa2759a910512abfe740e306a9b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-16 14:42:31.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2cfec242e1a136773f9710ee5d0754b23a0e103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2cfec242e1a136773f9710ee5d0754b23a0e103"}]},"branch":"refs/heads/master"},"9ae96ce28a86101cfd5a7dc1385108d6a7ae2877":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2022-11-23 15:29:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/50"}}},"commit":{"parents":[{"commit":"66cc89ab21f400f115052706e10cb886b91539c8","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66cc89ab21f400f115052706e10cb886b91539c8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-23 15:29:28.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae96ce28a86101cfd5a7dc1385108d6a7ae2877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae96ce28a86101cfd5a7dc1385108d6a7ae2877"}]},"branch":"refs/heads/master"},"73f873381fc94322b71c8c94674995702fe53742":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2022-11-28 11:00:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/51"}}},"commit":{"parents":[{"commit":"76743e238650b11eb54b146c429c7b925978b807","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76743e238650b11eb54b146c429c7b925978b807"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 10:59:34.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73f873381fc94322b71c8c94674995702fe53742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73f873381fc94322b71c8c94674995702fe53742"}]},"branch":"refs/heads/master"},"387256f1609b2ce222ba1d425af41058cb0943d8":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2022-11-28 15:46:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/52"}}},"commit":{"parents":[{"commit":"74b49e5aa7d139fce56c789a29fadf446ba68dcc","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74b49e5aa7d139fce56c789a29fadf446ba68dcc"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 15:46:47.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/387256f1609b2ce222ba1d425af41058cb0943d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/387256f1609b2ce222ba1d425af41058cb0943d8"}]},"branch":"refs/heads/master"},"3954e97e3899ec00b675cc4eb7750743e98ce454":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2022-12-07 12:18:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/53"}}},"commit":{"parents":[{"commit":"ce3796639c9bb3b82fa6894229ad705969f79d99","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3796639c9bb3b82fa6894229ad705969f79d99"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:18:01.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3954e97e3899ec00b675cc4eb7750743e98ce454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3954e97e3899ec00b675cc4eb7750743e98ce454"}]},"branch":"refs/heads/master"},"42740484d96de20235e7d3ba6e924e1f0ac17a5c":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2022-12-07 12:21:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/54"}}},"commit":{"parents":[{"commit":"97a2fbdeb764714b88a345c2892b84a4dd6ff263","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97a2fbdeb764714b88a345c2892b84a4dd6ff263"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:21:23.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42740484d96de20235e7d3ba6e924e1f0ac17a5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42740484d96de20235e7d3ba6e924e1f0ac17a5c"}]},"branch":"refs/heads/master"},"8d757c2e067bd54e1956136eacac19ae85ac36c7":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2022-12-08 09:27:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/55"}}},"commit":{"parents":[{"commit":"00b655c89e4878dfae34d0142205fbeb1aec0654","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00b655c89e4878dfae34d0142205fbeb1aec0654"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 09:27:17.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d757c2e067bd54e1956136eacac19ae85ac36c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d757c2e067bd54e1956136eacac19ae85ac36c7"}]},"branch":"refs/heads/master"},"401543236e64c0af3ee36967cfbd46cf318949de":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2022-12-08 10:54:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/56"}}},"commit":{"parents":[{"commit":"dd0964b4c0b5cc8a8d4d5fc1019076757da1070f","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd0964b4c0b5cc8a8d4d5fc1019076757da1070f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:51:49.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/401543236e64c0af3ee36967cfbd46cf318949de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/401543236e64c0af3ee36967cfbd46cf318949de"}]},"branch":"refs/heads/master"},"0b1b6c58f55cbe02bb9473c2a4e961d3f658f489":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2022-12-08 10:59:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/30/849130/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/849130/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/849130/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/849130/57"}}},"commit":{"parents":[{"commit":"c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1ebe18e2a4e144d9cf03568b43da91c9cbaa6f3"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-08 16:01:10.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:59:32.000000000","tz":60},"subject":"Amphora vertical scaling optimization in Octavia","message":"Amphora vertical scaling optimization in Octavia\n\nUses the amphora agent for optimizing HAProxy for vertical scaling:\n\nSet cpu_map setting so that HAProxy pins each of its worker threads to\none specific CPU (except CPU0, which remains reserved for other tasks).\nIn order not\nto change previous behavior this feature will only be enabled if the\namphora image was built with vCPU vertical scaling enabled (the new\ndefault).\nThis change assumes an amphora image with HAProxy version 1.5 or higher.\n\nStory: 2010236\nTask: 46043\nChange-Id: Ifbbe714c66117e57f96614534e6f20a9634c26eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1b6c58f55cbe02bb9473c2a4e961d3f658f489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1b6c58f55cbe02bb9473c2a4e961d3f658f489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
