)]}'
{"id":"openstack%2Foctavia~850241","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I5a749c751d184ec10957a31b569419d57fe8daf6","project":"openstack/octavia","branch":"stable/train","topic":"fix_healthmonitor","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-07-26 17:52:03.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-26 17:52:03.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-22 07:19:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I5a749c751d184ec10957a31b569419d57fe8daf6","subject":"Fix HealthMonitorToErrorOnRevertTask revert method","status":"MERGED","created":"2022-07-18 16:05:54.000000000","updated":"2022-07-26 17:53:16.000000000","submitted":"2022-07-26 17:52:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850241-fix_healthmonitor","meta_rev_id":"3cef7ccd3fc519d6628932268b168010a614c286","_number":850241,"virtual_id_number":850241,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-07-26 15:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-07-22 07:19:49.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-07-22 10:39:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-07-26 15:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 18:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 05:35:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-07-20 08:57:16.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-22 10:39:11.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:45.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"c6ec7b374d0c9bab3e63717c2338bab19a189f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-18 16:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51d48f874a3abd3719234611caf3fad360a98add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-18 16:09:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6263aca371970ef76ad923f2ab2bab291bed671c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:49:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b507036e0427440d978d5b8042059a5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ff316008177403cb6d194194b01b1ea : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76684163fc343c2bb29f332a67ea14b : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97a85f4ab0a141a0a70b3f50f58aedb1 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72a641c5d234970bc2baec5ac7abdea : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa30c5585e99460387256837a85d999b : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccd72f9583c4140aa2d8669cc2c3bea : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64b4c297b00e4567b6f574bdce83ba28 : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1490568efb0944a8afbf5949d5a85e30 : SUCCESS in 9m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d251d1eef2640989770b24b31f0f6e4 : SUCCESS in 8m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/58ad2a377be9412bafff404409977a63 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7fce6ad4bbaf42c5b4e1ca6343a34193 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93f09d9fafa1490e8048aaf3231036a4 : SUCCESS in 1h 21m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e89d19477a044fbf912799759d255828 : SUCCESS in 1h 51m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62a510b711f94dce916e414343b106d3 : SUCCESS in 1h 45m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c30d9c1783584669929ca4c5250f21c8 : SUCCESS in 2h 28m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b03b65a546304fe680355e0b8b7ee0a7 : FAILURE in 19m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce30b3bcab6642fea16920da96eedc67 : RETRY_LIMIT in 10m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/193ad7e3db5c44a2a9216936862e133a : RETRY_LIMIT in 15m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c51e124a9f8c4423aa0dbeb86efa1d3b : SUCCESS in 1h 04m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8a78f315df341bd9cef5b49d8886d07 : SUCCESS in 1h 40m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/727bec09c3944879a6370718c3a89290 : FAILURE in 1h 02m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87ca7f73c9d14620a077be79f17afbe6 : FAILURE in 42m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c42a3a2967134c80990d804dfbb6d72f : FAILURE in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f03d325f916377dfef223216ffb03c400bfac90","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-19 05:35:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f17d62c899afe323546cdaafe640b40acaabca7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-19 15:15:14.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35fb076eb474d688b0c7eed4724e0d04b5344f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-19 15:42:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"37e1bd8c8f94a6444fcf1dd877780bc2c5cef976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:27:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae52c8e606bc4fd783acfacb0718efd0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c52e6e00ff97422996bc14a878aace6c : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6f73a9393443378ba37e5036425caf : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01a5bf5c18444743a3796d203a6b3f3f : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dba1550a01e465684af877ac86dedde : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc5d956fe3e4ad19faa04e760388cb3 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1dccd5a92a4c82be13f6f298a7bed6 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80eefa93c5a34f19a37ca43840263155 : SUCCESS in 6m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7267f8d88a3c4296a70bc209f5743890 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b851790c9bd4494d822b872d36745561 : SUCCESS in 8m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3de9d85352cb4ef78c8b27578684d90e : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc49d641374f4d38811177bfa1271d36 : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96a978def830421091c54a88b224bfc7 : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/de8cb87038ed4637a5dfba769763c0dc : SUCCESS in 1h 14m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e10a52c3d8744b09155f3fd08fc3785 : SUCCESS in 1h 32m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1066aac86570487a8adc1af207260635 : SUCCESS in 1h 42m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dac5046b96bd4ec492f4e7067c0b1045 : FAILURE in 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e313523edb9a426da239e73f09c121b8 : RETRY_LIMIT in 10m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b84956e0aaee42b2836e43e83067b06d : RETRY_LIMIT in 14m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52e0250007c844feaf8b05a310b9b0c1 : SUCCESS in 49m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19875449460546718b0bcba6911e7137 : SUCCESS in 1h 24m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8863f6af1b734d468f383a4187b587f9 : FAILURE in 39m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/417b6b44632c4aa98d8872d76716a42c : FAILURE in 39m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/764699a2bedf436fbbff5cc042ffa342 : FAILURE in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1caa527d0d42e82bcbaf0aca5d22dfeada11ebb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 09:16:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16e7d4436a43fa85200c0d523d7c895f7c7f79ac","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-21 10:05:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"411fd70037f83d934d0b222ce64e59ea4bac304f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-21 10:05:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c4199b96b0abf5d5c50e7d1fdbd1b636ce2608e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:18:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/216dc6f57d6a4da7bbec8c273b806e6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e96445c440e54ca5be0c18e81440a3b3 : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ced942a26964ec8abfd722f859f169e : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bab594c692824197909c5bd4868b63a6 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc671b9234334d559b1818eba2940aa8 : FAILURE in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7591a75aa6e54d58a956f97752664f7d : FAILURE in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2842b0ae05f64c69b41274b4baffb908 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86929fccf19544b89361798979fa2d22 : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64a39d29048e458fbc20ccf7ddb97935 : FAILURE in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e7e0921895334ac691a1285e8ca83209 : SUCCESS in 7m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c7227eca98fc43ffbba46975b0b0f3fc : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be213143280347dab4341c66fb658afb : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/049fbbf87a4645b182526f6da92b076a : SUCCESS in 1h 31m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2fe5faa1a41c4e5baa45c615cae6c9f5 : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90f627b9b0b74dbaa78740d4caf21b4b : SUCCESS in 2h 05m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ea67c797369e44fcaf72fa1d1942d237 : SUCCESS in 2h 11m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d65d5c2da3474c2292806954527b01e8 : FAILURE in 19m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2764260c323a4314bb8ba61a85a685c2 : RETRY_LIMIT in 10m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5160e28d038e4eab86637d13ff6304ea : RETRY_LIMIT in 10m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ab5e9736254423f94e246bb667fe5c1 : SUCCESS in 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8c0b1fc1cb744740aedc5cea972bf2ab : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3d391d00c2a4c0eba2e8ce0ee34a835 : FAILURE in 40m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/aaf9988b71ed422fbdc241601c8198ee : FAILURE in 40m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/03ce8f84069e432ca83299685c2cab0d : FAILURE in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"185b0467d523ce604fdf2459f0711919b4d6e358","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-21 12:29:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9fcfd18cba52985c589a8f4d510f45b2bccf03c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-21 12:31:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ab2d998a7d1a45a229363420d272253a56e65779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:49:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54605aadd7dc437c820fe045b1763d93\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44b11235f05d40efa757403916ef2237 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8122b596254172a265210fbe70b4d9 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beca488cafe6407a993ed7eebdcf8212 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce3271a57c4c4f25b160e010be297c9f : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5c3685c0be9495ab0653608cf4d5ae5 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587b3cdefd1f4c328a4d98c0f8b6dc3f : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6e6c0d018643179ce22f715da1e026 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8d0fb5f5653241e1805067e787541522 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/401e21801cbb42aa97bb3eec8ba8acdc : SUCCESS in 8m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/02d6002d86df48a89edc35152301ca44 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b4ac52ad5cd466aa79d516adf915143 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4c56d01f80d47858ed89e157cfd8843 : SUCCESS in 1h 24m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cf168914d5dc40a28c9406e034d2e58c : SUCCESS in 1h 28m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35989fc8eaaf44a0a5e020e602379e34 : SUCCESS in 2h 06m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a5d1857d44c74da7b3ad49cf5eead91c : SUCCESS in 1h 44m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/44ac66dc60bb44e4912eb698bd3f020f : FAILURE in 28m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d78657157ec147b18f9c7cc10ef98258 : RETRY_LIMIT in 15m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c3b7d8421054984a219d5875e83fb2f : RETRY_LIMIT in 9m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3212bc2555f34eb48bf78b4cb21e1ca4 : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5417ce0d637647569b9b75c82ff2f803 : SUCCESS in 1h 55m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2e83183b0bf4f30a645bbca1241ec13 : FAILURE in 34m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7f0ee0afd65d448ba6fa0f7d8733b2bf : FAILURE in 37m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2111e32eaa154a8e80a225215eca0e53 : FAILURE in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"84c0f0abc5d8571ca9e103eb790bc76c2be1a886","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 07:19:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"59d497f076d7ebfc632a604716bef8864274c468","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-07-22 10:39:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f94d29e660bba4ecc039591f63a1a1c894be805e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:30:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"29092a2198708a6a92b994093206711532828419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:31:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f1badbf5a0a8708da7404a8afb788e20d8753840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:52:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/059dbbc370e049cf8800a0f32502d402\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3b8a8e69504cdcb2bd34aa4918e8e9 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cbbaca01214342b8bde63f77f32cea31 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a77bf79f734c8f9fba9dd46644c7e1 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70d09d468e314490b111661a33c07ce4 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b921b2cd0bb14712bfa377189f602407 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/328e34e4e3554a3ab8aa6ea22348bd33 : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/72d5eb4fd7bc4812a261feb388483d51 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b6e5f06c87144c59728f5e80b538ee6 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/886b7c5cd1f148c791ee3ad30048f17e : SUCCESS in 1h 15m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1021ef96dfb043c9adfce8ec54c32d8a : SUCCESS in 1h 24m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef1eb06eff3b49ffad150d68d5881979 : SUCCESS in 1h 46m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ea3d40f2d05e42c2217eae27680ff8b1ee9b0f07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:52:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3cef7ccd3fc519d6628932268b168010a614c286","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:53:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a6faf01b4d4eeaabe9a9c354a636a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1cc39e5af374acd9ba3161b86b44f0b : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e2d965d9ed49af8fc945c01b90424d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6e6f47d3e10820f813d95fc7f26357746cc66471","revisions":{"0b6f497d91f455cab5d974344b53864a56d05311":{"kind":"REWORK","_number":1,"created":"2022-07-18 16:05:54.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/41/850241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/850241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/850241/1"}}},"commit":{"parents":[{"commit":"f18a230e93f155816be3089068fc537c8aa7a19a","subject":"Merge \"Fix duplicate object error messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f18a230e93f155816be3089068fc537c8aa7a19a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-18 16:05:53.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b6f497d91f455cab5d974344b53864a56d05311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b6f497d91f455cab5d974344b53864a56d05311"}]},"branch":"refs/heads/stable/train"},"7e8a872787d5207a14edbc4385a4d1c0af512b7e":{"kind":"REWORK","_number":2,"created":"2022-07-18 16:09:51.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/41/850241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/850241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/850241/2"}}},"commit":{"parents":[{"commit":"f18a230e93f155816be3089068fc537c8aa7a19a","subject":"Merge \"Fix duplicate object error messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f18a230e93f155816be3089068fc537c8aa7a19a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-18 16:09:42.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e8a872787d5207a14edbc4385a4d1c0af512b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e8a872787d5207a14edbc4385a4d1c0af512b7e"}]},"branch":"refs/heads/stable/train"},"370b15456efa04fda06e32131c25ba9019ddb324":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-19 15:42:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/41/850241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/850241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/850241/3"}}},"commit":{"parents":[{"commit":"f18a230e93f155816be3089068fc537c8aa7a19a","subject":"Merge \"Fix duplicate object error messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f18a230e93f155816be3089068fc537c8aa7a19a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-19 15:42:03.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/lifecycle_tasks.py\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n(cherry picked from commit 96f2ca7cecfca43000ddba929fc1bc348c845b27)\n(cherry picked from commit 5ca3849a8f9d0aaeabece97f520d5656431594d6)\n(cherry picked from commit c7aa79cdfa498d11e601578df8daae73235a6fcd)\n(cherry picked from commit a25076a11aac871cd7d0e3bff538a220980325a8)\n(cherry picked from commit 6429b2793f9a1079cc63305f7d77c4da2a1c8513)\n(cherry picked from commit d16b00c842ef16f1b88d6f7b52b90c030182db3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370b15456efa04fda06e32131c25ba9019ddb324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370b15456efa04fda06e32131c25ba9019ddb324"}]},"branch":"refs/heads/stable/train"},"183c0c60a17f8ab0542fe867996c0b962fee7832":{"kind":"REWORK","_number":4,"created":"2022-07-21 10:05:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/41/850241/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/850241/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/850241/4"}}},"commit":{"parents":[{"commit":"f18a230e93f155816be3089068fc537c8aa7a19a","subject":"Merge \"Fix duplicate object error messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f18a230e93f155816be3089068fc537c8aa7a19a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-21 10:05:36.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/lifecycle_tasks.py\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n(cherry picked from commit 96f2ca7cecfca43000ddba929fc1bc348c845b27)\n(cherry picked from commit 5ca3849a8f9d0aaeabece97f520d5656431594d6)\n(cherry picked from commit c7aa79cdfa498d11e601578df8daae73235a6fcd)\n(cherry picked from commit a25076a11aac871cd7d0e3bff538a220980325a8)\n(cherry picked from commit 6429b2793f9a1079cc63305f7d77c4da2a1c8513)\n(cherry picked from commit d16b00c842ef16f1b88d6f7b52b90c030182db3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/183c0c60a17f8ab0542fe867996c0b962fee7832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/183c0c60a17f8ab0542fe867996c0b962fee7832"}]},"branch":"refs/heads/stable/train"},"6e6f47d3e10820f813d95fc7f26357746cc66471":{"kind":"REWORK","_number":5,"created":"2022-07-21 12:31:35.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/41/850241/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/850241/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/850241/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/850241/5"}}},"commit":{"parents":[{"commit":"f18a230e93f155816be3089068fc537c8aa7a19a","subject":"Merge \"Fix duplicate object error messages\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f18a230e93f155816be3089068fc537c8aa7a19a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-21 12:31:26.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/lifecycle_tasks.py\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n(cherry picked from commit 96f2ca7cecfca43000ddba929fc1bc348c845b27)\n(cherry picked from commit 5ca3849a8f9d0aaeabece97f520d5656431594d6)\n(cherry picked from commit c7aa79cdfa498d11e601578df8daae73235a6fcd)\n(cherry picked from commit a25076a11aac871cd7d0e3bff538a220980325a8)\n(cherry picked from commit 6429b2793f9a1079cc63305f7d77c4da2a1c8513)\n(cherry picked from commit d16b00c842ef16f1b88d6f7b52b90c030182db3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6f47d3e10820f813d95fc7f26357746cc66471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6f47d3e10820f813d95fc7f26357746cc66471"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
