)]}'
{"id":"openstack%2Foctavia~850367","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I5a749c751d184ec10957a31b569419d57fe8daf6","project":"openstack/octavia","branch":"stable/ussuri","topic":"fix_healthmonitor","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-07-26 17:52:10.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-26 17:52:10.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-21 09:01:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I5a749c751d184ec10957a31b569419d57fe8daf6","subject":"Fix HealthMonitorToErrorOnRevertTask revert method","status":"MERGED","created":"2022-07-19 15:35:13.000000000","updated":"2022-07-26 17:53:36.000000000","submitted":"2022-07-26 17:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850367-fix_healthmonitor","meta_rev_id":"cfd43c709ca25761f4297cf39ffaa22ff2ef885e","_number":850367,"virtual_id_number":850367,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-21 09:01:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 17:18:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 08:57:09.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-20 08:57:09.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-07-26 15:31:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"56b1bfe30ca0de382096df01c0b8fe166738f350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-07-19 15:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a33b6fb3645df0b370aaf61bc03b6e6d04be3ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 17:18:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776bf56d544746998d35433bcc86f57b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39f88d83b2a742fd87034c5cebf2cca6 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945f5f4be5304e038cac8da5da814f79 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9cc8e544dbc4d25a57ea7c4739fbdc3 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb9906d46254430a9c2f21b9088bf24d : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824c67699e34436eb91bf1ab3be8a93d : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/192ec2a358fb45078b41519dee2ded63 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/077ad66546ff4e039e36723bb0753813 : SUCCESS in 7m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a5eddb29d644bc180e5b0dc4e4a13cd : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fc5ef8f778844bea972f9eacc2201ef : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3668c3837c874d7ab60129451e2534ae : SUCCESS in 1h 15m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e4668b357e54058ac2a9be3fd2a27ea : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b64f3089950d4b7cab591c4a6218317c : SUCCESS in 41m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/77b8a75acdcf47db91a2da372d967c38 : SUCCESS in 39m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/230c479ac8654f2ca347d8cd56feaed1 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7bd65d9bc1b34be9bb1a8fe26ea1b09e : SUCCESS in 1h 35m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/75411abe78724c10aae7cab1cbf8a397 : RETRY_LIMIT in 14m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/631bcfe9a73b4821a3feb4b05add853f : SUCCESS in 1h 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56babfd89f24cab397c7ea2349ab1bf6dda1a20b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-21 09:01:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5c2efc63065604ec262cb4a76e047692f135763","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"491612785cd8d0d76f77f639d8e6d3f607476afe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddb4be593fc67cd32aead0b70ee98b8a844ba09a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:52:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/577f392c69334ef2883b91b45826c40e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae519e918ce49b3b2fed072baf6333f : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce07952957746439e350e493958667d : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b36b4540cdd4caabb4a6aab57ba01df : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f802420e9534640b2f72156f32ec7ec : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66f2eeeeed544593847b06e0c704432d : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e7a73ccfc0f49e3af974918cbd543e3 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19438917baf144a5a85db0deb24c9dbe : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0869643028cd481286cd5ed2e0e65b43 : SUCCESS in 1h 44m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d50e9f254a044818adc76559e2503d6b : SUCCESS in 49m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3f167784466044f8827ab1584e785791 : SUCCESS in 41m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a7f1a9e74564e8b9ae6ba773c352a64 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bbd1c6e83789be22d70e49f65b7fb32bf7605a16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:52:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cfd43c709ca25761f4297cf39ffaa22ff2ef885e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f0b2031d02464a98cbeaf94852f62c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1b618f2e494140a6bb9dc5086f426a : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f544ca340dc4a47af6c942bce5c3373 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d16b00c842ef16f1b88d6f7b52b90c030182db3b","revisions":{"d16b00c842ef16f1b88d6f7b52b90c030182db3b":{"kind":"REWORK","_number":1,"created":"2022-07-19 15:35:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/67/850367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/850367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/850367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/850367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/850367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/850367/1"}}},"commit":{"parents":[{"commit":"5838600e6f639ab520af01fea93370bc03bdf43b","subject":"Merge \"Fix duplicate object error messages\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5838600e6f639ab520af01fea93370bc03bdf43b"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-06-28 13:25:03.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2022-07-19 15:34:39.000000000","tz":120},"subject":"Fix HealthMonitorToErrorOnRevertTask revert method","message":"Fix HealthMonitorToErrorOnRevertTask revert method\n\nSo far, whenever we had to revert HealthMonitorToErrorOnRevertTask,\nwe called mark_health_mon_prov_status_error with the wrong parameter\n(health_mon[constants.POOL_ID] \u003d\u003d the pool id).\n\nThis would set the pool\u0027s provisioning status to ERROR.\n\nLater in the revert task we call mark_pool_prov_status_active with the\ncorrect argument (health_mon[constants.POOL_ID] \u003d\u003d the pool id), so we\nset it back to ACTIVE, and the task continues correctly.\n\nThis means we might have missed setting the health monitor\u0027s\nprovisioning status to ERROR, and therefore we might have missed some\nerrors.\n\nThis commit changes the parameter to be the correct one in both API\nv1 and v2, and it modifies the tests accordingly.\n\nStory 2010113\nTask 45690\n\nChange-Id: I5a749c751d184ec10957a31b569419d57fe8daf6\n(cherry picked from commit 96f2ca7cecfca43000ddba929fc1bc348c845b27)\n(cherry picked from commit 5ca3849a8f9d0aaeabece97f520d5656431594d6)\n(cherry picked from commit c7aa79cdfa498d11e601578df8daae73235a6fcd)\n(cherry picked from commit a25076a11aac871cd7d0e3bff538a220980325a8)\n(cherry picked from commit 6429b2793f9a1079cc63305f7d77c4da2a1c8513)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d16b00c842ef16f1b88d6f7b52b90c030182db3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d16b00c842ef16f1b88d6f7b52b90c030182db3b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
