)]}'
{"id":"openstack%2Foctavia~850481","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I1633c2cacf1c4dc5c0aa605635545fae8085e296","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2022-07-26 17:51:57.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-26 17:51:57.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-25 07:42:51.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1633c2cacf1c4dc5c0aa605635545fae8085e296","subject":"Fix update/delete listener CA/CRL error","status":"MERGED","created":"2022-07-20 13:20:23.000000000","updated":"2022-07-26 17:53:09.000000000","submitted":"2022-07-26 17:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850481","cherry_pick_of_change":845093,"cherry_pick_of_patch_set":7,"meta_rev_id":"8f172305691ed3086957cae860cbc1b5cab7b468","_number":850481,"virtual_id_number":850481,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-25 07:42:51.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 02:30:23.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-21 02:30:23.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:22.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"1d5eccfb90ccf1f9eafe408c9c494ea1ea8b3bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-20 13:20:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9870864f1efbebbc88c82e74cb19a976e0759707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:38:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0be28456994deaadfc26a61e1fdf54\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6178fedf7bde42b286de118cb72c55f8 : SUCCESS in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63e202bbe2e64e5cbead1a00adc295b3 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b144c15021642b0a64a76c8d303ab23 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7340fa76d934c179a866b584d1636a5 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071a83af964f4352bf449c7b2641326f : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0846b7d9da8e46eea84a7fd0eea4c77a : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29da2243cd52454893037dfc1d6e2294 : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/74165d50941c48f499199fbed40faecc : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ba93bdb417bd43a7a482dc70a64609e1 : SUCCESS in 11m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7cfb0fdbdbd34e319135cc658b5eab14 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cd00837baae4e72a917e162d5b329c1 : SUCCESS in 21m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f61c2efec30947a1b677d0fd0bae1120 : SUCCESS in 2h 13m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb4a90c21b4e4ca69439fff0d3e8b12f : SUCCESS in 1h 43m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51401ce9209440e29b35f045027703a3 : SUCCESS in 46m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ec28e3507074893af9e39047fb7e41c : SUCCESS in 58m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1289ca0f0dae431aa1b5688c357f18f5 : RETRY_LIMIT in 15m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f7fcb8f779342ce86ea7ec8d87efd44 : SUCCESS in 38m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95148ef095ad419599918afe514edfcb : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b661448adcb426fb2c39a0a1e2bca71 : SUCCESS in 2h 01m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a47f17d46c34439eaa69c7f2b8adf8be : SUCCESS in 1h 46m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e11b5440e5e4c0599a72e24afdd2c99 : SUCCESS in 1h 36m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b766b220de7247a9b34c656f3da2b4fb : SUCCESS in 1h 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/897bfc113dae42c3b22c1567a76cbb0d : FAILURE in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5789d1a9c20c4e453aee526916108948845c670","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 07:42:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92af4b1b56dfae2c98ebb833f58ea7e1fb705697","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:30:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb0e2071f1ad7aee3e62b3c3ebc3751ee139bd70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:31:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83be12e1951bb2afec0984f9131c4680e42ce93d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5253747a012348bcb58bc5cfac1ae844\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/330459bb62c34e81b8eaa4226a641f68 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce3aa25ba19470c9dc6a0f91b746951 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0f49a7531fb478daa02c54d66c5e061 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e666df4ff1564e4ca5bbb85156215b91 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9099abf26fa9428bad800a81aa23b210 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4638fcffdff94288b12c64d743dbefa1 : SUCCESS in 5m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8947fa3f87d4b7eb3651f86ab76fe61 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29433420438e4938b785b26f92950dbf : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0096ebf3fc9d44cd992970be8d175415 : SUCCESS in 2h 20m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8c4438329014e268b115984deb3ca54 : SUCCESS in 1h 57m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f24e94c6eca241d4bb728ef8b33329f3 : SUCCESS in 43m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/865633282a904efcb1d68c26a97a7c96 : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"889b0f7a2c7f6a96f9305107d1c3abd2c1d74ab6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:51:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f172305691ed3086957cae860cbc1b5cab7b468","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:53:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef4c3cc3cf2d49649c8c1a7a1004e3af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2badfd7577e34ed9af6656527394ad36 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d402fbb48f8b42848617056b2594be37 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74f2baf539359f5ded2bc01b28d570083bc0a797","revisions":{"74f2baf539359f5ded2bc01b28d570083bc0a797":{"kind":"REWORK","_number":1,"created":"2022-07-20 13:20:23.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/81/850481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/850481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/850481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/850481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/850481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/850481/1"}}},"commit":{"parents":[{"commit":"10f037c7e5cabf870957f71e894c4e78f26cdeae","subject":"Merge \"Set sensible nf_conntrack_max value in amphora\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10f037c7e5cabf870957f71e894c4e78f26cdeae"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-20 13:20:23.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f2baf539359f5ded2bc01b28d570083bc0a797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f2baf539359f5ded2bc01b28d570083bc0a797"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
