)]}'
{"id":"openstack%2Foctavia~850482","triplet_id":"openstack%2Foctavia~stable%2Fxena~I1633c2cacf1c4dc5c0aa605635545fae8085e296","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2022-07-26 17:48:54.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-26 17:48:54.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-25 09:56:10.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1633c2cacf1c4dc5c0aa605635545fae8085e296","subject":"Fix update/delete listener CA/CRL error","status":"MERGED","created":"2022-07-20 13:21:54.000000000","updated":"2022-07-26 17:50:03.000000000","submitted":"2022-07-26 17:48:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850482","cherry_pick_of_change":845093,"cherry_pick_of_patch_set":7,"meta_rev_id":"1a9b8d328fdc7a80c8188fa6232fead306841422","_number":850482,"virtual_id_number":850482,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-25 09:56:10.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 02:31:27.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-21 02:31:27.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"ae14e7831e28368735d339c5037aa7b18123b4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-20 13:21:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"92dc3b6d38b03633e21539775efd4ad8a0b3f092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c33c0c54a584451ad4f5186f9987597\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8d4cd55c954edfba0182116ccb6d1a : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dcb7e84e75548ee85209f2a6c8fd9d1 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20a4c736f1f4947ac695ccf243cb160 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc787049157948678d4d92abee29a817 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc88741b20947eda0deebfd098805df : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8df566ed1ffd4714b844eafdb98b08b6 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504d2e696db5452b847e9def63d41f42 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87bd51d7b251402385294054ddc4a300 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4d50bccf9bae45deb4a6b53928d16825 : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a762d6033e284bc8828cb75ab0d4c91b : SUCCESS in 10m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13e66003774345f0bb8370a9e74e5675 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60962ed4845f40bc97eb5d48ccaebe39 : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/717ed0e35f354b5aac5f9810739f3e7e : SUCCESS in 2h 08m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c85b2672b2b94c9d94f38efc44149228 : SUCCESS in 1h 25m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/406b0ff5409d487085421fc583a6c852 : SUCCESS in 47m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f736ffffe47d4bd480efd0112028c1d5 : SUCCESS in 1h 00m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/34a76dca606145038176820a5ce012e8 : RETRY_LIMIT in 25m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be07405c4326439f983fc4d2c0e597bf : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9ca1c2fa6256445a9630774fc76a5692 : SUCCESS in 1h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74f8bf0222b74eb3ad9710af58a281fc : SUCCESS in 1h 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/278d311003dc4c1781fcd45ee197af31 : SUCCESS in 1h 50m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aa493c49dadd46e589536424d3ba3dc2 : SUCCESS in 1h 39m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/be6da0ab8bd047d89b1bb032d29e9700 : SUCCESS in 1h 39m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/170519bb92024713b8a31f1939379546 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cfcdcb68f8898a7661520492ef2079f8345d715","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 07:55:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bb9e8746a0f5551197ff1e03839a6af25153141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-25 09:51:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5c8ba0978b1d7ba1db0531453f316313dfcd7757","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-25 09:54:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e447a872257f454f15d0223c7c3726361390dc6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 09:56:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11886f1b4c8d18c681b52d83184e562f1fe6932f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 11:58:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c8936c7584489d935e7292664a92f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6f50c443187497b870d4f3654ed7e27 : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4cdd9b0805c4fba9c0df31f68e29f17 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea00df88fa9a481ea5e7de5f8c8c72ee : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35586be6dccc471ca19d20908e4a64d2 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c93058b929649988a276dafcb426265 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f32464aa810405aab80a5f1d8e4fac7 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c371ebb84964a06ae105cc86177b622 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4bd1c609a5244659c666df309826816 : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/641bb66ebddf4288ac13e960d91e6951 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0d3c46f5dcfc4051af4f6ef5784902a3 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/224970f81a034037925fe180315632e3 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90a8980f94ce4c6089925bfe7fe67f3c : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2ad9a58ef3a41209260f9ca1157d32a : SUCCESS in 2h 01m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67be99fe74124bae904211936fb39660 : SUCCESS in 1h 47m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6aec23026054dd5954960d487734e55 : SUCCESS in 43m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0bfddc2d8cc43489c11e310a3d3624f : SUCCESS in 56m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/974c208b8c944f74a14ef40683170760 : RETRY_LIMIT in 24m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f310e8e87e444262a91c5aef12d0622f : SUCCESS in 37m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3d8ee23203a469c90879876f8d19d4d : SUCCESS in 1h 49m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3dfde96488a743c3a345b026df13598f : SUCCESS in 1h 57m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/837bfa4fec034174a354938afe48e76d : SUCCESS in 1h 32m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/069cf25d24504bd6a97cfe18bbcc7c45 : SUCCESS in 1h 26m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8e69b035aeb44362b9cffc58aef298ce : SUCCESS in 1h 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/050eeab0993341d587c846c54c2f2016 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dca6b0576afd069330bd38977c448cbcce2c955b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:30:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2869fd21a403f5915e7df12f7c04c3e7554ded1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:31:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee71c4b03bf8cfc2b0c7318a841082321327c183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:48:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59150a6c63ab40cf9912ab291f47dc1c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54dd66c8c65a463baed8e48b1188506d : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d9c2a90c01345bea4c47f037fdc2f1d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee8413800d44f969ced4d590161177e : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7960085fee2466ca11e9d88f3985520 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115a176c7c5645a8bfc7912f1580766b : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/983654ff0fe844ef893e1aa7b03f5ecc : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9892c4cddd7a484ebff74c92cf655fc0 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b43802101cf640689070f85adfe2986e : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/567f7924695c4ddf9d8f1a5d4c957ac0 : SUCCESS in 2h 16m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f76380198fa4eab819fce8d431665d4 : SUCCESS in 1h 46m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d872ae985da46fdbd50fb21f9ee7528 : SUCCESS in 41m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66b6a2f9d05b4776a63e7a1b34c6718f : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ab780006cce0ddce3248311d14aecbc4cc8165a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:48:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1a9b8d328fdc7a80c8188fa6232fead306841422","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a1f9b5af474608bddc8050b67a7070\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e274101e37af485087e081cc835ba269 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9181b9346e1f43468da4b0974e3dc55a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa1c69a58667fd53ede2f9e08deaed61545c0b32","revisions":{"9bb36ea2853afb8cd39af39a7383a44806ea34af":{"kind":"REWORK","_number":1,"created":"2022-07-20 13:21:54.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/82/850482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/850482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/850482/1"}}},"commit":{"parents":[{"commit":"3236435a2c2b6d9c2f3c7015ba473dea91ce91fa","subject":"Merge \"Set sensible nf_conntrack_max value in amphora\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3236435a2c2b6d9c2f3c7015ba473dea91ce91fa"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-20 13:21:54.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bb36ea2853afb8cd39af39a7383a44806ea34af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bb36ea2853afb8cd39af39a7383a44806ea34af"}]},"branch":"refs/heads/stable/xena"},"aa1c69a58667fd53ede2f9e08deaed61545c0b32":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-25 09:51:30.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/82/850482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/850482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/850482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/850482/2"}}},"commit":{"parents":[{"commit":"3236435a2c2b6d9c2f3c7015ba473dea91ce91fa","subject":"Merge \"Set sensible nf_conntrack_max value in amphora\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3236435a2c2b6d9c2f3c7015ba473dea91ce91fa"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-25 09:51:30.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n(cherry picked from commit 74f2baf539359f5ded2bc01b28d570083bc0a797)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa1c69a58667fd53ede2f9e08deaed61545c0b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa1c69a58667fd53ede2f9e08deaed61545c0b32"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
