)]}'
{"id":"openstack%2Foctavia~850484","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I1633c2cacf1c4dc5c0aa605635545fae8085e296","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2022-07-26 17:28:20.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-26 17:28:20.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-25 15:24:07.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1633c2cacf1c4dc5c0aa605635545fae8085e296","subject":"Fix update/delete listener CA/CRL error","status":"MERGED","created":"2022-07-20 13:23:47.000000000","updated":"2022-07-26 17:29:33.000000000","submitted":"2022-07-26 17:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850484","cherry_pick_of_change":845093,"cherry_pick_of_patch_set":7,"meta_rev_id":"f2a21ae0c33ae04b4e8bef1bfe062d6186f0df66","_number":850484,"virtual_id_number":850484,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-25 15:24:07.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:29:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 02:32:00.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-21 02:32:00.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-26 15:29:10.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"50e51cefdaec76fe7a3717990935a2b7f975f8bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-20 13:23:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3bb3ae7770cd4876cd302567ac27c9104a11225d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:52:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0390b907b74340a2c1d32e2dd1ce26\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d131df218384ff1b637fbd0fa6c4c01 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc32e33e40b40d9937efaa3dff72037 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77776f94377f470996e51ed34e189b39 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2247d6aca734fd28694b1a46d15039a : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f4f2d9e9c54b2b92f7b413185fc6e2 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b761a3e9a66d40d1a256b446f86a9e81 : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bfab04ba1cfd435dbaff2607fc46a12c : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8112fb5eb5d42bcb49ef2a54a869910 : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/baaa8d2c52c6493ea3da068651d6bc5c : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a945af3b80774bcea4ec22045887dcf3 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d27538e2ec144484ba7b5f047ef57e73 : SUCCESS in 1h 26m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4ea43925ba749d8bf398850d0eea7c7 : FAILURE in 2h 01m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29d9c0d894ce4064a1420e7691515ac4 : SUCCESS in 44m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/13c836161d9246c095736eaa693858f3 : SUCCESS in 37m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/357656f1a7e149ef913037484d388e19 : SUCCESS in 42m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6919909c175e419ab7a8a32fc4d64741 : SUCCESS in 1h 53m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43bae53654464d28b2b3aaa17857c1a2 : FAILURE in 35m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/251ee30799984230b963f36e58cca7d7 : FAILURE in 2h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a71d9e4c8b704442b6b92f34736d492e : SUCCESS in 2h 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cdca798fb0e6413a99f4852c6df3a290 : FAILURE in 1h 59m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eb711a7f94ab491a9fdf748369c9d410 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a05dfb40bf1b361e9466008fb3ef7308744a966f","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-21 02:37:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7b0ff6bee7b1c07ecebf5e062e12eceaab266f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8bbc19a3dd84151859699ce221ccc60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3c14f95a5bb44ae9ed5daf3bf350494 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84921614ed90434f896925ee9921ba06 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839da05e9c15417cbd5e0d3d3fdf4e47 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed30cda2d2ae45669df57d726e7bb51c : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a934a0742f74652b1adf5acb45faaee : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea1ed73c21cc4d42b1ea0ca3fbe3b1b2 : SUCCESS in 5m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/66bf215d13f64d718e42ae46486b6671 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c5367e89e6041b5b23c73fa7ae06058 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d34caaa93dcd41f7bf1da49044e2af6b : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7143fca6027b422ca737eef1b68f678e : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdf6a86d3b7c4f868267b64fcca19cfd : SUCCESS in 1h 38m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acb1c6315c57445b92c86d32cbc66de9 : SUCCESS in 1h 37m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a02864136ec455a899b111ed9fba8e1 : SUCCESS in 41m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/52d442a6473b432cb4caa6b8e1431819 : SUCCESS in 50m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/142086c73847433f9612f414a828bbac : SUCCESS in 40m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e56ebcd9cf53448284e4c6aefc27874f : SUCCESS in 1h 38m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df85f6d00d1c4e52bafa86333340fe3e : FAILURE in 42m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7894076e0e604940b14d22d4ff704183 : FAILURE in 2h 27m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b7466c5adf94bb98bb065f46099dcb1 : FAILURE in 2h 10m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f86d7ec55eb741d48b06984d82100c1d : FAILURE in 1h 55m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f3aa8b17ad624f00982eec273fafa56d : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"054b12bb325bcd5601529559e22c6c5f8704c4aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-25 09:57:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1a6c8da22e805880814fef5a537ae3ec59807b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 11:51:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9985d04d7f5c4fc895d5bf1882caff37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf4ab2c9f22e47fa9085e33a4a9f8c28 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55509da686b84225ab3b29e041fc4079 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2caeeb38efe94e8290401fd53df87874 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe189d739804e64a7eeadb5e56785b5 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890e92e1aa85424ba92e1429f3941c06 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a3b8841dee412aaaa19b405021ea08 : SUCCESS in 5m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c038b0af0fa1456996209650f3951334 : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/705aabcb25e14f428a95f3bb07ceb102 : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f2665adda2e4e2ebbc4f0a6f5e20290 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26e850d923984363b2630b605e0db9ae : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c95eed40ac74d29a4ecce3e4f612d24 : SUCCESS in 1h 49m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94568e6ac98a42f1a228f554e0ba095a : SUCCESS in 1h 29m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c33bb27c313341b6b2d96870a1c65133 : SUCCESS in 39m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ceabb6e12ff435c982111b77f7ea97e : SUCCESS in 41m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab34db2cf4fa4ee48fddf95685f12b13 : SUCCESS in 42m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50777bb5f3c34c0b8e41169a6dd65a4e : SUCCESS in 1h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f600b63fda114931a9958ddc4cb2cecc : FAILURE in 37m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cd3aaac098204e6abab8778d60957567 : SUCCESS in 1h 36m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3219596cd18a41aaa02b74843928b11a : FAILURE in 1h 47m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/057a4c55a3594d37bc637e157c070ed4 : FAILURE in 1h 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ec581a66e97e41aaa457da2a72918cc0 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ec010196cf68696a17d6822b422a44766a2e958","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 15:24:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"75ee7ff07dae1c04cd5403fafcc8ce242909e588","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:29:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c79da437ab0de42080dc6c38c206114e6b425a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:29:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d69dd80b03906d6e047b5199f0e1b2caaa6c317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:28:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c70a86eb83441e9e49f4c65d904428\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f7d6d38e9b4f44b80177fd9d1b703f : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb848254466a41eb801eb6c0fff00f07 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0401a7a2dadd452bb0a649b9734be57f : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6a199981504b54916a5d1c5295f47b : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cfff617a0d145afb4bf04968fc040b1 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6144d1a901b0420ab6f0c9e0b131ba8b : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/422e0f27bb1d459d9ae66a454af894d3 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0811b39f17d84decbc6fd8bfcd8191a0 : SUCCESS in 1h 39m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f19a88955f1461489e5e8d21084ef11 : SUCCESS in 1h 55m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3efe216d80f142388f66a95cfc49ea8e : SUCCESS in 42m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/52aca7f8164049c098a11ac26c115db1 : SUCCESS in 42m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35acc2553a62462caec07a5d6853233b : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3d1464751fe13f4f01236134b1b9f2ca82ba43dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:28:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f2a21ae0c33ae04b4e8bef1bfe062d6186f0df66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6080203ecd534f398c3da411c6c6399b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93051fe5bce4c05a18dc3ef67ffcb8d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf07ca79dc0d4a8eabce05d7826c93dd : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69909d976029ed8e3060d7d49e33a6ea1c1e855b","revisions":{"183b8a4d803ef3abdd73c86f56e4fabd28754681":{"kind":"REWORK","_number":1,"created":"2022-07-20 13:23:47.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/84/850484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/850484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/850484/1"}}},"commit":{"parents":[{"commit":"a1aed05cd34b6d7d935ca8a295d240f04e77a8eb","subject":"Merge \"Fix duplicate object error messages\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1aed05cd34b6d7d935ca8a295d240f04e77a8eb"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-20 13:23:47.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/183b8a4d803ef3abdd73c86f56e4fabd28754681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/183b8a4d803ef3abdd73c86f56e4fabd28754681"}]},"branch":"refs/heads/stable/victoria"},"69909d976029ed8e3060d7d49e33a6ea1c1e855b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-25 09:57:17.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/84/850484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/850484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/850484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/850484/2"}}},"commit":{"parents":[{"commit":"a1aed05cd34b6d7d935ca8a295d240f04e77a8eb","subject":"Merge \"Fix duplicate object error messages\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1aed05cd34b6d7d935ca8a295d240f04e77a8eb"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-25 09:57:17.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n(cherry picked from commit 74f2baf539359f5ded2bc01b28d570083bc0a797)\n(cherry picked from commit aa1c69a58667fd53ede2f9e08deaed61545c0b32)\n(cherry picked from commit 1028bc242261aabf0b9dff1709bf80afc06c4a46)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69909d976029ed8e3060d7d49e33a6ea1c1e855b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69909d976029ed8e3060d7d49e33a6ea1c1e855b"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
