)]}'
{"id":"openstack%2Foctavia~850522","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I1633c2cacf1c4dc5c0aa605635545fae8085e296","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2022-07-26 17:12:15.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-26 17:12:15.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-07-25 15:24:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I1633c2cacf1c4dc5c0aa605635545fae8085e296","subject":"Fix update/delete listener CA/CRL error","status":"MERGED","created":"2022-07-20 13:32:46.000000000","updated":"2022-07-26 17:13:30.000000000","submitted":"2022-07-26 17:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850522","meta_rev_id":"8faf6a090e9789bfb9547fee27c93e93117d8a6d","_number":850522,"virtual_id_number":850522,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-26 02:30:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":2,"date":"2022-07-25 15:24:24.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:28:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 02:29:18.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-21 02:29:18.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-07-26 15:28:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"8505d88481f9656ea42830ffbc6d3be59cc62c83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-20 13:32:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17fe25be795ae428a6cb1e4f4ae441e365543419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede3decec16042ae8b25f54ec6338e69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d6285ea35b4456386e99f5429874ed4 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f3d7e8f8b948a29a3a79eacd8650fa : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f14abe78f19245dbabc3452cd91eb7c5 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1110b620583b4091ab4c267496d046e1 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863d5dc2f74d41b28b5205b569b54f56 : SUCCESS in 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e4e209efe754ea3a23b88d29140a557 : SUCCESS in 6m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30438541a9054414ace8fd6720d016e2 : SUCCESS in 7m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f939188c7e5e4103997fe3ca0d3434f6 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c437a6cd471d488b935030bab0829853 : SUCCESS in 11m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c545e77ecfbe474ea40e4bee755c75d0 : SUCCESS in 1h 17m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fd270f2c4de473792f4af0f3a9b0b27 : SUCCESS in 2h 06m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d657fba086f499b8dbcb6b3556d3816 : SUCCESS in 47m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/757a16fdaf25435685c66701ed6fb5b3 : SUCCESS in 57m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de10909b8f24428f93a7e8fa7179f19d : SUCCESS in 37m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/485563b4648743feb9b45004139b15c5 : SUCCESS in 2h 05m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/51e6714d5e5b473aa5077beac345d58c : RETRY_LIMIT in 9m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5687a17b672a46b98d9a521a9126ae7c : SUCCESS in 2h 09m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20b3e3bf957219af83dfc002d88d553c80656496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-25 13:16:15.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6b4521234f368f0c63f6dbe510a29e6e69126332","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 15:24:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5d56a0de45f68782627b60b2d7537d7ab47f00e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 15:46:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec739de49dc4be3ad504f5409e3a31e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05abfd5e2cbf4cbfbcf2d7c0f01c4a1a : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60351b97c404e329f915d9b9793ecbb : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/155f0697cd6c4bab8dcbee655185cd33 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1968e2799fca4f738b378f5595dcc067 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d8a869179844a6bb79abff15753160 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f81c735ece7e4b6da7117de2c72dfab5 : SUCCESS in 4m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c01f61ca46c94a14b1b692af1f713f16 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2231d44b878d476783f2c1c31359d56f : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/982330dd14bb4058a1a3c92dc3223615 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03bc824134f84abab3cf9b896958e781 : SUCCESS in 1h 43m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80e32e7ca7a14b03bbb863acd7a5755c : SUCCESS in 2h 06m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8aa0b4981038457ba51d7894e9559e45 : SUCCESS in 1h 05m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bff22ecd609446bfadf7778aaf797c38 : SUCCESS in 59m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7af9f2bf059243719d646bd1ab4dc7d6 : SUCCESS in 40m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/473a62eac42c46b8aa4ce949f3aed214 : SUCCESS in 2h 23m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3701301f24234bddb458a0aee4787cf2 : RETRY_LIMIT in 9m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d524b273b6ce460c921e92726e14407e : SUCCESS in 1h 56m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f844a4d2af411a99cb687e3ce756fe10d97d93e1","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-07-26 02:30:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dfb824a83a5b6d9cc79d7bbde854c07456e3d584","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:28:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05cf4e1e5cb123326f2b56cff10b57f30973a9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:29:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"695bd8efb6b74f7ea07e6c50da599000f7092a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:12:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b42a70e808341e8a777e121f31d6f4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8208a4b5ee334e20b2e3342e51f29059 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d453027d9bf487da2d38da6121964f4 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbfc1051348748e8b99109ecd78dd2e7 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a566fbda1346558507f5fb80411a59 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b2d18fec6e6443f98adc30296c14329 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b828bbf53c5449983f60f079aaf6263 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47662df788aa4a44b00c07f75b048556 : SUCCESS in 1h 13m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6792019caa0045d884df2fe05fcc1cd5 : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d98ac6856c46422e9861c6c0bd7ba3d3 : SUCCESS in 46m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ff6ce34abcdb450887d44286f5da198b : SUCCESS in 38m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd3d9e0e556845b281981f45f3327683 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0483114661b6954debef40951df53437387da529","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:12:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8faf6a090e9789bfb9547fee27c93e93117d8a6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9280d5bccf8f4e6a9a1182b4ead3254a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1e792ce6f84d64b51beb1d9014e366 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f07acc9e4db74717a7cde0d64cdfde41 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14f131dbf2f8a72d5d1e7a1ece394a9b557e7c49","revisions":{"df2ee208ea03a54cc1c601d9b13ea2e679a969f8":{"kind":"REWORK","_number":1,"created":"2022-07-20 13:32:46.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/22/850522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/850522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/850522/1"}}},"commit":{"parents":[{"commit":"5838600e6f639ab520af01fea93370bc03bdf43b","subject":"Merge \"Fix duplicate object error messages\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5838600e6f639ab520af01fea93370bc03bdf43b"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-07-20 13:32:26.000000000","tz":480},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2ee208ea03a54cc1c601d9b13ea2e679a969f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2ee208ea03a54cc1c601d9b13ea2e679a969f8"}]},"branch":"refs/heads/stable/ussuri"},"14f131dbf2f8a72d5d1e7a1ece394a9b557e7c49":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-25 13:16:15.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/22/850522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/850522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/850522/2"}}},"commit":{"parents":[{"commit":"5838600e6f639ab520af01fea93370bc03bdf43b","subject":"Merge \"Fix duplicate object error messages\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5838600e6f639ab520af01fea93370bc03bdf43b"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-06-08 07:43:56.000000000","tz":480},"committer":{"name":"conna","email":"yang.shaoxue@99cloud.net","date":"2022-07-25 13:16:15.000000000","tz":0},"subject":"Fix update/delete listener CA/CRL error","message":"Fix update/delete listener CA/CRL error\n\nFixed \"Could not retrieve certificate\" error when updating/deleting the client_ca_tls_container_ref field of a listener after a CA/CRL was deleted.\n\nConflicts:\n        octavia/tests/unit/api/drivers/test_utils.py\n\nStory 2010081\nTask 45577\n\nChange-Id: I1633c2cacf1c4dc5c0aa605635545fae8085e296\n(cherry picked from commit 9a5273d3ea91109c8f22671fd7215eb7b953b975)\n(cherry picked from commit 74f2baf539359f5ded2bc01b28d570083bc0a797)\n(cherry picked from commit aa1c69a58667fd53ede2f9e08deaed61545c0b32)\n(cherry picked from commit 1028bc242261aabf0b9dff1709bf80afc06c4a46)\n(cherry picked from commit 69909d976029ed8e3060d7d49e33a6ea1c1e855b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14f131dbf2f8a72d5d1e7a1ece394a9b557e7c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14f131dbf2f8a72d5d1e7a1ece394a9b557e7c49"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
