)]}'
{"id":"openstack%2Foctavia~850569","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745","project":"openstack/octavia","branch":"stable/yoga","topic":"fips-compatibility-cs9-stable/yoga","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-08-22 17:12:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745","subject":"Change FIPS jobs to centos-9-stream","status":"MERGED","created":"2022-07-25 10:51:23.000000000","updated":"2022-08-22 17:14:43.000000000","submitted":"2022-08-22 17:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850569-fips-compatibility-cs9-stable/yoga","cherry_pick_of_change":831840,"cherry_pick_of_patch_set":10,"meta_rev_id":"90fbf73f4eadf394b9e6a0bfe98c8289273655bd","_number":850569,"virtual_id_number":850569,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 17:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-09 17:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-22 14:40:47.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-22 14:40:47.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 10:51:23.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"CC"},{"updated":"2022-07-25 12:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 17:59:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-08-22 14:40:47.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8b047d2e2a0158f8ff819bb0152e9ae3c9144efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-07-25 10:51:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5e3a599f472de213630a3ecfb27b032747323f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e5eeeed5d44bdfa2ee1bd7909f3311\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/901dd01d656442fdae1618fad4bae836 : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a17518ae87545f6bc99500d57ca39f1 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12c4bdfe3f0f4bb8b1072604f89e8e7b : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf99b559c1964ef584cf67ec714e4aac : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94e11d3422ec4255b1ef0a9e8fcac388 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa65a0a039e420ea9f901271584a7a4 : SUCCESS in 10m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/43ff5250b7f44ff89d604f4507c7ecdb : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/004bcf47c871429989473a0f0bcf6280 : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b76ff094e661425d8cb180dbe10c3567 : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ad9e81618bb4b288ec06daff04c961d : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ffa12fb660f4669a782fa725e1b38c7 : SUCCESS in 2h 03m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0256914b8d64f78ab5999d281429949 : SUCCESS in 1h 31m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab35f62dd3514ace88e85ec0959824c3 : SUCCESS in 44m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7999648c4a074ed1a4564e3d6611dd9a : SUCCESS in 58m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/748e0536bea840dba5d1ab0ea8e9fe3c : SUCCESS in 1h 27m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da2ef9d2355e4dd2a0590de8e2a276e5 : SUCCESS in 40m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/106cd23cecb34308b5dfb92673fd1f09 : SUCCESS in 1h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6947fe137db498b9c6ccd2e999e4bec : SUCCESS in 2h 04m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d0c929e1756433ca9758f34111aefe4 : SUCCESS in 1h 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/28fdefdb35fc4576907db1134490fe02 : SUCCESS in 1h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/299513bf466d421cbaab3202d4fa73d6 : SUCCESS in 1h 52m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/95994669d5a14cdd8e7f80bb8063c612 : FAILURE in 56m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"770caf60ee7deac0457bbc8161c6a4cc15fce279","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-09 17:59:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ed884f0f20ea77ef241a00e02d65f66d718c0b2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 14:40:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74c1fcf408c02e6c0ed4c9a7f779b1975e9571b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 14:41:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd665e2435a9f3d9bbff8c73ba4cca323be0b451","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:12:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af6c9e6a7c34b77a168374b3226441c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e223e046e84c43ba8c00b534ed622f : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807d2d0688454b50a178cfbd17e8e2b8 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d4329363394b939bbaba40b89660af : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fba33ba704144b5fb16ffe3b260d0eb9 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd25c45e49044b9a56ff70d7ed7d728 : SUCCESS in 11m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac83a3ba11634f0a8c5bd27f88e1fe8e : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eeb4b4c09bb847b29ae5cae999488be1 : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea9760dd3680483cae52c66d2e81622d : SUCCESS in 2h 06m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19dd1e7275f047b09a151c1464131735 : SUCCESS in 2h 18m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f5a31afbfaa40dc92d9a65cf53c99a2 : SUCCESS in 1h 02m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/704f8b92b947439b9b29f0c88c6cdf43 : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9866a18951ffcec107d0a32f0763ea79e34b07e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:12:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90fbf73f4eadf394b9e6a0bfe98c8289273655bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:14:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d6da32d328e437ebb37daede357be6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abf8fd702184cd690f26764db0f0d36 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef76cbf7dec92071f04f57f117a2d852218f9071","revisions":{"ef76cbf7dec92071f04f57f117a2d852218f9071":{"kind":"REWORK","_number":1,"created":"2022-07-25 10:51:23.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/69/850569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/850569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/850569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/850569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/850569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/850569/1"}}},"commit":{"parents":[{"commit":"10f037c7e5cabf870957f71e894c4e78f26cdeae","subject":"Merge \"Set sensible nf_conntrack_max value in amphora\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/10f037c7e5cabf870957f71e894c4e78f26cdeae"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-07-25 10:51:23.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nIt also fixed the missing nslookup_target variable for ansible.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n(cherry picked from commit 6d7464303ee554de2a598502486574e2a9a4e3f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef76cbf7dec92071f04f57f117a2d852218f9071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef76cbf7dec92071f04f57f117a2d852218f9071"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
