)]}'
{"id":"openstack%2Foctavia~850591","triplet_id":"openstack%2Foctavia~master~Ib704cdc69a4888381151f2d59dbfec5c00ef3c16","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-28 19:27:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib704cdc69a4888381151f2d59dbfec5c00ef3c16","subject":"Fix rfc3986 deprecation warning","status":"MERGED","created":"2022-07-21 00:30:12.000000000","updated":"2022-07-28 19:29:03.000000000","submitted":"2022-07-28 19:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850591","meta_rev_id":"764fdbfbda0b3635f7b34818d037935b69c74874","_number":850591,"virtual_id_number":850591,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 19:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 16:47:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 16:47:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 03:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 15:33:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-07-28 12:32:21.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"a28648d78e5919171ae076b6ad450a2f266862a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-21 00:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873d67794b877bcd512521c27c4c6c1414933268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 03:10:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5cddda7484a422b8739132388593297\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b60c8536d18847d38e2f388c0b3e7370 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e434385c49dc4daeb2f784d884a60630 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/800eb98301ea4ec7b5787411b16bba6c : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/801e50bf78c14026b97c7c4f74de65ff : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56d252a76a0b4bcc9edf520c2c79eb2f : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2a58a278404631a92b080b30820ee2 : SUCCESS in 13m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c8120613ce8a4848aaf7c9c193b0a846 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/805e51c8681f4842b36cb84f48fd8055 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/92711577e0ac4ae9a8eeaf328bf4b119 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f58b8524fa0493291f29be1dca280cd : SUCCESS in 8m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a4b71f4e6484cccb9cbc4c626bd37ed : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/958427f588994bbc877604a99194251d : SUCCESS in 8m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de96d675126f44c3a82a5ecad49e31c8 : SUCCESS in 2h 23m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/732e7d5f6517424aae155ea105d03f17 : SUCCESS in 1h 41m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7b5792ad7974e1d93153c4c84e652c2 : SUCCESS in 43m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f4ffb778f9b4fd5b72620dfd3d75801 : SUCCESS in 1h 12m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/10efba0daadb4d89a3a8d581ef73e0ce : RETRY_LIMIT in 24m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/256977aa0684481b90f365a7f83bf368 : SUCCESS in 38m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95f6bad6313e448c8ade87bfd256c6bd : SUCCESS in 1h 54m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7b7bc3eb1c148e7812ab0264f630f8d : SUCCESS in 2h 38m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/553dc11264ce4dbf8214fed30343bb2c : SUCCESS in 1h 59m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9974a8761bd04412b8f4697511650043 : SUCCESS in 1h 53m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f990f05e9faa4bec8dfc607286d47a74 : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9e544955ab44be6a963e23f2232b7d9 : SUCCESS in 2h 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d4c890f224e5d88f7f0cc672430ead9c59d2772","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:33:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe4ddf3c7fb9559b59df03d2aba620d1bccc30a1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-28 12:32:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38359b78e284cdab4c4e407706c50a717fdca4da","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-28 16:47:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13e6f593a29de62b8b7f2a16aca01839111e55a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 16:47:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63e9c132718ccf9a9a472b737198af739b8694c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:27:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e73a8f5e094bb78bb68f974a94698c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45959a5f6ba447a87a7f770f3a54546 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99305696fa044fa597d905343cb1fea3 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d51ecc85ace4190a6a0603b7776bf0e : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e1e6cbab7a4743959c7867bec34a30 : SUCCESS in 12m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8697062ab93c48769af26c873916a69c : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9aa9cbd67253415cb86a60b838758e45 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1892335e01e0484cb0f50ccb3ef42748 : SUCCESS in 2h 33m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c789f964ca741b390d060c0f4fa0da9 : SUCCESS in 1h 28m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77be935aabb34a5f81acc66ea95e031f : SUCCESS in 58m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd3a1df9f8bb47638a3927f0222b00d6 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8f1ff16832bf46267945f879bef54ef7ce0ec703","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"764fdbfbda0b3635f7b34818d037935b69c74874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:29:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e141b0017324f128b6954509648cbbc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad411f97f1064b269fb6e8ab8a0853d5 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d7fb4eca622b38132f3cd80a885d72daa40f46b","revisions":{"8d7fb4eca622b38132f3cd80a885d72daa40f46b":{"kind":"REWORK","_number":1,"created":"2022-07-21 00:30:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/91/850591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/850591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/850591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/850591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/850591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/850591/1"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-07-21 00:28:07.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-07-21 00:30:10.000000000","tz":0},"subject":"Fix rfc3986 deprecation warning","message":"Fix rfc3986 deprecation warning\n\nThis patch modernizes our use of the rfc3986 module and resolves a deprecation warning: Please use rfc3986.validators.Validator instead. This method will be eventually removed.\n\nChange-Id: Ib704cdc69a4888381151f2d59dbfec5c00ef3c16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d7fb4eca622b38132f3cd80a885d72daa40f46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d7fb4eca622b38132f3cd80a885d72daa40f46b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
