)]}'
{"id":"openstack%2Foctavia~850722","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ida33a1a24eab2706902b1ce3cca704d51fa6a599","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-22 17:12:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida33a1a24eab2706902b1ce3cca704d51fa6a599","subject":"Fix driver-agent cleanup","status":"MERGED","created":"2022-07-22 06:56:35.000000000","updated":"2022-08-22 17:14:23.000000000","submitted":"2022-08-22 17:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850722","meta_rev_id":"75537f2aaad40fa72e4f3dae02c618184898b660","_number":850722,"virtual_id_number":850722,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 17:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-22 14:38:57.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-22 14:38:57.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 09:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 15:31:20.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-08-22 14:38:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"24a9478acb02ad1542a1ae590cf24eb6a90a189c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-22 06:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8befb31f20802ac353f94512d7c00a11c49bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:27:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd5cc846a864acd89548971c1ae6b92\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e5fbc4da1d84cf3a689e3b878faa2d7 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/538a2884492343c8aac08927d8742c49 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d441e76b5bca4e198bd222831cb4004d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac4181d9f51413382f7c0a5964f2339 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c01853e6f614f5f8fd857ff20a79912 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5548a49b3cc543bc849e40c5969f78b8 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082a716dc5504240bfd4076dfefc63d3 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72445fde6cee44a59839f9ef8fa6f5cf : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/523c0015842648698d61cb9752410371 : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f8e0c3782cf4970802b7ebd92c075a4 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1347f0d35544473bed9c8ca4f2a41f0 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c53bd14b0eff4f96baa06f7d20c9f4aa : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed12420bb9814119ae2b8aa80c333388 : SUCCESS in 1h 40m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f623689b95114e808bcd4dd52656a6d9 : SUCCESS in 1h 52m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/748b4371129547099d83a39680fd34db : SUCCESS in 39m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bca8f63a86df4b058bf9111ab95b1381 : SUCCESS in 52m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8498c2c3e4d430fb266d77127bfc24a : SUCCESS in 37m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ff286650012149a1854a16cd0ca3bf2f : RETRY_LIMIT in 13m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b89fa4324754fdd85e90fc258cd5d64 : POST_FAILURE in 1h 56m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d60d51de314147c8b290010982cec1be : FAILURE in 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/42ae4e9fa76041e48c83f2749aab132a : SUCCESS in 2h 10m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/21d1565338e44fe391c2a8ebb4e19eac : SUCCESS in 1h 42m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/182234f744d74d248078d4a0e1319058 : SUCCESS in 1h 54m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0c860f0584c94f889377c2f12e20192d : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b8990089cb4b6ee0fb03e8ffc860cd49ce9e452","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:31:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f20d863941598cac493805296949aa6987307eb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 14:38:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22b3ddbff9f77cfda60e186b8d9863c8e818c8ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 14:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdc0bd23ce41976b2e86d34b897c8454d7f7dadd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:12:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4924ca5109d7487b94692075a667445d\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e3ef10a05b7489e82aabe8cadeb222d : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1d1afc8fd14fa4a9556d424bdbd9f1 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d539746a944a48ddb81653d3fddfda7a : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccbf664004cf42be8e1ccc7405c81956 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea97867cf86485997a976a90bc8b140 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ba33e4f880446a286e84ea7cca6361d : SUCCESS in 6m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/14267ebed67242c981567baabd493a5d : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3984ebc4230f4da396848caf5f0fc4f6 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a0403d066ec4f25a4387868c3c37ebb : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cae9499fb9848faa07bb14ed03b0eab : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5da601880f1c4c658809fe9a9a07e676 : SUCCESS in 39m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ae33d62e73244c3398f105adabad8447 : SUCCESS in 35m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f6160c3ac6f4c5c9a0ef3685e00cec0 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ce9ad6f47feb96c13fbce1d68bb85b3109d0f7de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:12:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"75537f2aaad40fa72e4f3dae02c618184898b660","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:14:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2319769da0d451b8d01792e8e9ce323\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632ca3fcb773462facd68664e83d74a4 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/416cc378a8f64b8bab0f97b4c4918c5d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84e2e36717f26bd95cca1af0be635a0920e2800e","revisions":{"84e2e36717f26bd95cca1af0be635a0920e2800e":{"kind":"REWORK","_number":1,"created":"2022-07-22 06:56:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/850722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/850722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/850722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/850722/1"}}},"commit":{"parents":[{"commit":"94e5a6f99d752848a0bbd16ddb10e76fc53e0179","subject":"Merge \"Set sensible nf_conntrack_max value in amphora\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94e5a6f99d752848a0bbd16ddb10e76fc53e0179"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-25 08:08:11.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-22 06:56:28.000000000","tz":120},"subject":"Fix driver-agent cleanup","message":"Fix driver-agent cleanup\n\nThe cleanup function (that handles graceful shutdown) of the Octavia\ndriver-agent didn\u0027t work properly, the threads that were spawned for\nstopping the Unix listeners were stuck (deadlock), so the threads leaked\nand the join functions in the caller didn\u0027t return.\n\nNow the code uses a thread for handling the connections, and the\nshutdown function is called from the process.\n\nStory 2010052\nTask 45468\n\nChange-Id: Ida33a1a24eab2706902b1ce3cca704d51fa6a599\n(cherry picked from commit 51eef0d6d7ac6f511e23a8ceb59dc0f0dd7cfbdf)\n(cherry picked from commit 1a1c5d421f8650f675ab90ceb45570cfbde83981)\n(cherry picked from commit e33414dba652893cf4a2460a0381232c18a92e37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84e2e36717f26bd95cca1af0be635a0920e2800e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84e2e36717f26bd95cca1af0be635a0920e2800e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
