)]}'
{"id":"openstack%2Foctavia~850894","triplet_id":"openstack%2Foctavia~master~I4600bade69f27c670e58077ded5b7646d2e3f786","project":"openstack/octavia","branch":"master","topic":"image-build-debug","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-08-18 15:18:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4600bade69f27c670e58077ded5b7646d2e3f786","subject":"Fix stat error in periodic image build job","status":"MERGED","created":"2022-07-25 14:05:19.000000000","updated":"2022-08-18 15:20:34.000000000","submitted":"2022-08-18 15:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850894-image-build-debug","meta_rev_id":"8eebaa9c9f3a87a500d54788c65ecd8feed17128","_number":850894,"virtual_id_number":850894,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 15:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-08-18 12:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-10 16:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-08-18 12:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 14:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 16:39:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-08-18 12:48:36.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"4ac7490eda781edcfc1c1955aafd719527ed2c18","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 14:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48873c376bb4322f2e1e259bf05fec46097671d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:05:51.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-openstack-octavia-amphora-image-focal","accounts_in_message":[],"_revision_number":1},{"id":"7004aff96a5426f64d5b2e63efc9590bab463403","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 14:25:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b13c58e426d95a2f52a388ec6ba624eb5b2522a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:26:00.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-openstack-octavia-amphora-image-focal","accounts_in_message":[],"_revision_number":2},{"id":"e8bc574f59b2de57f608f4795d3fb6e683fbb0ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 14:27:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeb7a4e04e97195d81daa63cda38b75086291def","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 14:49:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b868c93ab87dcb8c8ee18b729a91b5cd956985e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 16:05:22.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"aa5adcaf72f302a4c3e2e9d9b0ce669871e6019b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 16:18:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c7bd56f39b9e6dafbe4d7a786c6660fb3547141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-25 16:20:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cda98e7b1b7237e684fb533cf0868ffc62240d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 18:41:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d039c56d3c684b83b341635b4485e899\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c07713f1686e483ab60b4ed2b359f84d : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fddfe20d1ea4ec0be2d695bee369563 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ce13dd96e6c4a428015367c1a8ddc8e : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f426035ef89e40f59518b9e260c4a116 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de6db53aca6549038bdfd750f7811151 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660ddd74d84f45c782b3a0e802a4a672 : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35f0fb4fba334977848db53300966a72 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/93aeb52468ae4b16a424111c314756be : SUCCESS in 6m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/19272e4aba0f46e9aca48d85a61033bd : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7917c978dd1450da01fc50249217ce6 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50ada98304fa410b8805772d413b6c2a : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/68de37ae822540d8bb1a918970837e3a : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ac06faea9f3464db762a7e83e5c3796 : SUCCESS in 8m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d861dbc3452044a1adf699498383608a : SUCCESS in 2h 15m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e56ba15d980c443bb381918a30c65921 : SUCCESS in 1h 34m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abba3ecb6b014c2798596e261a1755e2 : SUCCESS in 50m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e0377d1a9074a60b5cbcbfd50a2923c : SUCCESS in 58m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ea0b47e871e64e5892fb9e5b6f84d60b : SUCCESS in 1h 22m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17c63da409074885a8e275183f9c71bd : SUCCESS in 42m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38fea857a1ba4b849d1e126ed1cc96e4 : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8ccfe4964c04a778fff8443dd61d43e : SUCCESS in 2h 14m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a1d8de7555d64ec5bf0becd3a68e5ade : SUCCESS in 1h 43m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/785f17e21bf947f9ab836d35ad140c20 : SUCCESS in 1h 43m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bbdc6f1807be4316b294f9a6125eaf0c : SUCCESS in 1h 22m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/40a346a3ba254f1ca7313f1d146da7a0 : SUCCESS in 1h 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a52e349591e17a620ecdb59e848da5aa2c7f5f93","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-10 16:39:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac28d700791eac9850b4da9f4c9718cf5bb84d89","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-08-18 12:48:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6e8ac1c1911ab786abf81f17da2e00f9804304ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 12:49:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"12ebd5f65d7880f882b5bb96bade85c81cadff5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:18:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa0e161ba884bfa8d28d780638633f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c05e68e1a24df4863e69d3fb0d0e11 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81d023de802a49a6b655c4f7179a47d7 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85bcadaf0d8e40578b7e21365bf912c2 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4fbf8080eb4862ba281baa1f1e6078 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deb0a391b7394168ae856b6c1dbcf7e4 : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8cc1e0e0c35643f1854b1c253e1cd6f6 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7b00d3303ae94bb281c2310207502f31 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7912260dd71b4c96b086881d98221363 : SUCCESS in 2h 24m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dd139968aea4048b8fd267b7179c998 : SUCCESS in 2h 21m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2d33ff86a494c658abc49ec746cb32f : SUCCESS in 55m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2ee2c19ae4245d797ff995d6d68ed93 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":6},{"id":"24c6e5740e04806bc3034d81a5e05270e0e24375","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:18:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8eebaa9c9f3a87a500d54788c65ecd8feed17128","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:20:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5b3ad2224f40ff8dabfa3ed3514b7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1724579040f244cd921fcfefbb1ceaab : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/858fb48f1ec04c2e9ca31048106e9585 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6db364087304b10e71fbf5742bbc62bcaafc02c6","revisions":{"b36a6a93720923e62db14288ae750979d0f066d1":{"kind":"REWORK","_number":1,"created":"2022-07-25 14:05:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/1"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"subject":"Debug disk image build error","message":"Debug disk image build error\n\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b36a6a93720923e62db14288ae750979d0f066d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b36a6a93720923e62db14288ae750979d0f066d1"}]},"branch":"refs/heads/master"},"dc4e9d92771a6606a6a61b5fd7dce14dfd27351c":{"kind":"REWORK","_number":2,"created":"2022-07-25 14:25:26.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/2"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:24:53.000000000","tz":120},"subject":"Fix stat error in periodic image build job","message":"Fix stat error in periodic image build job\n\nCurrently the publish-openstack-octavia-amphora-image-centos-9-stream\nshows this error, even though the job appears to succeed: [1]\n\n2022-07-25 02:21:52.929693 | centos-9-stream | Successfully built the\namphora image using amphora-agent from the master branch.\n2022-07-25 02:21:52.931356 | centos-9-stream | stat: cannot statx\n\u0027/home/zuul/test-images/test-only-amphora-x64-haproxy-centos-9-stream.qcow2.qcow2\u0027:\nNo such file or directory\n2022-07-25 02:21:52.931640 | centos-9-stream | Amphora image size:\n\n[1]:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_741/periodic/opendev.org/openstack/octavia/master/publish-openstack-octavia-amphora-image-centos-9-stream/7414761/job-output.txt\n\nStory: 2010187\nTask: 45876\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc4e9d92771a6606a6a61b5fd7dce14dfd27351c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc4e9d92771a6606a6a61b5fd7dce14dfd27351c"}]},"branch":"refs/heads/master"},"eada2d8773b51c28d272dfb5be1ad5c01b5bb7dd":{"kind":"REWORK","_number":3,"created":"2022-07-25 14:27:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/3"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:26:56.000000000","tz":120},"subject":"Fix stat error in periodic image build job","message":"Fix stat error in periodic image build job\n\nCurrently the publish-openstack-octavia-amphora-image-centos-9-stream\nshows this error, even though the job appears to succeed: [1]\n\n2022-07-25 02:21:52.929693 | centos-9-stream | Successfully built the\namphora image using amphora-agent from the master branch.\n2022-07-25 02:21:52.931356 | centos-9-stream | stat: cannot statx\n\u0027/home/zuul/test-images/test-only-amphora-x64-haproxy-centos-9-stream.qcow2.qcow2\u0027:\nNo such file or directory\n2022-07-25 02:21:52.931640 | centos-9-stream | Amphora image size:\n\n[1]:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_741/periodic/opendev.org/openstack/octavia/master/publish-openstack-octavia-amphora-image-centos-9-stream/7414761/job-output.txt\n\nStory: 2010187\nTask: 45876\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eada2d8773b51c28d272dfb5be1ad5c01b5bb7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eada2d8773b51c28d272dfb5be1ad5c01b5bb7dd"}]},"branch":"refs/heads/master"},"ac9a8fec4ff4796e0f227c39b5b4fbe6cea63510":{"kind":"REWORK","_number":4,"created":"2022-07-25 14:49:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/4"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:49:03.000000000","tz":120},"subject":"Fix stat error in periodic image build job","message":"Fix stat error in periodic image build job\n\nCurrently the publish-openstack-octavia-amphora-image-centos-9-stream\nshows this error, even though the job appears to succeed: [1]\n\n2022-07-25 02:21:52.929693 | centos-9-stream | Successfully built the\namphora image using amphora-agent from the master branch.\n2022-07-25 02:21:52.931356 | centos-9-stream | stat: cannot statx\n\u0027/home/zuul/test-images/test-only-amphora-x64-haproxy-centos-9-stream.qcow2.qcow2\u0027:\nNo such file or directory\n2022-07-25 02:21:52.931640 | centos-9-stream | Amphora image size:\n\n[1]:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_741/periodic/opendev.org/openstack/octavia/master/publish-openstack-octavia-amphora-image-centos-9-stream/7414761/job-output.txt\n\nStory: 2010187\nTask: 45876\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac9a8fec4ff4796e0f227c39b5b4fbe6cea63510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac9a8fec4ff4796e0f227c39b5b4fbe6cea63510"}]},"branch":"refs/heads/master"},"274eee37531fdaa056dfd98237a36e285d44be8a":{"kind":"REWORK","_number":5,"created":"2022-07-25 16:18:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/5"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 16:18:06.000000000","tz":120},"subject":"Fix stat error in periodic image build job","message":"Fix stat error in periodic image build job\n\nCurrently the publish-openstack-octavia-amphora-image-centos-9-stream\nshows this error, even though the job itself seems to succeed: [1]\n\n2022-07-25 02:21:52.929693 | centos-9-stream | Successfully built the\namphora image using amphora-agent from the master branch.\n2022-07-25 02:21:52.931356 | centos-9-stream | stat: cannot statx\n\u0027/home/zuul/test-images/test-only-amphora-x64-haproxy-centos-9-stream.qcow2.qcow2\u0027:\nNo such file or directory\n2022-07-25 02:21:52.931640 | centos-9-stream | Amphora image size:\n\n[1]:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_741/periodic/opendev.org/openstack/octavia/master/publish-openstack-octavia-amphora-image-centos-9-stream/7414761/job-output.txt\n\nStory: 2010187\nTask: 45876\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274eee37531fdaa056dfd98237a36e285d44be8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274eee37531fdaa056dfd98237a36e285d44be8a"}]},"branch":"refs/heads/master"},"6db364087304b10e71fbf5742bbc62bcaafc02c6":{"kind":"REWORK","_number":6,"created":"2022-07-25 16:20:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/850894/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/850894/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/850894/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/850894/6"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 14:04:53.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-07-25 16:20:07.000000000","tz":120},"subject":"Fix stat error in periodic image build job","message":"Fix stat error in periodic image build job\n\nCurrently the publish-openstack-octavia-amphora-image-centos-9-stream\nshows this error, even though the job itself seems to succeed: [1]\n\n2022-07-25 02:21:52.929693 | centos-9-stream | Successfully built the\namphora image using amphora-agent from the master branch.\n2022-07-25 02:21:52.931356 | centos-9-stream | stat: cannot statx\n\u0027/home/zuul/test-images/test-only-amphora-x64-haproxy-centos-9-stream.qcow2.qcow2\u0027:\nNo such file or directory\n2022-07-25 02:21:52.931640 | centos-9-stream | Amphora image size:\n\n[1]:\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_741/periodic/opendev.org/openstack/octavia/master/publish-openstack-octavia-amphora-image-centos-9-stream/7414761/job-output.txt\n\nStory: 2010187\nTask: 45876\nChange-Id: I4600bade69f27c670e58077ded5b7646d2e3f786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6db364087304b10e71fbf5742bbc62bcaafc02c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6db364087304b10e71fbf5742bbc62bcaafc02c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
