)]}'
{"id":"openstack%2Foctavia~852269","triplet_id":"openstack%2Foctavia~master~I8f90cb3372fb33ea5601d33677585aec8080e985","project":"openstack/octavia","branch":"master","topic":"story/2010216","attention_set":{"34512":{"account":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"last_update":"2025-10-03 12:33:45.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"31976":{"account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"last_update":"2025-12-10 17:51:40.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},"37241":{"account":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"last_update":"2025-10-03 12:33:45.000000000","reason":"Votes got outdated and were removed: Backport-Candidate+1, Code-Review+1"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-02-14 10:37:10.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"36795":{"account":{"_account_id":36795,"name":"Gabriel Hartmann","display_name":"Gabriel Hartmann","email":"gabriel.hartmann@netways.de","username":"ghartmann"},"last_update":"2024-02-14 10:40:12.000000000","reason":"\u003cGERRIT_ACCOUNT_31976\u003e replied on the change","reason_account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-12-09 16:00:17.000000000","reason":"\u003cGERRIT_ACCOUNT_31976\u003e replied on the change","reason_account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-02-14 10:37:10.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"38031":{"account":{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},"last_update":"2025-12-09 11:16:36.000000000","reason":"Votes got outdated and were removed: Backport-Candidate+1, Code-Review+1"}},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-12-10 17:51:40.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2025-12-11 15:00:24.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"37685":{"account":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"last_update":"2025-12-15 14:48:33.000000000","reason":"\u003cGERRIT_ACCOUNT_37685\u003e replied on the change","reason_account":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}},"31754":{"account":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"last_update":"2025-12-12 12:09:53.000000000","reason":"\u003cGERRIT_ACCOUNT_31754\u003e replied on the change","reason_account":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"}}},"hashtags":[],"change_id":"I8f90cb3372fb33ea5601d33677585aec8080e985","subject":"Health sender: support to send via TCP","status":"NEW","created":"2022-08-05 13:18:15.000000000","updated":"2025-12-15 14:48:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":48,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d01a6b41000dabf9dc762e0795f165e82964c063","_number":852269,"virtual_id_number":852269,"owner":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-09 17:43:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":1,"date":"2025-12-10 17:51:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-12-15 12:20:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2025-12-15 14:48:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2025-12-12 12:09:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36795,"name":"Gabriel Hartmann","display_name":"Gabriel Hartmann","email":"gabriel.hartmann@netways.de","username":"ghartmann"}],"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 17:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 07:15:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-14 11:31:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-11-28 11:56:23.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-02-13 08:27:00.000000000","updated_by":{"_account_id":36795,"name":"Gabriel Hartmann","display_name":"Gabriel Hartmann","email":"gabriel.hartmann@netways.de","username":"ghartmann"},"reviewer":{"_account_id":36795,"name":"Gabriel Hartmann","display_name":"Gabriel Hartmann","email":"gabriel.hartmann@netways.de","username":"ghartmann"},"state":"CC"},{"updated":"2025-03-12 16:47:35.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"CC"},{"updated":"2025-03-18 13:31:10.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2025-03-21 16:24:49.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"state":"CC"},{"updated":"2025-03-21 16:25:12.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"state":"REVIEWER"},{"updated":"2025-05-15 08:39:25.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"state":"REVIEWER"},{"updated":"2025-05-15 08:57:46.000000000","updated_by":{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},"reviewer":{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},"state":"REVIEWER"},{"updated":"2025-05-15 09:08:00.000000000","updated_by":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"reviewer":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"state":"REVIEWER"},{"updated":"2025-10-03 09:48:32.000000000","updated_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"reviewer":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"state":"REVIEWER"},{"updated":"2025-12-04 12:41:52.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"b22cb5891177a87ebbc0f466d967c9db54ed6c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-08-05 13:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8845258651fface4216fb25df43c55bfa25001de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-08-05 14:31:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8bff5cd1223912f63e4dc9262af30637443c8c1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-08-05 15:01:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b481c7723ead66720734efcf79d6435d91860da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:18:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30ac6048fa014cd582f996bf67922d11\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe0108116409469c910e526c62e21fef : FAILURE in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588f891454da400bbb9890b753624354 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d016b2e271ff485985f04a626a948eed : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc961f3a2deb4f29960137108a28d100 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5dd9d5dc5264664a3221c252bfc1195 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d2ef605a0e4a7087223189ced06201 : SUCCESS in 11m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/805c27d68cdd4469aa62b4635e43eb62 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c396ceed001045b287ebca97d0cfaa3f : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0cb4b26c13034f8688b34239433dd9d1 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cc698f3b5f9497293aca8aaef6b8c6c : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba2e0f60d76c41039c44f19cb5e8dd07 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e406ff18208a44a1a78a14fac36b8fa9 : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84aebb327b9244539937b0208324e59a : SUCCESS in 2h 07m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b907ff1f25c47868aa79c4fad8be1b1 : SUCCESS in 1h 36m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8102b2137934d8c859bdafd0f719796 : SUCCESS in 48m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/feeb86f05f1943aaac315c78b7147da2 : SUCCESS in 1h 04m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0c55e90ecf804e74b8fc79726e6594b4 : SUCCESS in 1h 09m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0b5812e53a74ca8a9acfb9497bfeb7b : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/98ad825733054aea9a6ebd6eb4625a14 : SUCCESS in 1h 50m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aeb2e7162431408fb2aa60211ff7189e : SUCCESS in 2h 06m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5763ec05f1a43368dcce44bbf1524ea : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5bdb01a35bc841568f8145611efa2244 : SUCCESS in 1h 56m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7ecff673b3504c2d8e73f100b5e008b8 : SUCCESS in 1h 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/af6fecc2c38d4d7f8e5235a6ab3b4783 : SUCCESS in 2h 01m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98b6f9032072546c058b19720481c0087acad427","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-08-08 08:02:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e09427a630f3956c0ae8d77476f9bb244a449609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df41b6f8e0ce43a093d9f0561588d671\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a351c7576c3046168a87ac4751b2f8cd : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9518ae584546ea8bb0aca63eb6da7c : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4e42efad1f479e9709bfdd9caaaead : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a70a540aba471385ac65f1e04c5b8e : FAILURE in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10bef2946c3b4b4badf442ee7846181c : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3316769b4a44935aee1a3f56f30876d : SUCCESS in 11m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f5f199617ad24272b6aee5e2f2fcea29 : SUCCESS in 6m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/186f702ca746491ca86cbf8a3e7f1db1 : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5ed0010c5a724d2cacb65e41171f350d : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/517b69fe84e94fd6a4ed2166447887b4 : SUCCESS in 9m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be2dda5cd38e4b349e110950a0454d95 : SUCCESS in 10m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/68b9ed15aa61401386e236f5b15fe85b : SUCCESS in 7m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dee6616ec034477585a8ff507cabc665 : SUCCESS in 2h 00m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f8f5e823a13434da7e165af19d3c5bd : SUCCESS in 2h 28m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3329d556b35245738cdc5a37622f80db : SUCCESS in 44m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad9972672bbd4ee69321a18272fcdc23 : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7950d8d0f932420a89a56683ef35256d : SUCCESS in 1h 16m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1662289a40184ffebab73828db2f58a9 : SUCCESS in 54m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/20c7fdb2e88643088d9a3869e346a701 : SUCCESS in 1h 40m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82733d90cde94f63943acfa7b83e094b : SUCCESS in 2h 27m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6f3b6d3cd39f419b95577463e4aba1c3 : SUCCESS in 2h 24m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32c11bff69af4e4cb720af40371f1adc : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/36d9f1c1a8f748e88ba9c900253cfd36 : SUCCESS in 2h 25m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b267d47a752741bfb85c765d5d1f2652 : SUCCESS in 2h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0d5c071f38d7ea650dc5e902fb7d1121da0813be","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-08-08 11:05:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b9a40d6c3753e140afebfcdc3b100d8020df2d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 13:15:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21cbd3276ab4e4db4cf46dbe89e37cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c20b18a1164c4b8fcfd223fb767d78 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5e2002e1b6408f831e1029703c4d53 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b638b588c3d346fc9571927c2046fce7 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d593ed1d6f24be482ce398dfd529ee6 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dc6e785405d4c76b9862f71320683aa : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ae3f2330dc49a1ac3eaff58ded8b01 : SUCCESS in 10m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6e8dcfc4380a4674bfe534ea0bf4253e : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e723448705a947ae8ce8623d3f6e462e : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfbf13cbdd8148699206278921a9a7e6 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0713cf6368034bafaed8db05833b2315 : SUCCESS in 8m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa0b6d9a180c4038816c8cc96dfb746c : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f90de63c89c844a68a5a2df22e54627b : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9a3c2b554b14497a6d79038262387c4 : SUCCESS in 1h 38m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7080d996fee540058243ac05e5bcecde : SUCCESS in 1h 45m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7df3e021555a47699a6c534d6d111a95 : SUCCESS in 38m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3aa7cb5ca24444589093da45b0f17e6a : SUCCESS in 1h 06m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/10a05c47cf5e4b019f7bea9701312bc4 : SUCCESS in 1h 13m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe3632b0a4ae487f8e11ce13bee7828b : SUCCESS in 38m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b4c958e9d55456d8a21891adb2c722e : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3492b142ac45431a96966c279fdaf282 : SUCCESS in 1h 51m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60960b38db84464bb89516f66a8561a2 : SUCCESS in 1h 57m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6c56eda478044b67b0ec7d9365c570c3 : SUCCESS in 2h 07m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bb695998030b41e78232bbd4bfe34e30 : SUCCESS in 1h 51m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f287de798374b71bd1b8e3130b9630b : SUCCESS in 2h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bfa2cf8354911d1a2cf3dae8e6b550c49b925b5f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-09 07:15:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84c2c9a1e6db21a8dec85d8ca80afa36d1c1a609","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-09 07:24:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe92865291c236cff8c586c996717a2faac7f29a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-09-09 15:33:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"565a2af054a3e12518d50ad4a7c1071a77b67527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:45:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6c2c0aa208476d8e3973bfd2c198b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6053e369c5148a1a587643ab069c3fc : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a1f8daca524de2aa253517e417e947 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0c5d184771849dc8158b32d44ef918e : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35ac3fd6c0a4e9e8887523a3cebe4a1 : SUCCESS in 9m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7b6cb59dcc24eb28bd4eb7316ffbb34 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8a1ca387b041689ecc573ac98ead10 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c88184df6b4bee82f5ec6b38559a8f : SUCCESS in 12m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/291e70eb6856421ca6e883b08fb7cb1c : SUCCESS in 9m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3655a565fd5046248bd755b5abd8a80a : SUCCESS in 10m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ee75b85412c446eb1cf23989113d629 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fdff9e726bc548bfa3038520c152d9ed : SUCCESS in 10m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/71fd302de3064b5682a5ee90ce8abe6c : SUCCESS in 9m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b00642eed3664fb39f96dbd55c2e8c95 : SUCCESS in 12m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24796f803c8e4231bfc0b711e787ebc2 : SUCCESS in 2h 35m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9c381bd006f4dfdb92441dba6bd5c88 : SUCCESS in 2h 23m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37006d370f1b4780b50a653f6171421e : SUCCESS in 59m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad4f9c89e5574143bfb1736258af16c2 : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c150cd0561944fa7a5ead5d1e070f688 : SUCCESS in 1h 36m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/790ee847e2184be1a9db041a64bebad2 : SUCCESS in 54m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b85374203c64a0ba9c0b7f94eca9916 : SUCCESS in 2h 28m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/011f84c3cb174ab58c5c967ed0aba742 : SUCCESS in 3h 08m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a90642889e7744ce90631e0b94d5eabd : SUCCESS in 2h 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd294d840b514b77b35b747498acf570 : SUCCESS in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e791449fd24c42de9951e8e2014d551d : TIMED_OUT in 2h 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/440d5e49870a49db8fc3f5f0ac029f49 : SUCCESS in 2h 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3e9052d1839f6b20788e01d3e51983398606afe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-09-14 12:36:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"996e5c782ec30c1dde092f6b993424406ca4433e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:13:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/257d26b9918542fcb4289f3537ed22fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbea8f77733418a940a814d2be0a7bc : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d19b1bff164f5d9e41b88c8c1b3f65 : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e07c77844b9407b8c51f33c7d4d8628 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/839af9b70cdb48a68be7526b6de9d386 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3decdf2fdbfd4100b204d858e668cfc6 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336ab8c59e8f44708654d7d2053b8280 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215f6fb8fc7842d6952b2a5b28e27a59 : SUCCESS in 5m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/84991e39336f4aada14557ff22b4a6e4 : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dd9f661f2a4a44df94249feb18c6005a : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/60722d75d5244a6ba8c238150a98b917 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c3eee33817f4ccbb9aa3f2f6ff8165a : SUCCESS in 10m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85e35809081b4d2dbe4b2565f93bcc9c : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/757c478368b347d1ad81d0abaf2736ee : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df05a64a7dd0413f97f3b9c5546e1929 : SUCCESS in 1h 34m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f54dfbd70ed847aea8607856f50d94c1 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09c328f1bfcc4a4e9341b9ce4cd396e6 : SUCCESS in 42m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eac6c2e8a74c409c8a6ac2cca1513144 : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0850014341014eb2b61b3d59be22ae46 : SUCCESS in 1h 17m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5c4229c838b4a41bfb26f7c3acda485 : SUCCESS in 59m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91b363c05c314195b5b9685a0788e732 : SUCCESS in 1h 47m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6b460c88b1f4b10a5294a53ea05c52a : SUCCESS in 2h 02m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d489ca21ba5a468b85531ec0e4918ee9 : SUCCESS in 1h 52m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efa939cfaf8645ecb26383d94d8f843a : SUCCESS in 1h 39m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ff665e77ba1048a9803186dc995f28d3 : SUCCESS in 1h 45m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2da4fd5e51004897a39506374b4f3fb9 : SUCCESS in 2h 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7de0ff4892542864c218d2d8b94d095b5fff8d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-09-14 15:17:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"da828b9623303a4e0faf79450fd0dd8c88802923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 17:58:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/979eec84bd36483e8a29b5b955050fba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4ce96fc3d1f4ec39013bd721f3d4ab1 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636c5bb7793f4871a9b86abfb0fbc6e1 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6acd7787281f41239a16af782ed56431 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd8ba8d99323468bb81ab53d911c62d4 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7814adc4545847fc8227a9651928d381 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f415462b4914ad39e870d3815efd54a : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d0ff03f9b740bda8a864d433ce47b6 : SUCCESS in 6m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ae043d3c33f84e17afac4752ad5cb603 : SUCCESS in 7m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2edb8cad0ce44ebe9590cd9a7aba785c : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3b9397da915949599f059c5f7facb010 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5823519f7874a729659ceeb7ad0a432 : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/313d86d130a94b6d8ccad13c52ef6728 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1a6ab92710a4ce489e9fc0b6b49f54c : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5804c20a446e4293a3175dd13a37a677 : SUCCESS in 1h 17m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/adcf0b28f0324be589dc64790d1891fc : SUCCESS in 1h 31m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db8e78add2564337b8e5d8d07fa38eeb : SUCCESS in 43m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e66a5b22fdd8434bac6b238116d1aaec : SUCCESS in 55m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ee919d3653a9424788e84a2f4c8060a1 : SUCCESS in 1h 09m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14190ebce3ab4881a6a8a72374e9f941 : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2dcec6633a2420fbb6f65b6d35b2de5 : SUCCESS in 2h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d468b313c3954424b0466ee4f6691b35 : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/161de11f2c0f4c9b81d6181e9175854b : SUCCESS in 1h 53m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fdb46e0ec58f4a6b9a872e3f62714857 : SUCCESS in 1h 49m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d4ed92d58a4c4afb8969864931595c17 : SUCCESS in 1h 54m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bc04b51dd5440969fac1ed59823abe9 : TIMED_OUT in 2h 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8e52d4fde33e32bbe3b7426e399e07c6b6e781c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-11-11 16:24:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fdbaea60f32f48917cd632ff288636294a46d9a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 18:58:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9476ae14e5be45718ab63c25a44b3972\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/224024061eca41f1bafa052dd2332f37 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3852c9aa717441dd996eb6ca3a903a6d : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca99dc92d87f4fffb6ed8d45102f0027 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/063087a829224e36aad3ae74f47c95f8 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c28b062bb9499fb2fa8f0f6310823e : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15d6c82924a24926933a41043f5644d9 : SUCCESS in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/aea22bf9f9a54b91b4fb04d37438c6fe : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3104a3872b37456fbc12b7fa2f9c4fba : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/761ca66182bf459aa4d51aab8ee16ded : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf4bd23bafb2478aa0730c2992661a33 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17c798f2a6b44d409bfd220a3102c60b : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61422df775f548a099af9b8356927be7 : SUCCESS in 6m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0528d260d9574c69bbae495c9c7ae8a9 : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfc942b0d7c047e2acb82a9f05c76127 : SUCCESS in 1h 25m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f428f52b1e594c67b808f1fce9056882 : SUCCESS in 33m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e98c9be1c9d646c5b8150835b4ecbb95 : SUCCESS in 1h 06m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f45e0f9df1af4dae944278c2546cd1cb : SUCCESS in 1h 13m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cef74b1037d499b8c9bb33b505e3af3 : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a5960155e6704b558d468d7f37aaa1c8 : SUCCESS in 1h 56m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af49dc5ba4a149f68e715f080569335c : SUCCESS in 2h 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd370e4c7ca44c168e4082b28fc2d8b3 : SUCCESS in 1h 52m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d8418b850a04df89000775c19c20e76 : SUCCESS in 1h 26m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c4e0425c6d5647b48cfbdddf48c6b12d : SUCCESS in 1h 39m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/96f506023d714e389f393d271e05d5da : SUCCESS in 1h 39m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"215793c9d5ad91c872fb307c478b99972b234fd7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 11:31:24.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f60f65217150e423279255cce9038d6f0529f5f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-11-14 15:19:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"97739a0e1fde121698462fb36f7300c9f5505c78","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-11-14 15:24:20.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4de75e69418c6efaed24cc42601114716a37489b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 15:31:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ca67074baf232385bda5710d4dea1442198ab700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-11-14 15:48:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"424a7c75bb7726ed8fb1dd3419551b245542127e","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-11-14 15:50:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1e44d50c116f08a7dbd2475a5c9456a6031a4c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:48:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3145a21450684414a5fb576f00beea0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7134a270c5ae418da1a32e235c47bd2d : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b6ccdee86445359b53c9567fe1da22 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f3db883d06d43f1a686996cd046827c : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/039cdc1cc06f472aadb87de6e548ddf1 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76817d0974ab45268226861d0b7a1bc1 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f865e015eb14a3c9bc06132afedb7e8 : SUCCESS in 5m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3a6e4bc175e7459b827d9d598ec1b2ef : SUCCESS in 4m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3546410263ca46279c2aab8bc462580a : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/34984569d210441a85fd2ac9554802ad : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2cbfadbeab5488e9f3ba7a687c40d0b : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/185950d7384b40cc992ec146408519e4 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3fd14b02dc1f4fbf81d929f95dfa0219 : SUCCESS in 7m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/45558a27950f47e988e09a41d8c999bc : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4927a86441a41b0909aefc316e17e88 : SUCCESS in 1h 38m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9aaa635309fc4d4d8f504995ac40cb24 : SUCCESS in 34m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ec2a59ed2974fd3b34a8705e365f0cd : SUCCESS in 48m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/089cfece217d404581216a24b4985d67 : SUCCESS in 1h 11m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee1d33bf53314fb8ba2e5ca50bc4b249 : SUCCESS in 30m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/485198e02be94157aa3650d1736a34b2 : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc16613185e44cd298b6a840251f7c61 : SUCCESS in 1h 46m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/531999aa68994d10b08c6eb332711d36 : SUCCESS in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/83a3ddaac46f48d8978a24557bb5ec0b : SUCCESS in 1h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1df190a866a84c3dbe5d08d7f6d931be : SUCCESS in 1h 55m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fef2130e783e4fd2a91fcb16dd2e7d03 : SUCCESS in 1h 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e6ab8255a9907fe8b3b63ac94354a0d72b530986","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-16 08:45:47.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"6f61e7bec7114ab364df8df77389262330eb8891","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-11-28 11:56:23.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"8de236974e0d65ae60a60768f48cd28c668751a5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-13 10:29:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fa64496b20b8e59ef1468b2bd33d6e9c13d6ef53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-02-13 16:31:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9a813b605ddf7b98b094c867dbd574973c72b757","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-13 16:32:19.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"c91d129fefb10f3114f82b6eebba81370ce834a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:07:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99bb13adea04f3c9f6bd8b494c28ba8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f06902722b3a492fb25a0bf0c8ab59b6 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1cb020b5ca4d66bc8d426a74f3164b : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/488c97f8ed564dd487e63386a5d9cfb4 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ebf809a3e9f4d5ea97ab50d67bdd78b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2d91bddbac4b5ebaba4901a739a1c7 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/068aa8c0a1c94ba0a2e9e9af41a85266 : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/62f9f93bb6db48a3a4f0ab7ce433bcca : SUCCESS in 4m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1c5581cbfdcb4da3b7c1d31f0e1778ef : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fc3dc45988f2415fb0d8f7f84c84d423 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a67f7bd3b6f744aa871ea2e49590c280 : SUCCESS in 8m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09ee72bed89442f9b9f55e7ba178b8cb : SUCCESS in 9m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9038660aba904bb6a9623441e1cf6f27 : SUCCESS in 6m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25614943c50048a994f5ebf34b268eb6 : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9aca0c4cf084942bd0a06fac288b411 : SUCCESS in 2h 11m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9ad1c81e24346ff85348b8ebcc8dc67 : SUCCESS in 51m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b87da2d8eb4c422b9062b3eb757e850d : SUCCESS in 45m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6f4ca3afb82644c28a77e18bc9799136 : SUCCESS in 1h 23m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fc3d1ba7de34215b78624d8c64e66db : SUCCESS in 36m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f7df9003fe64ec1a17c0c780e71a917 : SUCCESS in 2h 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb10561fff0043489e233751f7e06e21 : SUCCESS in 2h 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f69c09eadb44adf87795e391c30682d : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2b7dda51fd1841d787ade0385579bb28 : SUCCESS in 2h 22m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/66aeca46046143978d12416e3d49627c : SUCCESS in 2h 26m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d749c73e815c48398a317725b36f90a9 : TIMED_OUT in 2h 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9c88e6c1d7a623331f03dd9b4e454e96a16d2d2a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:45:12.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"7d9c8d17f4327189b77b9e2e287999bba72fc11c","author":{"_account_id":36795,"name":"Gabriel Hartmann","display_name":"Gabriel Hartmann","email":"gabriel.hartmann@netways.de","username":"ghartmann"},"date":"2024-02-13 08:27:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1981eedd41590e1cfca1160af328c79b06f90d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 10:37:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"452f061f7a101002a537f183af16851f47eb9a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 10:38:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"737ce56848404b8c1217365123ce778d53626b9a","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 10:40:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"94a0e9a6af781f0a2308d035e140719ca590d5a8","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 10:46:46.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5301195b912351d970a61adc0b46fcdf0fe25859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 13:09:54.000000000","message":"Patch Set 12: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32cb43823d5b4b0fac6e468c679f44c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8218e3522fd7434eab1277b5a8931dc0 : FAILURE in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2e89097ba5451aa4c02f731af824c5 : FAILURE in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac970c1a2e8145c89108868417c5adb2 : FAILURE in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f8aeec529754d01b111f53cc1fa62a8 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d058bc556a4b9bab9f575ae352116f : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19860753817742d2942684b05496b65e : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ce3198f9865c4fa185c1b46ed5b9663c : FAILURE in 5m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dca54aa259214edbb4ef8bb31a87ff3f : SUCCESS in 10m 18s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1da9217ef9ca41f3b2b06797f91eb9e0 : SUCCESS in 10m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d325cd7a512a4cc08ecd7577ffd96266 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f13fae907d0460a9ae174c400d0346b : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb13d56304f3471294398ccecc8d28e7 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9aecf77e508341b39b6e5eb5280757a0 : SUCCESS in 9m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de853ce970f54002bb86b442795b6865 : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9ea5c36d0f3b45c382da672009f40fc6 : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/438d040f037a4c9cbac5ebe274373227 : SUCCESS in 59m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4928fde3624b4ce5beb347a3e74b5373 : SUCCESS in 1h 36m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e2200ec198d14d5c927ea3b5602ce9f6 : SUCCESS in 1h 02m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ee366737a744367a13f1259f6ddf800 : SUCCESS in 40m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52244ab9eede424995842a515cfee715 : SUCCESS in 45m 50s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cedeed6ee3d340c5b0b08482ea2186ae : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/42ed5dbde19d4ad88da0b24a75cb29ce : FAILURE in 1h 05m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/579d79a1fca746e294a1acad4b576b6a : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/630f38c5730f45ae9f910b99756e4444 : SUCCESS in 1h 56m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e5c032a11f248dfa36b75f67d8799ff : SUCCESS in 2h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5b11f74c87124c43a7a654b86f36aba3 : FAILURE in 26m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a4d2d6e31504c5c8da4433e0626227f : FAILURE in 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9052d8e573d5874ed6c0ee3033d35becfea38e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-14 15:26:09.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"84393362d997dbe9ae33de511f4dddbec358adea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 18:02:03.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82a5baf7406a4757a9f122da44bd535c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e8b500bf9d4d71bfc45ef03058f54a : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0178f55c26340d0908ed3e829fe3cc2 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e1b131dc7f48bebd2f96e3d9051e1a : FAILURE in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29b297bb96c641d2a3e25d61367d92a3 : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460afc07f53f4c8eb83d8d3ed7f19dcd : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f9c3b672b8474899a160402468f423 : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9551bb9bd317444bb7a5da3bf519e58d : FAILURE in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/71702ade453e4b409ebb0894817e31aa : SUCCESS in 9m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5b99fdcdc63d46bfa3f3311246c704ac : SUCCESS in 11m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c9dee65fb0d4d4897fb38ae2503be7d : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab55399a189f409fa4e058c26b1476eb : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b857af3a83c84f61a27ed99a291ecd96 : SUCCESS in 11m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53e96f57241c4d7f809c235d30138ee3 : SUCCESS in 6m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c826e79494704f24b3f1337bb10a0853 : SUCCESS in 1h 08m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/23021f432d874776a689e67f3c0ec007 : SUCCESS in 2h 00m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/967ac66a19124d5e8d5bb6b0205d7f60 : FAILURE in 1h 25m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9189b35f20724347947a0e8bdf1736f9 : SUCCESS in 2h 19m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cefd5cd190284081ba2fd36791aa1480 : FAILURE in 1h 25m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbcb02b569084c67999c35c52295412f : SUCCESS in 36m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77d2da3da80641189aaaa24779be910b : SUCCESS in 40m 24s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b37f560fa2e44d4db3602d7997313f31 : SUCCESS in 57m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c535605c3e9f4bd4b0427c49476a8500 : SUCCESS in 1h 02m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95c2d9576ac7435cb1f675a3fb7fa7d8 : SUCCESS in 53m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9fb7ef74ef0548558d2a26e065438bd4 : SUCCESS in 2h 12m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/615d9b904bcc4916894c03c0acd9ca8d : SUCCESS in 2h 23m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b21835ffece84245b0f497c77d41a1e3 : FAILURE in 30m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aba063b9fda74d12a8cbbc345b098fa6 : TIMED_OUT in 2h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8fcd5275b13be775f7a1e42faf23a8696de27fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-26 15:58:16.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"f2f5e96ca9194854332e6a37a61faa7d40b020af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:49:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c86fc4051d6d4e63bd8e56e6173de0d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63d6e3ed4bab4cf987f4b315162d5df5 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8fb0716c8c401eb50d624a7af00d67 : FAILURE in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f8f1142a241423ca0ca5d152638d973 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5488132cf5184d51a2dba8e91488878c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308d9819be1f4148ba9011804aed0cdc : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe33f4c8f96d4509b6c556527ec1e2ab : SUCCESS in 6m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6c717f90a5054ed1ab87446d44f1268b : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62431cc1a1744009bce6c40e9f0efe57 : SUCCESS in 8m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/78058a9f15b34632979b07c329bbc239 : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b0970c238e44fda8342a60515957416 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a746e22a0f8a4215aef7dd023d7c725d : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46c5015e02b447d3bb1618c715f29a94 : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5428fda7d5204cfb98a1f403eb20d16c : SUCCESS in 8m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3a4552eb305448ea76e74d6a7a27f08 : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2d6483c49a2649c2905b94e30bdea1be : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cc055dfdc0ca4eb4b46b667aafab29c3 : SUCCESS in 57m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/877e8336ba1b4416ab33f1dd797478d1 : SUCCESS in 1h 38m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2ae8b2d369ac40839f0f1dd7febd7481 : FAILURE in 1h 23m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c614f8b79864d9db97b10417eed6fe4 : SUCCESS in 52m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/112bad6b7c324a2b97fc0fb373396546 : SUCCESS in 53m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7337c377eb5740edbe8b70562e55366c : SUCCESS in 1h 02m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e98fd2ab682a4e3a97d0196aac4f1999 : SUCCESS in 1h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18334c2dc3104dde912e08afaf4bc128 : SUCCESS in 51m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af4af3ced98d41bcb0ae08194a6b4d7a : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/273bdeffde3546c7932b39f0c4e892ef : SUCCESS in 2h 48m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd94f62c4d1948398ec74e8f4aa51694 : FAILURE in 29m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0835081355943e6b0193e8e8469bc3e : TIMED_OUT in 2h 33m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":14},{"id":"d44869af16aaa8901d380f7bd7feae90a7a13124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2024-02-27 13:22:57.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"e369740ebb3b6d5df94c188bfdbb1cff04abb4c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:29:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7aaeae145042d9b21a8c697cf629fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafb81bbbdb448f589865a6c74293cea : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b025d6289c4ad8b20a628a1ed5415f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4648f8de6e5645cb83c44ca5946be4ea : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/907ce2a1a1154a41921ed31c2287d774 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5177987536be40c8a9437eb637398997 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc4382e566c0465eb78f1bdbbbd95ee4 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/11a4922c90924636a367403325b038b6 : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0c51cd69a6d645c1965d68644ce3046f : SUCCESS in 9m 54s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e972fe2383d847899b9e4619d7d64412 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a48d0fb0122b43f3ae454badc83acb9e : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d82b4fc6f71948a8b1ab8b6327977adb : SUCCESS in 9m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5143efbeb1846a6997d81b44546e48d : SUCCESS in 9m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/903f02aef53a4c48b64c1299e0982af3 : SUCCESS in 9m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24a4d855cfaf42d9970ee6b77403460a : SUCCESS in 1h 20m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4a88963e7c824a7ea154f8499030d1d3 : SUCCESS in 2h 19m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80db8d4c6ace4e578c5eb6fc7557d5e0 : SUCCESS in 1h 15m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c797b79d8e94d9b87c01cc8aa94e8dc : SUCCESS in 2h 26m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9438da2619db420493254091f2519b4c : SUCCESS in 1h 29m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/caf1929e567c4289bf48e881b75f5056 : SUCCESS in 43m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0362b58b10024dc4ae45bc164f62aee6 : SUCCESS in 38m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dc5cf5d4fe6545d58a2802e598aef255 : SUCCESS in 37m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/db46841089ff49e3a26aa48162ed56a2 : FAILURE in 9m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56f22b7e110c4f2591c546dea2a24564 : SUCCESS in 32m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4d5dd7860848411dbd94c6a13ffbde1e : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62dfe362dc734b94ac04fe8e28c62cbe : TIMED_OUT in 3h 03m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d34a180a6b844e7aef4367771e8658b : FAILURE in 29m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/832ea75a71454d22aaa81000c09d478c : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5e2130993b2a3d3e0a6e00c72a351d87f6910ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-01-09 17:30:49.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e116d9515c222e21a449e1cee88924cf5542df59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 20:41:25.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f184227e1f46be9fbf743972cf9bdc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c32e957238e54b9680144b1cec76662b : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2722294cc44089912d22106b413aa2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eba6ef47a1c94ababe546d185119c28c : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d7bdec278584f649a83800a7e7c812b : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976f09b34aa04481a5abcee8dfe60169 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8447a9c1adc245078d0671dba810765e : SUCCESS in 4m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f2474ba940924769b17b28549a403106 : SUCCESS in 3m 57s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/43a79476d5dd4de8b4935841df422c86 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6264261b944e4e02b5a4ce4f5e754602 : SUCCESS in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4185c05dd7be462eb4274cada8020074 : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0736f15b915e4514b96569d7492758a3 : SUCCESS in 7m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e53757cbbe914d57b3228aa8190b9455 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0182b863ef04e2eac4cd3bf922fb85d : SUCCESS in 1h 20m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fee9a9565c804fbc9a2eab86ade2b206 : SUCCESS in 2h 13m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/133ff83bc0b7438b96160ebacf1b3b2b : SUCCESS in 1h 08m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eeabf4ebcdef4f8693c8578e6c15ffff : SUCCESS in 2h 19m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f06be3341ed6404fb96f68a738f3e7f2 : SUCCESS in 1h 22m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9954dbc56f84ce09d96407c96d8de8b : SUCCESS in 36m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f90961a2dbc492483dcc298f4e5f798 : SUCCESS in 57m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/548a274cd7c2414cb63a84826e1dfeb2 : SUCCESS in 1h 00m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d40d9e1a8f8c4268a0a8704bed2f40eb : SUCCESS in 35m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/521f6b39f63d4ab7809e46687cbf56ad : SUCCESS in 2h 17m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05dcc0a0ff974542af3b2d8816d18c77 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b3a0756072a4232836828821806639e : SUCCESS in 2h 09m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/248198dca7dc46efa17f01cb951e5e70 : TIMED_OUT in 2h 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"fe5db79dad64e70d3ea508e274bbb056e41ae835","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-12 16:47:35.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"326646ffc656ed0942e630338c8c539ec911e6bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-03-12 16:55:25.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"da9d1aa3af270a9ad442829a8f22c39e52e81622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 19:59:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dac0c9993bf74f85bcda8a41c5f42f9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cfc99ab549a46c3878dee82b2ca8a51 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6cc3e4c3e94404af21f23854e9e201 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60ad44af46554360b30ecda45d19f319 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62d62fbbe8f3487596fd7f8ddc7f99b6 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98849d6c1e14abcbe9f16e29e45b21d : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce88ff548fd41d8982f59eeed2aaae4 : SUCCESS in 4m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d6c60cc8edcd4dc5afd6eecf811c8fc5 : SUCCESS in 4m 42s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1313583af8624c66aa3fca4c95a91b0e : SUCCESS in 5m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f1f4c3d33574245b971f7c412e6afb7 : SUCCESS in 2m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c47abeec778431c8162ca478da7103f : SUCCESS in 4m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6715c7061aa74c1dae7d484ae71e072d : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ee49756fe4c945b6ba664f04ae09fbbf : SUCCESS in 11m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c549a455515c4b3ba938b6f5ab40f01a : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/458d0a4f8b1b493b990936f9d2f9fe56 : SUCCESS in 2h 43m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5f406ee6d65244dd8142a3a36c502f2a : SUCCESS in 1h 21m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91faa9a5f18e40f0b65a4c9a5fb840a3 : SUCCESS in 2h 02m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f30cfe9254d3468a8371b8a290bbd327 : SUCCESS in 1h 20m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/17139a52b8fe4e35a01c774fe0c75819 : POST_FAILURE in 2h 31m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e228942736544e7a369cc7389b4c391 : POST_FAILURE in 1h 31m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a690043461a452aa7a2a52f38c199d8 : SUCCESS in 56m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b20c0999197f4fbc888c9ac4ee87035f : SUCCESS in 1h 07m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ed4744a83b7c4c33a0501205c6f75773 : SUCCESS in 33m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4a1b76b952343c18ee6ad69de30e33f : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96f6d9b66f5a44758dcb692ff7c05851 : SUCCESS in 2h 22m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1330426856314387a8c963074162dc7e : TIMED_OUT in 3h 03m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b094cf9b4d1b45aa909e5d4d27ef8b08 : SUCCESS in 2h 17m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9736628d65140a6aa4ed21ec83df0d4 : TIMED_OUT in 2h 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9cc8894e68e5e4a68855e9cc24374a8c85358930","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-03-13 12:06:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"85c342e40c76fa0a76f906b264a2671135fa4cc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 14:40:56.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f62ebbc31461473d931d8800586a6841\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f21b77852444e93aae88ed4e4437c30 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7519a9df62d450abb870b6a56c82fb1 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05b1639f7f084b2c9417d4c88db525f3 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/601362c3da9d4741b0988326dccea479 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee6562055c04103b62524eba67862ac : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3f87f48c2c94e099f533d8b18c4360d : SUCCESS in 4m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/709377a9e957407dbb72a9646f6a27e4 : SUCCESS in 4m 23s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f87421ef8e2d40a2b95d7cdecfb02ffc : SUCCESS in 6m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/798503017ef34e45b08cef687bcbdf30 : SUCCESS in 3m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff3cc059136344caa5410d7e94c224c3 : SUCCESS in 4m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e50376de8784d3aae21ea6f25b3269f : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32c49da98b2a46a0819152e5868efe03 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f45d09a72a7c4b24b94b1bb82c460143 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b9eaada2d884444faf5c710b7043f4fd : SUCCESS in 1h 49m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6299a2b9353b46b5a71d989482d49591 : SUCCESS in 58m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37108b0ed3f5442cb0ea12f00bdf540d : SUCCESS in 1h 49m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48c96ef8a6cf436c83eabba3f4ddfaeb : SUCCESS in 1h 13m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5136009a72f4420c9fb34879c5d0945e : SUCCESS in 1h 52m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a7fff688465544f28a0bae6a7f85645b : SUCCESS in 54m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5410cd17fd0a46c181aad38667a8085a : SUCCESS in 27m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2829f33cb6804200acb825478d7fd95f : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3b0eb0a568ea410cbdec6283ea9da004 : SUCCESS in 1h 07m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3558b4650a774db9881ba9f00e227f04 : SUCCESS in 37m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4f73d3520ea44d6498a1efdac2d92821 : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e111f09fad2476d829e1cb7c374897f : SUCCESS in 2h 20m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/590d6a9108f14415b800595f55643871 : SUCCESS in 2h 30m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a01740f9d537409fb28334f4b0d24966 : SUCCESS in 2h 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"5bad0eab1c2fe6eed18de1443d006e34575ab52a","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-18 13:31:10.000000000","message":"Patch Set 17: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"146b1fc7c1134d8c569e174e221b0463537fd251","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2025-03-21 16:25:12.000000000","message":"Patch Set 17: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"d11ee61a23eca1b8579626cf3d96633abf50be72","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2025-05-15 08:39:25.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"4bc33019b4abd5e2e39ac8f43df3571f3612b680","author":{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},"date":"2025-05-15 08:57:46.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"881c3b1c01e314f6fdf5e279463f9315df6d4be7","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2025-05-15 09:08:14.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"7f20fef49c327409b28308dcdf536148c9948960","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-10-03 09:48:32.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"5a3ace615e78b5afcb7b796776d791b61f05030f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-10-03 12:33:45.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"4260d05eb2d1ace0861b822826a2cd9b7bdd1415","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-10-03 12:39:22.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"5487127f5895b7a3f99dd7b467ffe4b66629d67a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:38:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b92a2f4ff3e74e649ca54b303ae7ecae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd611c282c904b0ca308a1fd2b5bcdb5 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7cae3129d124e8e95fc908a586ec923 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f0553876c1f469fb06cf2a0f9accc09 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fefd0f7383c4020893fea58c9cf5adb : FAILURE in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2c6903e67124e42b4f770b3e26e304a : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363a2eb09aeb4be6911f6bd54f153cb6 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/621bde9fe75348a297f1b69259c8fba2 : SUCCESS in 4m 58s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5aefb96e216841ed93238afac1997392 : SUCCESS in 5m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9a601e74dfd1417f95fd7671afa2332c : SUCCESS in 6m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7a63fa5ea9634f138ffa46b68cb6ffef : SUCCESS in 4m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0deb73355684f70b21b2e6ce398457b : SUCCESS in 4m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a65417ec685c4671bbfb59c0f03a1928 : SUCCESS in 4m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03815523f5bc4648a0f5381b3cb47f35 : SUCCESS in 40m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3991d1ddbda04a2b847a90da6b2be8d3 : SUCCESS in 1h 58m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dbc97b44318b4ad8a03cca9094f66352 : SUCCESS in 1h 11m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/644405967e4244ae8aa2b9ad4b852d60 : SUCCESS in 2h 03m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0e35fe4244b14ce6ab0c0cfe7f5996ad : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e5bfd61c96bf483ead15c79115e8a46a : SUCCESS in 38m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d4db53d906ef4afab191d6696ced96d3 : SUCCESS in 42m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/579f075da7484fccbfeca7e6e7006b7c : SUCCESS in 40m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5472623841f54cb69f81b0ca9b080d71 : SUCCESS in 2h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d8f85d1a2b7473c99278707c0475c4d : TIMED_OUT in 3h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/786780be0e5840a7969810bafc197a87 : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1751247ac5e887f79c496258953c4b3d0c84dfeb","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-10-03 18:14:31.000000000","message":"Patch Set 18: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c87ef05668c7ff16177f4d73c097b2f5e6311238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 21:16:59.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1ddb007ede4730a3d00a571dbf3067\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a085fd6d51947a2a6292ec9d8bf390d : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad7e011a79a142b0b20a66d4becbde52 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31f1ece87a034cd7bc268cb97cab826c : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15089045df2a4b11a9ca5b49eb02c40d : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e062b2cae864e37b77efa441b386a7a : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6058f4ded19141d5be2dc80618780d9d : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c7f63495f13434cb1c042a868194707 : SUCCESS in 5m 41s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d08c53f8c5ca4e0795f0cdba62f597ed : SUCCESS in 3m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/50485fb4cc364ec5ba11d4c7ce3b99ec : SUCCESS in 5m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b7d6cc975e24adcb73c934bd2f8eb16 : SUCCESS in 2m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a54106c380a646e7b926b2b969d3801f : SUCCESS in 7m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a05f4861ada242c090436fbf76ccfb90 : SUCCESS in 4m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13b406b4ed454af3869ff2da21d22d0d : SUCCESS in 1h 11m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d7096768fc954416a2ebffe2f3437d5b : SUCCESS in 1h 53m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e2758b848de74287a091dea7ab325e34 : SUCCESS in 1h 09m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1280f15a554f4c35b3636a9a866f513c : SUCCESS in 1h 55m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0f80ea9b9c7f40aaa12f28426516c44d : SUCCESS in 1h 20m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48099269a4084165831da67680702c5d : SUCCESS in 38m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f4b48d2bba04c8f964334ed1d4e21bb : SUCCESS in 45m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4304e42779ed44a4ac8ad419abd5dd6c : SUCCESS in 34m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/32a241937d7244aea109c572d1868455 : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/99f7712c252045ba87cd66edfccaaa0d : SUCCESS in 3h 01m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91f14aec998b434eb753d1ef49e1a425 : TIMED_OUT in 2h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"93ee71936bd37071a077ccd34506f8ceb0d0c30f","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2025-10-04 09:40:21.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"4668b142af4122c253a94b9ef1f553d91093a00f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"real_author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-12-02 10:41:36.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_37685\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}],"_revision_number":19},{"id":"205286d59c64d52b0d58021f37d880f84a926f1e","author":{"_account_id":38031,"name":"Maxim Tsibulia","display_name":"Maxim Tsibulia","email":"madkokozzi@gmail.com","username":"madkokozzi"},"date":"2025-12-02 10:45:32.000000000","message":"Patch Set 19: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"1efd2abe856ee42195a17824a7a94c4ac7f18e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 13:05:02.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c96abac2ad4c4840b719a53e22306b65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/109b1385cbaa468bb78eee7f79d12d3c : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00747c04a903474fa6ddfebaa9d3ebb5 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c315a64ddef3491ca5b9ed5d48d74d11 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/269a787199e14f538dc2cb27cbd4b20a : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ab3821168b44e7b6737426ae0d06ff : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab8673582e6544eeaffdcf99b1df6176 : SUCCESS in 5m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e599b9ec22a94cda8aaf2f142d1c9304 : SUCCESS in 4m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/db3c9ba6037f4031ac07690770663795 : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a36d42f3444c4d11982af77acaf008d8 : SUCCESS in 4m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf6db997e2034a83b59fa691d53f8d29 : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3364afa0058f4a0fa5f3d6c8a7f98a4a : SUCCESS in 5m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc52c7a1900645269dfbae58f7d68df9 : SUCCESS in 35m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bd50793cff16437dabec32bee2e12a05 : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80ca7faaa2ba4087a4ea8e58170e9aa3 : SUCCESS in 1h 00m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c73b70b205014e198f2bc4912b94fe40 : SUCCESS in 1h 41m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf35df9656e147c1bc427d69669c618f : SUCCESS in 1h 09m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e432c657b194a21b7c8adb87ed9a49d : SUCCESS in 35m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14d3714b21c048aab2fd1aa88c2f4998 : SUCCESS in 37m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a21fc47681540efa24fe65e74ec7e4f : SUCCESS in 35m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/93dccc918bb64f9a95e776ca136198cd : SUCCESS in 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9300e9b08bf04138b62f5e3e3f0be180 : SUCCESS in 2h 22m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/30f2b83d315544d89688f7def45ecb0b : SUCCESS in 1h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"328491f7742667616fcbb4e34c25717f5fe51449","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-04 09:56:48.000000000","message":"Patch Set 19: Backport-Candidate-2 Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":19},{"id":"450e16ac00672fb46f271273ec032b8c07331830","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-04 12:41:52.000000000","message":"Patch Set 19: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":19},{"id":"cb972f10f14f9b2e4e55533da7bd07aea3cefe7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-12-09 11:16:36.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Backport-Candidate+1, Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"45de32950394b5066498b0fc36145ec80ecf53b4","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-12-09 11:25:13.000000000","message":"Patch Set 20:\n\n(10 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f4941d5d3a4c8d7e3fcf26bb83f8c0d35aa13ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 14:05:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ede2da60ad7423999955ce44d0b7b6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c51054d5d92f41e19ba4d8977485b4fb : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce477c9ecc8c40109c7fcca22ac3caf8 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8b1cf6d74544a0ea1efff8b2b160b64 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2cba7466d8b41f4ba9ca07106dee8b3 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbded105309c4f04a5b1eed163be14f9 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89e3c6226a3a48149bd5256654e1fd18 : SUCCESS in 5m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9c80d1350af74a7ea8cf745ce662ed8b : SUCCESS in 3m 09s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/05270d5d5d164df587253dfc4b33f8f6 : SUCCESS in 6m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/edd09296621d46d992a23289c2626459 : SUCCESS in 3m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ec7ed9c8511459f99527d110558a8cc : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/42a5dc28f28a4368b08fcd345d0f1696 : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6bcd5410c8442ad8f645d41e61b24eb : FAILURE in 1h 08m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c85b89f815124553b919f5aed8284783 : SUCCESS in 1h 31m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d5f27f4f262b4378bd3c2e4546139682 : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6dd3ca692b7e4901ade21c0d9c09cf51 : SUCCESS in 1h 40m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/81feb249073e43fdbc8017a2af2988a8 : SUCCESS in 1h 06m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e5657f14a0d4506bea68790bf93046b : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3765fdb1e9bc439ebb9e7ab636e70083 : SUCCESS in 41m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f0d3bcffac44d4bb8ebb15cf46fd049 : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b6332110720b471fabb07edf8e354f5a : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0119417d4544877b575ae67fecd0b53 : SUCCESS in 2h 45m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a8f65a945fdc4038b548a2d3599f2a05 : SUCCESS in 1h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"24265c6a3a7bfa94ae81d15b4e5f44c838be90a4","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-12-09 14:39:06.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"570f7d8d7901a0e735855f7b4d2375be222c078e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-09 14:46:53.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a3f8f2b3dda995314c34a1c7efd74dfbf88f3d8f","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2025-12-09 16:00:17.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3bdee924a1cc68729c17a11cfbd020e8d4f2142d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 17:43:16.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62d2b9d5b6b4efea572644bde34e4e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea839d33f26e4b2dbcefc15046176c81 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba3c89ca4e8433cbc86bb293c6c65ce : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db3e9a9a69c1430c9024703ff5b535f1 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f84228684fd4c7093ae2210feaab525 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b3586eecd04864bc9842784f31017d : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/982cc9bb293f42a3b07556c15a205b96 : SUCCESS in 2m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c08bb5c1c0d043dca66d072dbfbec256 : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8b8552e2dc1c48efa0b231b7938bda35 : SUCCESS in 4m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/502e9df075314d8cbae8743e5745cc69 : SUCCESS in 3m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8f0e4238c1049e3b1422b4c00951765 : SUCCESS in 7m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e2f13d4b7a8e419c8e2b06cffa30f691 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cbc23666d2946d08d7b2c93aa6063fa : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0c1ea38c4819416ea6d60c82958f7450 : SUCCESS in 1h 39m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/837136d29df947228d1eb6080e5a608f : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5852d18f9c9c4e8e8a23eefb910aa5a7 : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8f66a4c0218243ed92c862b580b51b77 : SUCCESS in 1h 27m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3315e24e63bd4828ba35a93e40662c50 : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e00ecdfff05948409d36fc2da1bac3d6 : SUCCESS in 48m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ffecdec254547cc86ba81b7c023d6d5 : SUCCESS in 39m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79dacbeae0df483683567119c931278c : SUCCESS in 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1fc6b788cbb483e97980c4e578a94c2 : SUCCESS in 3h 02m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b8d41318a88e4a2aa053140697497496 : SUCCESS in 1h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"07bb26cb4f4707f46319640e17dbc78f8ad596e9","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-12-10 17:51:40.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"a29095e81005e0df49d32cfdf122110b9e1e2ee4","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-11 15:00:24.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"1e25b4626a9491e13371836c55d4ce2f32aea831","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2025-12-12 12:09:53.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"07032f8cd8ed0f77e5e9e57aa9db68bf8e454ec6","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-15 12:20:57.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"d01a6b41000dabf9dc762e0795f165e82964c063","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-12-15 14:48:33.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"df4a10bf1b9b3b743dd42f1b83f872f04f81806c","revisions":{"8a0d864bdb5db7beb5b708a2f8d6a8da8a89a95e":{"kind":"REWORK","_number":1,"created":"2022-08-05 13:18:15.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/1"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:18:11.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a0d864bdb5db7beb5b708a2f8d6a8da8a89a95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a0d864bdb5db7beb5b708a2f8d6a8da8a89a95e"}]},"branch":"refs/heads/master"},"a4070c81b118fe2e775e879148bbf7cf747d1035":{"kind":"REWORK","_number":2,"created":"2022-08-05 14:31:52.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/2"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 14:31:38.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4070c81b118fe2e775e879148bbf7cf747d1035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4070c81b118fe2e775e879148bbf7cf747d1035"}]},"branch":"refs/heads/master"},"4666115e92ad2173b8259ef5f436b5e3d84b5587":{"kind":"REWORK","_number":3,"created":"2022-08-05 15:01:20.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/3"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 14:51:57.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4666115e92ad2173b8259ef5f436b5e3d84b5587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4666115e92ad2173b8259ef5f436b5e3d84b5587"}]},"branch":"refs/heads/master"},"1961da02a5c24cc8e8d85c3a275ff754fb976066":{"kind":"REWORK","_number":4,"created":"2022-09-09 15:33:29.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/4"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-09-09 15:33:10.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1961da02a5c24cc8e8d85c3a275ff754fb976066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1961da02a5c24cc8e8d85c3a275ff754fb976066"}]},"branch":"refs/heads/master"},"1ce45f8510f673dadfd10d2d7f3ecfc8272ef440":{"kind":"REWORK","_number":5,"created":"2022-09-14 12:36:43.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/5"}}},"commit":{"parents":[{"commit":"8caca9219f36c32a4ed7fd28d03206a49a15a083","subject":"Merge \"Add missing parameter in amphorav1 batch member update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8caca9219f36c32a4ed7fd28d03206a49a15a083"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-09-14 12:36:13.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ce45f8510f673dadfd10d2d7f3ecfc8272ef440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ce45f8510f673dadfd10d2d7f3ecfc8272ef440"}]},"branch":"refs/heads/master"},"b87de150f44e6244cf2fdad098b547d3f47c06a9":{"kind":"REWORK","_number":6,"created":"2022-09-14 15:17:24.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/6"}}},"commit":{"parents":[{"commit":"8caca9219f36c32a4ed7fd28d03206a49a15a083","subject":"Merge \"Add missing parameter in amphorav1 batch member update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8caca9219f36c32a4ed7fd28d03206a49a15a083"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-09-14 15:17:14.000000000","tz":120},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b87de150f44e6244cf2fdad098b547d3f47c06a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b87de150f44e6244cf2fdad098b547d3f47c06a9"}]},"branch":"refs/heads/master"},"d9a8c88628d0a8d051bee011e8a2e72ec6ae7bd5":{"kind":"REWORK","_number":7,"created":"2022-11-11 16:24:46.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/7"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-11-11 16:24:33.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9a8c88628d0a8d051bee011e8a2e72ec6ae7bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9a8c88628d0a8d051bee011e8a2e72ec6ae7bd5"}]},"branch":"refs/heads/master"},"ef3dd2a63ca68f9a534c868eb61fa3df68684bfe":{"kind":"REWORK","_number":8,"created":"2022-11-14 15:19:24.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/8"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-11-14 15:19:11.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef3dd2a63ca68f9a534c868eb61fa3df68684bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef3dd2a63ca68f9a534c868eb61fa3df68684bfe"}]},"branch":"refs/heads/master"},"f1d6124e40b51ead3ff8fd31defb5db9b82787f7":{"kind":"REWORK","_number":9,"created":"2022-11-14 15:48:53.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/9"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-11-14 15:48:43.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1d6124e40b51ead3ff8fd31defb5db9b82787f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1d6124e40b51ead3ff8fd31defb5db9b82787f7"}]},"branch":"refs/heads/master"},"755b42567e03f9e05a3178ff763d978272562068":{"kind":"REWORK","_number":10,"created":"2023-02-13 16:31:08.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/10"}}},"commit":{"parents":[{"commit":"55b2d70c4d75566bd5eedcb7bb53b78c361a50a0","subject":"Merge \"Ensure options are dumped in debug logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55b2d70c4d75566bd5eedcb7bb53b78c361a50a0"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2023-02-13 16:31:06.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/755b42567e03f9e05a3178ff763d978272562068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/755b42567e03f9e05a3178ff763d978272562068"}]},"branch":"refs/heads/master"},"370c6613eae3e2e9f7fdf9ba798514b6035600f7":{"kind":"REWORK","_number":11,"created":"2024-02-14 10:37:10.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/11"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-14 10:37:08.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370c6613eae3e2e9f7fdf9ba798514b6035600f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370c6613eae3e2e9f7fdf9ba798514b6035600f7"}]},"branch":"refs/heads/master"},"1691d538a40590caf0eeaaca3d6596c9d0e37c7e":{"kind":"REWORK","_number":12,"created":"2024-02-14 10:38:44.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/12"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-14 10:38:33.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1691d538a40590caf0eeaaca3d6596c9d0e37c7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1691d538a40590caf0eeaaca3d6596c9d0e37c7e"}]},"branch":"refs/heads/master"},"8facdd0999c17615e4069bf5e0ba52eda8d15bd5":{"kind":"REWORK","_number":13,"created":"2024-02-14 15:26:09.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/13"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-14 15:25:17.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8facdd0999c17615e4069bf5e0ba52eda8d15bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8facdd0999c17615e4069bf5e0ba52eda8d15bd5"}]},"branch":"refs/heads/master"},"27792a0989655efdf696c836933ac9913e980f94":{"kind":"REWORK","_number":14,"created":"2024-02-26 15:58:16.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/14"}}},"commit":{"parents":[{"commit":"46f4020e7031a1032d3fb69ab3964a060f4da506","subject":"Merge \"Bump hacking in pre commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46f4020e7031a1032d3fb69ab3964a060f4da506"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-26 15:58:05.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27792a0989655efdf696c836933ac9913e980f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27792a0989655efdf696c836933ac9913e980f94"}]},"branch":"refs/heads/master"},"48491ddeceeb4269081a29b1690eb0c1bb511b3c":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-02-27 13:22:57.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/15"}}},"commit":{"parents":[{"commit":"ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d","subject":"Remove some unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2024-02-27 13:22:45.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48491ddeceeb4269081a29b1690eb0c1bb511b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48491ddeceeb4269081a29b1690eb0c1bb511b3c"}]},"branch":"refs/heads/master"},"66f2e04535bc71d0a51d124eb9a3a373aaab38f0":{"kind":"REWORK","_number":16,"created":"2025-01-09 17:30:49.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/16"}}},"commit":{"parents":[{"commit":"9f0634c23a5b20352ba27eb971550409c91a7880","subject":"Merge \"Unpin pip-extra-reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f0634c23a5b20352ba27eb971550409c91a7880"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-01-09 17:30:46.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f2e04535bc71d0a51d124eb9a3a373aaab38f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f2e04535bc71d0a51d124eb9a3a373aaab38f0"}]},"branch":"refs/heads/master"},"1e8e89d1ffc0e78d0c3126c5df4443c7227e433b":{"kind":"REWORK","_number":17,"created":"2025-03-12 16:55:25.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/17"}}},"commit":{"parents":[{"commit":"27f5dd211d95679b57193b27ba74015d2e1bfde5","subject":"Merge \"Drop VIRTUAL_ENV\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f5dd211d95679b57193b27ba74015d2e1bfde5"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-03-12 16:55:22.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e8e89d1ffc0e78d0c3126c5df4443c7227e433b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e8e89d1ffc0e78d0c3126c5df4443c7227e433b"}]},"branch":"refs/heads/master"},"27ef4a5f410247cbcb4285ed37da475dae1f5e36":{"kind":"REWORK","_number":18,"created":"2025-10-03 12:33:45.000000000","uploader":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"ref":"refs/changes/69/852269/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/18"}}},"commit":{"parents":[{"commit":"40e13e517b3000a0e4b305fdfebe0c12d339f2bd","subject":"Merge \"Remove unused amphora flow and status constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40e13e517b3000a0e4b305fdfebe0c12d339f2bd"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Konstantin Eremin","email":"k0dt@k0dt.ru","date":"2025-10-03 12:33:10.000000000","tz":180},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\nSigned-off-by: Konstantin Eremin \u003ck0dt@k0dt.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27ef4a5f410247cbcb4285ed37da475dae1f5e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27ef4a5f410247cbcb4285ed37da475dae1f5e36"}]},"branch":"refs/heads/master"},"14834069fcdc0dd5846dfdb8bdcd3a0241815263":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-12-02 10:41:36.000000000","uploader":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"real_uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/69/852269/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/19"}}},"commit":{"parents":[{"commit":"bc9a76dc072254d0173185d9791551b4cd9c232b","subject":"Merge \"Remove option for apache \u003c 2.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc9a76dc072254d0173185d9791551b4cd9c232b"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Konstantin","email":"k0dt@k0dt.ru","date":"2025-12-02 10:41:36.000000000","tz":0},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\nSigned-off-by: Konstantin Eremin \u003ck0dt@k0dt.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14834069fcdc0dd5846dfdb8bdcd3a0241815263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14834069fcdc0dd5846dfdb8bdcd3a0241815263"}]},"branch":"refs/heads/master","description":"Rebase"},"df4a10bf1b9b3b743dd42f1b83f872f04f81806c":{"kind":"REWORK","_number":20,"created":"2025-12-09 11:16:36.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/69/852269/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/852269/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/852269/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/852269/20"}}},"commit":{"parents":[{"commit":"bf16995711034b65a761eb85fcd786eb53edc0d4","subject":"Merge \"Unpin pylint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf16995711034b65a761eb85fcd786eb53edc0d4"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-08-05 13:16:17.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2025-12-09 11:16:14.000000000","tz":60},"subject":"Health sender: support to send via TCP","message":"Health sender: support to send via TCP\n\nAdd a feature to send larger health status messages via TCP,\nin order to avoid fragmented UDP messages which may not arrive at the\nreceiver in some networks.\nThe new configuration option heartbeat_use_tcp_threshold determines\nat what size of the message TCP will be chosen (-1 to never use TCP).\n\nStory: 2010216\nTask: 45958\nChange-Id: I8f90cb3372fb33ea5601d33677585aec8080e985\nSigned-off-by: Konstantin Eremin \u003ck0dt@k0dt.ru\u003e\nSigned-off-by: Bodo Petermann \u003cb.petermann@syseleven.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df4a10bf1b9b3b743dd42f1b83f872f04f81806c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df4a10bf1b9b3b743dd42f1b83f872f04f81806c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
