)]}'
{"id":"openstack%2Foctavia~852459","triplet_id":"openstack%2Foctavia~master~I0b8cf3cc8dbe5bc2babc469e73e32b2ed9ccb98a","project":"openstack/octavia","branch":"master","topic":"story/2010220","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-09-01 08:25:13.000000000","reason":"Change was submitted"},"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2022-09-01 08:25:13.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-01 06:06:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I0b8cf3cc8dbe5bc2babc469e73e32b2ed9ccb98a","subject":"Fix listener creation allowing pool protocol","status":"MERGED","created":"2022-08-08 16:48:02.000000000","updated":"2022-09-01 08:26:54.000000000","submitted":"2022-09-01 08:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852459-story/2010220","meta_rev_id":"1dbdc710577f0c3c3826ed6649a6e84d1abe876d","_number":852459,"virtual_id_number":852459,"owner":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-01 08:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"all":[{"value":1,"date":"2022-08-29 15:14:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 16:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-01 06:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-01 06:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 19:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 07:34:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-08-11 12:23:41.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2022-08-11 15:30:05.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2022-08-18 12:45:27.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-08-31 16:27:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"61255a382131ec33bc3343b93ffb90ad27dfa25e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-08 16:48:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d201a0d385cefdd26252658b86298e0d9f9e9648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 19:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd49e47ef46c495e92bfe8a90a4e8ce7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/360ee3d0e7a44d389266da2b4d0939f8 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6a368035674a29a56b1894a855e0f0 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e203ff9655c4d43b714c016700ab062 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b0c64857cc4c4bbe19ce057f673329 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05d659e58450476d92fae3285e63a553 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a10655316c84ac0bb4c7874ad6a64ca : SUCCESS in 10m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2a0ff01b7c1d43f18df5397efbc828c3 : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/44caab6b62c54eff8ef0ed4e3dd9131f : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25e3feacd2f14d309b181ad5f8d9c4e3 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39fdbca8ec77417aab6ef51983647afb : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6690d5be5aff4ea18290cba9e4589999 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc1abc48c1604dbfadbf3f54bb382fdc : SUCCESS in 8m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47aabc1bb9764703a893c2362307df62 : SUCCESS in 2h 00m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52fdb4f16d8248c6846a7ad8cda1e42f : SUCCESS in 1h 51m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36b7ca8e3855467eb3b443dcc5674ba0 : SUCCESS in 42m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f134846e3bdc4facbad2e98be14c5b14 : SUCCESS in 1h 10m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/73fe775d45024e8b9412bc8998ed201c : SUCCESS in 1h 04m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93cd318369bc48be8982d69308cab20c : SUCCESS in 39m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/23a753a279dd4ee89ee7d86ac87d21b2 : SUCCESS in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2374034b1f946b8a81d814f0c3f0e34 : SUCCESS in 2h 18m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/753a05d4a56e43008d8ab630209d9ee9 : SUCCESS in 2h 11m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b39f4283ebd4b95b22dcf34c5da6521 : SUCCESS in 2h 05m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5f69ad1427e24f4fb8fcb6d09c0f88be : SUCCESS in 2h 01m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5f69e2b06649d7a02b5993d557b06d : SUCCESS in 1h 51m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32e6e23f133c7418b49cd208d3e2fc929b083113","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-09 07:34:38.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"61923fc7fde3a20c8d809090c405d1f931ec48c6","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-09 09:00:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"633b3ecc9047fae446afbf3912a5943db10a6807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-09 09:48:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c38d2c0c8d88842ffadc64c99d10a42eaa28ce9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:31:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07b75c03a91348e9a47116e8bce9a9ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e373eb7b8f544ff391ef5e1b73161500 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a329bf4688a74cefa77185bdbd67f019 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b42bfb4cc3b47d9a0b95035f1f47601 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1730b36a784449f5bd2c180d98f506a5 : SUCCESS in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af001703067242009594f184b0297d37 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd63af25878473cb5d45867cead4240 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e540b207f114b37a5cf281d937798af : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6404d5f87fb94aaebc03af9020150ae4 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b1c72b3d25e844bca10e52f4ebb010f5 : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/749cfa975d194cadbfab157d08fe9fa6 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7590badaf374d8ab5cb8b73f0c2176f : SUCCESS in 8m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ac35e45c15d40a6abd2707ec9cec1c6 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b4c375692e148339cb883e4b0a3d926 : SUCCESS in 8m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1676f3eb681c495a82ff22d19901131f : SUCCESS in 2h 38m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c82afcfaf6214064a5362d7022f8f751 : SUCCESS in 1h 30m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7d372c8c58347f5924ecd028f13cda3 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0286beb6cddb40759c5175810c00627f : SUCCESS in 56m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/82fd55d23de1486894a64e9bda894fc4 : SUCCESS in 1h 09m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/469c1e46c5394c58b024d49c8b4157b4 : SUCCESS in 45m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b471ebedce8b428baca17874c3bb6316 : SUCCESS in 1h 45m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e8e4f1115d74e38bab9db6a699521c3 : SUCCESS in 1h 54m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac0bdfdb46cc4b29a1defc8a1bd1b5b4 : SUCCESS in 1h 52m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aa2dabd8e72848f7a290b9702c021a92 : SUCCESS in 1h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/acbb17e2785347838b3da384f3070596 : SUCCESS in 1h 35m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9ff752fc76d48e4866c8545794ef9ce : SUCCESS in 1h 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5f23d4cbc23335a9c8b62eff821d2293fa506d3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-08-11 12:23:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b5e6923f11e13f27392199c77a57b841726d0a6f","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-08-11 15:30:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"99d79d1fa7efa0673b6bc0b2f128a9231edb8297","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-08-18 12:45:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b28998ac3964deadbdb40b45482c017f0cd3b62c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-29 15:12:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1abc8764e9d555ab5ef7e43ce7d3bbb931221a42","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-08-29 15:14:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"29e33d912d84d58ff563403e32748e7c362f98a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 18:01:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5bb02a1f43443284447a2311772e5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/626cebd2b56f4319964489cdfd7e8202 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d042695946e4202be5e692bcad1da7b : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/189c33a214a14fd1bfa06a815beb16fe : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfbea903303e480193b8b74c91e43871 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c9fb584c55d401a977369fa3927a1a7 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32792f33e1d74b40b3246865738c47d0 : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95f5021856d841d3bd31b063cd63c2b2 : SUCCESS in 6m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a2d86817c46f495f85014cd9cd321258 : SUCCESS in 7m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/086508adbcde42148b596554a60cf284 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd0a66e2add74d6fbeac5831a33ed2b0 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5868e700233343839ed3a3c8bebf5b3c : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/338d57126d2149dd9ecdd23e0c794edf : SUCCESS in 9m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1bd697f01234cfe94e6bd36ac1c1b65 : SUCCESS in 8m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8424890aa6fe4b16a229d8d08d2d2b40 : SUCCESS in 2h 36m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be16241e63184ebea6b48cd79e64d7fd : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e128162e11674c18b20e918f7582130d : SUCCESS in 52m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01d32c2d39e648c599bf07d0523556fa : SUCCESS in 58m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/24acd4de51e04d24be017103b3b370f3 : SUCCESS in 1h 11m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a1a3843c0c043f98b6f8c45288a2f01 : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e499c6461a2e48f79c8373993d792380 : SUCCESS in 1h 39m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6f175638e4048fe900adf5f2007c787 : SUCCESS in 1h 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6a5ee83d84d468283bda4d79c4a964a : SUCCESS in 1h 39m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e58bdf19d6a74b3988316563d1e3f0a7 : SUCCESS in 1h 40m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd75b3328e2e470f812629b8c34916dd : SUCCESS in 1h 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7067006d101541a4a5b55e39b74ef6f2 : SUCCESS in 2h 07m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68a52b8f6bca156b8742a5a7b68ca86e4c6707e8","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-30 07:13:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f244477944afcf89dedf44c4340d1e11fdf8fb72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 16:27:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20b537c2508769613874de1472b200851edfa85d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-01 06:06:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2a201cb7e021ddde84313fbda6a5908005dc894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 06:07:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3caa6841c018cba5c0715d205bc8d18b775aa56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:25:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad7f53455d3e482ca674741d6032aa4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518826fb0ea04785bd63c07b8b3555ad : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766ecea163e947ae9dda3d10904c4e7b : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5c01fdbeb244633a4f6308c7554bd65 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67d30b50b9e4d3db0f81abe22f6b778 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/061a29b70fff459985b66c1362fa2099 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1067fc7750f14f7eac4b98c8a8a31b57 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb01bff795f3474984ff22154d640836 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a2fd75ae4fa47b6ac787321396dea7d : SUCCESS in 2h 13m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c7b970abca948e3b40ac0f7af449e84 : SUCCESS in 2h 16m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19eb50390b8b4b538c722fa2e28c841f : SUCCESS in 45m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c310dc1c7e648a1ab841851376074f2 : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":3},{"id":"62763a5a4d4a19c6397cd70272df2be8165192d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:25:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1dbdc710577f0c3c3826ed6649a6e84d1abe876d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:26:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23dabf827b7949d78ee4e421c481788c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef713a4e81746518c708cdc80e095cf : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dea9546c16544f2a2435cf57bd9e770 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"05b4d361721264dbb0ff9fec4bc9fa87779745a7","revisions":{"2efcd820b5d1889d5701ac8c60ae129a26995575":{"kind":"REWORK","_number":1,"created":"2022-08-08 16:48:02.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/59/852459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/852459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/852459/1"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-08 16:40:47.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-08 16:47:47.000000000","tz":120},"subject":"Fix listener creation allowing pool protocol","message":"Fix listener creation allowing pool protocol\n\nListener can be created using proxy and proxyv2 protocol which are pool\nprotocol and not listener protocol.\n\nStory: 2010220\nTask: 45968\nChange-Id: I0b8cf3cc8dbe5bc2babc469e73e32b2ed9ccb98a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2efcd820b5d1889d5701ac8c60ae129a26995575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2efcd820b5d1889d5701ac8c60ae129a26995575"}]},"branch":"refs/heads/master"},"6c712a299ee43b9345db9f2c7ca3e5f0f981c06d":{"kind":"REWORK","_number":2,"created":"2022-08-09 09:48:19.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/59/852459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/852459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/852459/2"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-08 16:40:47.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-09 09:47:23.000000000","tz":120},"subject":"Fix listener creation allowing pool protocol","message":"Fix listener creation allowing pool protocol\n\nListener can be created using proxy and proxyv2 protocol which are pool\nprotocol and not listener protocol.\n\nStory: 2010220\nTask: 45968\nChange-Id: I0b8cf3cc8dbe5bc2babc469e73e32b2ed9ccb98a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c712a299ee43b9345db9f2c7ca3e5f0f981c06d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c712a299ee43b9345db9f2c7ca3e5f0f981c06d"}]},"branch":"refs/heads/master"},"05b4d361721264dbb0ff9fec4bc9fa87779745a7":{"kind":"REWORK","_number":3,"created":"2022-08-29 15:12:29.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/59/852459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/852459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/852459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/852459/3"}}},"commit":{"parents":[{"commit":"e96240158558133c2ae5b3c26e729c466270b36c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e96240158558133c2ae5b3c26e729c466270b36c"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-08 16:40:47.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2022-08-29 15:12:20.000000000","tz":120},"subject":"Fix listener creation allowing pool protocol","message":"Fix listener creation allowing pool protocol\n\nListener can be created using proxy and proxyv2 protocol which are pool\nprotocol and not listener protocol.\n\nStory: 2010220\nTask: 45968\nChange-Id: I0b8cf3cc8dbe5bc2babc469e73e32b2ed9ccb98a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05b4d361721264dbb0ff9fec4bc9fa87779745a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05b4d361721264dbb0ff9fec4bc9fa87779745a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
