)]}'
{"id":"openstack%2Foctavia~853606","triplet_id":"openstack%2Foctavia~master~I87efba18017aa3e8b9b3cc812664efc11af1c4c4","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-12-08 16:01:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-12 11:16:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87efba18017aa3e8b9b3cc812664efc11af1c4c4","subject":"Optimize HAProxy SSL cache size","status":"MERGED","created":"2022-08-18 09:10:44.000000000","updated":"2024-02-06 11:07:05.000000000","submitted":"2023-01-12 11:16:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853606-amp-cpu-pinning","meta_rev_id":"23bd22f625e3da769d26a3abd0bb98d3c86fb02e","_number":853606,"virtual_id_number":853606,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 11:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-12-08 16:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-01-12 08:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-01-12 08:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 12:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 20:26:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-12 08:50:07.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a6cf6533860198a5f4d5e170ba4422661d78383e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-18 09:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9a905ea311f2bc4d5a6ab0a2059a867b14956d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-18 10:02:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf84bc4054e139652d925348fd3bfee9f542e146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 12:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbb74c0b70c42efa04873294a4d6a06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d9df414a43048c18c724b707d11209d : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae86316fade4c4f8d90774feb5e573a : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5393b11f6ebf4b41b6166d4cfd60a9ee : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efde7a6f8c7e4f1cb2e447223e93722e : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbaee32fe21b444186bd0e862a9d94c5 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448f87d270174802801694b84318ff44 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57172789a3b24dbbab549c995f151e69 : SUCCESS in 6m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/beaa2ded37a346f18ff86b63fc66da16 : SUCCESS in 6m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/da4324d744384f7a95dcfcaef7bc4ab0 : SUCCESS in 8m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a19eba103f148709d33089b5d5a3692 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3176956c4e9a4348a737b9a836652e53 : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb79a0d85fab4ecaa07e01f739b7494a : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/069ee695c75c4a56992b82039c4e5ecf : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d76dce52a274399aa4c6f670fa2c1a0 : SUCCESS in 2h 19m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f68d837b887440a58754bd6f2d2d3758 : SUCCESS in 1h 58m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d64ad1ecb0ac4610872e643b27e069cb : SUCCESS in 41m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6ebe4af5a2d42538e1a72ce9815ad2a : SUCCESS in 1h 14m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/01771949663b4c9796c293bea7f40ef9 : SUCCESS in 1h 17m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/551d78d2f1884ae1aef1c313cc4b4ddc : SUCCESS in 39m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3de247c77625430189c578dbc638b948 : SUCCESS in 1h 55m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/537380e62fbf4fb1a43f86bcff61133f : SUCCESS in 2h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b90aaa9941314595b6592573fd8b9f96 : SUCCESS in 1h 49m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a2ae185839084f998d7880f3086b0bff : SUCCESS in 1h 29m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7da3ca5ee475486895971e99e1b46f98 : SUCCESS in 1h 51m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c59c6d8c6fe04fc2af824b14f2bef74a : SUCCESS in 2h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b21ae6b97657ab0c6fb6a4732a3b47731df9b308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:44:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9e11ca5915fb2c24d1d2bd796d57556fc98160c6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:45:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"2706e97bf320adf827878236e508e51e94f813f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:37:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad2debd705aea603a92f7be20d6b1b4636c3ae16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:47:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9bd9672b078eb58b4df122657c89d1b29e2b758e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 13:22:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"adfe39ae5cad7a4d4d58e80dadc18930b9a58a7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 14:46:32.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"45b802862dbb04ec11e0bbe10a09dd3672ddc093","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 15:36:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e0f53d1fad41c8568de30dd4f79850629abd0db","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 15:36:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"93160ccc65507e865b0715da65bf5a91ee24a17c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 15:38:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e1a7668440fee57f97b365f75e4bdf679943d936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 15:42:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a8e05007d1b7c1d3f23ecd2a79834c9f8bce12ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:56:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/991fb9a870ef40d7b1f70ce12677fb0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61437de1af5049c0a09f922b339e36ba : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd32ee6ba99425294b60a66dbb651d1 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9071b7f3e490439fb2758055f7ea468e : FAILURE in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017f214935da468fbbfbf875327ef265 : FAILURE in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/044432e6f576455aa593f55aea2e220f : FAILURE in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e754f1d6ca41e09467bdefa3f73b32 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4524710bb0844bca2afe5169166671b : SUCCESS in 5m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0efe4d86348c4ca4b17ae2727ecf7422 : FAILURE in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7926fe22ab7b4415a90ea005cafe4a66 : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc8c49f2fb804477bb35c06ac33411e2 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/efaf82851de34a6eba87752e83118289 : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/998be263a4b147e0b3a768af7ce8c72a : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88bcfea5e8ce49338598979e1770740e : SUCCESS in 7m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/595d8de09033448da5f07969a505590d : SUCCESS in 1h 49m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb06fc71976a414d850c66b55bbf49db : SUCCESS in 1h 41m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7b8a8fb7abb4ca49782d529340bd211 : SUCCESS in 39m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d89e2e7d0694589ac3cc6d2317091cb : SUCCESS in 1h 06m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b9441ab34f54c26b67ccc2b69c396af : SUCCESS in 1h 14m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f1b1c92e4b54460ab396d743f0f8416 : SUCCESS in 37m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a4e836d331044bd4ba60633d0e93c769 : SUCCESS in 1h 45m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f933302fe59845c7ab9b7902fa60006a : SUCCESS in 2h 12m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/165d206ce8e84578ab358720d1d26f32 : SUCCESS in 1h 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e52c4d964f84472a6cd397283b8ac0d : SUCCESS in 1h 45m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/90cbfb13aada45b580cbd23dbf021f8b : SUCCESS in 1h 44m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bbb4e7091a24222931a76b053ed584c : SUCCESS in 2h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bb360c2352b683b549ca833b3598c8b3228aded0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:00:58.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"6bd99f7ed40b55156c735c9d1a0eea9e46e951c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 17:45:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f6f59fb4fe54005b62f0a9ab3cc1873\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e5efddd3be42b59fcb18eb74e4f111 : FAILURE in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76b14a316224126bd802657259ff216 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a3df844775a40f998a47b2c66657721 : FAILURE in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebc63b0013c94384b80455ecad91a524 : FAILURE in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baef522bb0674f008b479b42fda3879c : FAILURE in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118e48c74f4842a884d5fdac955ef8d3 : SUCCESS in 14m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a43dfc335e2d4a97b01b4623476c5987 : SUCCESS in 6m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8dbbb0e946914a7fa6dc06128e28ae59 : FAILURE in 9m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ad2afecea725481dba61239d74486e24 : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68f62d97778242c6b42343fdf70edd60 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/48bac7e5d9d24aab9e86c17d70c56ac1 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b67518e0c77a42c1bb79c1f098c57a43 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c373809419b64f0882f7f15f5d096efa : SUCCESS in 7m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6fc9841e6f2c4dce9fdcdaa3bd3b0706 : SUCCESS in 2h 14m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6d48b6dfe194bb6b65634240f8fed5a : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30837d87db984913a686f64b51d5dc24 : SUCCESS in 48m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/589bad364f6f41398536c0abecf2e721 : SUCCESS in 58m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5a8d7f789b2b441b8323f75d2c4774b4 : SUCCESS in 1h 11m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ad360b5f0624bb9af6797c888a32163 : SUCCESS in 40m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5cdabb230d9b41bbbfa21049c92326f4 : SUCCESS in 1h 30m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/313cf5bd53084c55820eb9f21749a572 : SUCCESS in 2h 02m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4d178dd37bb347279830e601eb0fcd88 : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/940a29c9b547487b893d35a5783713c1 : SUCCESS in 1h 39m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e2ff88c1f59b4786b130b5b0e351a6a9 : SUCCESS in 2h 00m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4886d900f3c64265b7a173111292d983 : SUCCESS in 2h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e71ebb93ce4c18c4e692a8a246b458fcecb9cd04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 08:08:17.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"8cff079462aebf97894849841ed8f6f4384cec25","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 08:20:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":12},{"id":"5c1d5be83a9a1c1117fa70cb798ab7b7509140d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:46:16.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4d670b2dda348deba52fcd24dbf48e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e048f4137e43b2abc42e1b642a6158 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0120efc2d37044b7825a3f674169117c : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f10c6fa3e2c4d2caf5f26dae45c485f : FAILURE in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fac63062c14b45e889c9ef849696d41b : FAILURE in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c1e6eab07fc4c6eba539752e5fcd6eb : FAILURE in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfc6c483921406fb96789a45090bf77 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1685b273b30041be80dbce7ab305ae04 : SUCCESS in 4m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f310672166304f02bb1f27e00a83f16b : FAILURE in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c174b503868a48a097558d10f75037f3 : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09ac2f9101534244aeb19b8b71e19fea : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fc265dcc67045a9a64be82b6c6df231 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e03c7ce215e045cdaa9763e17865e672 : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7493b53e5e4849db9f6091edc9c7c929 : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f34bb166c36c4298bd7a2992ff7528c5 : SUCCESS in 2h 21m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2907ad5f116d421090739d6b283f6677 : SUCCESS in 1h 44m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dfedfae43f7472b8240f2a30e9b5d8d : SUCCESS in 47m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c0b8b748cc84f69aa5a219bca508132 : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f49a12a2a696459888d65f7462838d79 : SUCCESS in 1h 20m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6947f18ee47e4baea355a18fbfd52840 : SUCCESS in 57m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3da0e53d63934f5fa1b719b873e4de1f : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5cc3e3d71cf144a491bafd3ca385ca02 : SUCCESS in 2h 29m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6715a80e401c49b9b48f8ff4da7951cb : SUCCESS in 2h 02m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec21765ce63749bdad1b68629ab89450 : SUCCESS in 2h 11m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/919603acecb5456a9dfe8c8ade606df9 : SUCCESS in 2h 25m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/05c07fe154f6470fa3dacb8307ea8254 : SUCCESS in 2h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"20916da6cb999bc56e6763c5632cfc35260af06e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 13:52:53.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"5afca2ef7df8bf6b5404524c4631487b702f942a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:12:55.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a06b798dc2446848de147901d885134\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b9a2d694f84b919740d981a849d493 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0e106445644b9a9958664007a63a58 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86835cef3886488cafe6a046db2ddeb9 : FAILURE in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2eea24b85354979b5115fddec44e1f7 : FAILURE in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b92de05b5534d1b9a520ca87888546a : FAILURE in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77dd6b9654b548e88d7f433722b96c17 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61ff844480f747dd8e82e40116bfee77 : SUCCESS in 4m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2c68aa1d2e7441b8b0f2d3c4ab6be82a : FAILURE in 5m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b19a6ca82b1a436dac315522b8776588 : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0a4799f5020c4df2a469aea69cbbfcfd : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf336d29c0424f4db4b112d6edac8670 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a7387d675184d90a200b364cbac404a : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a5574eed13142538fc13f2cbe074a8f : SUCCESS in 6m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bbf95ba9635c42dc8f093a41d96bf47d : SUCCESS in 2h 12m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ab0081b57c04073831a0ce5fc18e123 : SUCCESS in 2h 03m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c39d4a88d2b2470fb225428d742bb6df : SUCCESS in 58m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b34ed2c7f224eb89f1540dc61a69d1e : SUCCESS in 1h 31m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/30ba8913b40b4f5fbbf84f7e196ac4b0 : SUCCESS in 1h 32m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c6af1b3428f45198890001c3d3a6abb : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad0826347c5044dc9c86e12279293220 : SUCCESS in 1h 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c8df4b429fd4f158710baaf89bc4d95 : SUCCESS in 1h 55m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2988febc06b42a28ed14d7e820e0168 : SUCCESS in 1h 48m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c123b6172f6744eeacc26fa20a859f7b : SUCCESS in 1h 43m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8252493a155f41068678f3d9f9acabfd : SUCCESS in 1h 55m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f2dfa37de2a49a5b6a768cea3d1f9a9 : SUCCESS in 2h 08m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"886b1d7957876f024c0b54cf670cb2ec058e6fa8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 10:07:03.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"e8685e5c457cac9e237bd2d86b0b230a291e7120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:32:18.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a8de9355b92423bbabc88463be0ff66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6cdff6bc0d459d9c65c2eba17f40d9 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e334c1ba900444c3989a820544bd866b : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac7c0380b2994c6bb78ba6fffdf15be7 : FAILURE in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40cc39324fbc49a18cbc7872c59498a2 : FAILURE in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afd1dcf554dc4662860da6217b35a142 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21434b5ebc94a01abc2a15ca24c0fc7 : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53f559ae00a64a9cb1f1f37b980e3b1f : SUCCESS in 4m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/50085bd415e34103b4c3e763de8ca77a : FAILURE in 6m 16s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1e8ae2d536ee4b3a8264cb2fc06021fa : FAILURE in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a50ab0d045d4414e8cd8f722c5abcb4f : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b17fc5ef7724a94a76d85680ed32b43 : FAILURE in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8b643494af4841df80f5765f47645850 : FAILURE in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b87f1c03127f4a799643239f8fd80b11 : FAILURE in 9m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/669473633a9241bdb5912e772bdd59e1 : SUCCESS in 2h 11m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b97d5e8d1aab41ddb2a6bb8a48a6881a : SUCCESS in 2h 22m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a11d1d8121a6412dbd2e8d14762172ad : SUCCESS in 45m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c63d1e9fab5d4da2836131468d7f3591 : SUCCESS in 1h 25m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/62a2d4e76cae46ad8946034a4022df49 : SUCCESS in 1h 12m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b68743907e754291bcb1e52bd8e3af7c : SUCCESS in 41m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/defc35b5fa964ed28a4cc18f5ace72f8 : SUCCESS in 2h 12m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/83aa4d9103aa42879080386a716ff9fc : SUCCESS in 2h 10m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/530240ebf8674d88b2e26376dd5c2276 : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0db684e8d9374a21a01e4c54f44156e7 : SUCCESS in 2h 05m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ffdcc46e6852488b965a5415adb6b747 : SUCCESS in 1h 52m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/826a8b230f8d48008a396c22083eeeaf : SUCCESS in 2h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8c23f20b025be732189e6407c00f190062e3cb2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 12:45:59.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"09242c4e41f47fe2187caa1c4990ae312bd66c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 14:57:57.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/797deadb23db4ce88a4eaa34e45a1eb0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ca9248c806c42eaa87517ce53b5952a : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e119d6601a8d4f8791676ed480f1a708 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba1462b5f80b45aa9a3234688440f981 : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2e818fe96544006a06e04d63ae939e5 : FAILURE in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed1744ad7cd04520891cab4a70555270 : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d05d1a1aca347cd84a855d37604975f : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3faad434e60a46a48e83854657f39ce5 : SUCCESS in 5m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/df66080f2d47423689ecb441671d8227 : FAILURE in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9fc735f31d854301b30ad49d92ef42e1 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6883e7deb76540dfa441a711e8cd7d50 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e317fb598af2445c93eab238aa0a5029 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a09577d92380436ab0e8dfd1df58f1f7 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7934fae8e1ca4d96ba5cea3528034b94 : SUCCESS in 9m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d34e8c1d0f524359a8b5640aa40da8a0 : SUCCESS in 2h 06m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a4c179d771a4a33978b85f1753b6da9 : SUCCESS in 1h 56m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dff968cedb5453b8fb2de7709b25217 : SUCCESS in 44m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1f6bf8d88b04c439493dbf2b82ae641 : SUCCESS in 1h 09m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/30ae61e93dd34788a5fe312699f9df62 : SUCCESS in 1h 11m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c491397d4a34489eb817fc06b99a7a9c : FAILURE in 1h 00m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb70a1f73c7b41468d88c084a78057d4 : SUCCESS in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da765c9ad6c34c1b90329f8a5ce87742 : SUCCESS in 2h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23e56db6cd694470b53dbb557ccde578 : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6bc7f06a2a2a43abb549c41a2c4d1b96 : SUCCESS in 2h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fe64fa5b3b674cec9f2692172bd35c0c : SUCCESS in 2h 01m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1c37a16db05496ebce866365fd15634 : SUCCESS in 2h 09m 12s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6015e0f903bc22f7235bb0dbd049a9815c09b0b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:21:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f21e672bf19941f1a9ff1cec5a9a5a05d9cf92ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:35:14.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"598fd8d7aac04dfd45664dbeec99ffc13a18cb06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 13:50:24.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de1ebba0bb4448718b06ff22eab481e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e25d3f08687149018eefe3fc77ef07e2 : FAILURE in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd662cba6a74bb78676b1ad131f4e6a : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7332dcd960da4bc784b7574b9e2b050a : FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c978dc75e6c3485388889e216eace221 : FAILURE in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365321f8ec094526b0c931842454149b : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcf007021ff489190e1d7ce0524c03f : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc4474669bef4d19bdba524ac34f5ab7 : SUCCESS in 7m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5fb06e0c4e1a4fbf91fda52a4930c8ee : FAILURE in 8m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0a8f0933c02e43e8976dcf95eb890d82 : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/411ee08b570f4b74a3f32d065e5263f2 : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92a6b82ea8314200bec83495fcdf98f4 : SUCCESS in 11m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dc49992e6ef46a8ae6a692b7c856293 : SUCCESS in 11m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f010f6ec4b874ef08e9adc8c7d7afa84 : SUCCESS in 10m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e966756e7aaf44aba62a5eba5b1e007e : SUCCESS in 2h 04m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa1d764af52543848b97f96b7ae89368 : SUCCESS in 2h 12m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10df1bd55d124e809927d2bf72323ab3 : SUCCESS in 58m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/400b6f22ddf04e258dbecf8b989500f2 : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bc4982e7f42b4e09b51afb7dc0938aff : SUCCESS in 1h 06m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdf9fa637ba24981965cd4f5f3a8644c : SUCCESS in 40m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f52c30f95504433792188722412eb3cc : SUCCESS in 1h 54m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96f71855c0c54c808625e12542ecd8d1 : SUCCESS in 2h 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c52361b99e2b4ac5835281c0e5cfd163 : SUCCESS in 1h 38m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1691138ba6b34cd48781a94984021ac5 : SUCCESS in 1h 40m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f81cff7707484f8a9912075116798d03 : SUCCESS in 1h 29m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/62d97e1de6184380b9fd19ce7a3f4b01 : SUCCESS in 2h 05m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"292e6f0d715d9b8f56e529707fe206a08dd310f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:22:33.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"198a23ad1f8eafbf37d02cc17de45cb755a8468f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:53:12.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"cd1de8c63b971319d1496a08b33e6d65d726837b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:09:53.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a33b767029f47e88df26fa8cb5f346a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9bf42f717bb40238aa134288491cd69 : FAILURE in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1db0cbec244ff088e970603874429c : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/742b4d2fbbce4494b7435e18b32ab18d : FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70a67d580fd94a959503c1af71b48c43 : FAILURE in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f4d211d149e4c33905e0e76ef2808ed : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c6fac1b31ed49079bc26fd7b2cd49f0 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a589b494ca24d91978f806000cf6cdd : SUCCESS in 5m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7c847a09e74144fb95c6976d89db8ef9 : FAILURE in 6m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/39a1c21491d0472cb66ddbdc6ce6c010 : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1643f97049e648ec9c53ed64484b18e0 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/155e21ca9ea24fdbb994c18c4cdcdd26 : SUCCESS in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10ccf5e45cc64de69cd12279cef509ab : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ab5a623c36949348ac11c8e8d37b1ad : SUCCESS in 8m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1205b5d05599453a913e8426c941f496 : SUCCESS in 1h 59m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f2eb16d38ec440bb0dcbec0f14bef10 : SUCCESS in 1h 48m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9d2c888db4a4256b7a9c0697e42f18d : SUCCESS in 44m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1556ef87451b4a478a6c288619c9c7bd : SUCCESS in 58m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e5e42ddffec8437d94906646bf60a605 : SUCCESS in 1h 20m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0c1185231de4cfab511bfdf861340b2 : SUCCESS in 48m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e8426cbfd2740bea1c02cb570751b50 : SUCCESS in 1h 43m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65e8da6f4e374ad1aef6d1f4864f889e : SUCCESS in 2h 11m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65ff892d96204e048f4a440daf2f55fe : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57063562e65f46908fd57b9dc4df7cbd : SUCCESS in 1h 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee290eab507a41b9b3b7115523fcc855 : SUCCESS in 1h 28m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7acb503a81a34149aa572e84381c497a : SUCCESS in 2h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e50cc86b7b6fe5752e6012b64ee40e704b4c00cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 07:29:59.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"f2d15552016424ed523d176698404c4857fa223f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:09:52.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"21963ca00887661d3b1e4c31af7c861eef264cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:08:11.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74494e55a2b44073ad54a69f2a8e6e10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/383d139c403747689eddca23b3cd1830 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/620c4b3cafe845fdb821989af2181d9d : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0efa55a770584045a4486c600ac0d6a6 : FAILURE in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70c298ba18384a01acbfe129b04b9730 : FAILURE in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7feb09911c13476a9a749d2b3480fa61 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c052235e8841e194ec0248ff26d6b0 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d872553919c341deae79ecee9c351372 : SUCCESS in 7m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/590e4b3e5cc44298832d1803c2d98b2e : FAILURE in 5m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e5f3c37cd21048239ee475913c8e8259 : SUCCESS in 11m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/774552abc613450687b452339d803ed5 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e1b878d121e4b21a566ea2496b1073d : SUCCESS in 9m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73555793aea74932a786d23de9acd518 : SUCCESS in 10m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/166a9ff4767943af95435a16709ea3b0 : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c729ef5d486d42d5ae1a26a120380e36 : SUCCESS in 1h 48m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e40baf4d47f41d68c9e5cbefc9deb7d : SUCCESS in 1h 33m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9db69f28f3164367932dd414c2cda040 : SUCCESS in 45m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24b02866af3b47859affc9a4bd9c13e2 : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f6490db912a146cc958a4fcc1d55671f : SUCCESS in 1h 15m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32184b3d2f154863911f8b3e828b0d3a : SUCCESS in 57m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4f97e799e42546a697df03fc650cfe3d : SUCCESS in 2h 08m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15e29dfcde474a97996a9e0ab9a8b332 : SUCCESS in 2h 05m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e39c76b2a8c04104b126e3fd882f4b5f : SUCCESS in 1h 46m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c481ca57845f448888f3f4369b064fed : SUCCESS in 1h 37m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23ae80d3d5084f5585f32234f72537b5 : SUCCESS in 1h 35m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3de490438db0477888492d9d5c910b9a : TIMED_OUT in 2h 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8c5c22dfd9bcbceb9e43b18a50b1bad0fb75df2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-01 08:56:27.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"eef326dbd2c6d49c4d4c782952a5e43b17bd9b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 12:08:24.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b48194175bd14478b144c810a4bac9ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af4e57b3b6574d28a042cf6212cb5f84 : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04abb44d9b6849ea8a5c80b53d00ff3a : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d249878f54d14b39b1ef04ced359b081 : FAILURE in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1fccec3d52a4c398847b2f56f26b199 : FAILURE in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a66ed1b10d5e4718a3b22795f4c1ce0b : FAILURE in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a038eb809de4f74ac8a5a6f36a07072 : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bce2f4f3d8104592b041020e1085b7fd : SUCCESS in 4m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e482262ccd4d40bd8a3b0f48851ecc70 : FAILURE in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bf3793f74fdc4940a40172019f67a5d7 : SUCCESS in 12m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c3f4adaed97e4cbd9d2ca4af36c6c543 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8b52799f02f46749d1808352cb2e174 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f0e2346167f4cd7bfd9fbac07b69b85 : SUCCESS in 10m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59ca6cc8fecc4c0797e372416b7d122c : SUCCESS in 9m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/613b9550d967460f937825fb2d1f8e88 : SUCCESS in 2h 25m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53c3db2f10224b69ad08624cc8a64282 : SUCCESS in 1h 53m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/109adbd548694304a5ec3f7158cae3a2 : SUCCESS in 42m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48b81211b62745569964c5d5448e7ae5 : TIMED_OUT in 3h 05m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de2db4e2b5cf43a993f2fd29e5da3071 : SUCCESS in 1h 09m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90bb83049fcc48d0a0dc76ed04f75f69 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4b5288c8f704c3ba431ba008db2b77f : SUCCESS in 2h 29m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a6cb5703dac477581a3d897beb44447 : SUCCESS in 2h 20m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d21e5ee49c54a189c3a2fa055d29e45 : SUCCESS in 2h 23m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4ef62d2807d47a6944853de86100ae1 : SUCCESS in 2h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/44e0e7400d124b8b867349188bfc719b : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/10fc5dd059e349ef955e60a165d0acf6 : SUCCESS in 2h 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"9cb13f943b24e70073bdf909832fb6fea4f30724","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 09:57:43.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"3db121de31f785b39015f2265df2037bebfcb355","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:03:44.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"4626497498edc8948fd3764f34649c4a1431277d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:08:17.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"fbba396bf40e2841f465b2af8bc7e190cffe398b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 12:28:16.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"ef53eb9862974f28f09fc34aa612d8a49c51b168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:17:08.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fb912424e5649da9f602b3aab39062a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e31ebc7b5624efcb3f7b50b337f2cbd : FAILURE in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac8ea47fe9242258e7d141d2adc0cba : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad1d808e9a234e6990cbd2cc23f82add : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217d60cf2a0e4800ba9a4a295a9dc4fa : FAILURE in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6cdf99d9ff645cf9e3e203ebf4a8328 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adac62a8c0a54fcbb8a94027317374e3 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71250dec2c4a44439eceabb2f68f7847 : SUCCESS in 6m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d1c4105989da4ea7ace50e41fab579be : FAILURE in 6m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9bbcb038846c4f05a07e468a092936a7 : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/676575a91e1a472885d4e4a3bbdf2149 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c69e977e09f491bbdd3606e207ae7af : SUCCESS in 8m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c27186afeaa49f69fb4c2fd52dce260 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5e0e264e4fe43fc871e01193605a639 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10e94c8f37bf40e19ee79a13cf0fbd2f : SUCCESS in 2h 40m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81ebe77a29ce4a4d993ae66c8056bd59 : SUCCESS in 1h 46m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74dd6a1684b343189f32c6bf6917d827 : SUCCESS in 1h 02m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/930842b5b14a4d0c9fb7bc473fdfde36 : SUCCESS in 1h 30m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3acb4760cdee421f941b1342abfd1fdc : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7575a84eabe04848817229738378cab6 : SUCCESS in 39m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fe360766a052478f8c0e32ba36a89e4f : SUCCESS in 2h 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/28e3fabb125a4929b87219abf340252c : SUCCESS in 2h 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/463ff1641f4c4d749109b33552ebe64a : SUCCESS in 1h 50m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/37b93efdb84644caa34f71298fb8414b : TIMED_OUT in 2h 39m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/37fbcd7ac18c4c48bbf4c433121ab689 : TIMED_OUT in 2h 39m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ba00c9ffed9443f80b2e44eaaf18177 : SUCCESS in 2h 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"b94063f7c8c1afe1989b916d098f21bf60daaf46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-14 10:30:25.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"2d9b5a2a569ae45f0e744d994d20cc5d2d4fda43","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-14 10:32:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":27},{"id":"90ab7882010ee2e19a73abcd1906507751ac76d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:06:23.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d68560daa04bfa927ded46e424b64b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82280a1bc6cc499eb57567b6adef6da4 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f93afbbdd9e45fcb1582fb59b7e3a78 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42851c0ecbe7464aaf50bba7630ddeee : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26deaf6841524fd79d119602c6b5eaae : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/987c489f1a384db99611dbfd232f71ce : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466651f230de44a98e07070ca2e80b43 : SUCCESS in 14m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9be1ee7de0364abda6f862549934471e : SUCCESS in 5m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/79da5217dff14d5c97da00305f9af912 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa33adffc3314791a6b629d5574fd5d2 : SUCCESS in 8m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/08cb4b8889064095a885fb4a3d81d50e : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6b56609b2054b3ba9d1441f4f61508a : SUCCESS in 11m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8600ab53294455a84978f57535640ee : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/261dbd353d8a4a9498fc873dcc50da10 : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad6d4331f153470497cccce6dfde51b9 : SUCCESS in 1h 28m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4ae3ce0f56b45e2893d0e38f4699ffd : SUCCESS in 1h 55m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef063550347a4012860dd293bea0109d : SUCCESS in 43m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d11cf23d94804022908e8ba1592b4849 : SUCCESS in 1h 06m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/41a86e607dfe49998201a938eff575d3 : SUCCESS in 1h 33m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/554d58d2d0a84f4d960790bad3106f49 : SUCCESS in 38m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30b7c53c671f4cd5a5d4812e207a6ecd : SUCCESS in 1h 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a77dbcb30d0544b7a3b31935cb84bb2e : SUCCESS in 2h 30m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/99786dfc74d44b77a6f33166b1c3d380 : SUCCESS in 1h 50m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7858385feaef4281b8a7d39bd2a3ae93 : SUCCESS in 1h 52m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f501c61f1e054b6eb3f2e6ffb0e1ade6 : SUCCESS in 1h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bf94b08627140d1b47b42ffe8be8f89 : SUCCESS in 2h 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"2c1128895953690cab73ea555123cac67bedf19c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:09:36.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"c39a352ed1ecff36b3b296775c86e9e36585487e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:11:01.000000000","message":"Uploaded patch set 29: New patch set was added with same tree, parent, and commit message as Patch Set 28.","accounts_in_message":[],"_revision_number":29},{"id":"68f86c09e51f74c8ea593b873b0068efab6751b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:41:05.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4675ced49a464ab9a6ffabf11c09a557\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c555051784a4454997fe2d48dd3468c1 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307d7ed368534bf8aaa1e2b57bd52f87 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c213acab0774223b3b0bdd1adf887cd : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d324be2de334049b7a837a119f33f0e : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5f55c425d34edcbf2d42fac7de7f28 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c022e39e8afa4566b413748d3699cb7e : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c029619fb041437697072cbce2e09451 : SUCCESS in 5m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/27097a21c9b64ac3a430951eaf7a5960 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f2eb77a5c004f1bb18b1dde62647dab : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dbc3cd441d964625ac05589cde692c1c : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7aa73e4b33e4f989328b763fb04fce5 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3dd004a5d4e8416595c62d875c9f5470 : SUCCESS in 6m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/436f5c2b28544b708399a445b0759b55 : SUCCESS in 1h 18m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ba218e757c04ac28d97d567ad87b95e : SUCCESS in 1h 46m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8befdaddcc7b43e98b112e1d52a6ab4b : SUCCESS in 38m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c19ed79b552947bda1a2177250f0840a : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/236fad195935499fb1d5ddbb15080cf9 : SUCCESS in 1h 19m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08fc2d88a28e4422a0ab104363a16065 : SUCCESS in 36m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3146d0a21a9d4778816457b3d75d3b25 : SUCCESS in 1h 38m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40d84abb5db74dcc811e691254197ea6 : SUCCESS in 1h 57m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba7591458e754da19235f51a9ffa704c : FAILURE in 2h 10m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c7ff24dfeb1b4556935578e7712bbf74 : SUCCESS in 1h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e607e77ad17244b9bdcfe0dd5eca0b88 : SUCCESS in 2h 01m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/12c1f0076f6a47289ebe40b0007d61fb : SUCCESS in 2h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"5e7d782aa5780597103fc8f982dce3db5955e8bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 11:30:48.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f4f871efca4b1a52a7d12638c2cdc81a0b9179b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:22:40.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"8d3762c6e72cacbf51d8a59281f17b4cf4eaa416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:04:30.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"faa1df24c64e50d871fcf1d03eeb88f0cebc5598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:08:24.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.","accounts_in_message":[],"_revision_number":33},{"id":"79c689f68c87a094bada9fe5369bc28c5c00f654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 14:41:34.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.","accounts_in_message":[],"_revision_number":34},{"id":"11f02eafcb1e47af425ed4ba1a24c9d5bc2a92b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 16:19:29.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"9549ad919e94ea76b15169fc41f45cc4d68d897e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 19:14:40.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1bcf1ed5524b078463d3682c794b77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85bfef77fc96416b860bf26c299b0244 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a64a01e578f409e9889b97eb8b719f4 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/783bef03b82f4c9a9f66a117bca3b241 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d48cbceeafd640728ea2cb4520a070c8 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92383343d424c918a8f87f0469d2a06 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f49adc1b57574f44843ea82d1277bd7f : SUCCESS in 5m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4944ce78f8e54088a27de3738855ef32 : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/83bf41ad397440dd81f24b71fe50b784 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a334c188c69147d49c764909c7acf36e : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5c34a32ec3243e6b9eab963ea2e5dd7 : SUCCESS in 10m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f101006825c540338dc08f65b0301873 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79a1dc8553ec4837ac4542eb84015003 : SUCCESS in 7m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3440b7c46e8e40db86cc1f0186533939 : SUCCESS in 1h 13m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8407aa609d64f099c0e1398e855895f : SUCCESS in 1h 45m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ce49578b8214ba2ab51723ef04a45d5 : SUCCESS in 40m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d179f1340934eb3880ffdd4a4e04d6a : SUCCESS in 1h 01m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/17b6c5502df147e98fd71cb303a5c99d : SUCCESS in 1h 15m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8958214de1324ec7b66595c77e28a040 : SUCCESS in 44m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3feae0e94a948f3a389b85ceec7557c : SUCCESS in 2h 02m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfc4dcee29ad4b7798c4652969603b2b : SUCCESS in 2h 50m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/afbe7e302863441493aaa5561b9b9ba6 : SUCCESS in 1h 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c616bfd0f91e46b490d2571bfd5ea68e : SUCCESS in 2h 08m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c460ffdc37604a3da74009b5e0506972 : SUCCESS in 2h 14m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2aa2b198b0049d0a52a1f20d5d32c38 : TIMED_OUT in 2h 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"77e770b60d27e7593181f18fc0673c53b7883a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-28 09:35:53.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"6fe1b66cf4b17edc7134216bec1c210e3d8c3f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:28:07.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc69147f8e44e939bba117eaee70d75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4696fb42be624a2697b69e10088624ce : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8aa677181b843be922912538e81e028 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/091f7bbe668045c49a48d7cab8f60832 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9150fed40dd4f189cc4681d8676f2af : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a264b8b7a6b41d2bb187d456a09e6c1 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cf1dfdc9ab5469ea5eab5b800b9bc4c : SUCCESS in 6m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ed02f549c5e648bcbf1a67152d675baf : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/386d0c3b2a0849c881ea31ffa7f7a06b : SUCCESS in 6m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/43dcc465bc1e4d5c8502e63fed709291 : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cf7023c0edc41f4be93b5f0fa31ad3d : SUCCESS in 7m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aef2eff0a214c0c948f8851741e8b6e : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15eff39531064593a35018a7fde4618b : SUCCESS in 8m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6edc4ae329d34a7e9fbd3ca61cab5278 : SUCCESS in 1h 36m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/814dae9799394a0da73e41ad60d00625 : SUCCESS in 1h 57m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a794e56ab584b68b1aa22b117393818 : SUCCESS in 40m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e4a3593b20d4b528415e8da2f386990 : SUCCESS in 50m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b0d6556ef7f441bc9dfd337a091d4b16 : SUCCESS in 1h 15m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c90a443d17c413089db2b87ad88aeef : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba06ab6d2d22474895a9226c455e8e49 : SUCCESS in 1h 51m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32df22fbf8f3476f93c4964993eb3452 : SUCCESS in 2h 38m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b13a985d8d964b238103b40af41a7eea : SUCCESS in 2h 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/590377afb3df40a68fcc7b2feffa1319 : SUCCESS in 2h 05m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ed2401a89e404b97b3fd03d6926eaeff : SUCCESS in 1h 55m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c6e648b801345fd977b6e43977b09bb : SUCCESS in 2h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"47ccb94e6be265dc8bf7cb7a36247143c48af4e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 16:24:16.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"920e33e9368ed6526ea534a8d7d036dc1d98d915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 19:02:44.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/324ed9e5d9f0447195f3002180850da7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65223961d2564aa3a53c48425ca4f477 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2aac289a7b4cc0b6b6d60e128a8a75 : FAILURE in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/976d1f782ca4448389df4a709489e321 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70b5354cf7e04293b302c0c2c0b6f694 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52c1156290749d2b5e37f3068ac29c8 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d124bf1a2793410398cdd7d0068e0984 : SUCCESS in 6m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/df71b3830a124ae9a91a77e0487ee8ff : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/76ba891f4ba04bc992621ae4d4ea60ca : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c67e98ba5df84a80b51528ef70b69eb5 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29ad460780234d848ba5d8ac6226b133 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07aac897391e4c709499c01a849a3df0 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8d9e62e0dbb4622aced43e1449182eb : SUCCESS in 9m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d59bba955c7400b8462b9fc7da283ba : SUCCESS in 1h 14m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aac2f187575477487ce06eee758bc5a : SUCCESS in 1h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/988b88a952f24af79f8673f3214c7310 : SUCCESS in 34m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50aacf4006bf4102b0a0e5ea4ed1db5e : SUCCESS in 59m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3303e48d113648e395949cbc459638a1 : SUCCESS in 1h 18m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e65677073d44d77b45ab475c7d0c7bc : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b25175e52abb40959f79a73a761be818 : SUCCESS in 1h 29m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0cefeeda74f748998f07bee4bf1b64d7 : SUCCESS in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bbb93efe1af048019ff6850348c54c4f : SUCCESS in 1h 50m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c679e3fcfbf49b580fe0f045acbb7a2 : SUCCESS in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/af4491def01543959c83387837f7b7e7 : SUCCESS in 1h 48m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/558b887e8ee046c8a29cf0d0ba61dbb8 : TIMED_OUT in 2h 34m 49s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":37},{"id":"6f81a867a0f2355512942babea56579b578d9c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 19:03:55.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"7f56da02f5cdc5c6a3f744fee74855e2a68ac651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 21:46:37.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf829373af5645cc84e37fd02046d561\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/133672aa973f4be3afff31a2b348b926 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43891a9789d44df4a625cada2dfba3f4 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0de5b1752fda425ab2928745a43c2ed7 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0165d233b4aa42baa1e93d9d7e97c3ed : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d0dcc32a664a8388adec8e46f69b22 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9991c88f93a24e27837d9cddd3900edc : SUCCESS in 5m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f679779e8611462d84ab59cd0b58ed1a : SUCCESS in 4m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1b9bb0da1d4c403db9ea2bbc3af4cbd2 : SUCCESS in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d49809b478d3416491fbcd39ee1bdd90 : SUCCESS in 3m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97e91b83f9034641baace712082e2d10 : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c9db2acc4b94932beea1f9bff92a5d7 : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/171c1fb05d404220aee0e0bf0d38a93d : SUCCESS in 6m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5e6649a61784387b5debcdf87a53720 : SUCCESS in 1h 04m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/359c0633fffa49c39fe816e029b0d65a : SUCCESS in 2h 04m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e286d204e7b74899b0a01a75aea5851c : SUCCESS in 37m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82f0d77ebf614a6ca04a86e1e9aaa9b6 : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bfeb91b0706745b891c766511a303a15 : SUCCESS in 1h 08m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06e9461fd2d041e6b10a0c39a4cd18ca : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e774e13f5660462eb5a47dcf718f7b99 : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f06071435774f7e81d9bcbaa0ae0328 : SUCCESS in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4acaaa89908843f4892592f5c51fab50 : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/85543705deb14bb6a0a4fb4629de29af : SUCCESS in 2h 00m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9bc62201f3024f49bdb9fc287f13677f : SUCCESS in 1h 48m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/867ad21bd4f948369538e3efa5fab777 : SUCCESS in 2h 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"9a7d982fab26bee508bbfc828ccfd68a1b5e60f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-16 14:45:18.000000000","message":"Uploaded patch set 39: New patch set was added with same tree, parent, and commit message as Patch Set 38.","accounts_in_message":[],"_revision_number":39},{"id":"bf23e8be63a5184bc956ef73748f78a348ab4ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:12:19.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a90e7ed4de4b7cbafa4bc953978eab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feacb6a650bf487998ada8dada653ffc : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df5275b44f344a38a02b88dce38350c : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d41587cbf5e64c16aa424a6f3545e510 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/247f8761bb024fcba17abb8835f8a2a7 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddee05d99ebf461ba927920dff42b7c3 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d3d69a746ce4f539eb1efc73fd3d455 : SUCCESS in 14m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/67dbb10d25704086b14ebb0794baf156 : SUCCESS in 8m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2ad2935c426943f5ac0e13dcbfa7ddd1 : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c93dad0e981541e19d02a2206de86dc4 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21a9e213c58b4a1596e733bddf55a525 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e45c60b13c8d4ce1aa28cbe491c0ec3e : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12df9bbdf73e4e22b47aec208584bc51 : SUCCESS in 9m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ade879997204189a19a686d4ca843fb : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcab8599e3d7474e97af4ee8e6c02a29 : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8fb74605eda4042a848abfc28ce2802 : SUCCESS in 35m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bda0e2d1b024424b387fac764c35632 : SUCCESS in 1h 02m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e18243e2d91b45caa7f062f86e87555e : SUCCESS in 1h 00m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e281c8fc3e94f898178e330f97f65de : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8e7fbfc5157416fbd9d82e8ef3c568c : SUCCESS in 1h 53m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e76c00b0165844e9bc0d98b5ed975665 : SUCCESS in 2h 20m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c7a12cea2254a6e95f6d6002273ef5f : SUCCESS in 2h 08m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4032f9abbd9848ef983b5151fd4365ea : SUCCESS in 1h 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6a7f27ea7ee5489ea41216b1ef190351 : SUCCESS in 2h 06m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/93d8b667d7884109a9587dbd3c05f5b0 : SUCCESS in 2h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"7c963611ace35c8f1d80e8d63662ee8414131113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-23 15:29:39.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.","accounts_in_message":[],"_revision_number":40},{"id":"43e3cd812ccb0d7266364bd722690c9970f5d18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:08:55.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/691a346ef0de4daaa643cdddd14dfafe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b484e2b4a4eb41ca99510322dcc042c4 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb10879bc6448d093349d0447389686 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28a2ebcab42949d6913116f5e817c2c5 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/846db6cec6b84fe3b99d1118ba191bbc : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0ebbc816be4a9787197d2c1f478487 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52ee272060704f38bb76fbefcda3ec2a : SUCCESS in 4m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c7677983a38e4f72bc86cdd6ff3ea999 : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3fe6bf0fc53b45aab900cc26fa1c19f0 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3cb158f5fe874dab8b0bb5d6b1689507 : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d29ccff634ac43f4a59ed379d3b99a4e : SUCCESS in 7m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/441c2ba5a77f4a46879c32f268bcb1af : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/730f2bbe6d1b4905b86784728c89f986 : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3cfd2eb154c149d89f93b63cbf823688 : SUCCESS in 2h 11m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f21bc0e3c097409b9be06a901de4c0bf : SUCCESS in 1h 48m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a9e1ba1c07c4090851b5050bc0d8021 : SUCCESS in 53m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/718d78f9ff2d44fdaf3f3f0288dc78d7 : SUCCESS in 1h 09m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/84e800ef5f64491ab2899c4b688521bf : SUCCESS in 1h 29m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d523f9856bde49039d54c4fa6a586268 : SUCCESS in 35m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fa7b32e18f734b62a070f9a92fa6677a : SUCCESS in 1h 46m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/859fe25996f1470db7b6ef31efa24150 : SUCCESS in 2h 04m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f77a62cf844d769e98165309b77207 : SUCCESS in 2h 30m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e6c18357ee1440dfbcb846a1a48386fa : FAILURE in 2h 30m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/103de8b3cec6432884acfadd710573e7 : SUCCESS in 1h 59m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/990d612e6384436caa84c98c44f60b8e : SUCCESS in 2h 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"b4d874dc66ecf2a0819faee140c20432b06589e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 11:00:01.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.","accounts_in_message":[],"_revision_number":41},{"id":"10d2edeb1a601e3e0122277b1187c76e6ae4fef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:42:46.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd88fb2e2e934cc58ad0cdeee344e832\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fac1d39cde724588a63de638497f3b5f : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e6e6b1d57e486582aee77c79cdb928 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f82bc6cfb4b4f74a22f2db8b607df00 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df118b50b7d842cba7f49903868ff526 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3bdbd920be24a5fbc068c85e90baa65 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18f7854535f14ce181edc3a42e856ae2 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1f06ca36a9f34e8a915ded96e74666ce : SUCCESS in 5m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fd4819093dd9492dafb1b565f5f5a05a : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c74bb88157e54356b85eb7fe898860c7 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e6d14218e5d4c828b45e01be3b0deb3 : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bab0c4095dea43b88a54660abd21656e : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/836786f9f0634335a26e84e21bfb036f : SUCCESS in 7m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7559b0a23e724c418b58c9a1cf4c8317 : SUCCESS in 2h 15m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/603ee30803e04c4d895e4a40e775f407 : SUCCESS in 1h 40m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/23ead309105b4c48a3e0c4032c2fe13e : SUCCESS in 34m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b2f60b744b64a3ebe0dd00720ecc225 : SUCCESS in 49m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8c51e8a28c2148d28cffc19556ddf96a : SUCCESS in 1h 07m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f86dce6fe07245be901d746a7ca0b3e9 : SUCCESS in 33m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6bf7f4f7724d44c794fd997294e63fde : SUCCESS in 1h 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0dacdf1d64f49b281c51c76f0031ef4 : SUCCESS in 1h 56m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71ba85f79e7a457aaac7e9cad8b8742b : SUCCESS in 1h 54m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29705adebb1c44febdd8af852d517d09 : SUCCESS in 2h 01m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c4c3d45cf09d4d3daa66b4ed3f927eec : SUCCESS in 1h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e46b9e53d36d4541ac6a9df8a4587973 : SUCCESS in 2h 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"a61832c6fe767e184e490ec5b3c374c77d5f104c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 15:46:59.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"10843ebb2af849c2c401975f85d68a6aa1480950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:22:04.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c0997472364f87be8b90d031a40396\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1eee88c0000495b922b70d27503e008 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01da966be1d84639917badc3b82cbcfa : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2f7dce201b84400b5f063f9b4f64aae : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68704b25218842e9bac1716a0dfe7d39 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff309ae901f4b6aab0061723ffd9198 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdaac73b94504a89a62a9194b7d85639 : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b951487bc5db497e81c6ab7adc6c45d0 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/05c5f0c7ad874ad7978f01da02589f48 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7e2a927493d489795340739f4d87cf4 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e00bf37beb364952b070694fe640c2e6 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0befba7379e24512b727000f48020d36 : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/761b13c8697a4cbc8276ac10f9c546a3 : SUCCESS in 7m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf5ae83f6f664d64921fb527ae94eb09 : SUCCESS in 1h 23m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8f4e2686fce4916a5ce0a5d5b79ef98 : SUCCESS in 2h 09m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d76af6c3042643ea8d344c3460c9c3ef : SUCCESS in 35m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08992df5251247d0af79142ef90db3d0 : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/03e583f279f141b5a3c0bb92e23dfd3e : SUCCESS in 1h 20m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48a635280f3e49f7ada8694b12ea87f2 : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1c612ae5aece4ed0a2e57ff7ff743cf3 : SUCCESS in 2h 09m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54c477197cf749088ab7701de188d571 : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/01645500a33042c58f7b2182813f9aab : SUCCESS in 2h 04m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a886564f4094927b5391956996f42db : SUCCESS in 2h 21m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/69c502d580c148c185e4513ff1412caa : SUCCESS in 1h 40m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/38b08429f9a6439eb01f41e7d4f4c760 : SUCCESS in 2h 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"cb2acf8e6cd88a544983b6cc47311b56be0d6e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:14:50.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"d35fd8c64aac5277498db27c3b069fdff249c5bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:18:13.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.","accounts_in_message":[],"_revision_number":44},{"id":"fcf5ecd34b6750877c1367cd1a357ba29226030c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:21:35.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"5f18932bf151046267fba376b51a5726ec898a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 15:26:22.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ef4f9f83be45c6bc795bcf246c00c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dda8d1e6740d4d1a815e6b0db08263f8 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8b0722451a408588dabea43afbd4eb : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/579523df008440389ba912156a4a83d4 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccc7b67f88fb4c43a07efb1d5c917c87 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd15b2017364d0ab6a9e525cdd24720 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc4dbe712dfc41329da1fa83ad958c0a : SUCCESS in 4m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d1246ac831d848d3a59f1c2ad9bbe94f : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1ce49a65a078496e9be671b409b81a02 : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4329bbb889574a10acae6b894ab8b392 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58322a72c6d0427092d09770d6f0fc59 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab6ec1fb419d415aa049d54e4a37c323 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/225115cff6da4888855f515c6e9171e3 : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c09a329dc4b45b1a1a249ee539b3912 : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/369e7bc9d26343d4b61dcf94c76d312e : SUCCESS in 2h 20m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5402306315e4215903a8b632cee02a8 : SUCCESS in 43m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bbb07ced60524934a2342b09fda593c8 : SUCCESS in 1h 07m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f831a234355a4b56baf4525bb2088aca : FAILURE in 1h 30m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26cd394b58a343349940bee17fc753a4 : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e71541750c1479087d9f9e07d489fa2 : SUCCESS in 2h 14m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cdf93c7dd2f8441abad38598c226cff2 : SUCCESS in 3h 01m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29f7ab8619844f33a6993ebe3f8b5fd8 : SUCCESS in 2h 16m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/21ffc8598dff4fbb9d3b2411efcaf0b1 : SUCCESS in 2h 06m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/65391cf135a64682ae3d682955cb86d4 : SUCCESS in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9b3c1c457df4968b9b460523ae5d90d : SUCCESS in 2h 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"513ac514e8604c174f7c7df4fa8ba349a6b59896","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 20:26:15.000000000","message":"Patch Set 45: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":45},{"id":"e16f8365aa4807c356e9ba9977314d1679e810dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 08:30:13.000000000","message":"Uploaded patch set 46: New patch set was added with same tree, parent, and commit message as Patch Set 45.","accounts_in_message":[],"_revision_number":46},{"id":"3ddf4bdb38a0dd2d4c6ba658ffdb549e43eb25b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 09:27:30.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.","accounts_in_message":[],"_revision_number":47},{"id":"cc34a2010f86d68a65c2b357ed7039816c6115b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:54:33.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"55b58b82a079d0fb1f8e59e7f215f5c4e0345516","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:58:25.000000000","message":"Patch Set 48:\n\n(2 comments)","accounts_in_message":[],"_revision_number":48},{"id":"67632a88b1603c7a4d45d2aefa6337aee50089ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:59:41.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.","accounts_in_message":[],"_revision_number":49},{"id":"2d1473e65de8084fc7ff8804190d3b14b1996403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:32:10.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/022a7d2bbac3433d84c4cc379c075a2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0600bb8f89c49bc9ee5648840d5aadb : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427d6c413702465b9d164db881836632 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7edb5c3696cd483f8119836d184a523c : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/018484a24faa4955a11f0be9f649b230 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8bdea60d984c8a8b024e1cec4f711c : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8424b02287948fd928efa286dc8c644 : SUCCESS in 5m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/182237216b304c14afb9eac5ca0d3be4 : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bea89e592add404c99a904cda016bb25 : SUCCESS in 7m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5839a27eaacc423cb2fcdcd57b2aee0c : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04534100ca1a45fc87ae759804c93fad : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d06e2039041942c68193868ee052d7b0 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2aeffead75b74ad592e8f061f198d646 : SUCCESS in 8m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd44a3763d684747b0234bdef1af6de9 : SUCCESS in 1h 36m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20faa019b2b540c28476ad2d8d3c2b9c : FAILURE in 29m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/733f631a8c1645cba5466a859059020d : FAILURE in 29m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/977c44ebd8dc4366b94097f563767d15 : FAILURE in 26m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3e871473aa204e1ebc70ddf0250b4498 : FAILURE in 1h 23m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f7188b7f9bc4f0d8a3540f0dad73be0 : FAILURE in 35m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/642056c592634511916b2df899909366 : FAILURE in 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f7e977c7ccd4a7da14fbe261b2c0909 : FAILURE in 25m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c84e1a0ba2324e31a86be1bef5d79969 : FAILURE in 24m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bcf40742a642403587dbcf62c7799299 : FAILURE in 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cde06b616520469489557120874b3bf1 : FAILURE in 25m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ea50defe00b45ac8c37b8b126adbdfb : SUCCESS in 2h 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"2f9432247a421743dfb8599ee5f80c73bc4d4e05","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-08 16:01:14.000000000","message":"Patch Set 49: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"7e51ef4e1d703280b35951b8d41b5314e4014927","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-09 17:54:37.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"98324042727dab27568793065c8e99e6b4e6b967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:35:29.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b868b318810a49ed87371b230ca01aa0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57ebc3571c04aa9b13605f9a776fab5 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faaee6d2a2ff4857856ed202fb1f0dcc : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75e03e856bc84958818d413dd5560583 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d2f357b113a4dde906d51e976186cc7 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4e609d94bd439380c6fdf673b82779 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef795c72b27f4765a87f8d091b948012 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/70741d7bc2b64d9297201dd3fb853fea : SUCCESS in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/324801cc8b9645dfaa6d33fb50d8fd87 : SUCCESS in 6m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/91483825036d4b7480434bc7f7e1fdfa : SUCCESS in 3m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d715ac3220d2472898bb668915ec8754 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e5ba33c6ef14573b1a4ab58de1571e9 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69f0f9ee4e764089b924c9fb436a820b : SUCCESS in 6m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e72662d92df84a879391820dcdfeaaf4 : SUCCESS in 1h 12m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e764f6c9fb9d4265b47a71bb5cbf1268 : SUCCESS in 1h 35m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1430f2218e7a40eab5401eee589f7434 : SUCCESS in 44m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19df85c877f14f5a9e52c69911b75ad7 : SUCCESS in 55m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d4f769a31a0744a9b05809208d7708bc : POST_FAILURE in 1h 21m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/604b834181e64d53a707c536e8dd2532 : SUCCESS in 32m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f049c8c9d7624c6781392e718dae5799 : SUCCESS in 1h 53m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2bd47bac619b4bbb9789a624ceed843d : SUCCESS in 2h 09m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29d15cd23b554e75b12e86c0d2511473 : SUCCESS in 2h 21m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8db2597c0aec428888d78226e3a381dd : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9635ada4cdcc454aa6a0487918d4851a : SUCCESS in 2h 02m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/27d42197997a409f9359903b3daf5edf : POST_FAILURE in 1h 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"1b179706764ce710c5d8e0553cbc9a89456281da","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 08:50:07.000000000","message":"Patch Set 49: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":49},{"id":"9c8b5f4747ca5be289e81bb7327c8b19bf7d464a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:50:53.000000000","message":"Patch Set 49: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":49},{"id":"60ff1e3b72fabc7cdaff515d5bd0c209d2f81c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:16:18.000000000","message":"Patch Set 49: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3c5d0604734c5d96e4645fbdada465\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c26c0f9f914b999d0ebbb4ec7f78bb : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b2463fcb1d74c438e77c335024b39e7 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe4b9b4872bd4ecc8d745a0c331f7170 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7111b3d8c9f412fa5b91b473efc8afa : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d25447f5ffcf4098aaa77f08ff680dee : SUCCESS in 5m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf407997710d4cd2921da9a40cbacb09 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19a7d0883920481a8d95d59083674306 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba7df489837340c799d9517ab7319c79 : SUCCESS in 1h 13m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/731849334f974aafa5cade604d3394e2 : SUCCESS in 2h 16m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3eb576f57bc4619a0ff440e48933e0e : SUCCESS in 47m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd49e057d92c409aac379dfe0a3d795d : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":49},{"id":"e714cb7d93f0bc44c9a2927d0464990f24f0eea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:16:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":49},{"id":"f74eb1c1ae24fde28db992b3625bd5c1eeeba9e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:17:27.000000000","message":"Patch Set 49:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b7f0131f2647f9b0dfc06327fc9216\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a313041bcdfb4462936aeef1be87860d : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74b72d0254bb4547ae9d0f448d267a1a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":49},{"id":"23bd22f625e3da769d26a3abd0bb98d3c86fb02e","tag":"autogenerated:gerrit:revert","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 11:07:05.000000000","message":"Created a revert of this change as Ic263c3e316f7bdab832b9c7d22e9258924f8f9c7","accounts_in_message":[],"_revision_number":49}],"current_revision_number":49,"current_revision":"454cff587ed10b5e504da93b074b77cb85055b13","revisions":{"874a48a92162ef01e9b5800bdfc23a297125baf9":{"kind":"REWORK","_number":1,"created":"2022-08-18 09:10:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/1"}}},"commit":{"parents":[{"commit":"5a4abb615fa9ead5dc7bfbcfc793a31612f50581","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4abb615fa9ead5dc7bfbcfc793a31612f50581"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 09:10:18.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874a48a92162ef01e9b5800bdfc23a297125baf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874a48a92162ef01e9b5800bdfc23a297125baf9"}]},"branch":"refs/heads/master"},"15066c5ec665435c98219580755b2cf7fa93e9bf":{"kind":"REWORK","_number":2,"created":"2022-08-18 10:02:20.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/2"}}},"commit":{"parents":[{"commit":"5a4abb615fa9ead5dc7bfbcfc793a31612f50581","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4abb615fa9ead5dc7bfbcfc793a31612f50581"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 10:02:09.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15066c5ec665435c98219580755b2cf7fa93e9bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15066c5ec665435c98219580755b2cf7fa93e9bf"}]},"branch":"refs/heads/master"},"ac61735bf0c084245bf62883c737b639e2de4184":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-19 09:44:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/3"}}},"commit":{"parents":[{"commit":"624de65e52f6e2793643e842796520ab2045efbf","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624de65e52f6e2793643e842796520ab2045efbf"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:43:47.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac61735bf0c084245bf62883c737b639e2de4184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac61735bf0c084245bf62883c737b639e2de4184"}]},"branch":"refs/heads/master"},"64ae1cbd10639451d16f9c13ef1ec9bb6091dc63":{"kind":"REWORK","_number":4,"created":"2022-08-19 10:37:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/4"}}},"commit":{"parents":[{"commit":"97e195ff825775a21cc2f8b12e55cce507561ccc","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97e195ff825775a21cc2f8b12e55cce507561ccc"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:37:12.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64ae1cbd10639451d16f9c13ef1ec9bb6091dc63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64ae1cbd10639451d16f9c13ef1ec9bb6091dc63"}]},"branch":"refs/heads/master"},"e14f66591e86f330562713f8bd7d7439aab42d59":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-08-19 10:47:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/5"}}},"commit":{"parents":[{"commit":"cdb963e959aada6b94b93a6c6eee25d9a112a1e4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdb963e959aada6b94b93a6c6eee25d9a112a1e4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:47:40.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e14f66591e86f330562713f8bd7d7439aab42d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e14f66591e86f330562713f8bd7d7439aab42d59"}]},"branch":"refs/heads/master"},"177d05d9b26e3cbf758099f9802fb5e669ac6f4f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-08-19 13:22:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/6"}}},"commit":{"parents":[{"commit":"cda2f0008f378967cdd1e3e879f596644a302b57","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda2f0008f378967cdd1e3e879f596644a302b57"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 13:22:10.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/177d05d9b26e3cbf758099f9802fb5e669ac6f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/177d05d9b26e3cbf758099f9802fb5e669ac6f4f"}]},"branch":"refs/heads/master"},"ecc4dcdc02fbfd0ce200de2b424238ab098694b1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-08-19 14:46:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/7"}}},"commit":{"parents":[{"commit":"2ab7cc53424cd81c021ef5d9deda5aa3057537f4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 14:46:17.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to 500000 which allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecc4dcdc02fbfd0ce200de2b424238ab098694b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecc4dcdc02fbfd0ce200de2b424238ab098694b1"}]},"branch":"refs/heads/master"},"2947cb8f35f445e3fdd85bfb4ee194b19c15b9d6":{"kind":"REWORK","_number":8,"created":"2022-08-19 15:36:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/8"}}},"commit":{"parents":[{"commit":"2ab7cc53424cd81c021ef5d9deda5aa3057537f4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 15:34:58.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2947cb8f35f445e3fdd85bfb4ee194b19c15b9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2947cb8f35f445e3fdd85bfb4ee194b19c15b9d6"}]},"branch":"refs/heads/master"},"fe9e193baaeb154b6ba15c4c43bbde047b76192c":{"kind":"REWORK","_number":9,"created":"2022-08-19 15:38:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/9"}}},"commit":{"parents":[{"commit":"2ab7cc53424cd81c021ef5d9deda5aa3057537f4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 15:38:16.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cache) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe9e193baaeb154b6ba15c4c43bbde047b76192c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe9e193baaeb154b6ba15c4c43bbde047b76192c"}]},"branch":"refs/heads/master"},"dd8209e3007d0bb0ba54459c5aea244f8e7fb438":{"kind":"REWORK","_number":10,"created":"2022-08-19 15:42:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/10"}}},"commit":{"parents":[{"commit":"2ab7cc53424cd81c021ef5d9deda5aa3057537f4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 15:42:04.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd8209e3007d0bb0ba54459c5aea244f8e7fb438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd8209e3007d0bb0ba54459c5aea244f8e7fb438"}]},"branch":"refs/heads/master"},"2c77caa527da0fcb37d9800f387410224363dbb2":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-08-21 15:00:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/11"}}},"commit":{"parents":[{"commit":"201bc774e9d485e3020fdaeccb99ac7242f235ed","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/201bc774e9d485e3020fdaeccb99ac7242f235ed"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:00:45.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c77caa527da0fcb37d9800f387410224363dbb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c77caa527da0fcb37d9800f387410224363dbb2"}]},"branch":"refs/heads/master"},"1b7f605d009ac2c06fff25ff9e2598e8b772e4e5":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-08-22 08:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/12"}}},"commit":{"parents":[{"commit":"5b408565d5181b3d45d67c3041e4adc8e0e94aa4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b408565d5181b3d45d67c3041e4adc8e0e94aa4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 08:08:06.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b7f605d009ac2c06fff25ff9e2598e8b772e4e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b7f605d009ac2c06fff25ff9e2598e8b772e4e5"}]},"branch":"refs/heads/master"},"a038d9479a2e884d480a293ac6e450caa08a5f52":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-08-22 13:52:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/13"}}},"commit":{"parents":[{"commit":"d6d92cf61e4c1f84d69497bf1fbeb7326ba0fb40","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6d92cf61e4c1f84d69497bf1fbeb7326ba0fb40"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 13:52:45.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a038d9479a2e884d480a293ac6e450caa08a5f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a038d9479a2e884d480a293ac6e450caa08a5f52"}]},"branch":"refs/heads/master"},"c4efac5f8805c0a51d813841af1b233e27e9c4f9":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-08-23 10:07:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/14"}}},"commit":{"parents":[{"commit":"28d4350b3224e13863e231fd99b71d11bb12e86e","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28d4350b3224e13863e231fd99b71d11bb12e86e"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 10:06:51.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4efac5f8805c0a51d813841af1b233e27e9c4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4efac5f8805c0a51d813841af1b233e27e9c4f9"}]},"branch":"refs/heads/master"},"002cc2c732f023356bdfc63a1c5740966acf38cf":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-08-23 12:45:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/15"}}},"commit":{"parents":[{"commit":"a006ae159521b3670ba48f73a1b2cf355164ddb4","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a006ae159521b3670ba48f73a1b2cf355164ddb4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 12:45:50.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/002cc2c732f023356bdfc63a1c5740966acf38cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/002cc2c732f023356bdfc63a1c5740966acf38cf"}]},"branch":"refs/heads/master"},"f10f9bb7670f049ed63731f2fe71a94f4bbbd94a":{"kind":"REWORK","_number":16,"created":"2022-08-30 11:21:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/16"}}},"commit":{"parents":[{"commit":"895b1038690070b447fffbe502f73d39888a8d51","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/895b1038690070b447fffbe502f73d39888a8d51"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:19:48.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f10f9bb7670f049ed63731f2fe71a94f4bbbd94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f10f9bb7670f049ed63731f2fe71a94f4bbbd94a"}]},"branch":"refs/heads/master"},"43a7fa87ced8b147438fa8c0b4b260d06ec16386":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-08-30 11:35:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/17"}}},"commit":{"parents":[{"commit":"d2112ea041a03379b68342976026bf05d5f74700","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2112ea041a03379b68342976026bf05d5f74700"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:35:03.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a7fa87ced8b147438fa8c0b4b260d06ec16386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43a7fa87ced8b147438fa8c0b4b260d06ec16386"}]},"branch":"refs/heads/master"},"6a6c85a0ccd76f9b53ad3f8d308c91c0bb801b1c":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-08-30 14:22:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/18"}}},"commit":{"parents":[{"commit":"28f966e0795b375f5a0c20ea8146757f3e069f4a","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f966e0795b375f5a0c20ea8146757f3e069f4a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:22:20.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a6c85a0ccd76f9b53ad3f8d308c91c0bb801b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a6c85a0ccd76f9b53ad3f8d308c91c0bb801b1c"}]},"branch":"refs/heads/master"},"3915fb0c5aebe97e1e4d280fd9923e19f22db974":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-08-30 14:53:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/19"}}},"commit":{"parents":[{"commit":"ec0b9462aedb3861578e09b2e9e3db0b8d49fa7c","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec0b9462aedb3861578e09b2e9e3db0b8d49fa7c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:29:31.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3915fb0c5aebe97e1e4d280fd9923e19f22db974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3915fb0c5aebe97e1e4d280fd9923e19f22db974"}]},"branch":"refs/heads/master"},"565866e62b9ff78faf03c63f7e72066a9b71e33d":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-08-31 07:29:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/20"}}},"commit":{"parents":[{"commit":"893ac804510df9ee18dbc714bd683961efce6779","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/893ac804510df9ee18dbc714bd683961efce6779"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 07:29:43.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/565866e62b9ff78faf03c63f7e72066a9b71e33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/565866e62b9ff78faf03c63f7e72066a9b71e33d"}]},"branch":"refs/heads/master"},"bdb0ff5c613f95e3337508e99c0096d1cf72accd":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-08-31 09:09:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/21"}}},"commit":{"parents":[{"commit":"95556136a681cb79a12e91b96a191ae4ac5501f1","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95556136a681cb79a12e91b96a191ae4ac5501f1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 09:09:40.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdb0ff5c613f95e3337508e99c0096d1cf72accd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdb0ff5c613f95e3337508e99c0096d1cf72accd"}]},"branch":"refs/heads/master"},"aeb47d0cc947848a489a92096431ef841a455d5b":{"kind":"REWORK","_number":22,"created":"2022-09-01 08:56:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/22"}}},"commit":{"parents":[{"commit":"abed41dd40f63f30a926ce5d7306fefe3bab7106","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abed41dd40f63f30a926ce5d7306fefe3bab7106"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-01 08:55:12.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aeb47d0cc947848a489a92096431ef841a455d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aeb47d0cc947848a489a92096431ef841a455d5b"}]},"branch":"refs/heads/master"},"98e29369c96dae1e710f431681a90340c293d401":{"kind":"REWORK","_number":23,"created":"2022-09-12 09:57:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/23"}}},"commit":{"parents":[{"commit":"2a851fb5aa839edb7a003138f11eed8246465651","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a851fb5aa839edb7a003138f11eed8246465651"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 09:57:08.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98e29369c96dae1e710f431681a90340c293d401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98e29369c96dae1e710f431681a90340c293d401"}]},"branch":"refs/heads/master"},"8931fc6437062bb2cd45bc28ab8ebabf2c52e62d":{"kind":"REWORK","_number":24,"created":"2022-09-12 10:03:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/24"}}},"commit":{"parents":[{"commit":"350636ba1acb61e12111003bb7d12ed0ba05bdf7","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/350636ba1acb61e12111003bb7d12ed0ba05bdf7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:03:29.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8931fc6437062bb2cd45bc28ab8ebabf2c52e62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8931fc6437062bb2cd45bc28ab8ebabf2c52e62d"}]},"branch":"refs/heads/master"},"dee97a2474774383fc05935d46b1261c39850db0":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-09-12 10:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/25"}}},"commit":{"parents":[{"commit":"85005f5e5713050f7fbf3014f16c58651e5137c5","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85005f5e5713050f7fbf3014f16c58651e5137c5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:08:05.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dee97a2474774383fc05935d46b1261c39850db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dee97a2474774383fc05935d46b1261c39850db0"}]},"branch":"refs/heads/master"},"89d2434bed892adf69b50a1428cff272d0fff728":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2022-09-12 12:28:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/26"}}},"commit":{"parents":[{"commit":"e0d6d7420829f1d787004d664499f31aaeb497a9","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d6d7420829f1d787004d664499f31aaeb497a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 12:27:46.000000000","tz":120},"subject":"Increase HAProxy SSL cache size for better performance","message":"Increase HAProxy SSL cache size for better performance\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about one quarter of the available memory\n(free + buffers + cached) on the amphora.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d2434bed892adf69b50a1428cff272d0fff728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d2434bed892adf69b50a1428cff272d0fff728"}]},"branch":"refs/heads/master"},"e9c01b04bfa5f6eaacd58140cd35c4c3dd69fd25":{"kind":"REWORK","_number":27,"created":"2022-09-14 10:30:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/27"}}},"commit":{"parents":[{"commit":"e0d6d7420829f1d787004d664499f31aaeb497a9","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d6d7420829f1d787004d664499f31aaeb497a9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-14 10:29:03.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9c01b04bfa5f6eaacd58140cd35c4c3dd69fd25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9c01b04bfa5f6eaacd58140cd35c4c3dd69fd25"}]},"branch":"refs/heads/master"},"0dfc6093ed6688e557bd8083fedd3116d5ae86fb":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-10-19 09:09:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/28"}}},"commit":{"parents":[{"commit":"5d56bd0021ee7e9035d5af6b11528400186d21bf","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d56bd0021ee7e9035d5af6b11528400186d21bf"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:09:22.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dfc6093ed6688e557bd8083fedd3116d5ae86fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0dfc6093ed6688e557bd8083fedd3116d5ae86fb"}]},"branch":"refs/heads/master"},"7b248b4f4b2882072e4366c3c5b5a910e16cdcd4":{"kind":"NO_CHANGE","_number":29,"created":"2022-10-19 09:11:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/29"}}},"commit":{"parents":[{"commit":"9fc07a3fa709b3a242a506e6f82cbf461c853a89","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fc07a3fa709b3a242a506e6f82cbf461c853a89"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:10:48.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b248b4f4b2882072e4366c3c5b5a910e16cdcd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b248b4f4b2882072e4366c3c5b5a910e16cdcd4"}]},"branch":"refs/heads/master"},"ffee59f150f820882100eb3943753b9f91bf69fe":{"kind":"REWORK","_number":30,"created":"2022-10-27 11:30:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/30"}}},"commit":{"parents":[{"commit":"05f8ba4ffd5ce6833b17cf0a5f5461ede0457d83","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f8ba4ffd5ce6833b17cf0a5f5461ede0457d83"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 11:29:16.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffee59f150f820882100eb3943753b9f91bf69fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffee59f150f820882100eb3943753b9f91bf69fe"}]},"branch":"refs/heads/master"},"f5cc886d1568d6d137a947a76671b6b6eb8a38ab":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2022-10-27 12:22:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/31"}}},"commit":{"parents":[{"commit":"eaeff20819ca9278ec1142d1080506178f4c72c6","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaeff20819ca9278ec1142d1080506178f4c72c6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 12:22:29.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5cc886d1568d6d137a947a76671b6b6eb8a38ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5cc886d1568d6d137a947a76671b6b6eb8a38ab"}]},"branch":"refs/heads/master"},"43aae3f2217b80d407c793753779ab548bcadc26":{"kind":"REWORK","_number":32,"created":"2022-10-27 13:04:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/32"}}},"commit":{"parents":[{"commit":"def0d0c65280650151442de434fe4539e9d7a68f","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/def0d0c65280650151442de434fe4539e9d7a68f"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:04:10.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43aae3f2217b80d407c793753779ab548bcadc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43aae3f2217b80d407c793753779ab548bcadc26"}]},"branch":"refs/heads/master"},"57684a535530e2d9d13a6e68ffce490a5b6826ea":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2022-10-27 13:08:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/33"}}},"commit":{"parents":[{"commit":"5bb3f40633cad37f87471e5d21bda5b5a2598b61","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bb3f40633cad37f87471e5d21bda5b5a2598b61"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:08:15.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57684a535530e2d9d13a6e68ffce490a5b6826ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57684a535530e2d9d13a6e68ffce490a5b6826ea"}]},"branch":"refs/heads/master"},"503d998d3b8e5d03de9167a48655731c0a3135de":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2022-10-27 14:41:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/34"}}},"commit":{"parents":[{"commit":"8de49d1479790bb31e228b7ba524dbaa7df2ac1c","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8de49d1479790bb31e228b7ba524dbaa7df2ac1c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 14:41:24.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/503d998d3b8e5d03de9167a48655731c0a3135de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/503d998d3b8e5d03de9167a48655731c0a3135de"}]},"branch":"refs/heads/master"},"d0c6088f284355dc80b58b64ce4a75cacba1c850":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2022-10-27 16:19:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/35"}}},"commit":{"parents":[{"commit":"09f2f57b991d766e7dd8b6afee89dd7358b2dd37","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09f2f57b991d766e7dd8b6afee89dd7358b2dd37"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 16:19:16.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c6088f284355dc80b58b64ce4a75cacba1c850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c6088f284355dc80b58b64ce4a75cacba1c850"}]},"branch":"refs/heads/master"},"a421a1444bbf63a6473b839caaef36946348db88":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2022-10-28 09:35:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/36"}}},"commit":{"parents":[{"commit":"836c4f092d19ba1c7b26f08168419dbc83e119b6","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/836c4f092d19ba1c7b26f08168419dbc83e119b6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-28 09:35:36.000000000","tz":120},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a421a1444bbf63a6473b839caaef36946348db88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a421a1444bbf63a6473b839caaef36946348db88"}]},"branch":"refs/heads/master"},"e9ab441ee52aec454f457dd2e45c268500e05507":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2022-11-14 16:24:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/37"}}},"commit":{"parents":[{"commit":"1f4172e13c227fef647a8a194c037c06682cb84b","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f4172e13c227fef647a8a194c037c06682cb84b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 16:24:02.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ab441ee52aec454f457dd2e45c268500e05507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ab441ee52aec454f457dd2e45c268500e05507"}]},"branch":"refs/heads/master"},"930d98c49a29651e7eb3182f6680bdc226267f76":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2022-11-14 19:03:55.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/38"}}},"commit":{"parents":[{"commit":"0d22d9567c637efc429cde4dca37471a36e126b1","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d22d9567c637efc429cde4dca37471a36e126b1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 19:03:40.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/930d98c49a29651e7eb3182f6680bdc226267f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/930d98c49a29651e7eb3182f6680bdc226267f76"}]},"branch":"refs/heads/master"},"df93af1902cbcfdff707ccca027c4b6fd838a683":{"kind":"NO_CHANGE","_number":39,"created":"2022-11-16 14:45:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/39"}}},"commit":{"parents":[{"commit":"b239d0a51abff31b39b8b0a285844cfa98ebfde3","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b239d0a51abff31b39b8b0a285844cfa98ebfde3"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-16 14:45:05.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df93af1902cbcfdff707ccca027c4b6fd838a683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df93af1902cbcfdff707ccca027c4b6fd838a683"}]},"branch":"refs/heads/master"},"6097ba1ca6d0e3b8e388e7905fe6eea823c8445f":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2022-11-23 15:29:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/40"}}},"commit":{"parents":[{"commit":"d53bcfe90b514571dce586a8b3939d9d18c1427b","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53bcfe90b514571dce586a8b3939d9d18c1427b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-23 15:29:28.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6097ba1ca6d0e3b8e388e7905fe6eea823c8445f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6097ba1ca6d0e3b8e388e7905fe6eea823c8445f"}]},"branch":"refs/heads/master"},"1ed9be0eb46202e406d6936a272f106afde60c33":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2022-11-28 11:00:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/41"}}},"commit":{"parents":[{"commit":"d0c42e2b8bba89394c868e37abd7bf041252ad81","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c42e2b8bba89394c868e37abd7bf041252ad81"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 10:59:34.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ed9be0eb46202e406d6936a272f106afde60c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ed9be0eb46202e406d6936a272f106afde60c33"}]},"branch":"refs/heads/master"},"880ee448033d42b79d51cac64bd36d711091e278":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2022-11-28 15:46:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/42"}}},"commit":{"parents":[{"commit":"699cb1745005e89ef780a73f038d71502c10ab5c","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/699cb1745005e89ef780a73f038d71502c10ab5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 15:46:47.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/880ee448033d42b79d51cac64bd36d711091e278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/880ee448033d42b79d51cac64bd36d711091e278"}]},"branch":"refs/heads/master"},"15ab46047ea5c8cb2cba706d368054ccb888aef4":{"kind":"REWORK","_number":43,"created":"2022-12-07 12:14:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/43"}}},"commit":{"parents":[{"commit":"699cb1745005e89ef780a73f038d71502c10ab5c","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/699cb1745005e89ef780a73f038d71502c10ab5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:14:39.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15ab46047ea5c8cb2cba706d368054ccb888aef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15ab46047ea5c8cb2cba706d368054ccb888aef4"}]},"branch":"refs/heads/master"},"3324833a0523913673ba8e169cdec3b029c013a1":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2022-12-07 12:18:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/44"}}},"commit":{"parents":[{"commit":"412f1d79c8c734661fc0ed33c2d79405eb6bd5af","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/412f1d79c8c734661fc0ed33c2d79405eb6bd5af"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:18:01.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3324833a0523913673ba8e169cdec3b029c013a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3324833a0523913673ba8e169cdec3b029c013a1"}]},"branch":"refs/heads/master"},"bfcafe51b2bbf05d37f0a4cd7ffa5b5c076e574d":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2022-12-07 12:21:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/45"}}},"commit":{"parents":[{"commit":"9e4a3ceddf38e850e1a4ef6437906a6d601b14d9","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e4a3ceddf38e850e1a4ef6437906a6d601b14d9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:21:23.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfcafe51b2bbf05d37f0a4cd7ffa5b5c076e574d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfcafe51b2bbf05d37f0a4cd7ffa5b5c076e574d"}]},"branch":"refs/heads/master"},"7ca2ca9efc754928d06fc7071a644d6c52cba130":{"kind":"NO_CHANGE","_number":46,"created":"2022-12-08 08:30:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/46"}}},"commit":{"parents":[{"commit":"b527b22ca504f673637b6b03817d7d5d0b97407d","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b527b22ca504f673637b6b03817d7d5d0b97407d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 08:29:55.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ca2ca9efc754928d06fc7071a644d6c52cba130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ca2ca9efc754928d06fc7071a644d6c52cba130"}]},"branch":"refs/heads/master"},"5742f832839fea16ebc89b35c85156d126ac8ba8":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2022-12-08 09:27:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/47"}}},"commit":{"parents":[{"commit":"a459cea3c33ca16784c64643079faeaf19f9c4c0","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a459cea3c33ca16784c64643079faeaf19f9c4c0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 09:27:17.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5742f832839fea16ebc89b35c85156d126ac8ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5742f832839fea16ebc89b35c85156d126ac8ba8"}]},"branch":"refs/heads/master"},"e3d28e2fa6324ce1fd09a80a7a2afd03230b264f":{"kind":"REWORK","_number":48,"created":"2022-12-08 10:54:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/48"}}},"commit":{"parents":[{"commit":"8f8d603037407d9652fdad2fa8350148cba318ab","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f8d603037407d9652fdad2fa8350148cba318ab"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:54:07.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3d28e2fa6324ce1fd09a80a7a2afd03230b264f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3d28e2fa6324ce1fd09a80a7a2afd03230b264f"}]},"branch":"refs/heads/master"},"454cff587ed10b5e504da93b074b77cb85055b13":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2022-12-08 10:59:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/06/853606/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/853606/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/853606/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/853606/49"}}},"commit":{"parents":[{"commit":"baefc3b4643d9e61361cacf46dfbeb95ce96933b","subject":"Set nohz_full in cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/baefc3b4643d9e61361cacf46dfbeb95ce96933b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-18 08:54:45.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:59:32.000000000","tz":60},"subject":"Optimize HAProxy SSL cache size","message":"Optimize HAProxy SSL cache size\n\nIf a HTTPS termination listener exists, set the tune.ssl.cachesize\nsetting to use about half of the available\nmemory (free + buffers + cached) on the amphora minus the memory needed\nfor network sockets based on the global max connections setting.\nA larger SSL cache allows for more resumed SSL sessions and\ntherefore less computationally expensive SSL handshakes.\n\nChange-Id: I87efba18017aa3e8b9b3cc812664efc11af1c4c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/454cff587ed10b5e504da93b074b77cb85055b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/454cff587ed10b5e504da93b074b77cb85055b13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
