)]}'
{"id":"openstack%2Foctavia~853802","triplet_id":"openstack%2Foctavia~master~Ie009aff52d572873da0bb5cc43fedd7f3325dccc","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-12-08 16:00:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-12 10:47:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie009aff52d572873da0bb5cc43fedd7f3325dccc","subject":"Set nohz_full in cpu-pinning element","status":"MERGED","created":"2022-08-19 09:44:17.000000000","updated":"2023-01-12 10:49:44.000000000","submitted":"2023-01-12 10:47:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853802-amp-cpu-pinning","meta_rev_id":"291c9fc6ff612908e1c294c78fb8eb9f5d76ea5d","_number":853802,"virtual_id_number":853802,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 10:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-01-11 16:50:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-08 16:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-01-12 08:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-01-12 08:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 17:28:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 19:59:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-11 16:50:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-01-12 08:16:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"914765cbc99d0804eee2d8eb15f960ebdcd5aff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 09:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbd2cb71edd80e655396a6a743509d6d5c174e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:37:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"56b250fdd3824784e413370ac35a31e71b662ec6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 10:47:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"201c2e97d153c60b27875c8b700aff3f42f7ef07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 13:22:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2321736669116b807b20de7ce5f0501b13c6bfc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-19 14:46:32.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ba651b6b242c74686ea66022980f51b7bc816edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:28:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/def9d3391eab446089242242a3fb18a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6639a8fe0ce144ae8be8de3597ec8f70 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a3d0be89f1492ca6abe14f276d6d46 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd1e19ec7364f129d9f9193de48d9a9 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c74a5dd9b1a146e3af295ff69d64885e : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/690804015a974735a393f161663e607f : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3775431b0bd54dd0b296f11ac34d446a : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a6d1796bfb46a0984548d3dc6b7b42 : SUCCESS in 5m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/36ec23bb7fd542f79134796eec31bc61 : SUCCESS in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/34cedeb477984153afde8ef29cc3f23f : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c69d96d8073f4800987c3164073cdf58 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bdd907176d7745ad82e9273c9e556597 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6454b659b9f43eeaba10450491f129f : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6a95ca341ad40f29d667b771d770a6c : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aedc71bee00e452c85c1bf1342daf49d : SUCCESS in 2h 22m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc0c8c3004b943db88b4c7654f9c4445 : SUCCESS in 1h 49m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c08145caf4f447c896dea7623ac875d : SUCCESS in 43m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/368f3165ad034045b7ca8f7c42db5bf5 : SUCCESS in 57m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eb4e36a362fa41b8bdbc393463507cc6 : SUCCESS in 1h 32m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54d35e8103cd4e2a91c5e78c41734033 : SUCCESS in 46m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/08fe3bb5ed6f42948b5070cda4d2afe6 : SUCCESS in 1h 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7df6ab5d07804ab0a22059015facc1db : SUCCESS in 2h 21m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f847dacce2aa4e2a9d627e6dbc7480e0 : SUCCESS in 1h 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/09112b1518df4d4ea2744ef365abe7de : SUCCESS in 1h 50m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5258ca8ed80d4691ad459935de78f9f8 : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a33ea6b0aa934b188e1ba6ff03818dd1 : SUCCESS in 2h 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"300f258260c5ad35fdd184772ca3b4707dd2f8d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:00:58.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f40109430f02e5874b7e7ae9aa3aa454722011f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 17:52:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16c7f30b7aa4411a56a4b041f89fa1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cda2c9c24a94dc4b4708742779f7cf7 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b0e5cd654544ee866385ae9dc2d610 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e170dc558bb747c1b48e729f46829434 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40a4b9e645f74983b4aa23b199f54cf8 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc848047a99b4887a8f393447ffde17a : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf664afadf9344928cc86193f7198ba3 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8691185321f44336bd6c61ee36a690cf : SUCCESS in 5m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/54b44ef413224ccea366215f372350ca : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3717bda446a944468d3233d451460c4c : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3eaaf8c05d7941ecb4dbb851e5ef7dc4 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2df9885381674436a886891620fac269 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/282b5527d39d4093bcf6d949211c6915 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54adef5df26c4b159d1603cadc90d4a1 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53f8a31d58364cf1ada5bbd4ee0966df : SUCCESS in 1h 39m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/348ffea051d64e3da9e3847f1b4b405d : SUCCESS in 1h 49m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b5f8fccac7b4cb5a26793b982dfb7e0 : SUCCESS in 56m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/06a3f73520634e4eb78e049bea149902 : SUCCESS in 1h 08m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0ba55e62ea2f4c1f9bfa27275fe9d152 : SUCCESS in 1h 33m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/686ad74dfee841608d69c938afa955d9 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bc8095e63f6448cd92b1ce316dec1bf8 : SUCCESS in 2h 11m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c29330a3a9af4045b327ec54506a3fa7 : SUCCESS in 2h 17m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d0b08beb8164c7280e348bb1dec627b : SUCCESS in 2h 11m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c682b04a3d644d9a8797c5df91405d6c : SUCCESS in 2h 11m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/31f10d8cb0f94ec18f378134ae98f72d : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bef04ca2c0744334bb155d67356de0fe : SUCCESS in 2h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0f79a25568eceb58d5126bda13c6e71df89a5ee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 08:08:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ed02903ff5d932d6710af6d028510cd9bfbc7af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:27:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788e7ddfdc574b0da12b92214a1eb3e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/319cf0b0757f42b79a595c5649fbe2c8 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0753dab5d234f94ab358a2b7223b34d : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f7795dbf67433eaffdb7325c9e4d4c : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f33a734622e442db80977569c6e2662 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33bd26a823c54539b69663e7dd1f285c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2309339c3c440b1b78af52a33bd9da5 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35f758e454db43b29ee1a5325e13fdea : SUCCESS in 5m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f2fbbf2b8a2d4dbb848cec833be3dce7 : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a46587e448d248fabe47f9d21006fe0a : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f2f41772f28a4b2d911c164c8b91fc47 : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1384795f3782427898a63cde94468bed : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/89d57da249204ba886299b85e5139085 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d36e1c2b171d4e8098f37630f1c15991 : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da7469633c4549bc9f36572a22882c3d : SUCCESS in 2h 14m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3e16f6b566f497e851357a6b7246044 : SUCCESS in 1h 31m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35d57cf4bcb34cbd9cf510811f236a58 : SUCCESS in 40m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8bdb7f06173041c2b4860760c0c88f80 : SUCCESS in 58m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/23b5fae0b9bf48d4908d41fb56b21bdd : SUCCESS in 1h 30m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77e3fc111cb2440a82e941775bcab8d2 : SUCCESS in 38m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/51fec070df18412fa87d4af4721a9d91 : SUCCESS in 1h 55m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5d429c448f04aaa96374f7e78a93ac6 : SUCCESS in 1h 53m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1ac05681af9944a1845367153843b4a8 : SUCCESS in 1h 38m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c203a12a7d5e4eb585327962763ecf19 : SUCCESS in 1h 45m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/876f0db8a5624f3ab8959399b35caace : SUCCESS in 1h 30m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e1cc65f2dba42e4b99d16e238a70292 : SUCCESS in 2h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d8b252a7424345b32477b453fb6eaf38b365a256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 13:52:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"4cd31fa9ffd8668b32dddfeac0b403392286db60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:13:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47301ea5c14d4c49943d4407ba3853aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb1c69f04315452e96b0134fc31c7c24 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384f23f7d4d04df0808635057bc5ab06 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a05871b6df64abf855a0d185f5cf120 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b52b00bd1394f69a505ca4b772b9060 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fef41ceb9c464820828ef1ceb8479576 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57aabc3fc009458a8181fba81a183387 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b26f154fba46ccb5055c65975f1e42 : SUCCESS in 5m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/79d8a563846c40279edd6b10138a3885 : SUCCESS in 5m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e7ba0cc8c71645ec9912a6e93b8ef3a3 : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a77b0bb5dc00416fa50b482352cd483b : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f627b49af7904f5ca1a9394720142946 : SUCCESS in 9m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6a270735c1349209b2f654865dc7149 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/329c83dee6fd4c1993101d083e7cb5d8 : SUCCESS in 8m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f529984374234ad2b2bdaa44d12affbe : SUCCESS in 2h 13m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ac7e51a213644bbb02a8b844b5186ce : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc3b2e3fbb264ca4a018bad7c0fc69d0 : SUCCESS in 43m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2028dd010ff4eb4af7223ef2b541375 : SUCCESS in 1h 10m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf000d11ca084fb188cdd9996a52b32f : SUCCESS in 1h 14m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b71965c00b784039ae122881e62c38fd : SUCCESS in 40m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb71ec3e6ae54225a660dee6e19b3a57 : SUCCESS in 1h 57m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5d4ae1718734c0e99dafbd89a5db62a : SUCCESS in 2h 16m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/44c44eaee85442c1bf7a2e90845aa0cf : SUCCESS in 1h 49m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b75eb9804b4340e29d9b7ed462cab904 : SUCCESS in 1h 50m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f8ddc0f8b9e446e48d636aa34af00c25 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc24021df6234e238e2239528deddf2b : SUCCESS in 2h 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bbc251b89289ad43525ed40c5dc8ca166d8253ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 10:07:03.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"a553af8365dea8b3a264731a9a0125f924d2d343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-23 12:45:59.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"69dc5c1903fcae2fbcfa7c7bad6f9f5617ecd0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 15:26:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a5c1bbefb34b2794aa3f920a73b871\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d92e0c4ade848b99a0d6ff14320b7ae : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e932f6fa8eb486e97827079d5443169 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19cc9e0b7f04436fb1efee6c70deb4f4 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35fe49150e02409986e0316ec26c47de : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/887baeeaaaeb4e868aca0db2fc5a79cc : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f13fabc9fe45d296d080cc3b059e70 : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148fb2a2e5644f16a7fb9fe4e689b6ab : SUCCESS in 6m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dd8ffbf135dd4c66a27cd8bc07c06c7c : SUCCESS in 6m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/00bcb9e8908045d986d4b25917d5b8eb : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9bb389c38934afb9a655b165ea60464 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8053460cd78d4e61aec9bf9c7a6bf1da : SUCCESS in 8m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d4157b6848244aea7200455354947d4 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fbb15e38782455e9eaa383f143a9c25 : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2750fbbe826041a3994d4a7f1266d511 : SUCCESS in 2h 13m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3f09c2c1ad64ac6aa98aef31d9b0cfe : SUCCESS in 1h 42m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19b0d94f6e774c698d782d2e57a4ef7a : SUCCESS in 44m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a377f4706c7041aeab90feaf2e4dfa9b : SUCCESS in 58m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7725434c382c4e2a9414592d2c3cd8d1 : SUCCESS in 1h 07m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b574068cf8914adfb1acc78df079ed9f : SUCCESS in 42m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e1ca8313ec44be2839ade9ae76c5f87 : SUCCESS in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46a3b7b069fa42cfb22ce37f230e6290 : SUCCESS in 1h 47m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/17203b247cfe4846acddb5fc6a1ab423 : SUCCESS in 1h 58m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a957ad409c9a42e0b5304c26ed232761 : SUCCESS in 1h 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/10d9c46ac9304bebb617087c17f406a9 : SUCCESS in 1h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/baa779b5121f4f90a22e857684a44b46 : SUCCESS in 2h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"872c1971b6871e2bd821bd299d254560d08ee8b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:21:15.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"967688759032077aa5a298f0c56e8704c6f9aa51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 11:35:14.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"8a90167e6577de8a1ebc19a25f67ea223befd79d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:22:33.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"fb8e43ee205f3d7ee5abd42dab79d85c3c86369d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-30 14:53:12.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"dbcca88552e57ecdbd87a7b780e855b3205bcf91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:07:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe7f4d865764ae4b799e18dc9334cd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9564e6f5797f4af78d12475a0c8cdfbe : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b89dfd3eb7ca414eae967f4bd675508f : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e9e565cdc13413facf07a12b77601ab : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81e9e03d9f28467ca828bc41795b1342 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1d554e484a34eafb95c369d64855ff5 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5e0c164618460e9e049eafbacb1c59 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6d9a967ac6047a7a9ecff7172cd7c11 : SUCCESS in 6m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/44fe449ee18346c8938af7a30b6e50f0 : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4c741e2059354e4ba03e29c3f33c3cef : SUCCESS in 9m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e83f94cc111f4ec3afba78d96da044a6 : SUCCESS in 5m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3ad716e9f94443696e2b0737ad3977c : SUCCESS in 8m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/264aee61875f4e388656600bec106fb9 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1599ff12bae942239749159f55f1ca19 : SUCCESS in 9m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e761f06685ef42b684fe927dbbad7b20 : SUCCESS in 2h 10m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cc26b4bb4ab42a9882e8a710b9e569d : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57eb105e4dbd4fea927a82b3b8da341a : SUCCESS in 41m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9153ef05e2f84b829d1f4dfa58b3abea : SUCCESS in 57m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f0e551f9a09e44199326912dcaf9282f : SUCCESS in 1h 11m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b5d1d1f1e9948e6b69b167649cb92ad : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d7de578c06d3456e82e55c7ed97ddbb5 : SUCCESS in 2h 03m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbe71408e245481dbf0b8d0cd6a79f56 : SUCCESS in 1h 58m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f50a5385edca47c68c694507d26bf714 : SUCCESS in 1h 44m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/59f5f7236790466190b1ec710384a4ea : SUCCESS in 1h 58m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6470671e140b4c8aa644de9e2c2b8d96 : SUCCESS in 1h 42m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7789343129d41a9a2de2a806d64df51 : SUCCESS in 2h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"87bc43fef46f720c4a3c75060030f21a1022321e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 07:29:59.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"eea9b2d68f79675374b14ad2b7f576366dc2f028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:09:52.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"1e153f92cf27f1a449f91a3eb2eb88969e4ccea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:46:10.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4be04b01d0a4688a539b5efc8635dfe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dc74628d3ae4debbd74bc27614d210e : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d0cae8432e42119832e61f398ccb5d : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6985b94e364e469456e62f1f7a4f3d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fd84045f211479d80db8a0311dcf5a6 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66d2645b00864117b3e6eb3530d63e5a : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef0416d825543c9a825725d9e0a49c5 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6181b6ec94e4a22b5178a4c5bd3f0b9 : SUCCESS in 5m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4920bcd3d38642d480a4374ae91c841d : SUCCESS in 6m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/58d32c731df344be8a06368b17409fa7 : SUCCESS in 10m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b877cb1aae3f4539858e87606cc1ca87 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b0f9aeaf2d74daf97df6dcf21bacd97 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6e9ecffde52449999c8b0bba4fc507f : SUCCESS in 11m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cd532ac9440d4ea08fafa8bc892730fd : SUCCESS in 10m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15086a41515645e69de07f94a5016c5b : SUCCESS in 2h 12m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f62170c797049adbe20b6753afb4bc7 : SUCCESS in 2h 15m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a45339b9f9724722a3422c17ac00ff17 : SUCCESS in 44m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa4598a5bb00470b81c4a5b6c9d9ba90 : SUCCESS in 1h 07m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a6a6b831a23d4f1aaa8bc8591831cfe1 : SUCCESS in 1h 33m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa9e38f4152b42458960ec16cb5fcea7 : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b25539b0911460a9a7145059e46db3a : SUCCESS in 2h 21m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57a974d931ad41c2a13fe49990851ab7 : SUCCESS in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f2593dced384350abdccab0fe444dad : SUCCESS in 1h 52m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d29a4ad419314ac186c2d2aac9bd4b5a : SUCCESS in 2h 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1553475c58d64785b7ee2803d18a3b19 : SUCCESS in 1h 57m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bf42de3f3154f15856da9e4895c019b : SUCCESS in 2h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"61b0a8a433d19e1c65c8af148fe6b9be6eb3bb6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-01 08:56:27.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"766a9530cccf037b634345531172d85afa7d8b93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 11:47:14.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f36dc03d183246ba8db38189cd6528b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604cbf3a5d2442c7b5a8876215944cab : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c75de4ba6b46498ea3b76329af0e87 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a8306df5e63433bbc0f2d9a843c7f2a : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55ca9b100c3b4015984cdc91793bc3ed : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/825f7a198a564e9fa36946c3517a7a2e : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebf837cde804445ac27a52db4bb4a22 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d9a767e03a2434487d7dd1b250a99d6 : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/08a13946ec6b4468b2a96c7a5e17b519 : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6165c08967aa4976bea5395f6f33165f : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b53f22a8617a482aba342f940df712b8 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83f054155ebe46dca6a4767fc8417335 : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29f5bf1efb6f48438a52da9b0399a52e : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c47969fd2ab646c8b0d37fe23f89dac0 : SUCCESS in 9m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2451cbb577c455fb475313f73f4bb4c : SUCCESS in 2h 36m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10276825c68a4bbfbe00816e4d5f1a3e : SUCCESS in 1h 57m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21c19586e12d43828ddbb6a8253f1baf : SUCCESS in 46m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d8181e13e56947daa384b19d8c29197a : SUCCESS in 1h 13m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5eee071b593746eab710b0e5133966fc : SUCCESS in 1h 35m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cdd93bc00ba41d59827bef134e469ad : SUCCESS in 56m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ebebe7272f9497daa0df80b0aaa5c92 : SUCCESS in 2h 07m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/178113a16c9949fbac20ed32ff9c4312 : SUCCESS in 2h 24m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a413f54e0cbc49efaab2fd7a15109e69 : SUCCESS in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd6e50e42d8d4adda60a0a4a3c3572d5 : SUCCESS in 1h 50m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6602e82ea8aa47c1bbee7f12f1bdc519 : SUCCESS in 2h 24m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e9fb795e9a54fdab3108d6c6acd8d2f : TIMED_OUT in 2h 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8e762dd348543bd7d8f50cba126065d4e21eaa4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 09:57:43.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"890070fa22a3eb1942c109580cdbdcb3a168df53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:03:44.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"e7923d198956e4f74335e105234deef787b33375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 10:08:17.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"e7abefe06a9677a30577ad352e966ecfb9195f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 12:28:16.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"247bf8bb6d322f48c288636d2704609a764b2e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 15:24:37.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7091f422db0404c89b8b0829828c4fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89e67ca5e1094d18a205b52a7498c1da : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f40a16ef564935b047fc1904f4e5b8 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2c03b26963478d9b83317d44a33449 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c08c639eac9b4805a37eed0152156abd : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5ced94528c641268e5fdd71a83b1db7 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7238d5a326c9434b80f916c89d0d965b : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/773d6777816b4d688e2e6c090acd2103 : SUCCESS in 5m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/809c148e254d409cbe2bfa22ef2ca0d2 : SUCCESS in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/96a1533aca7042b990c915cd58f02aab : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd6a5ad0079445d58778dd14c6b0ab43 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3e86520ca79449395bf6c799661d25f : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/192a5d45f0eb41019dabcf4f101bd41b : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/665f9561a38f4a35b907851db5659771 : SUCCESS in 8m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9711babb11d648028dab917cc757ffc7 : SUCCESS in 2h 49m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7e0cadde4eb428bad8ad5d48d107719 : SUCCESS in 2h 00m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8dd6b169621240bfaab2de4c2b3b3a95 : SUCCESS in 59m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/383da8fa7855405b9b6b3e88b74779d7 : SUCCESS in 1h 09m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8ad46f862a144195bc98b4cf1ce54588 : SUCCESS in 1h 13m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/166d8ec89b6849eaac0e52cfebc55c49 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2e4d15132a44bbba0740ebca50c335f : SUCCESS in 2h 08m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fcafa6ce34448cfada86cf161021fff : SUCCESS in 2h 22m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/095398d2d6014f22b2c0ad447870afcb : SUCCESS in 2h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19ad4f3367ef44158971ef9c197de790 : SUCCESS in 1h 59m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8b5f8bbd8cf748a2a8c078cd3ab45382 : SUCCESS in 2h 06m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5df71b39d56434e9c9c0b991afba51a : SUCCESS in 2h 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"aa5d00d964ad9b8c9fb91b26d0da188cc8a5e28b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:09:36.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"5e9c31fb5c32f2d83d2a77e93bde5f4b982d068b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 09:11:01.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent, and commit message as Patch Set 22.","accounts_in_message":[],"_revision_number":23},{"id":"a2121becb1fe74ee8fa0b9f41cc418a70aca57fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:41:36.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67c96f7c2f84236b34917e68cbe4813\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb1c862e4e14a0b9b2904baf49dfa3a : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954cd749094a4f0380d3db15378ec0ed : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abff455aa2274c4684786188f63bd174 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9374c63b98245888ab018296a6f25c2 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94eb3202cdd421d9037d2727d0fa1ed : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/473a59d0375745debd572a715966e0c0 : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/45da2d58bd5b403995ac647251dbc429 : SUCCESS in 4m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3dd999d0868642bcb658abcc95ed0199 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcd1b1eddf2541d0b04c15a1aa963f40 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/212ac569a35745a9b4dc58f8a7275867 : SUCCESS in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd306590e49a49509cd8334e4d30c6e6 : SUCCESS in 6m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf8e77d2e20945c68bf79df339c2753a : SUCCESS in 6m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c33400f5f4d248bda4729b8bd47a4c2b : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/108760bfa16a43b4a7622f65b5d53329 : SUCCESS in 1h 37m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57231fbb09654b28b20ea05a9eeb6d93 : SUCCESS in 38m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50aedc1f07194449a224f5fcc6ddd967 : SUCCESS in 1h 05m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eced9406d2994b72819ed6cf8e6d9dbd : SUCCESS in 1h 09m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcd32e0994cf469cad3f8d844aa8702c : SUCCESS in 32m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7fbbcf79948046bf85aa9d0b2eeab515 : SUCCESS in 2h 15m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48e91d8f2a5b41f98fdd5612a04fbe6d : SUCCESS in 2h 14m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8578911efabd4a86912146865904ce01 : SUCCESS in 1h 50m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d8eaef5bbd146af9dc996f19de7106f : SUCCESS in 2h 09m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/390a7dd52e1c4b549ed686c84060fca7 : SUCCESS in 2h 09m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/30631a12e7d54e5f9c412568ee8c3694 : SUCCESS in 2h 04m 49s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"0fdbf0a8bb4c1c92ceec3124f4c2e2254443ad0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 11:30:48.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"65e3bb9129ebf416f7620867e81a1287c36326f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 12:22:40.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"67d960fb7250656ad448f60061b5524eeaf2666b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:04:30.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"9c60f66b3f5e1e196ff588ca39aef275f1b3967a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 13:08:24.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"0a2bbfb9244f28922a65444d071f5a048ea85e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 14:41:34.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"88780432cdee5aa433dd3e71c98dd20e0352b5b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-27 16:19:29.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"b90048fe8658f7014c1f46309e9073030414026e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 18:32:19.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29708efce7442eb948f673ac2f8df21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74f6e1be472b4064a27222c8a31e34ac : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed31768530fa4c3a9ac4b18493ba50e0 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95ccab6c4bbe4aaeb0fcafecba92fce7 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfbd8bb9f8774b5baf1a9683294dbe31 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f3549b81cb4e78a3c46a40a36f39e1 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2200b4c3f0554e40868cb91894cb86fe : SUCCESS in 4m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/54b9e4f75a394ff0829b6fb9bea04b64 : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f245ca4dd3414d7ea5f4a8eb3fdf48ae : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00440713f1bb4ae399dff745bad65ff3 : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8f2c2d6455e44feb6063bb0efa80e5b : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a53d030150db43c1a46c0ef3d36eabf9 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76913dd055784360af341c5420ad7be5 : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a1c0bf14b2b44f59fbd9a2bf0505653 : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26061435b0fd470c8a4023ba93367f98 : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed3d2258e9474df1a4908c67211fe3d4 : SUCCESS in 40m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15d85dad5da5454dbf97fefebc3ab122 : SUCCESS in 1h 07m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5a3c00f3408f419a839a169a55cbd412 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c36745f0c6d04e45852065dac234b731 : SUCCESS in 28m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50666fd32f984c4ab0f8e874cc638312 : SUCCESS in 1h 54m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8206f7dca4ca499eb683672fec4c3b07 : SUCCESS in 2h 01m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9358cb9202a84cf1aab21c9f127c453b : SUCCESS in 2h 08m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/476c12f5625a4c59ab737683899a65f5 : SUCCESS in 1h 54m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a9f3709ab044d5fa7f2337b57529ff7 : SUCCESS in 1h 53m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a75ed7779d2a4fc08222eb3ead25bd14 : SUCCESS in 1h 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"29d1ec0a9ecf61bb3253ecd2214c4bb8dc80ecc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-28 09:35:53.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"8ad0947747a9bd91995ef0049740872a142ac225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:51:10.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13fed8287dbf4162ac5704dca1f5b51c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d8280de407442b78ed5e3d8e0b8efb9 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6faf05c2c8e1423593f4f68df8f1bf7b : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f476213518d45c9a8d6367f9648390b : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92d15e2104fe4c84a376ba1b6eb409fe : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d4a2aa611845d187df2efd41e5f295 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36b81664bc724603811c3040aad8219c : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/483a79e51016449c965afbb169d2d417 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/54e4a0029ee14290bcd5ae41d90e180b : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d1d178a90af94b2d88811a5fb2788704 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cb562f13d364e7e94d0819ef6dd4489 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dcd310a8b7a4669b8b2ec241769a8ce : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1154f9863bc4af1b87d71c8cb4d896f : SUCCESS in 6m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74c745944efc4cb8ad587ae689e56ae3 : SUCCESS in 1h 54m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/462dfe5bd3604a8e99f9128aa4d21578 : SUCCESS in 1h 50m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28f306770e334f0eb46cf1e918227457 : SUCCESS in 44m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a81a8e2885044503b0eda9ca89f816a1 : SUCCESS in 54m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5c72b7ee73db447c8dab7d7677fc32b9 : SUCCESS in 1h 12m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27379cfcc5e54fa8af8b38c64656f81f : SUCCESS in 41m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52c663019cfa41d3b8946dde22aaa450 : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/69e9d4134fab433faee6822cb7e2a1e1 : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c097f6b1ecb24afdadc72601e430f0d7 : SUCCESS in 1h 47m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49806426497248199822545943f60c54 : SUCCESS in 1h 30m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2513f318992d4830a84a80930ac8063c : SUCCESS in 2h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2880e771120c4f5dbfd7ba1a0a52dec4 : SUCCESS in 2h 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"0f69d19d225515dcd3582586c89ea2866e45ceed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 16:24:16.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"47ab18cc4dd009d53ab7a116b0ef8b2719ae0ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 18:58:31.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a974d75dd1f4be5abfbd6462d86e215\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84b7c7b64883450386625d216a0ec8d1 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc089f1ab9e448df99cae2fe9856b681 : FAILURE in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0d067f6b1664f0d80ad0fb332f21435 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89d4afa8f564404298dfe79855e27542 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7442cb2e05084205ab85e04762e58316 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e2c7788e8a40d1ba5a7ec80ec15906 : SUCCESS in 5m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f970590078364a308122ba3bdd1b4f03 : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/603e003c63cb4e5a8f41c7066bd7e6ca : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cdf7f72c3839484ab859708e19d952e0 : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4390739d0fb346cda9558dd08098e164 : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ba1c022854647c3bf3eb4fa83c19f9a : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01b0e53f980442c7a12a62702bb69fda : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0ce502971e0484ea3539833cdb573dd : SUCCESS in 1h 09m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/375c682a0b6f482f8af4d85e30669ad4 : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/672106b3bd3d40adb9ed5376745f20fc : SUCCESS in 34m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72065bead925431795f954fc20eaa2af : SUCCESS in 48m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b2851fc404f34a0db10587e41cdae9ab : SUCCESS in 1h 22m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/495059631c39462b8eb59d7e825bedec : SUCCESS in 31m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e1dc38b0d04d438ba2466d4d1e5a7b44 : SUCCESS in 1h 35m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aad8b458f04d4261acf1c304dd848c3d : SUCCESS in 2h 16m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/22466bcf68a542f78f68ae905ef1c0f2 : SUCCESS in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c43b5e61d6184634a355fdbab65c35fe : SUCCESS in 1h 29m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6d31fedb2e4e4f039d6e783ad0ea4842 : SUCCESS in 1h 50m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/44323779bafb488f98bb685bbb32b898 : SUCCESS in 2h 27m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":31},{"id":"bd10acdd43a214ea0e76e97513cfafcd37877c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 19:03:55.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"8cfcad7a8a235e38ab31a741d0d997ee5b9a3702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 21:33:59.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9dec458dd8488f9cfdd2be7b0f56f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd8a02729464450b35ec093065f9c10 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbb6cf95462b4854a7552a15e1278a7b : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/549f3bec16164e37bc5daaa3085f364a : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75043aab13ec4e2e8c5e46ba04e74eec : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dca795b0514254854ef43ac9b62921 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99e4a474d194472e95d2162e985d22ef : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a3756d20a4c94cf4a2fb51b96704da79 : SUCCESS in 4m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c8197acb9f004625ae4acc21626a7892 : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a70431e90bab4afbbfc111b91be28be3 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41f2486164be417eaa4391d19a393b86 : SUCCESS in 8m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bedf5403e53341658dd9f2bc35b5cbbf : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/993b92fd633946de9caf59e38831a436 : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77511d4ac5f3498f9621d38fa9b9f801 : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e91957be7a249d8b8d81c2234a5b5f4 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ed7b04f2d854bc987fa9bb849929ede : SUCCESS in 36m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/45ba1738b8a04b969a9fe5e1c48216ad : SUCCESS in 50m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/07874513670f4d02b4eef397f0e3ede5 : SUCCESS in 1h 08m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de18d18effc740bcb279d1ab212e2dfe : SUCCESS in 32m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9c1f69a85924632bc9ed48cc37e435e : SUCCESS in 1h 29m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5362d7d9a0f14f778190f2424a52375a : SUCCESS in 2h 03m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37b74102e4704d57878ea7791c5a1e4a : SUCCESS in 1h 57m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/123ae36d591e48f9919c80ed99eabf01 : SUCCESS in 1h 29m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5b94d604d6824f298aade3f9dd6ed16c : SUCCESS in 1h 36m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e1ed386fe26442d886387fe37f9fe33 : SUCCESS in 2h 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"8bf99ba4ac045d4b91226d4fff96ed8b70ad817a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-16 14:45:18.000000000","message":"Uploaded patch set 33: New patch set was added with same tree, parent, and commit message as Patch Set 32.","accounts_in_message":[],"_revision_number":33},{"id":"bddd2aea3ae8be60a9ad63d37da9a6b34c4f5848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 18:02:09.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da18bddd77c8434db3cd0c2c940c0587\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcca2a0387d44a9e93cd2aa29f7f53ee : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323d1ee31a8245e1a287d71e262adfa4 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce9d8271299c4cc0ac6472b1fb01b072 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bac3f2809e54725a2c4e8de6bbe1340 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8851e0b6b664981962fa68059c90cc7 : SUCCESS in 18m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9619c9c1faa245268739ad8aa2412b98 : SUCCESS in 4m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6cef3e7382984072b4dba1b31f3bb841 : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3fae73d5837b43da9a86c0336d56d0fd : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b952ea10b1514634aa6d2bffd7ffe09f : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c5b19df528949a4adf5b21351b1bd92 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e83f948972f4f168e29032b6ecc33f9 : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cff052d821441ad810b4929c41ee1f8 : SUCCESS in 7m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1102eee3a6a49029a1a4d2756639a1b : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13d1bf2589e24d6b8ec18e2fa9c96859 : SUCCESS in 1h 40m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a79026139ac040efb7929c7e04c07ae6 : SUCCESS in 42m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea8fb882d0b7466db6aac4b46984c6d4 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6cc4eb9d6fdd48008a98d6cbdaff3e4f : SUCCESS in 1h 11m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82ab0bb74dea40b0a748dcae5bd2e91c : SUCCESS in 41m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d8061d42b36544668315a49585129143 : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b541f193f6b44b3bb1a230b0ab4ddba : TIMED_OUT in 3h 04m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89783b2d1f1a486bbd50ac567af254ee : SUCCESS in 2h 10m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc6d45bd3c894eda8c288f603caf4e09 : SUCCESS in 2h 21m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5bb6dd3f03164aa6b8b0cbc84741490c : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/44c20caa4f58479dae30570e79585447 : SUCCESS in 2h 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"0f938de316632faa24cb3365a3b967824c03f310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-23 15:29:39.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.","accounts_in_message":[],"_revision_number":34},{"id":"89085a79111aaa04488074847600169a6acb7793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 18:10:38.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5953cfe99a71469283aff8992a533611\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdb9fe6142ce458db9b54d9a73ef0317 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1fc1853bc94667ba712338a63422c2 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca0a7b8fb75481daea6640c0012a135 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c476b0fee6cb4e238ebe3c19e7d4b864 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5e5d14a58944f09d3da2a1274de6bd : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd7960e02f241b2952f12ce1574c5e7 : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/65dab2ab0d3e449f9e8b16f9b6fbe6b0 : SUCCESS in 7m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cfee66a2a0b544e5a48fab9219a7d274 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13b80aa71d82411999074256e1f1be0f : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc0673ec65c74bf3bccdef9f0205b296 : SUCCESS in 6m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5fc0de4ae31445478b227c33ec411a33 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e38136cfb5b4a13a587e66eff7389bd : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90a159e397854181a6c3349fd56527d2 : FAILURE in 16m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f00383146e8465babead761c8321aad : SUCCESS in 2h 31m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18b2b52334204d90904cebf03458a177 : SUCCESS in 47m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b1508449a31412f96d8245a3efb7ae6 : SUCCESS in 50m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c9985de362d4024a88a75e50fdf5495 : SUCCESS in 1h 19m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fe10db3dcb84dc9b778fa1cc32ced25 : SUCCESS in 31m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4c277bd8b48c4f0689f9bb40753f7723 : SUCCESS in 1h 50m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3fe722a0f54493e852a76d3f2a3d6cb : SUCCESS in 1h 48m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2fbc10cb644c4b30b76a5bf4ebc89337 : SUCCESS in 2h 31m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c1239eaff994507a2414c6f2651f2bc : SUCCESS in 2h 15m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e06cd22532a64b1ca0ca921dd646bb40 : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/70ca4775c9744f82b2d3d2c4194a00ab : SUCCESS in 2h 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"ef14fa99f9ff4d974532905a4cafef6b25653f61","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-24 10:40:42.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"8a95d774ac133b340c3200d45c494c7bc9487ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 12:53:41.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49903e4b48854678b96563d631482ba0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a236963453584a47b7e914264f45c864 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2509337a596e4cf2b7ae76f297d089ae : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5a8842cc94340cfa6feedbc5fd3b09f : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a19f6a198ba64eee8c9825045abba70d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a601222d217e49888f5c5a042379247d : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/450daf00ba124873ae00d50ff9cef1be : SUCCESS in 6m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3a0e6102a17a4b0ca241d82b2772f453 : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/10fb7fd9ecbc405abf14b0c666415e92 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d4e63c47768d46a289cb5d4acdf4179d : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4bbc73969d6747d49e55d8a2d2940069 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c794a59078ae4c418e4d6f1185fd3ff4 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb46026556f94078abfc3d3d9c567236 : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b844bb3a71c44e5baca89551a22a59a : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d95cfb045e124fcca07bc229769ab981 : SUCCESS in 1h 44m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58d466927c2e4de2bf1668fb11490a15 : SUCCESS in 32m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3f8b0e0d6624e5aa211ddd51ffc2cb3 : SUCCESS in 1h 11m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/df785da097a94d6281955806e7134710 : SUCCESS in 1h 15m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9caece601df4644bd150da19618f820 : SUCCESS in 31m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/13965e35e0a140e0a9b91840d9b324c3 : SUCCESS in 1h 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/967d49fc12cc4d9e9eb96df7dcc3773a : SUCCESS in 2h 10m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3dd0c7af8d44a52970cf4cc71132970 : SUCCESS in 2h 09m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a2e261a670d4269aea37da6ed9dc4ab : SUCCESS in 1h 39m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5df8b82980374d3db7901ab8533862dd : SUCCESS in 1h 47m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2459fc77fad44720beaa8e187be8b31f : SUCCESS in 2h 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"42d3d73672529f1c7ddcd210c8e13b0f26f7687d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 11:00:01.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"f67d9be4df5aab01ebafede2387157a250e5dc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:45:57.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158eb0a243be40268c8cc975baab11b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ac679d299734869855f2740aa5d6048 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c98ef568954ef3b23f3207eda5f59f : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0966906b7bf44efbb3c51b6072d6ccc7 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb4403d822774e9dbf6ef1e27e7ad29e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c19fc2238df24d37886022f2954a6eaf : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7def30ba5a8245c98707afa2415ac32f : SUCCESS in 5m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d3886dafa33b45168485894ef082895b : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c10e5677f8ad48288519cedce6954936 : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d153cc2cb3624edbbe3bdb482a07bfba : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/90e9df71ef044ca4972a2128240728b3 : SUCCESS in 9m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8366ecd3c8f482eaac49d074a9ace05 : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0e5c968f4a44c24a0a52ad0b5cf621f : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8619117fad9c4bf0b5c7ecd84949068d : SUCCESS in 1h 50m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22053aa38f9b4356a2d951d50748414f : SUCCESS in 1h 25m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1f6a777732d4f7f8f5d222728b8e718 : SUCCESS in 34m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3057cd7390b45aca49e2e9300f5ff22 : SUCCESS in 48m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c5186187b5db402383d1c4a34d456363 : SUCCESS in 1h 11m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/736362fc7d7545308fc670fd9b6a912b : SUCCESS in 30m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d33bac9f971645b2ac7d0ac36ab12f7f : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55a6e9378e1248f7813614a21a9367fc : SUCCESS in 2h 01m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/76e5e361e2b84772abf29b2e946e75b7 : SUCCESS in 2h 04m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3c44211fcc74f04a956dd9ff9c6f953 : FAILURE in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/32a396b9b0474620b40a8efb2b8d830a : SUCCESS in 1h 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1070e8a2c65e4347ac46221dfeaf048d : SUCCESS in 2h 37m 03s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"538528d687842af30254497f93a7f9234c87c82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-28 15:46:59.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"2cc0ffed09e98832da8e85512808af94085e5c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 18:06:55.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddbde9507983425d96c2ffd613faed63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ba133c411704ab3bfbc8823781b77c0 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fac4fb45664a409e70d3c7e03d4aa9 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccfd5b7e12ed425f837d56d722cf7518 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b932bd533014dd29ac24d8bdf7a776c : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c3c7f167c44b359de59bdfa249133c : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eae9ebd4a734dee9359e9ca037594bb : SUCCESS in 5m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c800d151e33a4127885c5b0057ef25b6 : SUCCESS in 11m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a5eb6a203c7541328f523c3a227623f0 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37a7e8b4307a49a984fe4987f1af7671 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d81fc1a8a7049ae9b8606c364412dbf : SUCCESS in 11m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f013ff47b5e4add92b1b1cc51524737 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fea28f2d2bfc49aa81f42d16c09c8383 : SUCCESS in 8m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2e9b38a9d1944779ee55652b8be2778 : SUCCESS in 1h 22m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01720dca4f4e47e1ae39d565cd99241a : SUCCESS in 1h 41m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9278fed08d0a4659be8e0a7fdd67412d : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74ba4ff9bb174c86b04b1a0ee2f42ff6 : SUCCESS in 48m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b8025e159efa417e921b564c9892a666 : SUCCESS in 1h 11m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac1e63aa96944813b08dd0b5ee512a7c : SUCCESS in 30m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c230dc41acde4d83b6b64b016fbbf8e4 : SUCCESS in 1h 46m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e318f91331394e1cae9fec3a5ba96005 : SUCCESS in 1h 44m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bb67a306f41a4bf8bbfd6212451692cd : SUCCESS in 1h 49m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0117b3460d044bb1a067a3519616a665 : SUCCESS in 2h 07m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cb9817621f024a14b55172a755b14d7f : SUCCESS in 2h 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c14ed0aeba94c84a5ae5def4b86783c : SUCCESS in 2h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"999197b0de17680ea4aacdff4c242e7682f8d2f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:18:13.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"baf835cf95a301f6b7c194c49a8dcae2b9115d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 12:21:35.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"d6e26484d57c3210f2cc324892432527e5947ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:35:26.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8914c33fabdb499caf40bc55734938b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ee7c8b66bd4f1ab5e1e8be09393177 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0f786822324523b82db2509778038e : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/639d28ecb87242229f15f8c15a8a5453 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32dbb976a7854b6d97073e952961e6e1 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cd9fd75d504d3888d28fa2099a2501 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bd0d30a103a4fd6a835ee2bb2fa130c : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/25e51ce0da4947c0af4945f3aa9bb89a : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3ae1d76097b34edbb51c884eb5b3b2d0 : SUCCESS in 6m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3164b04a4134357bf446f7c6e249444 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3488769c332a4615a10129a665285160 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4eec4ba9e1154ae59686a91e9710acf7 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2b55cd85c7e4b9e89813fdadcf56b7d : SUCCESS in 7m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88c32877a38e4d458c0816733ade5afd : SUCCESS in 1h 31m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90567ce980e34fe8bc1e2787cb7817cf : SUCCESS in 1h 57m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0291828caea4d43ae2a9de268a43570 : SUCCESS in 33m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf3a6dde906e4e21bc6a066bb5fdba19 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5222fea1d6134c09bf4b7aaca723ebb8 : FAILURE in 1h 21m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a00d5d3249224dd7b5ea7c488c5ee6b7 : SUCCESS in 30m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b4da54c573e41e487570f843dbd1025 : SUCCESS in 1h 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d754fe1c98754883a2466543066541f4 : SUCCESS in 2h 11m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e98a031fe29d45c3a912fea012c7b3f0 : SUCCESS in 2h 08m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a4473d20a86493f923248e86d47db2d : SUCCESS in 1h 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f9327681b6da4280a0df788b8443cec0 : SUCCESS in 2h 01m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d20c48426b9f424eb01d071bc0bf32d1 : SUCCESS in 2h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"b2183b2eb1da8287069945f2638cc61f8e43e4b2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-07 19:59:55.000000000","message":"Patch Set 38: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"8ae2fa2a06241d4e8b29d207bd9e87f1f69b5d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 08:30:13.000000000","message":"Uploaded patch set 39: Commit message was updated.","accounts_in_message":[],"_revision_number":39},{"id":"74ae6fb22edc473233b75b910873d2899e1e22b2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 08:30:39.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"7e0213312c8f1dd14f0ca7137de1250755b2917e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 09:27:30.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.","accounts_in_message":[],"_revision_number":40},{"id":"4e9996ee022d9916df8c49361b757b5ebb21df81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:54:33.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"d9a1d15fb9d05bbe39a30964e7689346aa4f55d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-08 10:59:41.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"bef838e6e455c7fd963b192a81e07a51f1d65983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:42:12.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1daef965d55b430dacd38e4368e4a57f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13fa6103e6854920b23d70aeb20849ce : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738b8a5780be40ec9f6a4205526b22ca : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/035eb38c588c4a4cbd96f4e739c9190b : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/413f8468a3af463b9359acca279a8edb : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebeab1918634bd086ac77d1d5bcde86 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05fd05c8dbb24a9293e46cde69d879b2 : SUCCESS in 5m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e89b58016790467db1cbe19d1e660425 : SUCCESS in 6m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/592d7c4f683546e1b680cbcbb5b23922 : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0551059ac71f451290b7c87f637f54ee : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cc6ce964d7f4a9f91ee0a2bf46f4af8 : SUCCESS in 9m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de74f613fa2941409195ff5c93b6dd08 : SUCCESS in 10m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b5d94a9320446a790d46ea54cec1374 : SUCCESS in 8m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8ef452d3e0f4a8aacd92ba4278ffcd4 : SUCCESS in 1h 22m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/834308e31ca8477989d5d5441f1c6da6 : FAILURE in 23m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e64a17fd92f4a93a48c08145b5a6cfc : FAILURE in 26m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ab1898faaf64c229a1eee2d7d0ce3c4 : FAILURE in 24m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c953bfa64db4c1e9e456ec523299000 : FAILURE in 1h 29m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44a285627c2e4c5cb8dd8c744a86f250 : FAILURE in 23m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5516636620844ae39814bdb141792025 : FAILURE in 28m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72cb70c58ca842da962b3149f14a11a6 : FAILURE in 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/169f7551bd38426fbe653e24231a061c : FAILURE in 28m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e4861c6e87b84c27b6917a2534c43333 : FAILURE in 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/98d183e1607c4deaa5c024be7d555b67 : FAILURE in 23m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0966362d98649c883a1d32057a9a07d : TIMED_OUT in 2h 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"962a0cae718573faaf3b69f2076992f5cbcf5ba9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-12-08 16:00:50.000000000","message":"Patch Set 42: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"7d686c3186c6512b40cc62a0c72292257b8a296f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-09 17:54:30.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"9d656f5d220afafb5e4ec46d8fb9433533eab29a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:37:21.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c7f522e4834460a77ee29b17918b7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b2268b5a09447baaf2de501ffa626bc : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89923165ec7d4b4cabcdd0aaf38c7ac3 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/666f1139b1b34c5a91079b812ada3446 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e40e28c552e425594bd725d34bbd979 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e34abf90dc4cde9b9f1b69e81656fe : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2357bf3013644031b6c727e8bbe1c76a : SUCCESS in 4m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1740e404130f4c00a59243b1c0427175 : SUCCESS in 6m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c4c78d73d90b4afda6453cfb66715a40 : SUCCESS in 6m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bde19e58928a417c9ecdb5879205cc7e : SUCCESS in 4m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b6567b3010040af9d7009dcdb57fc99 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1c33fc4567040d99da8f8c347754644 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f093f8e29fcd4425bd9b6eeacd36fa47 : SUCCESS in 7m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d9e8204312542d68e8c2c8be3cefdd1 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2132e5e5f0c437296032a440885751b : SUCCESS in 2h 07m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0906a0a25f240c98b9eae80c82d7621 : SUCCESS in 45m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f2483347e8b4514a1b84fb0f4db886a : SUCCESS in 49m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a18c6af8272d48e9bc1824e4fcc876f5 : FAILURE in 1h 28m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd00d67681fd4dc1b36434cb13e76bc0 : SUCCESS in 31m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4a5467ad8fc448e697da0409bb5169a4 : SUCCESS in 2h 23m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14296de97350402c86ec2111b550a129 : SUCCESS in 2h 11m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85ade64482da4a02bb47fab22c92660e : SUCCESS in 2h 21m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f01c060dc2d0400fa32a508000c8d3ee : SUCCESS in 1h 52m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e6e62ffa7dae4a2b9b10c28e99cb5597 : SUCCESS in 2h 09m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d12836e049ac4759ba7f12758b9c7b05 : SUCCESS in 2h 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"ea95c19c7e1311398ab03fdcc75a758f17e5ee29","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-01-11 16:50:25.000000000","message":"Patch Set 42: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"6b2e20a8eb02f208add59f8aca253030d933a5c0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-12 08:16:15.000000000","message":"Patch Set 42: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":42},{"id":"644d09912f712ed06b43f4d05066615929ad96ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:17:00.000000000","message":"Patch Set 42: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":42},{"id":"d1779d61d629d25aad7e36466ee86135a9d2d48e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:47:22.000000000","message":"Patch Set 42: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b1c3ce2b874907ae71e5b32cde5f53\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de37afc07ae042518c43ca8185b98236 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1fe6fa402b74f1e9282fb5bb1cd9385 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9d90cbfca38421d821b06342b21f43e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c064ba2d984d0a9b25d4c3aa42b85e : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fa667e640034a788db01bb4c2c5b3ae : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b6dfad3bc77b40a0a78d1e0d92ba82a4 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d94daccde33344ec97d4ff5957b42cf3 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/33a54f1ead3440fbb5f40a36dae73649 : SUCCESS in 1h 26m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5608864e048041d7bd8dcd57a5cd4c56 : SUCCESS in 2h 11m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0fc312147bf49de885833577772a591 : SUCCESS in 42m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36ce5bd922fc4d1c8e3c9be51b5d0074 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":42},{"id":"79b35425eecbfc9c107096bdd5f380549e764184","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:47:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":42},{"id":"291c9fc6ff612908e1c294c78fb8eb9f5d76ea5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:49:44.000000000","message":"Patch Set 42:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0199a6b9ba174b58b125125f7fd1c26a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e4ed69da674252bd2fa8ba5c10bdb5 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5463e5362f544b6caa8db82b61a3c86d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":42}],"current_revision_number":42,"current_revision":"baefc3b4643d9e61361cacf46dfbeb95ce96933b","revisions":{"624de65e52f6e2793643e842796520ab2045efbf":{"kind":"REWORK","_number":1,"created":"2022-08-19 09:44:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/1"}}},"commit":{"parents":[{"commit":"a6bd897fd35b5af9cf6dfca6615106a5bffe5c2c","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6bd897fd35b5af9cf6dfca6615106a5bffe5c2c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:43:47.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624de65e52f6e2793643e842796520ab2045efbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/624de65e52f6e2793643e842796520ab2045efbf"}]},"branch":"refs/heads/master"},"97e195ff825775a21cc2f8b12e55cce507561ccc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-19 10:37:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/2"}}},"commit":{"parents":[{"commit":"71422d3536886b575560662bf55f6ac8b3679ee0","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71422d3536886b575560662bf55f6ac8b3679ee0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:35:48.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97e195ff825775a21cc2f8b12e55cce507561ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97e195ff825775a21cc2f8b12e55cce507561ccc"}]},"branch":"refs/heads/master"},"cdb963e959aada6b94b93a6c6eee25d9a112a1e4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-19 10:47:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/3"}}},"commit":{"parents":[{"commit":"2aa91bdea18e9b5263ca3412963c521b149a17f1","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aa91bdea18e9b5263ca3412963c521b149a17f1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 10:47:40.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdb963e959aada6b94b93a6c6eee25d9a112a1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdb963e959aada6b94b93a6c6eee25d9a112a1e4"}]},"branch":"refs/heads/master"},"cda2f0008f378967cdd1e3e879f596644a302b57":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-19 13:22:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/4"}}},"commit":{"parents":[{"commit":"bcf07e0af3d5347dd5cb9000c3f0af96c6869416","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcf07e0af3d5347dd5cb9000c3f0af96c6869416"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 13:22:10.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda2f0008f378967cdd1e3e879f596644a302b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda2f0008f378967cdd1e3e879f596644a302b57"}]},"branch":"refs/heads/master"},"2ab7cc53424cd81c021ef5d9deda5aa3057537f4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-08-19 14:46:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/5"}}},"commit":{"parents":[{"commit":"60c5917cc40b01e5772ab35018b9f49757d3520a","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60c5917cc40b01e5772ab35018b9f49757d3520a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 14:46:17.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ab7cc53424cd81c021ef5d9deda5aa3057537f4"}]},"branch":"refs/heads/master"},"201bc774e9d485e3020fdaeccb99ac7242f235ed":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-08-21 15:00:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/6"}}},"commit":{"parents":[{"commit":"24ea898125a01b7f23f0f8a166e0a0cf522443c7","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ea898125a01b7f23f0f8a166e0a0cf522443c7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:00:45.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/201bc774e9d485e3020fdaeccb99ac7242f235ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/201bc774e9d485e3020fdaeccb99ac7242f235ed"}]},"branch":"refs/heads/master"},"5b408565d5181b3d45d67c3041e4adc8e0e94aa4":{"kind":"REWORK","_number":7,"created":"2022-08-22 08:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/7"}}},"commit":{"parents":[{"commit":"24ea898125a01b7f23f0f8a166e0a0cf522443c7","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ea898125a01b7f23f0f8a166e0a0cf522443c7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 08:08:06.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b408565d5181b3d45d67c3041e4adc8e0e94aa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b408565d5181b3d45d67c3041e4adc8e0e94aa4"}]},"branch":"refs/heads/master"},"d6d92cf61e4c1f84d69497bf1fbeb7326ba0fb40":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-08-22 13:52:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/8"}}},"commit":{"parents":[{"commit":"2dc7679b76d2b0a36bd8c601e4d462de08837e39","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dc7679b76d2b0a36bd8c601e4d462de08837e39"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-22 13:52:45.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6d92cf61e4c1f84d69497bf1fbeb7326ba0fb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6d92cf61e4c1f84d69497bf1fbeb7326ba0fb40"}]},"branch":"refs/heads/master"},"28d4350b3224e13863e231fd99b71d11bb12e86e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-08-23 10:07:03.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/9"}}},"commit":{"parents":[{"commit":"dcc945117ede18b498d362d9648107f2b2e3aedd","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcc945117ede18b498d362d9648107f2b2e3aedd"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 10:06:51.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28d4350b3224e13863e231fd99b71d11bb12e86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28d4350b3224e13863e231fd99b71d11bb12e86e"}]},"branch":"refs/heads/master"},"a006ae159521b3670ba48f73a1b2cf355164ddb4":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-08-23 12:45:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/10"}}},"commit":{"parents":[{"commit":"9514f5b9db9de4cc9dc267a9de4891358c859013","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9514f5b9db9de4cc9dc267a9de4891358c859013"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-23 12:45:50.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a006ae159521b3670ba48f73a1b2cf355164ddb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a006ae159521b3670ba48f73a1b2cf355164ddb4"}]},"branch":"refs/heads/master"},"895b1038690070b447fffbe502f73d39888a8d51":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-08-30 11:21:15.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/11"}}},"commit":{"parents":[{"commit":"1c9092d1944afbbf49b05097eee26e93235ddbdc","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c9092d1944afbbf49b05097eee26e93235ddbdc"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:16:37.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/895b1038690070b447fffbe502f73d39888a8d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/895b1038690070b447fffbe502f73d39888a8d51"}]},"branch":"refs/heads/master"},"d2112ea041a03379b68342976026bf05d5f74700":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-08-30 11:35:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/12"}}},"commit":{"parents":[{"commit":"d234f7e83dd5c2999a61088a5c5f4475def722a2","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d234f7e83dd5c2999a61088a5c5f4475def722a2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 11:35:03.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2112ea041a03379b68342976026bf05d5f74700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2112ea041a03379b68342976026bf05d5f74700"}]},"branch":"refs/heads/master"},"28f966e0795b375f5a0c20ea8146757f3e069f4a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-08-30 14:22:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/13"}}},"commit":{"parents":[{"commit":"7f5d541c0b1bb966c581d07576c074cb5f19e920","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f5d541c0b1bb966c581d07576c074cb5f19e920"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:22:20.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f966e0795b375f5a0c20ea8146757f3e069f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28f966e0795b375f5a0c20ea8146757f3e069f4a"}]},"branch":"refs/heads/master"},"ec0b9462aedb3861578e09b2e9e3db0b8d49fa7c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-08-30 14:53:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/14"}}},"commit":{"parents":[{"commit":"4048346a3121224f66da777202b1db24a862ba10","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4048346a3121224f66da777202b1db24a862ba10"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-30 14:29:31.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec0b9462aedb3861578e09b2e9e3db0b8d49fa7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec0b9462aedb3861578e09b2e9e3db0b8d49fa7c"}]},"branch":"refs/heads/master"},"893ac804510df9ee18dbc714bd683961efce6779":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-08-31 07:29:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/15"}}},"commit":{"parents":[{"commit":"799c16ed0219af080f57334d446778ce59d11506","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/799c16ed0219af080f57334d446778ce59d11506"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 07:29:43.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/893ac804510df9ee18dbc714bd683961efce6779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/893ac804510df9ee18dbc714bd683961efce6779"}]},"branch":"refs/heads/master"},"95556136a681cb79a12e91b96a191ae4ac5501f1":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-08-31 09:09:52.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/16"}}},"commit":{"parents":[{"commit":"bf6d4022f29636fa7f920e67b856a710e110b88b","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf6d4022f29636fa7f920e67b856a710e110b88b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-31 09:09:40.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95556136a681cb79a12e91b96a191ae4ac5501f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95556136a681cb79a12e91b96a191ae4ac5501f1"}]},"branch":"refs/heads/master"},"abed41dd40f63f30a926ce5d7306fefe3bab7106":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-09-01 08:56:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/17"}}},"commit":{"parents":[{"commit":"c8be509fbb5a7c7886a0b1c4b975122641ba2a98","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8be509fbb5a7c7886a0b1c4b975122641ba2a98"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-01 08:55:12.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abed41dd40f63f30a926ce5d7306fefe3bab7106"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abed41dd40f63f30a926ce5d7306fefe3bab7106"}]},"branch":"refs/heads/master"},"2a851fb5aa839edb7a003138f11eed8246465651":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-09-12 09:57:43.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/18"}}},"commit":{"parents":[{"commit":"91e7f0d1fa2d7220972c246e7eafb36f6d95b965","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91e7f0d1fa2d7220972c246e7eafb36f6d95b965"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 09:53:34.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a851fb5aa839edb7a003138f11eed8246465651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a851fb5aa839edb7a003138f11eed8246465651"}]},"branch":"refs/heads/master"},"350636ba1acb61e12111003bb7d12ed0ba05bdf7":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-09-12 10:03:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/19"}}},"commit":{"parents":[{"commit":"47acdcad26e52d352e9a6d685942fe40b05b3606","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47acdcad26e52d352e9a6d685942fe40b05b3606"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:01:55.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/350636ba1acb61e12111003bb7d12ed0ba05bdf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/350636ba1acb61e12111003bb7d12ed0ba05bdf7"}]},"branch":"refs/heads/master"},"85005f5e5713050f7fbf3014f16c58651e5137c5":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-09-12 10:08:17.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/20"}}},"commit":{"parents":[{"commit":"01888dcedb35e85b9a4a6fbfc71272f15e5bbf92","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01888dcedb35e85b9a4a6fbfc71272f15e5bbf92"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 10:08:05.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85005f5e5713050f7fbf3014f16c58651e5137c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85005f5e5713050f7fbf3014f16c58651e5137c5"}]},"branch":"refs/heads/master"},"e0d6d7420829f1d787004d664499f31aaeb497a9":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-09-12 12:28:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/21"}}},"commit":{"parents":[{"commit":"de7ea6e40a09a166829c88e540ae8c12c9728df3","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de7ea6e40a09a166829c88e540ae8c12c9728df3"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-12 12:27:46.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d6d7420829f1d787004d664499f31aaeb497a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d6d7420829f1d787004d664499f31aaeb497a9"}]},"branch":"refs/heads/master"},"5d56bd0021ee7e9035d5af6b11528400186d21bf":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-10-19 09:09:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/22"}}},"commit":{"parents":[{"commit":"deeda821420da1ffb8c5fca52cfed173019f30ff","subject":"Add cpu-pinning element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deeda821420da1ffb8c5fca52cfed173019f30ff"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:09:22.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d56bd0021ee7e9035d5af6b11528400186d21bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d56bd0021ee7e9035d5af6b11528400186d21bf"}]},"branch":"refs/heads/master"},"9fc07a3fa709b3a242a506e6f82cbf461c853a89":{"kind":"NO_CHANGE","_number":23,"created":"2022-10-19 09:11:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/23"}}},"commit":{"parents":[{"commit":"0929dc83680eb6d587ff4cd5addc15268dd760ed","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0929dc83680eb6d587ff4cd5addc15268dd760ed"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-19 09:10:48.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fc07a3fa709b3a242a506e6f82cbf461c853a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fc07a3fa709b3a242a506e6f82cbf461c853a89"}]},"branch":"refs/heads/master"},"05f8ba4ffd5ce6833b17cf0a5f5461ede0457d83":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-10-27 11:30:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/24"}}},"commit":{"parents":[{"commit":"431a94fffffce58f78d44cf87c2def79df132cac","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/431a94fffffce58f78d44cf87c2def79df132cac"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 11:27:41.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f8ba4ffd5ce6833b17cf0a5f5461ede0457d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f8ba4ffd5ce6833b17cf0a5f5461ede0457d83"}]},"branch":"refs/heads/master"},"eaeff20819ca9278ec1142d1080506178f4c72c6":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-10-27 12:22:40.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/25"}}},"commit":{"parents":[{"commit":"3c3ac18e3cb4d66ba0ee6b415dd91eace316641c","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c3ac18e3cb4d66ba0ee6b415dd91eace316641c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 12:22:29.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaeff20819ca9278ec1142d1080506178f4c72c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaeff20819ca9278ec1142d1080506178f4c72c6"}]},"branch":"refs/heads/master"},"def0d0c65280650151442de434fe4539e9d7a68f":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2022-10-27 13:04:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/26"}}},"commit":{"parents":[{"commit":"0666c752cf132524b8585d5a84ce67fe86e26498","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0666c752cf132524b8585d5a84ce67fe86e26498"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:03:51.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/def0d0c65280650151442de434fe4539e9d7a68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/def0d0c65280650151442de434fe4539e9d7a68f"}]},"branch":"refs/heads/master"},"5bb3f40633cad37f87471e5d21bda5b5a2598b61":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2022-10-27 13:08:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/27"}}},"commit":{"parents":[{"commit":"4a548aba8bd112624867369cbe5208fac4982d62","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a548aba8bd112624867369cbe5208fac4982d62"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 13:08:15.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bb3f40633cad37f87471e5d21bda5b5a2598b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bb3f40633cad37f87471e5d21bda5b5a2598b61"}]},"branch":"refs/heads/master"},"8de49d1479790bb31e228b7ba524dbaa7df2ac1c":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-10-27 14:41:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/28"}}},"commit":{"parents":[{"commit":"db1b365a0356130c7439279776defcdf816e8409","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db1b365a0356130c7439279776defcdf816e8409"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 14:41:24.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8de49d1479790bb31e228b7ba524dbaa7df2ac1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8de49d1479790bb31e228b7ba524dbaa7df2ac1c"}]},"branch":"refs/heads/master"},"09f2f57b991d766e7dd8b6afee89dd7358b2dd37":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2022-10-27 16:19:29.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/29"}}},"commit":{"parents":[{"commit":"abbc75d04c6b56803942fcc84b985b205e19bbe5","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/abbc75d04c6b56803942fcc84b985b205e19bbe5"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-27 16:19:16.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09f2f57b991d766e7dd8b6afee89dd7358b2dd37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09f2f57b991d766e7dd8b6afee89dd7358b2dd37"}]},"branch":"refs/heads/master"},"836c4f092d19ba1c7b26f08168419dbc83e119b6":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2022-10-28 09:35:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/30"}}},"commit":{"parents":[{"commit":"a85b78e5d78d1ebb01a94528edf98ee05c992cf6","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a85b78e5d78d1ebb01a94528edf98ee05c992cf6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-28 09:35:36.000000000","tz":120},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/836c4f092d19ba1c7b26f08168419dbc83e119b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/836c4f092d19ba1c7b26f08168419dbc83e119b6"}]},"branch":"refs/heads/master"},"1f4172e13c227fef647a8a194c037c06682cb84b":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2022-11-14 16:24:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/31"}}},"commit":{"parents":[{"commit":"8ffefcbb7d0ecbe4ca1578cc75a9a334cb46b6a1","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ffefcbb7d0ecbe4ca1578cc75a9a334cb46b6a1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 16:24:02.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f4172e13c227fef647a8a194c037c06682cb84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f4172e13c227fef647a8a194c037c06682cb84b"}]},"branch":"refs/heads/master"},"0d22d9567c637efc429cde4dca37471a36e126b1":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2022-11-14 19:03:55.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/32"}}},"commit":{"parents":[{"commit":"19c6774c1b488803659dbf41b875d8de7ff00e9a","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19c6774c1b488803659dbf41b875d8de7ff00e9a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-14 19:03:40.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d22d9567c637efc429cde4dca37471a36e126b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d22d9567c637efc429cde4dca37471a36e126b1"}]},"branch":"refs/heads/master"},"b239d0a51abff31b39b8b0a285844cfa98ebfde3":{"kind":"NO_CHANGE","_number":33,"created":"2022-11-16 14:45:18.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/33"}}},"commit":{"parents":[{"commit":"e2cfec242e1a136773f9710ee5d0754b23a0e103","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2cfec242e1a136773f9710ee5d0754b23a0e103"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-16 14:45:05.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b239d0a51abff31b39b8b0a285844cfa98ebfde3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b239d0a51abff31b39b8b0a285844cfa98ebfde3"}]},"branch":"refs/heads/master"},"d53bcfe90b514571dce586a8b3939d9d18c1427b":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2022-11-23 15:29:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/34"}}},"commit":{"parents":[{"commit":"9ae96ce28a86101cfd5a7dc1385108d6a7ae2877","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ae96ce28a86101cfd5a7dc1385108d6a7ae2877"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-23 15:29:28.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53bcfe90b514571dce586a8b3939d9d18c1427b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53bcfe90b514571dce586a8b3939d9d18c1427b"}]},"branch":"refs/heads/master"},"d0c42e2b8bba89394c868e37abd7bf041252ad81":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2022-11-28 11:00:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/35"}}},"commit":{"parents":[{"commit":"73f873381fc94322b71c8c94674995702fe53742","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73f873381fc94322b71c8c94674995702fe53742"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 10:59:34.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c42e2b8bba89394c868e37abd7bf041252ad81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c42e2b8bba89394c868e37abd7bf041252ad81"}]},"branch":"refs/heads/master"},"699cb1745005e89ef780a73f038d71502c10ab5c":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2022-11-28 15:46:59.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/36"}}},"commit":{"parents":[{"commit":"387256f1609b2ce222ba1d425af41058cb0943d8","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/387256f1609b2ce222ba1d425af41058cb0943d8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-11-28 15:46:47.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/699cb1745005e89ef780a73f038d71502c10ab5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/699cb1745005e89ef780a73f038d71502c10ab5c"}]},"branch":"refs/heads/master"},"412f1d79c8c734661fc0ed33c2d79405eb6bd5af":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2022-12-07 12:18:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/37"}}},"commit":{"parents":[{"commit":"3954e97e3899ec00b675cc4eb7750743e98ce454","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3954e97e3899ec00b675cc4eb7750743e98ce454"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:18:01.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/412f1d79c8c734661fc0ed33c2d79405eb6bd5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/412f1d79c8c734661fc0ed33c2d79405eb6bd5af"}]},"branch":"refs/heads/master"},"9e4a3ceddf38e850e1a4ef6437906a6d601b14d9":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2022-12-07 12:21:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/38"}}},"commit":{"parents":[{"commit":"42740484d96de20235e7d3ba6e924e1f0ac17a5c","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42740484d96de20235e7d3ba6e924e1f0ac17a5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-07 12:21:23.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nTODO: Stress test and measure difference in performance/latency.\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e4a3ceddf38e850e1a4ef6437906a6d601b14d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e4a3ceddf38e850e1a4ef6437906a6d601b14d9"}]},"branch":"refs/heads/master"},"b527b22ca504f673637b6b03817d7d5d0b97407d":{"kind":"NO_CODE_CHANGE","_number":39,"created":"2022-12-08 08:30:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/39"}}},"commit":{"parents":[{"commit":"42740484d96de20235e7d3ba6e924e1f0ac17a5c","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42740484d96de20235e7d3ba6e924e1f0ac17a5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 08:29:38.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b527b22ca504f673637b6b03817d7d5d0b97407d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b527b22ca504f673637b6b03817d7d5d0b97407d"}]},"branch":"refs/heads/master"},"a459cea3c33ca16784c64643079faeaf19f9c4c0":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2022-12-08 09:27:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/40"}}},"commit":{"parents":[{"commit":"8d757c2e067bd54e1956136eacac19ae85ac36c7","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d757c2e067bd54e1956136eacac19ae85ac36c7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 09:27:17.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a459cea3c33ca16784c64643079faeaf19f9c4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a459cea3c33ca16784c64643079faeaf19f9c4c0"}]},"branch":"refs/heads/master"},"8f8d603037407d9652fdad2fa8350148cba318ab":{"kind":"REWORK","_number":41,"created":"2022-12-08 10:54:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/41"}}},"commit":{"parents":[{"commit":"401543236e64c0af3ee36967cfbd46cf318949de","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/401543236e64c0af3ee36967cfbd46cf318949de"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:51:49.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f8d603037407d9652fdad2fa8350148cba318ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f8d603037407d9652fdad2fa8350148cba318ab"}]},"branch":"refs/heads/master"},"baefc3b4643d9e61361cacf46dfbeb95ce96933b":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2022-12-08 10:59:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/02/853802/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/853802/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/853802/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/853802/42"}}},"commit":{"parents":[{"commit":"0b1b6c58f55cbe02bb9473c2a4e961d3f658f489","subject":"Amphora vertical scaling optimization in Octavia","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b1b6c58f55cbe02bb9473c2a4e961d3f658f489"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-19 09:00:31.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-12-08 10:59:32.000000000","tz":60},"subject":"Set nohz_full in cpu-pinning element","message":"Set nohz_full in cpu-pinning element\n\nnohz_full (or full dynticks) is a Linux kernel setting that should reduce\nkernel noise on configured CPUs to a minimum. This change activates this\nsetting on all vCPUs except the first one.\n\nChange-Id: Ie009aff52d572873da0bb5cc43fedd7f3325dccc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/baefc3b4643d9e61361cacf46dfbeb95ce96933b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/baefc3b4643d9e61361cacf46dfbeb95ce96933b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
