)]}'
{"id":"openstack%2Foctavia~853875","triplet_id":"openstack%2Foctavia~master~Ifff0a4a5f78e35ecd722ecdb15edd08347230842","project":"openstack/octavia","branch":"master","topic":"gitignore-dotorig","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-09-29 08:20:39.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ifff0a4a5f78e35ecd722ecdb15edd08347230842","subject":"Add *.orig to .gitignore","status":"MERGED","created":"2022-08-21 15:04:50.000000000","updated":"2022-09-29 09:21:26.000000000","submitted":"2022-09-29 09:20:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853875-gitignore-dotorig","meta_rev_id":"c58655a0490d8e634c65c4794ca6087b3e92d6c7","_number":853875,"virtual_id_number":853875,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 09:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2022-09-28 16:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-09-28 16:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-09-28 16:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-29 07:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-29 07:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-21 17:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 16:27:57.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-09-28 16:28:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-28 16:31:32.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-29 07:17:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"bec85d4b1de5bf21ed78cb1b3fbe6b4734e48827","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c512cabb3e77d33f4b31d34624a1e4e1af286c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-21 15:05:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c8d8b1c533ddbea37d2d221144196db9cb5b2d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 17:17:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/236f2fbba08d459d86d268ea9a8d67ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db1d5388c60449f8673e8da0f18a5d7 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e97fbccbe1443fbbecf13032bc69b26 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b0fe38a9ce140baabf753f66fbd64d7 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940b8e542f80444b97d591ff2b88d696 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c622eb539a4441f8de038cc5ebad45b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75d52ef77924ba58dcb972ca9ab8e69 : SUCCESS in 12m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/40b07af6a8204ae58b5df114b2ac3f2c : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bd2ef1b6450a42068215260e34df8b89 : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/49a37668923e4de9b939ce6828f58981 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60299107a6484d7a9476110911b6b148 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41039b155e1e459c8fdd22552ace4158 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0398f24dc20d4c03b2e0c48c4dd2525e : SUCCESS in 6m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7eb527562d494c70ae8f9f2dc144cc17 : SUCCESS in 2h 06m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d948c71225e24a21bd7d5a0a3b1f4d18 : SUCCESS in 1h 52m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31c5bcb23ca640b1a7eaf9980beb847b : SUCCESS in 42m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbb322f727b44d09bc496fdfdd2b8769 : FAILURE in 16m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/40fa0847aedd4144ac60b1f8aafa29c0 : SUCCESS in 1h 12m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5ab87e87ac84aeb9d91f777d1c5c3fb : FAILURE in 1h 01m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c6d3efa3fa24af7864cb0ffb9a6ab80 : SUCCESS in 1h 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/631aa83ff04d4777bfde4a2ed1c20c8e : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c1b35ffe9ae4d9d9b56a79e2456e9c3 : SUCCESS in 2h 00m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3570ef09e630487999aa421f0b90e79b : SUCCESS in 1h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bd89e78149394ebda52483962b9b3620 : SUCCESS in 1h 42m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/108944ac4c6c4fce8e3e844f8b90ab5f : SUCCESS in 1h 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e922f1f1c16e37b0273172b54b4ae2398f6b08d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-22 09:07:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d506d5e4fadbc65d74e1f7a8dee75c3f59aecd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 11:37:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e156bc7df7440508fedf80e76c0407d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92497ef52f714653a1b882596f107292 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29af900e7e29434f85f3ac81b8300702 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24212880aed744c4ab8c948d20f4a0e6 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ebb0a3113bc4a1488d1b349f52b0b8f : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/087232574c044cbbb4efcf3384a9399e : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a5a3c1f0aa4f54bb72cb85971d7668 : SUCCESS in 11m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f0b8d9da2e4d46c68309bb90ffa874af : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1e20bbc2b49242f28aa122964742755f : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cdbd15bfb5204237a8db7f9feb6ec356 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a64f3c508dd34759aa9948e5fe2e6fb5 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4df77b01b47f4c119b76b21cb8fca8d6 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c924c72deb00439386889f824293d7ea : SUCCESS in 8m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eda933400ee741c7a6ead05c7788a7b7 : SUCCESS in 2h 18m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/123d21af23c84b0082c39feb78cc4246 : SUCCESS in 1h 38m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ade8a45b1ca24f708518f9c6d54281db : SUCCESS in 42m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65d41d2f6d9146e4990ed5ca2717790c : SUCCESS in 1h 03m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af9564208e6f49acb98ff76be997af9f : SUCCESS in 1h 31m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/929ab771a91447059e60507c16f6d097 : SUCCESS in 41m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6c3938a014d144249ca7a522f7c64a93 : SUCCESS in 1h 30m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48fdfdadecbd4622a65b5d4b1657abca : SUCCESS in 2h 12m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5cb550b908254783a5e80df366cb75e2 : SUCCESS in 1h 44m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a693616fd2df4971b82f598b24c3e51b : SUCCESS in 1h 38m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b87412e403444c2a866e7c26dfb3f975 : SUCCESS in 1h 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ed95a5d03874b9e81c377136b17b052 : SUCCESS in 2h 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88b1491849c8e261774f05a774736d8bd082f97f","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-09-28 16:27:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24112bd6dd56f63eb121f80dac386fe7f8a52153","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-28 16:28:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d6dbde544ca2745db17ebb9ba2c01e86b1d015f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-28 16:31:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8455dbe73c0280ee0126fdc31b2eeaff81bc2fab","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-28 17:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b61bd55eed101350087d348c207e1df2db6aa29","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-29 07:17:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d9ebee5ed7ea442c544276f2c5429008e80fe5c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 07:18:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2331677a2333d0927c14a81820dd422bd08cbc4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 08:20:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccd07e690ef147cf8f68a938364ba628341ee618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 09:20:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df57e74c3594ca4812bba251319bd85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875bf0923dd642b5aad1d66f9796e580 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3db6fe6e0ba46a2812b470f96b3e5fa : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e9d3535f0554b54abe8b747d314c2e9 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0861ddf4c34e73b79d9cdbe9ecd3a5 : SUCCESS in 11m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75a9931b15a347dd8f2b4b635a0e84ad : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a5bc9e53af94eddacb070d018855c7e : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bde22f8f9b9148eaa40c4389f1011987 : SUCCESS in 1h 32m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e89d2d1565424dcea5e2cc7f3b2dabfe : SUCCESS in 2h 01m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13401f172f704f7997035b7e1756bfd8 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":2},{"id":"aa2167ae9a27d9efd4dbd18ea313c95493692ee2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 09:20:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c58655a0490d8e634c65c4794ca6087b3e92d6c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 09:21:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c05f02510549039d947954121b97b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8415f782314b46b8bab7b9eb8c6aaa : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"54a778e8a598e5eca7269305ea1017d55d220fb5","revisions":{"08b92f51c81e79b91eb43afc3768a64bd6eb9104":{"kind":"REWORK","_number":1,"created":"2022-08-21 15:04:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/75/853875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/853875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/853875/1"}}},"commit":{"parents":[{"commit":"98ec46b8073bda426316235f743e9af004172389","subject":"Merge \"Remove netaddr module requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98ec46b8073bda426316235f743e9af004172389"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:02:11.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:02:11.000000000","tz":120},"subject":"Add *.orgi to .gitignore","message":"Add *.orgi to .gitignore\n\n.orig files get generated by Git when a merge conflict happens. Nobody\nshould ever want to add those files to the Git tree. By including them\nin .gitignore we prevent that those files get added by accident.\n\nChange-Id: Ifff0a4a5f78e35ecd722ecdb15edd08347230842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08b92f51c81e79b91eb43afc3768a64bd6eb9104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08b92f51c81e79b91eb43afc3768a64bd6eb9104"}]},"branch":"refs/heads/master"},"54a778e8a598e5eca7269305ea1017d55d220fb5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-21 15:05:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/75/853875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/853875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/853875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/853875/2"}}},"commit":{"parents":[{"commit":"98ec46b8073bda426316235f743e9af004172389","subject":"Merge \"Remove netaddr module requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98ec46b8073bda426316235f743e9af004172389"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:02:11.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-21 15:04:58.000000000","tz":120},"subject":"Add *.orig to .gitignore","message":"Add *.orig to .gitignore\n\n.orig files get generated by Git when a merge conflict happens. Nobody\nshould ever want to add those files to the Git tree. By including them\nin .gitignore we prevent that those files get added by accident.\n\nChange-Id: Ifff0a4a5f78e35ecd722ecdb15edd08347230842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54a778e8a598e5eca7269305ea1017d55d220fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54a778e8a598e5eca7269305ea1017d55d220fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
