)]}'
{"id":"openstack%2Foctavia~853994","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-02 14:08:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","subject":"Apply openstack-selinux policies in Centos amphorae","status":"MERGED","created":"2022-08-22 14:32:42.000000000","updated":"2022-09-02 14:11:39.000000000","submitted":"2022-09-02 14:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853994","meta_rev_id":"6b8865212e2c2bf7b225b6b3b18a7d1535bb3185","_number":853994,"virtual_id_number":853994,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 14:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-22 17:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-02 11:56:29.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-02 11:56:29.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 16:53:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 17:57:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-02 11:56:29.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8404ffbc7ec732e12f7660f1a22237fd08f13f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-22 14:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c5c92de72413c58234e348aada62910662d8fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:53:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e82a6bd20424d3c98830a9b2d1546fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ef77f3c4463484ca91f6a61e75774eb : SUCCESS in 12m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e5c156a20974438b9f40e4df8788d15 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6cb6e6004a74b8fa51b676d54c5780c : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54355db5f86d4020810de682658903f8 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66aa070b305f411fa150b2d8b349a37f : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dddd1060ae74673ba8d789e7a9008c1 : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d60c1f8b61471996f3aa7dc55bf565 : SUCCESS in 5m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b591f3a253494812a68319ae2c77b0be : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2283751aaa7042088847510910e16758 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/192f792eb2124ad39b11a26cf3575736 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/02e841ba76ac4adda5251cacfb9a4f84 : SUCCESS in 8m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d51d3a73d71428dba3781d3e50d98b8 : SUCCESS in 2h 19m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/029c3b096d724a96804ce77d6a026ca8 : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffd0363915dd48fb858e20751c1eba4b : SUCCESS in 41m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5c37fa2cfc942b8aa31439deebaf0b3 : SUCCESS in 52m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2eb91a8b458f442a84b83b88844513cc : RETRY_LIMIT in 24m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f171c92f68454a13b1e44bafa73ed6d4 : SUCCESS in 42m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/24b0c0656959492eb6fbefbfb62228e2 : SUCCESS in 1h 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/093c408af3de40c1b484c9d34e8eef40 : SUCCESS in 2h 14m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98906d2531754e0689e2b64d0d682745 : SUCCESS in 2h 02m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb68d8d80fbb41cfb82e6ab4f677909f : SUCCESS in 1h 51m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ac4c21b9a5f041d98bdedc8385654fdb : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb5b8fae750141c79bea5ab811e02c9d : FAILURE in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a66e1704c0d615b272e5109516cbf5feacb586fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-22 17:57:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8003ac963dfcf513c8d24766eddc6ea4d4572736","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-02 11:56:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0510c785f4e29181590f02bbb7b1f5fdf4da1e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"605fa78249df8d27ddad2a1520ecb03a79d893f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:08:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd64633f98fa4a1e8687cfa683e0d5e8\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f773e64a543542698c109e28dc068b65 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b09d46a8876f4439b1d26f6bfcdb11b8 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d8e4f16d29403eb3b48793edc441ab : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2fd194fcf394badaa70f953c60d2bd7 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccd7fb564a44572a70805864567b7b0 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00eb4581afda47a280ea3941f5416584 : SUCCESS in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0f208446cad4420bd3830bf0577c183 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e740e8a5db514e40bcc7c925d4d7e04d : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4cfb558f10ea4bc2805a27f4e483c3a4 : SUCCESS in 2h 10m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c65c399fb08f46b0aa08fb3cf57657a5 : SUCCESS in 1h 43m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a436766b0e34ac9a79213161b839945 : SUCCESS in 42m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7502dc36dcca450cbf19e32d2162c4e0 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":1},{"id":"79b4c1b5d8dbe54ad9072c606850c77f43b06f3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:08:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6b8865212e2c2bf7b225b6b3b18a7d1535bb3185","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:11:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f84bd909eef646c2b95ed3949aae9894\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0a1b409feb45ba81afc12bdf287703 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a400ff2d7fa045258366f7b5098aa2da : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"995fe216e247a5763fc6719a0a887c9998370c30","revisions":{"995fe216e247a5763fc6719a0a887c9998370c30":{"kind":"REWORK","_number":1,"created":"2022-08-22 14:32:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/853994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/853994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/853994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/853994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/853994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/853994/1"}}},"commit":{"parents":[{"commit":"9c30758dfbdc4d0dc5ff9a942ffece64e5cde325","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c30758dfbdc4d0dc5ff9a942ffece64e5cde325"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-03 12:37:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-22 09:17:52.000000000","tz":120},"subject":"Apply openstack-selinux policies in Centos amphorae","message":"Apply openstack-selinux policies in Centos amphorae\n\nWhen enforcing SELinux, the commit enables the following SELinux\nbooleans:\n- os_haproxy_enable_nsfs: it allows haproxy to run in a network\n  namespace.\n- os_haproxy_ping: it allows calls to the ping command from haproxy (for\n  PING HMs)\n- cluster_use_execmem\n\nChange-Id: I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf\n(cherry picked from commit 3115589f2158bfe0196c8f9d3c105d16c35408f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/995fe216e247a5763fc6719a0a887c9998370c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/995fe216e247a5763fc6719a0a887c9998370c30"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
