)]}'
{"id":"openstack%2Foctavia~853995","triplet_id":"openstack%2Foctavia~stable%2Fxena~I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-02 14:14:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","subject":"Apply openstack-selinux policies in Centos amphorae","status":"MERGED","created":"2022-08-22 14:33:05.000000000","updated":"2022-09-02 14:16:14.000000000","submitted":"2022-09-02 14:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853995","meta_rev_id":"88d1f53c3505fbfa5ab25161bae70e4ac2622470","_number":853995,"virtual_id_number":853995,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 14:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-23 21:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-02 11:57:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-02 11:57:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 17:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 17:57:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-08-23 21:28:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-02 11:57:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"b7931c0ae89ecaa9b0e39c8100b808e308ea83af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-22 14:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cc84f401ce0a77f1d0f2ab1aede274607f61853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:17:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c8084353be94874833028fa73298df0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69f6bf3d42cf4023b8a50647c54532d8 : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a6d3ae683234ecbbe595f223fe2df5b : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a656415d1f046149771f4bb25f7f702 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e17e0df24d246cea60252a13f045180 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90811499b9824a889de15c788a6e96aa : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31420a7ceb4c42c0b88dd4bed879d89b : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7ad94381574a6e91469382c5725b89 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/073d959547e240698089e538667cea2a : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a54efac6995e4de4be54f8a9e6087f25 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3811a6ac1e434e2d873d1bc196bcb4c6 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/576ff4d972e84b1c99a38276046ff061 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b1ed4607fb0433f87824939961ad387 : SUCCESS in 10m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9bd855aa34c402e816c06c398618d72 : SUCCESS in 1h 48m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13ac1a3de3f4417ba5d4f7eefb7bde0c : TIMED_OUT in 2h 41m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71af46876f3a41438bf985e2a4f48130 : SUCCESS in 40m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f126d49b9f6c46899c6a149b19da51e9 : SUCCESS in 58m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f4d8050775c64ea4a8fe6fdbe57d38a8 : RETRY_LIMIT in 25m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/168749438f0b4718b06730b8a2ceb3df : SUCCESS in 1h 20m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05258317f38a4c91aab1b278712a5061 : TIMED_OUT in 2h 41m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6645fce4ca342a1a198282d628f69de : SUCCESS in 2h 19m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aff860463d434ab4b4c836267227a4a6 : SUCCESS in 2h 06m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/56880a087925417ba46b76f3d4a6dc43 : SUCCESS in 1h 51m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cb1e85dfc9eb41ba933882acc5abbff8 : SUCCESS in 1h 51m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3e8209cb8b5043948d6a7cc0b72f3bc2 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12ee5f2bed40e97d2a75c51bba5e5e383b1b03a1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-22 17:57:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e7db31279acdfbe5ff919c84219a269d5bd74db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:32:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8fdd397983d4491bf50fd8f5932eb48\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f7b6411c21540ae85d56cdda3b29b01 : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2107dd6d954a4ebc83b968a4f56e2ef6 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595e4562490748ad91e8f50b6a7fc843 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af5f8f695a824840a99974db2ef4b5b9 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee2eba54b8bd4e8792a347a6192f787b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274b2c53bae942fa919a947cc7febf35 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/956ae6356be843f59d17df5c04bb5444 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cde5133b9bf419ba9c6d2ee78676068 : SUCCESS in 5m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/84341c384ec445bd9f1e4373d45e5fc7 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2aa0e34040ec497d899c7e3fb339dd31 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a51415e485b4ebcab9c03d3f7c75c34 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c92139212b5941bdb13b735905495be6 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e229029124b4656ac3ae83b31f2dcdb : SUCCESS in 2h 16m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfb5f207dde84c39acfb86eb2e9b651f : SUCCESS in 2h 13m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c662754ef084cdbbd8fed5055b0ab0f : SUCCESS in 59m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8da359768c634fe58b7f45928bbef559 : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1479a70b79b84ea38521ac8ac70e9fcf : RETRY_LIMIT in 23m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b46d5959a454a2396f52770d0bd271b : SUCCESS in 54m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/866e7272e131477f8751ab43f744ebde : SUCCESS in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1093ae1a03cc40dc93957bf209855354 : SUCCESS in 2h 29m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/27b1fa7ca1324c66ba1b9ba14d032ed7 : SUCCESS in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/13a3c2ca83404062b47c846867914dd0 : SUCCESS in 2h 05m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/011ffd24df3a438eac295a119c211a3f : SUCCESS in 2h 11m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e4c54eb83a984e499e98e0b66581601b : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d69a08f334e099ba2ed8345f32064124c1cefb1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:28:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cc1554d8271636f0ef7293751999e7188891ccc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-02 11:57:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a4c1f58e90c0a3337c5497581b3b29381e36482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:58:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62bff46425055124ab295b345e724891043ad2e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:14:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb4709076b547a6a24022cfe62dc01d\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6d3be8b9c4945f7946d16aeb97610df : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a6897b2c544e859407d26861db333c : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f391ffb7b7414b738dfbde696cd0e0ce : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42ef127025d7405ba65bd03b24e58fa0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc7c97101eb40d092ce59c819acd87b : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfb01d8c424c4e7e80922eb3ae579bc7 : SUCCESS in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23b2e28171874982994c71b207f0128f : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08ed76ade56c4565a0961d4eb01bc8c1 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c559d25b02684f31beacf2429ea4707c : SUCCESS in 2h 15m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df9579ab901b42afafe3ed595c4c70a5 : SUCCESS in 1h 53m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/864354f97f5e45e19cb5a27996324539 : SUCCESS in 41m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2831177aaf74304af02d797102a2a88 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fa18242c213bcefb5474b6dc88aecd33e328106f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:14:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88d1f53c3505fbfa5ab25161bae70e4ac2622470","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72767099f1d44765b6e7788d3f09516e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4653bbdde658447aaac7fa375156bb77 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3be8dafabcd41a596a5d58e3d3f6cf2 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"044de8e52f1d8647846c4662103dc0ef102119ec","revisions":{"044de8e52f1d8647846c4662103dc0ef102119ec":{"kind":"REWORK","_number":1,"created":"2022-08-22 14:33:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/95/853995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/853995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/853995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/853995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/853995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/853995/1"}}},"commit":{"parents":[{"commit":"bda4d7160aebf7549d72b049c8dca53f7d4dedb5","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bda4d7160aebf7549d72b049c8dca53f7d4dedb5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-03 12:37:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-22 14:32:53.000000000","tz":120},"subject":"Apply openstack-selinux policies in Centos amphorae","message":"Apply openstack-selinux policies in Centos amphorae\n\nWhen enforcing SELinux, the commit enables the following SELinux\nbooleans:\n- os_haproxy_enable_nsfs: it allows haproxy to run in a network\n  namespace.\n- os_haproxy_ping: it allows calls to the ping command from haproxy (for\n  PING HMs)\n- cluster_use_execmem\n\nChange-Id: I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf\n(cherry picked from commit 3115589f2158bfe0196c8f9d3c105d16c35408f7)\n(cherry picked from commit 995fe216e247a5763fc6719a0a887c9998370c30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/044de8e52f1d8647846c4662103dc0ef102119ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/044de8e52f1d8647846c4662103dc0ef102119ec"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
