)]}'
{"id":"openstack%2Foctavia~853997","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-02 14:07:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf","subject":"Apply openstack-selinux policies in Centos amphorae","status":"MERGED","created":"2022-08-22 14:33:45.000000000","updated":"2022-09-02 14:11:16.000000000","submitted":"2022-09-02 14:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853997","meta_rev_id":"8394c54fd88964bc5b0e846e28169f6d3001e699","_number":853997,"virtual_id_number":853997,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 14:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-22 17:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-02 11:56:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-02 11:56:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 16:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 17:58:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-02 11:56:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"63c4403fc07ab7af1e754e189a0a4902de11d9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-22 14:33:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"618041549f27b9d259cbcb3d47a77f2ec06c2ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9637f7cad69b4c98bd8aadd29b9eea40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ac4c17978e48128711f05c2a23bb7a : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab02bf716e3e4b47a642af4864b6b059 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f187863826d64f55ba0012358efbf7a5 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab1540e7745342d784903c7df938339e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b42652a82644106bcb7358d05801558 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebde767b48a243a6ad771eff8c4f5ae6 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5d3002ced68d449bb41353ccdb455137 : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9b2cd5dbf638485ab770a42e4e0564e5 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cea9cda024e347abb12550a94114e395 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/169f92a23d3d46a786e5d2c794c51185 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f8a726a0c3b402da6ee3a5104196a5e : SUCCESS in 2h 08m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ab0b2a922b6448c88b4c763b4ffb434 : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e7ed9a9bb164508af4f96388dddba65 : SUCCESS in 38m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b0eafc71b7dc4db6bf5e415ac3de26ae : SUCCESS in 38m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff453956e7a74ca983e8a4acfd148db7 : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61936c26e70d4cbe9cc02e9419c09203 : SUCCESS in 1h 36m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3777f61e1dd841f28822fcb1a126e9d0 : FAILURE in 1h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/46428b0be70542058d6de24f1fde46a7 : POST_FAILURE in 1h 58m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a35d25a35d745d9a946a493ab61dd89 : FAILURE in 2h 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6958dc7eb5aa4e01bc260335bb0f7689 : SUCCESS in 1h 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eb8064b3af91418a88b78ca83b3c04a6 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f107ea32b4e38b7d4d4a5db7a9ed3276e08e3477","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-22 17:58:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a485edde9c145deee5691954b7064ef61a11b6f7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-02 11:56:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbb8e3137c8c91b1c4458d0eb8633eb01fd63d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:56:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"620eb4f43f4128acd5f355be603d492f6b57f640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5592f93e81946c49adb7db0f10eaad3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ab517b8e8544908277bbc37d00b783 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22dac961a74e470bb949096dfa334c08 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7171f4758944eab6c26230a1fa99aa : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d038b4fd62484b5c9b3cf890997a10a1 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6247cb0fab904ecba9fd9538dce125c1 : SUCCESS in 5m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a59bc3192d88439a8f28f8e143ec5ef9 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78059af92ebb45469b3ba64d4e48d308 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8680ced0765457eb988195a85b757a5 : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26007645aa964b2a82dd0a3f4f0dee0b : SUCCESS in 2h 07m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29a6ec6e490740d29d1a448c5c8b80c3 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4df38001231c426780ae4d92ccebadfb : SUCCESS in 39m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7acbbcb7e3a450aac900f47d90022d0 : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c8d1fa8fb63feb2ac50cdf3094cb222fdaf0ca06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:07:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8394c54fd88964bc5b0e846e28169f6d3001e699","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:11:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5665dd6ba8ce49378307bfcd6f417343\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887a98c328a74576bd9d8943b283087b : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ee477a579984801862cf8a007d07b95 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21","revisions":{"7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21":{"kind":"REWORK","_number":1,"created":"2022-08-22 14:33:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/97/853997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/853997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/853997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/853997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/853997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/853997/1"}}},"commit":{"parents":[{"commit":"7eb6bf1681fb8538521def0395835775a180f9bc","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eb6bf1681fb8538521def0395835775a180f9bc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-03 12:37:51.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-22 14:33:24.000000000","tz":120},"subject":"Apply openstack-selinux policies in Centos amphorae","message":"Apply openstack-selinux policies in Centos amphorae\n\nWhen enforcing SELinux, the commit enables the following SELinux\nbooleans:\n- os_haproxy_enable_nsfs: it allows haproxy to run in a network\n  namespace.\n- os_haproxy_ping: it allows calls to the ping command from haproxy (for\n  PING HMs)\n- cluster_use_execmem\n\nChange-Id: I4b5199d13d122b56619c4a2d21e86a06bd0dc3bf\n(cherry picked from commit 3115589f2158bfe0196c8f9d3c105d16c35408f7)\n(cherry picked from commit 995fe216e247a5763fc6719a0a887c9998370c30)\n(cherry picked from commit 044de8e52f1d8647846c4662103dc0ef102119ec)\n(cherry picked from commit cf983b04cdae0dd20fdea06f50374d452d2fe6b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b4f4ea201b2d7c68ee31a53d0e4e2a7844e5b21"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
