)]}'
{"id":"openstack%2Foctavia~854192","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I8b1e3edfdc69f531b00cdef49121c45e944f25ca","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-22 15:52:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b1e3edfdc69f531b00cdef49121c45e944f25ca","subject":"Catch exceptions on I/O in driver-agent","status":"MERGED","created":"2022-08-23 13:45:48.000000000","updated":"2022-09-22 15:53:32.000000000","submitted":"2022-09-22 15:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854192","meta_rev_id":"6c3d1bcbe02de7699f937aa0c68f08af0db8472a","_number":854192,"virtual_id_number":854192,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-22 15:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-23 21:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-22 07:12:17.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-22 07:12:17.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:15:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-22 07:12:17.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"aded254dd8e9c51ab32bfcbe73c5a01f820b0e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 13:45:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a86ea2ec631cdc772f01140bf97a4fc145a292eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a87ae8c4c7c4a4496d3ac3f6897f84e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6bd39e5d122473ea8e4c185e5f84b51 : SUCCESS in 12m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dfa4af8fe5b4b0c8e7ddd564835f5f7 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58fd8e1f0b2046c39c88d7fa0cdf7bf8 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d6393d02f94c69a80109c9726c96fc : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf17471c5af42ca9acf02ad4cb2798c : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5157587410a840cfbf7577af36f9cc19 : SUCCESS in 12m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/862bc64a744f439c8f1b7f02f656d08c : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e53f819de896455592be2053425098c9 : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9c286c36881c4d93a416062ac7b2be35 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25078d0f7c5a4a1c9886cde17d7fb346 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0a4c3d1b7244194af19181cbbc71b5a : SUCCESS in 2h 42m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ecc8ea5957b420bb85cb649be85d2cf : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/439ffdb3e50c4070af1bb6e7bdeeb016 : SUCCESS in 58m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a54da52be1254717a40ae5da98314832 : SUCCESS in 1h 00m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ddd0e7602ef1471f8cdd4af64cad3117 : SUCCESS in 1h 31m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5656859790e64237900495b1b039ecde : SUCCESS in 46m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21345498bd2a4cde9d48b61a559832b3 : SUCCESS in 1h 41m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94bdbfd050284d65ba3cfbf7636a129f : SUCCESS in 1h 57m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ef8237f097d743aca7c9f8ef5e9a4037 : SUCCESS in 2h 21m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8dace07d281346299ac811f71c6d6d17 : SUCCESS in 2h 13m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bedc76efb62044caab82aaf7f1947a33 : SUCCESS in 1h 27m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/59ac4f8c1fb7452d81542e75401b0d76 : FAILURE in 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"436e6f13f816bf8d501048710438c560dd1b5942","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:15:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bf62673b5b5f97106f3702eb073bf119990abab","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-22 07:12:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbc92bf6466d2f44d7fe35d1ac504f65f1ba7b59","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-22 14:04:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ec152c2531e08e5b9004a5e397d157f7b5390ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 14:05:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16c6e6e64fdc500ad531ff1b3fde402b0ec723b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca8cb5cadad4b869fbe0ca195e2f15a\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c1c6a81d1e94974b613b9dfdb434e6f : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3405f2acb54c27b4f11351de081082 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/304025bb292649ff9328b1a487b6d841 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99131e896beb42dd8b0cb58a6a025877 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9409cbd6514f89bcfa2d951f08f9ab : SUCCESS in 13m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ba2d93a62a846cda5d5c4b616bc708b : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87a848fd846a4c968bd2ad5d6b6dd1d1 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e0d31cc77d04abca6e017d955c2c25e : SUCCESS in 1h 23m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff06331e03dc481ba2855400acb2a692 : SUCCESS in 1h 45m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56cc3ddc0e0f4e88bad9c0f30fc21a9a : SUCCESS in 42m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f8baa170e4e415b9338dfde8fea2b76 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":1},{"id":"76cb295e1448f4dbf246a406844752c9d60a2407","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:52:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c3d1bcbe02de7699f937aa0c68f08af0db8472a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14e06e756bb84d1a8cda71ad1cec95d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514b43c8137a4e3c93246d90dab4a4b3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d55654174ccec8d716333a2466ed92f4a8e2d7f9","revisions":{"d55654174ccec8d716333a2466ed92f4a8e2d7f9":{"kind":"REWORK","_number":1,"created":"2022-08-23 13:45:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/854192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/854192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/854192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/854192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/854192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/854192/1"}}},"commit":{"parents":[{"commit":"0bf3ae96d10b89b6442b0d01b834efe402f7d7ca","subject":"Merge \"Change FIPS jobs to centos-9-stream\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0bf3ae96d10b89b6442b0d01b834efe402f7d7ca"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 14:15:00.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 13:43:21.000000000","tz":120},"subject":"Catch exceptions on I/O in driver-agent","message":"Catch exceptions on I/O in driver-agent\n\nIn the driver-agent, send or recv functions may trigger timeout\nexceptions. Not catching those exception breaks the driver-agent main\nloop, forcing the admin to restart the service.\n\nStory 2009032\nTask 42772\n\nChange-Id: I8b1e3edfdc69f531b00cdef49121c45e944f25ca\n(cherry picked from commit 2886fc7224773dafbe44f95927476746bd7767e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d55654174ccec8d716333a2466ed92f4a8e2d7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d55654174ccec8d716333a2466ed92f4a8e2d7f9"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
