)]}'
{"id":"openstack%2Foctavia~854198","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I8b1e3edfdc69f531b00cdef49121c45e944f25ca","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-22 12:32:43.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-22 12:32:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b1e3edfdc69f531b00cdef49121c45e944f25ca","subject":"Catch exceptions on I/O in driver-agent","status":"MERGED","created":"2022-08-23 13:48:59.000000000","updated":"2022-09-22 12:34:23.000000000","submitted":"2022-09-22 12:32:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854198","meta_rev_id":"fcd69be1144e6e0055226bbae5d6705a9034b841","_number":854198,"virtual_id_number":854198,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-22 12:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-23 21:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-09-22 07:23:39.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-22 07:23:39.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:27:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-22 07:22:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"c50745a0ee69faf5b3cbafff4ea1129b45848279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 13:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dd64bae8c94a4dca5aa16f351b47ef648324dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:14:47.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f60a9037b8e14f0d98b0c0124f80084d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4e5cf92189245a4b2b5ecfec34a1ee9 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0f8f541e904ddca9d504f7f6fc3f1a : FAILURE in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2a4f3d614994dfabf8504df491cf361 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbbee97781b4466cae299a8699b1ad10 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42f59dae7fb541afbe7bb6c2935f296e : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24116bb42434cb69c3b3b1679f8f677 : SUCCESS in 12m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/837b3bf75f0f4d14aa0a6b708bfa80b2 : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb457c09c5d04b08b71fa7f1c3a1b40c : SUCCESS in 8m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e0ca39c05e7f4c329c2d8f16834500ef : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/595d756b642b4959afca8b27db629921 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cbfd6b3620a9420bbe554891ae8f2309 : SUCCESS in 1h 18m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e90251c730d24e5a8c48d772672a6a58 : SUCCESS in 1h 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcb179b891b1435b8c193bbfbe913f86 : SUCCESS in 2h 08m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7662e1b6dcc54afdb728e2bf8319b0aa : SUCCESS in 2h 12m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f3dcd252a57e42769410dbb812306879 : FAILURE in 31m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2cff974c333148af90326fd229848e77 : RETRY_LIMIT in 18m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d04d8ba306f42beb1116ff35c0f6738 : RETRY_LIMIT in 18m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55f660e34a114fccb29f22b5ac93354c : SUCCESS in 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcc37697afe844c79cfee0ad5ebf2171 : SUCCESS in 1h 38m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41650e1cd8e24830a85a11e934ccb205 : FAILURE in 43m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8723fd62b38742828a31a464a60cd220 : FAILURE in 54m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/650ef1f78e074c4c86bf401fa27a41e5 : FAILURE in 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1454f85a954659da2db65f35ebf63a03a8f1f746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:19:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ce2ac1271ae22cb6e3e6c39f13d250bfe2a25040","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:27:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26356faf663c89106def51ea509c0c6a77536d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 23:47:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221e9395ac3e4634b056b8f3ea4e8e10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1323b9753d5c44c4a87b455e1b5b1d03 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a488ae357d04c4d941ad0a58b358f2b : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f41ea939650b4d9eb7e920349ee32ce0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6ad4b877fd34377aa37de74799b1cc6 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccc1edaa01c04e24ae7572c8433cd946 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aab5eb6eb8040ddb56e1b3d51c7af1e : SUCCESS in 8m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/783b2db86ceb4e08b5899c82bd06e559 : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0b3d5e369b5e40fbbab5659ce1aa8e2b : SUCCESS in 8m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/74d6d95f119a45a39528b93d5892128f : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e84d20be496c4036be67f718e28cae33 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b583ea5a9f454043bc3f5562d171a6e7 : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/70981022438c4cc19a9b65d90002c1a2 : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f9133b33b374b73a37126d4687e094b : SUCCESS in 1h 56m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/718b26fe0a70412284ee93971f7f1c75 : SUCCESS in 2h 12m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/729b97a443cf4604b0b40a9986a0814c : FAILURE in 29m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/800b3325171148ec918ce7502ba7e7cd : RETRY_LIMIT in 9m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1405b7bf0edf46b3881b58a77d9e13ed : RETRY_LIMIT in 17m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9331a094022e4d28bd2e220de713f32f : SUCCESS in 59m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8c9c366a04d7485fbdae7962936b599a : SUCCESS in 1h 34m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af307b02e67943c09d9a0353eeab2c85 : FAILURE in 55m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a808176a1311430898064b861734553c : FAILURE in 39m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/837270ffff724ed58b5d9ce994a4eb13 : FAILURE in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0acb72ccbadd9edcd56af2e45ae23c7d5fd24f70","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-22 07:22:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96e680ca7658700076d87ee0ece0b7512ce5adc9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-22 07:23:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"343b66b63cc4d4cbbf2cd6aa734c020025fcc018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 07:24:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe752d8d50e715039d2d82dd7423dedb56f070f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:32:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0461747a49c14e85abc43099278ef1e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4079604b9fa543ffa01ff7abb8e3ebec : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/762eadf304e24fd8b3c364525437540a : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23e475af350347a99974733d79311573 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/387d7e536fbe4f3384058cb0bc1da2a5 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a405c530071481b98a1da2c991a0389 : SUCCESS in 12m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4b0c9f6a838b4fceb67b42a3e7dafbe4 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4794c483b4fa4726b65cb09189bcac55 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6cb94f54b8e1431eac74a03454c586fd : SUCCESS in 53m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dfda80ba784340fca09bef684cc67ea6 : SUCCESS in 1h 07m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/275110861c74429a943efdd5c212219d : SUCCESS in 1h 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"abd9d192d14546d3470ce74c3e5d1d009e8008db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:32:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fcd69be1144e6e0055226bbae5d6705a9034b841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:34:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f70e7590d734e38a198fa5421bd96a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8580bb463b34cc78b548d298bd1ba94 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af7eee2a2a477ee822377b0832e04a6ad6d4cafa","revisions":{"fd8cb1ce1624cddb9b84e4019eaf3e812b43a0a8":{"kind":"REWORK","_number":1,"created":"2022-08-23 13:48:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/854198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/854198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/854198/1"}}},"commit":{"parents":[{"commit":"58e2e13b2b107b92f5a9a20ffc83689731f16a38","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58e2e13b2b107b92f5a9a20ffc83689731f16a38"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 14:15:00.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 13:48:45.000000000","tz":120},"subject":"Catch exceptions on I/O in driver-agent","message":"Catch exceptions on I/O in driver-agent\n\nIn the driver-agent, send or recv functions may trigger timeout\nexceptions. Not catching those exception breaks the driver-agent main\nloop, forcing the admin to restart the service.\n\nStory 2009032\nTask 42772\n\nConflicts:\n\toctavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n\nChange-Id: I8b1e3edfdc69f531b00cdef49121c45e944f25ca\n(cherry picked from commit 2886fc7224773dafbe44f95927476746bd7767e8)\n(cherry picked from commit d55654174ccec8d716333a2466ed92f4a8e2d7f9)\n(cherry picked from commit 279bc0b68224eb550da6854df0d844f5d2dedc3d)\n(cherry picked from commit 656c0e48f60d2ccf1d1f587673781a3b54c4ee6e)\n(cherry picked from commit a63057dba45bed982df103b722e342ce491ad2f5)\n(cherry picked from commit 75b8f8838f922d5bc204bbd6cef3866feb2287fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd8cb1ce1624cddb9b84e4019eaf3e812b43a0a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd8cb1ce1624cddb9b84e4019eaf3e812b43a0a8"}]},"branch":"refs/heads/stable/train"},"af7eee2a2a477ee822377b0832e04a6ad6d4cafa":{"kind":"REWORK","_number":2,"created":"2022-08-23 21:19:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/98/854198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/854198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/854198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/854198/2"}}},"commit":{"parents":[{"commit":"58e2e13b2b107b92f5a9a20ffc83689731f16a38","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58e2e13b2b107b92f5a9a20ffc83689731f16a38"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 14:15:00.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-08-23 21:19:01.000000000","tz":0},"subject":"Catch exceptions on I/O in driver-agent","message":"Catch exceptions on I/O in driver-agent\n\nIn the driver-agent, send or recv functions may trigger timeout\nexceptions. Not catching those exception breaks the driver-agent main\nloop, forcing the admin to restart the service.\n\nStory 2009032\nTask 42772\n\nConflicts:\n\toctavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n\nChange-Id: I8b1e3edfdc69f531b00cdef49121c45e944f25ca\n(cherry picked from commit 2886fc7224773dafbe44f95927476746bd7767e8)\n(cherry picked from commit d55654174ccec8d716333a2466ed92f4a8e2d7f9)\n(cherry picked from commit 279bc0b68224eb550da6854df0d844f5d2dedc3d)\n(cherry picked from commit 656c0e48f60d2ccf1d1f587673781a3b54c4ee6e)\n(cherry picked from commit a63057dba45bed982df103b722e342ce491ad2f5)\n(cherry picked from commit 75b8f8838f922d5bc204bbd6cef3866feb2287fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7eee2a2a477ee822377b0832e04a6ad6d4cafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7eee2a2a477ee822377b0832e04a6ad6d4cafa"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
