)]}'
{"id":"openstack%2Foctavia~854201","triplet_id":"openstack%2Foctavia~stable%2Fxena~I876fd91fb40d9250381bbb37f8148b53692a7d8d","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-10-19 10:02:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I876fd91fb40d9250381bbb37f8148b53692a7d8d","subject":"Fix PortNotFound exception when updating a LB after a failover","status":"MERGED","created":"2022-08-23 13:55:08.000000000","updated":"2022-10-19 10:04:02.000000000","submitted":"2022-10-19 10:02:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854201","meta_rev_id":"375f7b69fde7dbf550f0c0522f931f853d3c6c7b","_number":854201,"virtual_id_number":854201,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 10:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-23 21:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-19 08:12:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 08:12:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:12:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-10-19 08:12:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9c21d52d970bd52484a06bf9cf9e771eae646f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 13:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e7cad461b6fd775f17ae64f275bc8b614a94c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173d5e62ebc540d3b71e11aa0e01c586\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b1b364a775a4d39b5fc128486962c25 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49aa0790781b440d8183cf202a723823 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de90c6e8dc75428ca62437cb3447ee55 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9e7a649aa04663a831d1ff7e7b31cd : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3b614bb3fc42aaa9be983f48aa4c32 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7bdf75104c14bb8aa24cc5e21a6d239 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27925cf3e6aa4d85b7d78df60c062641 : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7973f01388074ca7af55d433486c1476 : SUCCESS in 5m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db81c8e4591142cd85b5a05ea4d2a9b0 : SUCCESS in 6m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3e3610a745214833aa5f7d7ab9185489 : SUCCESS in 9m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ccb4a6e432014144bdd171c6ea9383e5 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63597e7752244628b82b5504dbecf852 : SUCCESS in 8m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df70488736d349e8bf6a441c2c01fbf1 : SUCCESS in 2h 10m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fb0e21bcaf049cda4643effbbc09a2c : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/358b67b5939c4e6cb0bed250661b6cd4 : SUCCESS in 1h 00m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e784cb5f27994b9f80d5071e4041642c : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b0d01c6441a4bb98c76a60cb1b92e03 : RETRY_LIMIT in 25m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe9f78e687c64ba99412124037502536 : SUCCESS in 38m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad21ccfde42e47fa83c79291c8473ee8 : SUCCESS in 2h 19m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e7d0cef17b447cba4dd243ba7fe199c : SUCCESS in 2h 38m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6bca079fe9b444d93e42c3975ba1ddd : SUCCESS in 2h 15m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/baf821711fbb46b8aae112f4948aae3e : SUCCESS in 2h 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3592143daa6244df8ebb4cdd89b04766 : SUCCESS in 2h 08m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b86047bdefbb40b59c6bb60f7d0ea64f : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02c0d2494e771b198aea1ee3e197675968310239","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:12:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7fa07b724f018af4e9f5c989d68c361e3953b47","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 08:12:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"471e999eb78c265b6aec0e6a750aa9f32bf8f6c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:13:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bc0302ff906fa8119d6fdd40973a2c1234d9d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:02:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e73e3de618440e996f4505de2a2b2f\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/080cc341113348d891bf1e0d6db7b11a : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f21edbeec4a4b7d85f4196e2bcbe6ef : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7be0f50c8d1499eb094757bdff7420c : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a01ce9d999714775839dfdbdd70e5601 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c88b8b86fe4196aaf7132784fe07f5 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811d74eea11a49298a957df0f36daa7a : SUCCESS in 5m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eecfb1a2c1384f3ab2b99442825b744c : SUCCESS in 3m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86233b760ded4185b2acf5181a101e0b : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae1fdde2addd463fad6ddbce0cf72991 : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de9fc6f7e0ec44219dd2517cd37ab765 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/488990c5e5244128ba6812c2344a9144 : SUCCESS in 35m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29fb87ece8fc4e03a96d289a2be54fe2 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3406298c3bc4edb01b8cac5a9f4328f3c97cce98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:02:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"375f7b69fde7dbf550f0c0522f931f853d3c6c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:04:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec20f244e59c49d191a4a802693860ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475d0b08a1ca4309a1c15f83cb752017 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4ce7e876d22451898e195bcf2afedde : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b28293ddbe6e74170c253b55fafee3e057c47b4","revisions":{"8b28293ddbe6e74170c253b55fafee3e057c47b4":{"kind":"REWORK","_number":1,"created":"2022-08-23 13:55:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/01/854201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/854201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/854201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/854201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/854201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/854201/1"}}},"commit":{"parents":[{"commit":"c51f15b49fadabae548ddc8a6d2863ac4a216837","subject":"Merge \"Fix driver-agent cleanup\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c51f15b49fadabae548ddc8a6d2863ac4a216837"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-09 09:56:39.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 13:54:51.000000000","tz":120},"subject":"Fix PortNotFound exception when updating a LB after a failover","message":"Fix PortNotFound exception when updating a LB after a failover\n\nThe ApplyQos task applied the qos policy on the VRRP ports of all the\namphorae linked to a load balancer, including amphorae in DELETED\nstatus. It triggered an ERROR when updating a LB (with a qos policy)\nafter a failover.\n\nStory 2009667\nTask 43894\n\nChange-Id: I876fd91fb40d9250381bbb37f8148b53692a7d8d\n(cherry picked from commit c3481d04a25c4e6f03ce4de720c620730188d99e)\n(cherry picked from commit 328ffc00cf5fc74eed6889bae2594cded1f8cd9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b28293ddbe6e74170c253b55fafee3e057c47b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b28293ddbe6e74170c253b55fafee3e057c47b4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
