)]}'
{"id":"openstack%2Foctavia~854207","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I876fd91fb40d9250381bbb37f8148b53692a7d8d","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-10-19 10:25:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I876fd91fb40d9250381bbb37f8148b53692a7d8d","subject":"Fix PortNotFound exception when updating a LB after a failover","status":"MERGED","created":"2022-08-23 13:58:39.000000000","updated":"2022-10-19 10:26:20.000000000","submitted":"2022-10-19 10:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854207","meta_rev_id":"9b7cf74087a32e04d95e19b6fd95c2f76255019c","_number":854207,"virtual_id_number":854207,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 10:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-23 21:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-19 08:13:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-19 08:13:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 16:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:12:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-10-19 08:13:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"ac19dfb39e550319973fc70a1f2e7ef1698faf26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 13:58:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6950b1fe3cfd9271583d037ba97290854240f9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f7fe2a2c686441598d77ec4e0029511\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b133bb5b384440d6a97da584b8691b93 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc91b9ab740f4d9eaaa4f17120964593 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1784771a57a84672aed2ced1b23792ca : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3a20cbc0d446c69f69610872793019 : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ea41f97256148cba102e7c83fc5ecef : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b2416d7d29434ba54dab5dd0ea829c : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/327cfa1f916a4bcfb898eefb4b967646 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0bb3e8777519470cb2491920396ed613 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/acf2eca3e17743e898b3c8a79bd23a1a : SUCCESS in 8m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6d98e2cf1945473a84312dc07292af5d : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4ffa3bcc31d48f68f78d320d7ac005c : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d65140c22b814ed8919b159668d10193 : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/09775ad0e0a14a279e7cfcd5f0daf6d0 : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7a5db05cc2a4c15a71c0fc1178943e3 : SUCCESS in 2h 16m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/629c15f225f9420391fa045557fa7268 : SUCCESS in 2h 31m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c119773a2e994d1cadfcd6f6af6b73e6 : FAILURE in 20m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca7df1f47d9d48d6afb367cf845819a9 : RETRY_LIMIT in 14m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4dd72a9044d4e798f7aae5b35067fe5 : RETRY_LIMIT in 14m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7df9454b0ff04288b478eec823313385 : SUCCESS in 44m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/07a616c5e0f24394a6f05b416df4b1e4 : SUCCESS in 2h 00m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fd994f4c0284388bf2d6e6666db3d2d : FAILURE in 54m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8a45bb40814b4700bc9274dfda9084a6 : FAILURE in 55m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/515097ff2d984f568a3c95bdbb6248c9 : FAILURE in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"357923e04adec119cd16bbd755e2b6f74d129c9a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:12:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"048fc7386e1d0385266d973f2d60f9d95a5ec1f3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-19 08:13:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5041a05095490b4633f9c39e1aaf16ae1e71201d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:13:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b53309710eb7636555cb28f9fabe5b4dfd3d124e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:25:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cddf0d4ebd4bf5bb39993a2bb77c6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ce5e55dd0a4332a291783bc56a7d9f : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d73bac863394680915d37f9b6a79da1 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49d58da11e64431a93a2e05fb3bf690 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e34a83f404734791be49466eecabd58e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a49de0fe6f41c599c2f6f761cf59c5 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eed37d29a64413da8881d663b11771f : SUCCESS in 4m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/754cd3be7f6b490083df8245fb02dc91 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2db8a0f839945018d024e4fe12082ee : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1c15e46a375473b8221bdb6120895e6 : SUCCESS in 50m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4eefb1fff7284f05ac276af1772ee192 : SUCCESS in 56m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9a3b20c06f34de2b0b7d5fd8c1a1426 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"39df9141d31cdad36edc474fa6f1346c9db9bf96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:25:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b7cf74087a32e04d95e19b6fd95c2f76255019c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:26:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a1b0a2a5a614c96a5461510aedadeeb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c890e066080d4d6b8a080429e2e7b5e2 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ea3b2ed64e43e291476170419e15e3 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fd073a7f3f78f06b40c418184ca5fa02e410199","revisions":{"6fd073a7f3f78f06b40c418184ca5fa02e410199":{"kind":"REWORK","_number":1,"created":"2022-08-23 13:58:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/854207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/854207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/854207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/854207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/854207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/854207/1"}}},"commit":{"parents":[{"commit":"58e2e13b2b107b92f5a9a20ffc83689731f16a38","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58e2e13b2b107b92f5a9a20ffc83689731f16a38"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-09 09:56:39.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 13:58:19.000000000","tz":120},"subject":"Fix PortNotFound exception when updating a LB after a failover","message":"Fix PortNotFound exception when updating a LB after a failover\n\nThe ApplyQos task applied the qos policy on the VRRP ports of all the\namphorae linked to a load balancer, including amphorae in DELETED\nstatus. It triggered an ERROR when updating a LB (with a qos policy)\nafter a failover.\n\nStory 2009667\nTask 43894\n\nChange-Id: I876fd91fb40d9250381bbb37f8148b53692a7d8d\n(cherry picked from commit c3481d04a25c4e6f03ce4de720c620730188d99e)\n(cherry picked from commit 328ffc00cf5fc74eed6889bae2594cded1f8cd9b)\n(cherry picked from commit 8b28293ddbe6e74170c253b55fafee3e057c47b4)\n(cherry picked from commit b8d68e6e3d16dd29e3b4ba58b0cb3b54780500f7)\n(cherry picked from commit 36a9c13e3b1898aaeb6c21f1e0d004533b711a27)\n(cherry picked from commit 77f5850b3c6b5f215c29ae704ecf5480062427e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fd073a7f3f78f06b40c418184ca5fa02e410199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fd073a7f3f78f06b40c418184ca5fa02e410199"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
