)]}'
{"id":"openstack%2Foctavia~854227","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I494ea357afc072360765d7a30a9488e08bfda52c","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-24 12:09:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I494ea357afc072360765d7a30a9488e08bfda52c","subject":"Fix bug when rolling back prov and op status for some API calls","status":"MERGED","created":"2022-08-23 15:13:50.000000000","updated":"2022-11-24 12:11:09.000000000","submitted":"2022-11-24 12:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854227","meta_rev_id":"abb9117dd33172f92996f4349d8c9a6a9c2711e1","_number":854227,"virtual_id_number":854227,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-24 12:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-23 21:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-24 10:34:49.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-24 10:34:49.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 17:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 21:11:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-24 10:34:49.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"8fcf5ab1dab92bd7447f19ab36cf9ed4ba5e40ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-23 15:13:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c39887a1522347bf1c2002ad4a62ceb8c8b09c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 17:16:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b58e4977dcd4288bdb91eeb057cd191\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad176cd0f5f04c479f90bfd1e69ab230 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f09a081e1d47259c3e0058aee819f4 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47d546bdf61e4fbe822b2c18cdc8bbd8 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39a18eb8f31f43e0b6fb93681a9827a3 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e24a05bed6245adb20226096799dc43 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ac1248208114f9994a84ec1ac8aba63 : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6eec12b824404879b26d2ec8c3087bf0 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6bb45e422c134694a2df3b49b4748dbc : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cde66077c2546a387b9e1f03d7e9cdd : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7edcbf8b40f146408ba59473da88ebbd : SUCCESS in 1h 24m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/272ff88c6b3d4614be1314c39275f310 : SUCCESS in 1h 58m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d56d1b0aa7264d21bf9844437793b18e : SUCCESS in 39m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/54291cf5dea940b688d9da5bec9d6a52 : SUCCESS in 41m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fdcf7445a2f4ed786b9c34f24e76e4a : SUCCESS in 46m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85b5e682016a4f6fa3cef2f8741e75fa : SUCCESS in 1h 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbc6c1153c38478f960b016979f8ee90 : RETRY_LIMIT in 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f5cc5d1d5f64fe5b301d233fab57812 : SUCCESS in 1h 52m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6b4c47ed3186062bfab8d3e9f32f85fe8356985","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-23 21:11:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e32a661b3fda5dafcba8125f93864716e5405d7f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-24 10:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6965d3ac3b5b9175dcf7f060afd6ef41048bfdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 10:35:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17360ee59eff4152d843d0ca10f548f3557b0bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 12:09:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6953d30b964f02a3085a2d14a85377\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737f7eff16d94720b73d1c45af6126e3 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc41aa81b825402b979b13e08915beb0 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/297429fbd7b14cdb960ca8011356245a : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aced77a4616a46379cc35f357b78cb08 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d337005715254cd3a24e659e51d0369c : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8620f78db28c42ddaa2ffa369db21d5d : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74e2888dfd4b453580a86f0b9aedd475 : SUCCESS in 49m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecacbf333d4148cbaf856c69697cc7e3 : SUCCESS in 1h 33m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e28855a4abf6405381c1a66c51b5acf6 : SUCCESS in 35m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b7e645db39a4866a20831509a20c9f8 : SUCCESS in 32m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15dc7669d6a644c6ac7b60a32996d496 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5c73423bd0cf711b840b060bfd3613bc44066d5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 12:09:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"abb9117dd33172f92996f4349d8c9a6a9c2711e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 12:11:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a908747d7cd40659b70a8b89355bf6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aa4f3904924300ba28b4a6db2dd7fb : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf33943440584807bc3f9d84c43c372f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64abc34c7325c695e84d87bab3d187e83e2cbd5d","revisions":{"64abc34c7325c695e84d87bab3d187e83e2cbd5d":{"kind":"REWORK","_number":1,"created":"2022-08-23 15:13:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/854227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/854227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/854227/1"}}},"commit":{"parents":[{"commit":"a84c88739a7ada3d5e806989e3d0e82413228932","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a84c88739a7ada3d5e806989e3d0e82413228932"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-08 16:14:45.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-23 15:13:44.000000000","tz":120},"subject":"Fix bug when rolling back prov and op status for some API calls","message":"Fix bug when rolling back prov and op status for some API calls\n\nPOST l7rule API call and PUT pool API call had a bug that might have\nkept a load balancer in provisioning_status when the call was rejected\nby the provider.\n\nThe code didn\u0027t use the right locked DB session to set the statuses and\nthen to rollback them when an exception was caught.\n\nStory 2010210\nTask 45947\n\nConflicts:\n\toctavia/tests/functional/api/v2/test_pool.py\n\nChange-Id: I494ea357afc072360765d7a30a9488e08bfda52c\n(cherry picked from commit fdcea400c6f591485fcb116cb47dc07c5229898d)\n(cherry picked from commit 3e6f180b77f1c58265ae9351151e32bb1d215fe7)\n(cherry picked from commit 81cb9d5efccdeabe67b7a79d8de43fdbb66fdd4f)\n(cherry picked from commit 173f024b59a0d25cb66e25bd4f9ca3e62d8ad9b0)\n(cherry picked from commit ab39cc3881a9a05147e620df61d3865b43418ce5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64abc34c7325c695e84d87bab3d187e83e2cbd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64abc34c7325c695e84d87bab3d187e83e2cbd5d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
