)]}'
{"id":"openstack%2Foctavia~854609","triplet_id":"openstack%2Foctavia~master~Icc7dc9bb22187f908852c20415227574ded347aa","project":"openstack/octavia","branch":"master","topic":"fix-server-state-file-handling","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-19 21:38:30.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-09-19 21:38:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc7dc9bb22187f908852c20415227574ded347aa","subject":"Handle feature compatibility of HAProxy server-state-file option","status":"MERGED","created":"2022-08-25 10:47:45.000000000","updated":"2022-09-19 21:39:36.000000000","submitted":"2022-09-19 21:38:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854609-fix-server-state-file-handling","meta_rev_id":"fd735c382e04ce5f02b922720d60740296f8842d","_number":854609,"virtual_id_number":854609,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 21:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-09-13 13:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-06 12:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-09-19 19:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-09-19 19:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-06 12:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 13:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 16:20:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-09-13 13:15:50.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-19 19:51:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"15204113ff6fba0b8caf7e05312779ec5ce9befb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-25 10:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bacf561460bf67bdba012f82eaa9beb1860fa080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 13:15:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0df0da3fb1f349b9b5079d590e69a3de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/106b8b429f924621b819aae57177695b : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba27d3905414fa0b8e8a703aa0f78e1 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37d68bf4238b40049c5c67657ede097a : FAILURE in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/644ac17fa4c742d6bc34b8770ff353e6 : FAILURE in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa3a83a65ba5442487437e7e3ee413d0 : FAILURE in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc56b16423c747ae88c4d9a722484d22 : SUCCESS in 11m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f10441e1649447b7b211d979ea9dee85 : FAILURE in 6m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ef4ea872123741048e6ef0eb251e3ed6 : SUCCESS in 8m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/688fdaac7a0f4e6ca4bae245ddbab1c4 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23c57837d6184f2f8e5ee2a1f15d39f6 : SUCCESS in 8m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d6b478c988b40d5b5d432d6287ad6de : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c25b4332a5b43cf872138fd89953a07 : SUCCESS in 7m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/903f1a7641cd4f4582fc8d3f371bc3d6 : SUCCESS in 2h 08m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fab7dd5bbc5a4ba6a177c81a460d26ee : FAILURE in 1h 40m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5a402e2e09946d680f32ebe38cfda38 : FAILURE in 1h 27m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54cfce525a62458bb965e718a84d0cce : FAILURE in 1h 24m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1caec6b91eaf4bb78c94af1a46e9dde7 : FAILURE in 2h 15m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e22172503064ce89cc73784ffb3d856 : FAILURE in 52m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df3a7eabed414dc98fb0b76756da86e4 : FAILURE in 1h 40m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/790a5cb3badc40378459e9f8a6dde5e6 : FAILURE in 2h 01m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1ce98181cc0641c79355afc59357465e : FAILURE in 1h 56m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef5c00f33626484aa3acdd25d2a66d28 : FAILURE in 1h 56m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/50ac02b16443409dbe2f9157e01f6fa8 : FAILURE in 1h 37m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/80014746292f42be93801336fceedab2 : FAILURE in 2h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a63966103d4c8534fb2f18795a15205a6338e948","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-25 15:24:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"512dd536f50dc4b67c282e16b5a7ca253c1bb228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 17:39:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/739cd2f1171547e7a3d45f29c97a02bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f72112e72b964c9882c63ae4e7c38cb3 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea834dd1b73d4fa3a4a80f4dfcdf2574 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51dbf800857d49a2b2b5dd63bb6a697b : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d11f3443394645cbae29369e6ed47c47 : FAILURE in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ceda50e0146418db217b570550fe3d6 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3607971bf96147e487612e9133739178 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19280876c60a4bd6a05db46c50e48f56 : SUCCESS in 4m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8be1feef76a44b7783eac4ba4d5b9db2 : FAILURE in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bb54235769064d2da689977f725979fe : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b5f77143ed7a4bdaa18397afdafbc7aa : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca93fe8c08f242d6ab7b106e3679db3e : SUCCESS in 8m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/126b5f8074924c388f065d1523b33875 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1383b06aeaa74dd395fede47d48c9474 : SUCCESS in 8m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/060b8c3017ad499484500621732f558e : SUCCESS in 2h 05m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da3959c180094907a60737bb5b27a63f : SUCCESS in 2h 05m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b04072f5cfa146409c7087d4552a4202 : SUCCESS in 45m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad791e045ee2485b81d77da1b119fc10 : SUCCESS in 1h 09m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/183c96e9c137471e8c814883a4a86b1a : SUCCESS in 1h 13m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c78f746e7ec147fe972cb899efc11989 : SUCCESS in 40m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85f304915bed419185089369f6542a53 : SUCCESS in 2h 11m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b95a089b53fb45bba399c57fd396ad7b : SUCCESS in 2h 06m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a5736a6bee4341beb25eb81eed58f723 : SUCCESS in 2h 08m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/817f1a85af7d4410a4d0eb8f2358c2be : SUCCESS in 1h 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f7b34c71c2624c6992a24642ca55e16a : SUCCESS in 1h 58m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a4d75309df04abc8d3b97e44713240c : SUCCESS in 2h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea5c0eef2d86eabee3c99951f87247a9597210e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-29 08:43:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f4708cfaf6d8029b8261bba9c62b8c7de74b660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 11:01:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4de11c53f7e4bc09da84e526d34d8a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9be0615fe184a05a3adda489ee48b5a : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ee6102135f423b823dcd15406f16a1 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e85986094e06496f936fde571700c1b8 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ce3f79401764103b30957f9df04384a : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c3855611fe445ddb129c7b1f07778e2 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563a13d7e8e043fcb5d9971b8b22ab22 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78c65207d8534035a9eb1410c7cc86e5 : SUCCESS in 5m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4dbbf12cbcfe41488b43338ab4eb8a87 : SUCCESS in 6m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1a7c5d2bef934416b77381ac0fbe0431 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d15cf4555546433bbe6746dafbf5b884 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89cf4c5ff98248d39977c37bdefb967e : SUCCESS in 7m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6f6b6b80abe402d930ea1e7259499b9 : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d3c8d0b974a74746bf8daafd4789aaa3 : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a992f32f0ac54f13a15e9ae46c03629a : SUCCESS in 1h 58m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c241cd1336f4bf7a09d6402ecd07185 : SUCCESS in 1h 50m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba82df1519504ef582e152be218139a9 : SUCCESS in 43m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0e8944a74f243ec99b5d5d7aa654d50 : SUCCESS in 1h 13m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f159addf3313480ead59e4c5eac474d4 : SUCCESS in 1h 20m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0af98af0657e464b931cd1e28d1930a3 : SUCCESS in 39m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/62b41cce1ccd4144902050bb7ff0a81a : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/18067c71fa5d4f01938d1e9b1b776d0a : SUCCESS in 2h 15m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f56f0d14f754ebf856b21b16b083cd3 : SUCCESS in 2h 12m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe3096259cc64350a66d345b21453709 : SUCCESS in 1h 43m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ff6c55f87654784bf2746cfac92b3ea : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8f61da005074cf5b5949dfeef2e1dcd : SUCCESS in 2h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d437ccc872c5025875d5548e63fe8e6e2e7b0cad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-29 11:31:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"874ab94331cce23d1caaedc55956dac18f033724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-29 12:45:00.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f14679e75740835e3ee42f03ee8d22d3bca1e6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 14:59:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb39bf8d34884b5398540c775e664ad3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed7ce6ada5ac47ac9186bbec52f8f164 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6ea0333ccc43b3a5f20ee786464e85 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b2d6c56f7774a8780d0376952d39533 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/115e655053d1478aa3ef6b1539f75a03 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f42e5c3e9a7d43359ffa4ab65ec83043 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae210fe5104147b3b2b040a7f3fe9d57 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45cb6682be124680880d7e3c1e64bfa8 : SUCCESS in 6m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8002aa31a5f2485b9c5f1160ec787053 : SUCCESS in 6m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62ec945309554b1eae46476c30132ac1 : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfa8161119714b83bbb4ea02cb1e04b9 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cfcefac81d4469daf45d19768fef39d : SUCCESS in 7m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65fd878588d04c1fb3741910b9ed8f2c : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fd9c3c499944173b91f49e6cdab04e6 : SUCCESS in 8m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82047aa379b3494c88b669900c913012 : SUCCESS in 2h 11m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22db2aec2d09491ca315c16509188159 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f65db1416734931a68f53b465ac3010 : SUCCESS in 47m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/94a9cf18445a42cc82fab2093edb068b : SUCCESS in 1h 03m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7959095eec7c4bd1957c78ada4942a90 : SUCCESS in 1h 35m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d121f852290d4455a61665ed7171bd90 : SUCCESS in 41m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/adfb737ce95b4cb0b411334df15817a5 : SUCCESS in 1h 45m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30cc773508554fb48398cdd56754cbe3 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6f15da29309147d2b601c9a2934e32cb : SUCCESS in 2h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3092d14281024430b4f025502ce9a794 : SUCCESS in 1h 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4d2a1128f96b497f83055c3e034a2d93 : SUCCESS in 1h 40m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cd1cba216bd49a9a2714973a1c9cf45 : SUCCESS in 2h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"347ff333b68c62a218554debbc2eee35e7817eb6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 16:24:28.000000000","message":"Patch Set 5: Backport-Candidate+2","accounts_in_message":[],"_revision_number":5},{"id":"1ff7d431476b2bfa91d0a2604651953e54fd4474","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-01 12:25:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"761b54feb902ef7edc26b5d76c7352db0fbb7329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-06 12:38:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c42c71398eb74b57301135a66b1f4b525cd49089","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-06 12:39:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0f03eb2e67288747b9ed487576a558ca81792a33","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-06 12:40:01.000000000","message":"Patch Set 6: Backport-Candidate+2","accounts_in_message":[],"_revision_number":6},{"id":"1439d078df36ce248c1be495ce140610e8dab2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:23:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3437e4b87e6041ad934104621572501e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bade0418f3f439fb7a0b0adefef7a63 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7815d178514c089b729bc15eb56393 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6262b97acc7421e82256af6e8f17d8a : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88e6f1e894e24cc2b30afacb87bf1528 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02840d71d445417e87fcc0ce2f6b0e83 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c14e47cc75c45ab83bcf2817162d76d : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c30082d1a763412b86036b9e28c2f9ea : SUCCESS in 5m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3c9265ca6cc54037a24352a4aba9b96d : SUCCESS in 4m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bd969dad014a4795ad174d3df3d0d9b3 : SUCCESS in 8m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3247a1c5bef4f95af66ab4539d608e0 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d75bb0e93a97487eba9233b8fd7c7c1b : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0aa04df066704d02870659af91abf57a : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3ce511a67ce41f1a9a568d7e3295d96 : SUCCESS in 9m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7056f6d4d2e2483a9a434f7a9b502721 : SUCCESS in 2h 38m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f94f7034a5b481b96dad69390f6b897 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6096bf6525a4688b6b8d90c558749fe : SUCCESS in 47m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d4a1f2885ed4b9c85205374b1a45d99 : SUCCESS in 1h 07m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7f1b46592125478792a03ee2136ea929 : SUCCESS in 1h 15m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e190624040854cce8a6f8ae7a445b446 : SUCCESS in 45m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0f682384a784999ba798956a7027d3d : SUCCESS in 1h 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a109c7cdecd948a4acae4e0698e016ce : SUCCESS in 2h 04m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c3b4a8fcec4f20a41844edadbc8679 : SUCCESS in 1h 52m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/425c13b5ba7d4f02b60b46cd02d5077f : SUCCESS in 1h 47m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3c2a8b1b800e4407bf38be1843630963 : SUCCESS in 1h 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8ef18f647d44941954371615ddd2f0d : SUCCESS in 2h 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46b81cfc318d4483ca21d351c2c4295416d4d599","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-13 13:15:50.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3a109c6cb5e6ebd269bad53bac0b5a02d0c1300c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-19 19:51:48.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"26a198762e5a0aa8daebd28ad27944a2c556d1a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 19:52:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4a8804695a460916009cba59a1d404229614a526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 21:38:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9016791e3614bc78aa4632a58e3784f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6598e3b5dbe64698b9651785b7edb42d : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/293dcea76dde4df4b85e2ed3d075eda7 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3267fb2e2c0f40b78193a06bbfba884a : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864f6af5fa804682a8df9eb6fe940d2e : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c37b4d85cec7467095268c7f47c9170a : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b259b2ce0bac4151ada9959f65b519bd : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8af4c476b1e4be78a9e3dce8bbd9cb4 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46594b5785ce45c2962c1faa8e09aab1 : SUCCESS in 1h 30m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a39da333d93e436698e36b3b59b3e609 : SUCCESS in 1h 45m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48b11ca21f254f02a0bc18d673d41005 : SUCCESS in 45m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0919c966843c4f3692e2a30a5f2378f7 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":6},{"id":"0ed0bcb34fd9fae50fb9998b5f046663918c1ee2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 21:38:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fd735c382e04ce5f02b922720d60740296f8842d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 21:39:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613e714844544fc68632d66fc0002715\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e71c11d21c84094af7299684689c5ee : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9b1d600e25c408e972cc58d493e6fe5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"69ecdb190880d0c49e9f0ef92fd53de14cf66720","revisions":{"b3fea4d442bad9ea7d85eaac6bf34a4220d763e4":{"kind":"REWORK","_number":1,"created":"2022-08-25 10:47:45.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/854609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/1"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3fea4d442bad9ea7d85eaac6bf34a4220d763e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3fea4d442bad9ea7d85eaac6bf34a4220d763e4"}]},"branch":"refs/heads/master"},"6a9c5756761c94d7482eb4643bbf425b3494040f":{"kind":"REWORK","_number":2,"created":"2022-08-25 15:24:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/854609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/2"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 15:24:19.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a9c5756761c94d7482eb4643bbf425b3494040f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a9c5756761c94d7482eb4643bbf425b3494040f"}]},"branch":"refs/heads/master"},"0c32b2c552eed7e3f06dbdbb3b091b31439948c6":{"kind":"REWORK","_number":3,"created":"2022-08-29 08:43:14.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/854609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/3"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-29 08:43:04.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c32b2c552eed7e3f06dbdbb3b091b31439948c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c32b2c552eed7e3f06dbdbb3b091b31439948c6"}]},"branch":"refs/heads/master"},"feabe945b73b55a3750114b98d0e493a825af5a1":{"kind":"REWORK","_number":4,"created":"2022-08-29 11:31:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/854609/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/4"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-29 11:31:22.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feabe945b73b55a3750114b98d0e493a825af5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feabe945b73b55a3750114b98d0e493a825af5a1"}]},"branch":"refs/heads/master"},"5789af988ba3af4280e3dda1f16386146426caa0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-08-29 12:45:00.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/854609/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/5"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-29 12:37:06.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5789af988ba3af4280e3dda1f16386146426caa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5789af988ba3af4280e3dda1f16386146426caa0"}]},"branch":"refs/heads/master"},"69ecdb190880d0c49e9f0ef92fd53de14cf66720":{"kind":"REWORK","_number":6,"created":"2022-09-06 12:38:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/09/854609/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/854609/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/854609/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/854609/6"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-06 12:37:19.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69ecdb190880d0c49e9f0ef92fd53de14cf66720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69ecdb190880d0c49e9f0ef92fd53de14cf66720"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
