)]}'
{"id":"openstack%2Foctavia~854679","triplet_id":"openstack%2Foctavia~master~I7951901f5ea8f97845f65ba486a12b7d6f6aa3e6","project":"openstack/octavia","branch":"master","attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-03 18:35:23.000000000","reason":"\u003cGERRIT_ACCOUNT_37685\u003e replied on the change","reason_account":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-08-25 22:54:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28008\u003e replied on the change","reason_account":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7951901f5ea8f97845f65ba486a12b7d6f6aa3e6","subject":"Stop logging deadlock messages","status":"NEW","created":"2022-08-25 22:37:02.000000000","updated":"2026-02-09 10:05:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e7d2c0295dd88d2b276baadfd14bfd3f1621850a","_number":854679,"virtual_id_number":854679,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-26 00:58:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2026-02-09 10:05:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-10-03 18:35:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":1,"date":"2022-08-25 22:54:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-10-03 18:35:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 22:54:47.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2022-08-25 23:54:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-08-26 00:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 18:35:23.000000000","updated_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"reviewer":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"state":"REVIEWER"},{"updated":"2026-02-09 10:05:48.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"}],"messages":[{"id":"6bbc74d15b0b263ec625cdd4df0529d3c86efe32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-08-25 22:37:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db4698c192bbd6302deac5ec16acb65ef94053a4","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2022-08-25 22:54:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9e63dc49a5432971ec7eb6e626b05b32679861fb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-25 23:54:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175de298c5b3cf61bc1b9c08757642587cccbf16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 00:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7a7d79f5c045068aec42ae31aa28e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60964a765ba14460813fa16f4b302b74 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a3d10a3db843b78b3a29326da3b8f0 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b084f0fd3ce34ae4a05621b2b5565ddb : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/390d3e086e0844ba8595101abd06d0b1 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0257bafede2b450498198cce2d7490fe : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c214a1d1e4e34f8c93b189e701d7acfe : SUCCESS in 12m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9ecc2a80cf714aa3a9fc823c9337f288 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/00bd968ca10c455199738b59d48ef65c : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a9b776bf20cc49a0af388cb7d317d11e : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1def626223ad450c87d48859cfb5b189 : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/006d908752a849329c4ab46b202da4c7 : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41ad9c581f924e82b477695791234aba : SUCCESS in 6m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d25133c737246d49d4b62f6500ccdec : SUCCESS in 2h 19m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98a4e4dff06c46e993b65ef0fa77df57 : SUCCESS in 1h 51m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ed29bb92d13416cbc1bcfc21982f9dd : SUCCESS in 43m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/07d4117842bf42f9886f18c61972141e : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a83d595d0fa2410c993c320c4e999243 : SUCCESS in 1h 10m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/732ab231f60447e5b5f0172ac425a6ec : SUCCESS in 38m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/112b3bd13f33496094d2baa118a13a11 : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/513d7323c33246f7822e38b4647d5758 : SUCCESS in 1h 55m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0060e9ffaeb846a79a16da5b5c2f9b60 : SUCCESS in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b446a985a254433f8c7939493268c9f2 : SUCCESS in 1h 38m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ecbc69c1786a4833bd619d5541debb3b : SUCCESS in 1h 50m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/72b16980bfa340c08fc85bec202b07bd : SUCCESS in 2h 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dd3125e29a6b6409a695d0ca95b41c41ba2f2ad","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-10-03 18:35:23.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7d2c0295dd88d2b276baadfd14bfd3f1621850a","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2026-02-09 10:05:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32446539e33d5f1a96dc75a8752feddb17c9d5af","revisions":{"32446539e33d5f1a96dc75a8752feddb17c9d5af":{"kind":"REWORK","_number":1,"created":"2022-08-25 22:37:02.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/79/854679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/854679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/854679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/854679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/854679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/854679/1"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 22:32:29.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 22:37:01.000000000","tz":-240},"subject":"Stop logging deadlock messages","message":"Stop logging deadlock messages\n\nAt the moment, if we fail to update an Amphora\u0027s health status because\nanother instance went ahead and updated it (with a different time value),\nthen we\u0027ll get a gigantic traceback in the logs.\n\nTechnically, if another health manager updated the last_updated_at,\nthen that should be totally fine and we should be good to ignore\nthat.  If we failed for other reasons, then the normal exception will\nfall through and catch it.\n\nStory: 2010253\nTask: 46136\nChange-Id: I7951901f5ea8f97845f65ba486a12b7d6f6aa3e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32446539e33d5f1a96dc75a8752feddb17c9d5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32446539e33d5f1a96dc75a8752feddb17c9d5af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
