)]}'
{"id":"openstack%2Foctavia~854680","triplet_id":"openstack%2Foctavia~master~I24ec7dc3b18111c126bcfdaa0780bfd7993020fb","project":"openstack/octavia","branch":"master","topic":"catch_validation","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-08-31 20:04:58.000000000","reason":"Change was submitted"},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2022-08-31 20:04:58.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-08-31 20:04:58.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-31 14:31:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-08-31 09:17:12.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I24ec7dc3b18111c126bcfdaa0780bfd7993020fb","subject":"Cache subnets validation for batch member update","status":"MERGED","created":"2022-08-25 23:44:20.000000000","updated":"2022-08-31 20:06:11.000000000","submitted":"2022-08-31 20:04:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854680-catch_validation","meta_rev_id":"da8a28c1b66aefcd3815f96036852c7beba0e128","_number":854680,"virtual_id_number":854680,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 20:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-31 14:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 14:55:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-31 14:55:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-31 14:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-25 23:47:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-08-26 03:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 17:25:38.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"CC"},{"updated":"2022-08-30 19:46:23.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-08-30 19:46:23.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"0f5087e75ef9c8ea97f3835cec8fb30c8bd28b68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-08-25 23:44:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2ce94b1a3c91c1af683baeb52e814fe20bb7e85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-25 23:47:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d494d4d49a16e69f30dcecdc34915684f5e9e716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:11:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49aaa8d0d5044ea886b1aad79ccc9308\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/043a4616ceaf49eda4114f061a549414 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/438578364fd54c98b7b370bf776daddd : FAILURE in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7c5f6a421c4d7aa2d072cea63017bb : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b40c5bd16714145ae6f23dd7f644d7a : SUCCESS in 7m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f34e97745744720bdbf8cc3eb710ade : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e0ff0d756d64fe796980820a4451bee : SUCCESS in 11m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9b176d605669490a87ae7b4e9e8e7b33 : SUCCESS in 4m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f969aef3ee8e40e2977d3dd507958839 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ab3f8f83fbd47228ba8143100c3f367 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1f25a478b4b417d87333830a4099493 : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/71857ce80e9d4e978e6e765e2c094665 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b281778cfac74a9c9d19939a787ddf11 : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ae0dde77e4e4bfb8869fc55ed04b69a : SUCCESS in 2h 03m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6ba67847afd43ad9914e629ad84c8da : SUCCESS in 2h 21m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f59de527adb943d3ba62b6f47cebc015 : SUCCESS in 57m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cedf928614514ed29383949ae40aa254 : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/36b4061ed8c8459f8e1736abec53270b : SUCCESS in 1h 32m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1faf7e65590b4403945920d7d36ddb6a : SUCCESS in 54m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0307b0df1c6442dacc95f41d480fc6d : SUCCESS in 2h 06m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f1b55a2ac8c4b8f86458906ad0fdbf3 : SUCCESS in 3h 03m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7131e9509e074de59c3b0b4f966e6889 : SUCCESS in 2h 15m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/65484dc1c7174af680f2bda11cc4c9db : SUCCESS in 2h 30m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a6e575c95640455ea6ed05c7eea9794f : SUCCESS in 1h 36m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fea8c0ca9b343f5806dbf140c4f78e0 : SUCCESS in 2h 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb11a6c5a73e1fe27979335cf40b184d3987cc76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 17:00:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15e3e2b1e3d5b53cc7fc3d32d8820e6ffaf61638","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 17:25:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"099a0a32c82e1ab1c07d75d078eabbead838e44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 17:51:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1675e02cb633e0b74ab6dacb11bd27904c8713f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 18:11:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec868637dab8e4eeaccb0f373d959561b6f8e0c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 18:14:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db9a64bdec2026ed6e01a7f1be92d72b8fa56216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 18:31:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9b3c9b6bedb2b1f704a0169db19e57de032afdd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 18:32:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4a16516e24a86778a7f3694a8c95a0a55d1cbfbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-30 18:47:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8be5cdef30cc3673364a2281c39ba5937cfd057a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 21:26:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacddff4df034d88bbb7bd08455ed4c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9219b79d7d7a4fcbadcd8ea5bb2a797c : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b48f8a7a364ca7970f43625a8b5616 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/422d012caaa547468965d3a89d65c536 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d512ba704e400d837442df28b1258b : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de6bd4e822d646058eccde683f20ce77 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c4e9741ded4ee4bd38cd8ee009c9f7 : SUCCESS in 12m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8e45023299754b08bf9bf9b45a183bcc : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bdce455bde0b4d57b4c7b979c6d69720 : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f2f44f6f92194cf7a7fbd4cf977307ee : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/afe7cb48b200430dadf008e3a20b1596 : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a97d560103e4901814fa08ab7330adf : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8cb23c6a6a84d7ca75485da145ee99e : SUCCESS in 9m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa1dbfabecea48eca0000d581dd98fd3 : SUCCESS in 2h 10m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/496a9e8c58594283b5c7b7ed5522f360 : SUCCESS in 1h 50m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6eea2692b482469bb0c9346a73fb55a6 : SUCCESS in 59m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9caf4040ce5644b2b01e239c7efaa24b : SUCCESS in 1h 28m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/33db9e2fe2e14bcba9560f531e1089df : SUCCESS in 1h 36m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d91b58b06071469f94c9433106ec786a : SUCCESS in 46m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27603cae438a42638ea1c340a1b1dd6a : SUCCESS in 1h 52m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5bbd06dff6754c1ebcfa180c3e115cfc : SUCCESS in 2h 25m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29df1acc88ae4ecdbd0645fa9d24733d : SUCCESS in 1h 45m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/53820fc7372b4a589f05b2ce7a152c20 : SUCCESS in 2h 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/14e85439d392481b82dfc979109bcb89 : SUCCESS in 1h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/047694cd8fd8454c9e19202f5eda24f1 : SUCCESS in 2h 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b83f559b5db07350be9ef3f5677ddfa508b98169","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 06:44:26.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"54935b3ecb56e49fe45cd8622e6c654478c34776","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-31 07:36:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0d95e94bad4a78aa7c2874a60a97489ec9ba9ed8","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-31 07:36:55.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2c1612fa85d0ec8892613bb423a11df4fcfde5f3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 09:17:12.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c08d034a02feef407949d707ff2487d365dddb5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 09:58:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7d3a4677e941d48f43b5b8e436352a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38bb0076d347413cba3ef75264ad6724 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf28b2d87c7407dbc541ac15d0d6fc5 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8e183a9e63471e877d60ee10d9d070 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0335897d3df7452c8656605cbab879b0 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4b0b1939cf44d99a5d1db841c989073 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd8e9bfc3934c7bb4c8b4d956ab17b7 : SUCCESS in 13m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6b20bf43f2b4474091f1125a83e78549 : SUCCESS in 5m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/507bba8bda244adb8a1411717ffde558 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/298e9a6265e649d9b885d55ebc89e199 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce5ff37c8da546e4bb91d57c53b1fefe : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5bb613b732d142abbeb613d14668d215 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea925803599a4c72aadb6f7f693ab2a4 : SUCCESS in 7m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/94a31534d90b47e4a430e4943a17994c : SUCCESS in 2h 15m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afe5d2182c2b47689f6fb10545d24f75 : SUCCESS in 1h 44m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/adacd6300de34e9fb1926f334daf5187 : SUCCESS in 41m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e9855f8620441b78f94bd867ed79eeb : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ffdbd3daf6aa4fe2b484018185c97edf : SUCCESS in 1h 35m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deea2863fcf94aa0b3f620d683a9787a : SUCCESS in 48m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22fb90d1edcd4189a86bc14023aee334 : SUCCESS in 1h 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1c97c107bc0497ea70236e7ebf91a78 : SUCCESS in 1h 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/42a627b1dd5947549118f7e907ebd3cb : SUCCESS in 2h 05m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a91d380e4aa491d8bc60639751a9e47 : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/83c3c4df26464d0fb5382a1780c6d1ca : SUCCESS in 2h 14m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e56e31c88d34a94877aed7c17743e7b : SUCCESS in 2h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b796a7fe5410e840b793cb52590ee71ec17d55f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 14:31:50.000000000","message":"Patch Set 9: Backport-Candidate+2 Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ee76104c3261914cccec04181998449cc3ffb610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-08-31 14:50:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3887b8252cb49ab8169ec2b70d41a141f8132c9c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 14:52:20.000000000","message":"Patch Set 10: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3b42239b70f54f7970891e5e9941ae0d795a0d68","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-31 14:55:00.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"8a554f828cff80a46abf1d995f19e4c9264de430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:54:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993203a71cba4cb58f1dc152eddbf79e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26db120e19864de5afe1a04962546112 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7edf3e69251472896111e030bbdd073 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df0e5eee1b5a404ca7bf4492ba8aef53 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56478eb3bcbb494d8405a07cace5b898 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c404684741a4416880f096f8adbeac3 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba38c20fe0e4fffbfb81f9f584d1ac2 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43d7e971763e44ef971ef4f4c083b708 : SUCCESS in 4m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0d52ac29ac1c419c865d5ea45144089a : SUCCESS in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/36422157f5f34f81915d5b07690063f2 : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/22cb5709f13143b7a5c9520a2db4d162 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26472ce3ef164c1ebfacb06025cdab7b : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4a155bf0aaf4cc192f104e04aeee43c : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c127ce568e094283be97bad2d46fe586 : SUCCESS in 7m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43792fa7f2cf4a03acf325957a9af03c : SUCCESS in 2h 14m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af854109dbac41a09ff40630a7fd86d6 : SUCCESS in 1h 36m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a1f46c495684c07b2300422f65fd373 : SUCCESS in 44m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7670f367403042ae83ec97b893b18dd3 : SUCCESS in 1h 00m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6d8b7ac56a334d7cb028924d08284b94 : SUCCESS in 1h 37m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c305d79f371a493c816abaf51c5d5750 : SUCCESS in 41m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3f680bd338ff499faaf05e0f69e2e689 : SUCCESS in 1h 30m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94156566ee8e4f6499d917dde3a594de : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/582707ed63604f4c955a9d29b5269005 : SUCCESS in 1h 54m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72a9c5d52c6f404d88172d04a1743791 : SUCCESS in 1h 43m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5399e56074504d82a8b5a72d7a960675 : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a228114131d4f6083c2ec74dbd147ae : SUCCESS in 2h 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e4882431cce684b108e967cc6e8c1f700e4c2ce8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:55:11.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2882727e34ce736456021cc3a13efdcbf4305bcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:04:58.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89b511beb2449efba36702ee5d44b3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/454d271523ab46afa5063b647da50d82 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b57fa0a6b0f4b959c1949489c441f88 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce2f8f5359b401780ca285272efb444 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a0b6e4e6dd45e583f63c27476209c4 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78e3a8666f54a55a8490f3327c907bb : SUCCESS in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1384395af5d54c25b5769338740bcaae : SUCCESS in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b696edaaedac4ba0bd442d957a1a14ce : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43eee82b41324f3885962ab33ad65f00 : SUCCESS in 2h 06m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d50645d9e34742008924b5ef887f78f6 : SUCCESS in 1h 36m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc8a300f6925426fb292e6c5457fa33a : SUCCESS in 45m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f7ff78fd7f54484ac756015c9dc1507 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":10},{"id":"bec8e599ad5730507f1332538394517c74352133","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:04:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"da8a28c1b66aefcd3815f96036852c7beba0e128","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:06:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d92f24e40c46b3b3e64887cfa75536\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0d75dec7c7427db2c01dd6b1f363df : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c8232466e44a9eae69bb6c343e6ac7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e0a9ba1560bf38d07d7e03bc0f36195703884db1","revisions":{"8646f49340f54f553ba81b5d3c2c54e16ba30ba2":{"kind":"REWORK","_number":1,"created":"2022-08-25 23:44:20.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/80/854680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/1"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"subject":"WIP: Cache subnets validation for batch member update","message":"WIP: Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8646f49340f54f553ba81b5d3c2c54e16ba30ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8646f49340f54f553ba81b5d3c2c54e16ba30ba2"}]},"branch":"refs/heads/master"},"4378e37d6224738f0231fa819cb3a37b864ab8a5":{"kind":"REWORK","_number":2,"created":"2022-08-30 17:00:16.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/2"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 17:00:13.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4378e37d6224738f0231fa819cb3a37b864ab8a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4378e37d6224738f0231fa819cb3a37b864ab8a5"}]},"branch":"refs/heads/master"},"12e579ba39033af64350d4699d58ecd8526cdf3e":{"kind":"REWORK","_number":3,"created":"2022-08-30 17:51:53.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/3"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 17:51:43.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12e579ba39033af64350d4699d58ecd8526cdf3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12e579ba39033af64350d4699d58ecd8526cdf3e"}]},"branch":"refs/heads/master"},"0f63c0062b71c230486d0a670d51f4930c01a2c1":{"kind":"REWORK","_number":4,"created":"2022-08-30 18:11:13.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/4"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 18:11:04.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f63c0062b71c230486d0a670d51f4930c01a2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f63c0062b71c230486d0a670d51f4930c01a2c1"}]},"branch":"refs/heads/master"},"d5d25948c3a26f2a8bb41488bfd01fce60e9d82a":{"kind":"REWORK","_number":5,"created":"2022-08-30 18:14:25.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/5"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 18:13:56.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5d25948c3a26f2a8bb41488bfd01fce60e9d82a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5d25948c3a26f2a8bb41488bfd01fce60e9d82a"}]},"branch":"refs/heads/master"},"d3d545fbb6dce3d0321394c77b9a1b5f8a5dfa55":{"kind":"REWORK","_number":6,"created":"2022-08-30 18:31:32.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/6"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 18:31:24.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3d545fbb6dce3d0321394c77b9a1b5f8a5dfa55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3d545fbb6dce3d0321394c77b9a1b5f8a5dfa55"}]},"branch":"refs/heads/master"},"8321fe1eafdb186b8c7d2cd70c88625746086d0b":{"kind":"REWORK","_number":7,"created":"2022-08-30 18:32:01.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/7"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 18:31:52.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8321fe1eafdb186b8c7d2cd70c88625746086d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8321fe1eafdb186b8c7d2cd70c88625746086d0b"}]},"branch":"refs/heads/master"},"c5edd0479eb3bbf980be9bc62397d643c5005980":{"kind":"REWORK","_number":8,"created":"2022-08-30 18:47:34.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/8"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-30 18:47:24.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5edd0479eb3bbf980be9bc62397d643c5005980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5edd0479eb3bbf980be9bc62397d643c5005980"}]},"branch":"refs/heads/master"},"c4ec0947b8ca34c51998eb0dac2ba4ff9debf4ef":{"kind":"REWORK","_number":9,"created":"2022-08-31 07:36:48.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/9"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2022-08-31 07:36:38.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4ec0947b8ca34c51998eb0dac2ba4ff9debf4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4ec0947b8ca34c51998eb0dac2ba4ff9debf4ef"}]},"branch":"refs/heads/master"},"e0a9ba1560bf38d07d7e03bc0f36195703884db1":{"kind":"REWORK","_number":10,"created":"2022-08-31 14:50:38.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/80/854680/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/854680/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/854680/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/854680/10"}}},"commit":{"parents":[{"commit":"34579fdc5e0aa61d0271914e602d0925bcd7f871","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34579fdc5e0aa61d0271914e602d0925bcd7f871"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-25 23:44:17.000000000","tz":-240},"committer":{"name":"ricolin","email":"ricolin@ricolky.com","date":"2022-08-31 14:50:35.000000000","tz":480},"subject":"Cache subnets validation for batch member update","message":"Cache subnets validation for batch member update\n\nAt the moment, we are actually going ahead and sending\nan API request for every single member when doing a batch\nupdate, this can be very expensive in large number of\nmembers and will cause a big hit against the Neutron API,\nit potentially also makes those API calls timeout.\n\nThis patch instead implements a simple cache (per batch\ncall) to simply store the subnets which are considered valid\nwhich means that it will make significantly less calls,\npotentially even one.\n\nIt also adjusts the behaviour so that it only runs this on newly\nadded members which should avoid running it for deleted and\nupdated members which won\u0027t have an effect in this change.\n\nChange-Id: I24ec7dc3b18111c126bcfdaa0780bfd7993020fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0a9ba1560bf38d07d7e03bc0f36195703884db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0a9ba1560bf38d07d7e03bc0f36195703884db1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
