)]}'
{"id":"openstack%2Foctavia~854927","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ia15a110c6c5a9b83a29bab56538fa425d994d477","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2022-09-26 16:03:17.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-26 11:48:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ia15a110c6c5a9b83a29bab56538fa425d994d477","subject":"Disable conntrack for TCP flows in the amphora","status":"MERGED","created":"2022-08-28 19:23:48.000000000","updated":"2022-09-26 16:04:26.000000000","submitted":"2022-09-26 16:03:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854927","meta_rev_id":"848b1e07166e34a18206a1da769543fb799c98c7","_number":854927,"virtual_id_number":854927,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 16:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-08-31 19:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-29 09:48:45.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2022-08-31 19:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-28 21:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 09:48:45.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-08-30 12:05:22.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-08-31 19:36:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"56329ae2a19fbb6712d3baeaac26cb8648a4dc8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-28 19:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81f6f4a80f3790883272d8c17d52d66643b9a3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-28 21:22:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08145f945761496b89b641b52204f2b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/040ed3e0cf0040e991625fc3364bb456 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1e30b9d4534d00b7e279fc434475c8 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8406a1f06c0349de9164518a5fe9b0ee : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c588e2f0011447c997f2c75685abc527 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b71c0143f5e4ebcbea51f818e788bff : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a2fcae76d594886be3ae028f31ef101 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7eddfe69b58145a581941bc83c9e252b : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a532531c0eed4a2998db25b6c382dfb3 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25371e4de4104224a443dde0d3541459 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b5f202a43034587a7087ee4d1d2e2c8 : SUCCESS in 1h 17m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/add16c71f7944697bb3d21ae294c01a0 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0b974d5132e4be98185a654cf5fd6b8 : SUCCESS in 50m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/47d1f6568e3a4e8187d043a5e4f046e9 : SUCCESS in 44m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be6030b59e44494990672eb3b6e8de22 : SUCCESS in 34m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79962407bffb42b4be6d9c9d4bcc3524 : SUCCESS in 1h 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/624ca4f8bb4b4fe7b67b50b7f43ee6b4 : RETRY_LIMIT in 16m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73b9090a96d549f7b134aed0fa85b024 : SUCCESS in 1h 56m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86b76af15e1c7766f8eccfb3b5017459b0a9e1c4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-29 09:48:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"014cc29f85f3b3305c42a158117accef5129ca23","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2022-08-30 12:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"594ab370d69b0989927bcd0add2eebbf7f82e342","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-30 12:17:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e33183a53da464e0e74a058b364fa88d12b97ae5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 19:36:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c169052c8a535446c3e482f49482d5a1a30cd7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 19:37:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f42416fe13457de9949a5001acbaaae69e7a54d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 22:14:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c502e9ce91a4ce9b49d380d6d2bd839\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0f87bf5ea7429780ecd11b56e22027 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6730747828d409e9a1080c6a52b1eee : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9129d440725644e48b8e0ea101880419 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f830696c8aef41bd84262600191b1a12 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f35320974c4e89ae5356d6c2562c63 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6be43fe8bf534077b139a91c48dfdf28 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07701d41b26046ad9296fae79d6e64b6 : SUCCESS in 2h 08m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fda75114e1d4805aaae83dbcc1c027a : FAILURE in 2h 15m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ea356eedce3432d9969445e350f7d47 : SUCCESS in 1h 03m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a4db27b56bb451ab9756efaf8af6390 : SUCCESS in 57m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/238c3571d5fe444a9fc0141357644552 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b0d744fa9c7adc01c4a3582cbc9f928015084186","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-01 06:04:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"792a4a44b8dce1d52943118279e1fc97e1a8b7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9241c5a56502471da85d8e756efb4d86\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e8793d0645c452787f06e28ad55c3df : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c9326595f14389ae0880a706c7d142 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0251c318db33482a8d95aec8d3d564e0 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce5a575702a44436add6ce55c9a66e81 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d13445bd6e9406e9d1b92ac4ed7aa95 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f738ff525f09422ca98d2c4e836c7aea : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ce82547c3f042cb8fec9c3a9b92b557 : FAILURE in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6ea90efd212f4f269914aed0e0fc6f5a : FAILURE in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c68cc48f7f3e43fdb3b4c6c0aad485a1 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/39eea4c19ab14160a53f6aaee907ba32 : SUCCESS in 1h 10m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b272dd4963840f28c8f063652b2081f : SUCCESS in 1h 50m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c413de2e8a74468d8ee5978c0600c727 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ddc1c5c395844f16af333fc95af4a1d7 : SUCCESS in 44m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff6cbd55eaac4a38ba5b7eaa0b1587ca : SUCCESS in 41m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a032b4a8b28641d395fd405ba417e436 : SUCCESS in 1h 35m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2639ce172b5641419508333f7ba3df1b : RETRY_LIMIT in 18m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24bccc1946064169994f95db6a246a6b : SUCCESS in 2h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a91d35e56fff3b17d529996b6a8e7efad7a2c92","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-26 08:30:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eacecdc236f5c29689cccfe758f1866c1630a8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 10:36:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3266dcf5a6d94ad28d3f97468d1421fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a82701df3544c0b0e87f543bbc108a : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c5a558fa5d4dfba82ab4221010fcc0 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74a496e0f454c6a929a7739abb3e4d6 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e059656fdc045d1b38b514b026f7a9a : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56feddaf59b742fc9b89a40a5d964f44 : SUCCESS in 14m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/becbd2b517de4530a74c4be4cc4b9d6e : SUCCESS in 5m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef118afb43df470fa6cc7547e4746edd : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5aade6b9ca5a48349563548283ddfb8c : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/142efdc21d724b349aa215c50f8f0df9 : POST_FAILURE in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/894fcdb626bf4b8b95cfb3cd94c788d5 : SUCCESS in 58m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fcbf7f4aa654acea6c9e4eb4f2c7f0b : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/603f48eb8f78437b9498fb3d21be148d : SUCCESS in 42m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f695ba1511b34c19919cc21f092437b0 : SUCCESS in 37m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46bf21e863ff46b0ae068853c1437240 : SUCCESS in 38m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d26c20506eeb4db9802298f20e0b4580 : SUCCESS in 1h 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea55451b596642f8ae52b9b3b3b75832 : RETRY_LIMIT in 19m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/28a4d8fbe983460c9f3bc73ddf260458 : SUCCESS in 1h 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"051f0cf2deafe0e7ecd3abd2d631551565645484","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-26 11:48:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ce573127a0519995bd3d2f299c3d8b6b03370e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 13:45:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5b47d20b7b4ccaa3f9683f9da2f65a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb3727fca4a94d77be864f80711836cc : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1536e517224bc0bde068c3aeb35e05 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1781797a36214f37b942bcc523cdf6de : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a9a0aaab5df48f3a50945cc5ee6b5a4 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2cdbf994c54fc8ad172f2cb4af7202 : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24571e66214f4bb28e3998f7e9892244 : SUCCESS in 5m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d5a609c1903a44ddad75850f6e153932 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1d392ba930341a8967cbbe006547c98 : SUCCESS in 8m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a3b5f294b75484eb0a5edaa686db465 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb185fe2aee5434dac25bc13dad93eb2 : SUCCESS in 53m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9df87957bd11413e8a0ceb4ede61ed2f : SUCCESS in 1h 31m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df03901290e742a68093ea404db1e48c : SUCCESS in 45m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f9cbb070ebeb4ad8890628d88c6a68cf : SUCCESS in 44m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d10564b7d24486abed4eff88befcdf5 : SUCCESS in 47m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7089a006f103409ba572252f4fb9ccb6 : SUCCESS in 1h 55m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca9ee09d90b746af874cbf1c158d8136 : RETRY_LIMIT in 8m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d7fb91c7e54044b2ae3eed291a670da8 : SUCCESS in 1h 53m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ffbf70c57ca9fbdbad6e507fc9231473e6b74df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 13:46:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04c6d986b3b4993838919628fb247c0dfa623e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:03:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/363cda72bb834ab9b882d4dab7eb3d44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f67bc669b743cf9afa4112cd9e35ed : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e21eb1f260a142fc826b96b476f2422d : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57c61f283e7e4ce38b759aeec734786a : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93431160b3b4a49aca99e426b523aec : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/104ae349968c4ffc86866fb63e89eb1e : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4b4d97c42bc460a9f121213a73c1cf7 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10fad8a4b5a040e59500dfb401a0b6c4 : SUCCESS in 47m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/944e2822d0a74864b077c63dd91dcbea : SUCCESS in 1h 49m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e0a3f1444924e1292614fe16ab86393 : SUCCESS in 46m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1b9d3f8ee51e45e9b5b7c852b03eb067 : SUCCESS in 50m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01073ae8f4cc4a2b94a923050e7bd12b : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c7563530215d965fdb93b6190b1093c0573086d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:03:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"848b1e07166e34a18206a1da769543fb799c98c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:04:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d9076e571fd40508f13c883f2c5c1d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ab42b005dd40ddbfccb8f5c276eaf0 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9184afe7f5d144e8a2d5f1a81768f620 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fe7ca572275c695be2a6c4339b31720e8952b8e","revisions":{"0fe7ca572275c695be2a6c4339b31720e8952b8e":{"kind":"REWORK","_number":1,"created":"2022-08-28 19:23:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/27/854927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/854927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/854927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/854927/1"}}},"commit":{"parents":[{"commit":"a84c88739a7ada3d5e806989e3d0e82413228932","subject":"Merge \"Fix HealthMonitorToErrorOnRevertTask revert method\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a84c88739a7ada3d5e806989e3d0e82413228932"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-15 07:00:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-08-28 19:15:01.000000000","tz":120},"subject":"Disable conntrack for TCP flows in the amphora","message":"Disable conntrack for TCP flows in the amphora\n\nconntrack is used in the amphora for LVS-based listeners (UDP and\nSCTP) and it\u0027s enabled by default.\nAdd nft/iptable rules to disable conntrack for TCP flows, it reduces\nmemory usage and prevents getting \"nf_conntrack: table full, dropping\npacket\" messages.\n\nStory: 2008979\nTask: 42623\n\nNote for stable/ussuri: this is not a straighforward backport but a\nreimplementation of the bugfix for old branches.\n\nChange-Id: Ia15a110c6c5a9b83a29bab56538fa425d994d477\n(cherry picked from commit b3f59eeb73289306dd5fd2e9d664903911dc614c)\n(cherry picked from commit 1905859662a75403705db2fca6619b929814fc6d)\n(cherry picked from commit c0ad3b347d514d6851c1c9234af0dc2d5c2a6952)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fe7ca572275c695be2a6c4339b31720e8952b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fe7ca572275c695be2a6c4339b31720e8952b8e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
