)]}'
{"id":"openstack%2Foctavia~856689","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I8c92761b05ba01bff00b16215477e216e4b97065","project":"openstack/octavia","branch":"stable/ussuri","topic":"cap-virtualenv-py37","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-14 15:36:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c92761b05ba01bff00b16215477e216e4b97065","subject":"[stable-only] Cap virtualenv for py37","status":"MERGED","created":"2022-09-09 09:59:59.000000000","updated":"2022-09-14 15:37:51.000000000","submitted":"2022-09-14 15:36:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856689-cap-virtualenv-py37","meta_rev_id":"624391fa555e71a62488e1f53fb85c68a3bbede4","_number":856689,"virtual_id_number":856689,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 15:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-14 13:31:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-09-09 15:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-09-14 14:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-14 13:52:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-14 13:52:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 12:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 15:30:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-09-14 13:31:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-09-14 13:52:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-09-14 14:18:31.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"00b52ab4449a15a07b4e43c63e22d0c0a786b437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-09 09:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdc3f0b21a135905f8a802ab254538e2ab06810c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-09 11:16:51.000000000","message":"Topic set to cap-virtualenv-py37","accounts_in_message":[],"_revision_number":1},{"id":"20000edd614bfd5ed3f44a56dba0c317427b30cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:25:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96fbf8655a846ada240d83221108309\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de5cc77bf9146b5ac0f7b4dbe9e587f : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24daf0f28ae24a91a4ac1adec86c04f7 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce965079b24e42048e429c6e40ea0fb8 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd8f605b15114009a4f7c2e64cd5375c : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ea76a466ae4578af9d36e41278b314 : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecfcb51729fc45c7b3c4821f7855e44d : SUCCESS in 5m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8137caf4b2174bc8869915bddb847bb1 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/af00ec9ab2f84e4bbce1739c40de4c57 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f161fe8a57b54da2b4819726a204dba5 : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7279a818fb24c5f952ddf2962a03004 : SUCCESS in 1h 15m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f817e1ba7c640ba89550db5ca16823c : FAILURE in 2h 03m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81243e8808ba45f8aca9e90d2a5a30de : SUCCESS in 57m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e56c33bd1f984bc8a6903389c9fc14ca : SUCCESS in 53m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53e109a1d83b484c80eb6d3191ca55e2 : SUCCESS in 55m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c16b520fcc464fe18b8a9f88f7593867 : SUCCESS in 2h 16m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da37f1fb39a04f64a0ccccc49a80664f : RETRY_LIMIT in 18m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/262c78274aa84caabec3c7480f6127da : SUCCESS in 1h 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6ff733342589cfde607e47de14b27ad91a55a0a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-09 12:40:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"897f03c1f4df53aff28e2b4b883bec485503412f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f1dea4b9cc4c08be4e296750072c31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19b72f63943d4cb981c373d46f6cabe6 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3427c2f3bb0b44fd894a7311267f0712 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5caf86ec724e41f18e87257525b27fd1 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed833fa7c60545d69a23b7e74968c653 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d1868f8ea64ff085205d72f9242d0a : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07a366383b4e4766807e9f81ec5a401e : SUCCESS in 4m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/62754860ce0348ed86995c4d62781b9d : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f31edcee6e394ff79904df8f5bded59b : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cc727e28eca452b90a50e441764dd9c : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09238373497c4f1c861d82b85bff59a4 : SUCCESS in 1h 13m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46a5efb9b7234867ad102c24fd4f2808 : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9757ccfff24045f5892c869216c32811 : SUCCESS in 43m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/74ecd4be564b4b639aee72f998fbbce8 : SUCCESS in 43m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/762341714ad84ac69552f6bb02e9e2cb : SUCCESS in 42m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/438173a3d07f4c848c1f46ec7ed05839 : SUCCESS in 1h 41m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b173da5111748238bb328bff9c18e91 : RETRY_LIMIT in 9m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc50fd4606914af9a0ef98f75cd13351 : SUCCESS in 1h 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e88c0149034b610951edbaae8998f61e5fd3b3e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-09 15:30:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b188fb442ee3c9e01a13724d27ff0ed151dd062","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-14 13:31:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c974c79539abc8cc985527c04712628171748184","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-14 13:52:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36722b9597b459c6eea433d7e3ee6a26741a8e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8852c65c587618a5783fcc30e1a4a5cb4d1867b","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-09-14 14:18:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e1344612df6d28716b71bbeae08ab873d2004fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:36:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8e616fbd57e41628efb30b43f42ed03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee75f23d70b47948872043e45ea9dea : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d81c254559e649498510f952157ed114 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb2afb9b89a84da4b4abfb9a0223bfdc : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/214fe1a0eace4ce88a19fe096a1354ac : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b6832979c44453d9b2660262f1250bf : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62b57885a57a4112b9297b3fe931e50f : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79506046b13a42d3977b83e6bebd161e : SUCCESS in 57m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9e637e85ee04f6ca0d3555576c3a67e : SUCCESS in 1h 39m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8f145357ae74dde93faf8452fbc5fac : SUCCESS in 45m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ba7af856c53841d3b6c1324826b141e0 : SUCCESS in 45m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa8647d3bf09460aac6fb5f383f87d27 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4cd128e5f806c68f5f6462abc86edb534e371463","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:36:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"624391fa555e71a62488e1f53fb85c68a3bbede4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:37:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5aa2a88ede40509f02072ee4a03105\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee0a35f11db4c16882fe361d87bf774 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24a8fb5ffa2e474db87fd882b4d46229 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f996e7b7aca51278431b4cd2335e30941c5b6a93","revisions":{"f996e7b7aca51278431b4cd2335e30941c5b6a93":{"kind":"REWORK","_number":1,"created":"2022-09-09 09:59:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/856689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/856689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/856689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/856689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/856689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/856689/1"}}},"commit":{"parents":[{"commit":"a87f01e42f77d33e4d0c22a4126dcb2becb4299d","subject":"Apply openstack-selinux policies in Centos amphorae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a87f01e42f77d33e4d0c22a4126dcb2becb4299d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-09 09:58:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-09 09:59:12.000000000","tz":120},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nCo-Authored-By: Elod Illes \u003celod.illes@est.tech\u003e\n\nChange-Id: I8c92761b05ba01bff00b16215477e216e4b97065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f996e7b7aca51278431b4cd2335e30941c5b6a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f996e7b7aca51278431b4cd2335e30941c5b6a93"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
