)]}'
{"id":"openstack%2Foctavia~856916","triplet_id":"openstack%2Foctavia~master~I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1","project":"openstack/octavia","branch":"master","topic":"story/2010284","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2022-10-17 08:40:36.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2023-02-06 09:19:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1","subject":"Fix wrong assertion methods","status":"MERGED","created":"2022-09-11 04:07:48.000000000","updated":"2023-02-06 09:20:52.000000000","submitted":"2023-02-06 09:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856916-story/2010284","meta_rev_id":"a525add085296341990874b9c595129d262cc930","_number":856916,"virtual_id_number":856916,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 09:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-02-05 22:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-02-05 22:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-06 07:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-05 22:34:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-02-06 07:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-05 22:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 04:21:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 12:51:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2022-10-17 08:40:36.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-11-14 13:11:12.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-11-28 11:56:06.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-02-06 07:27:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a04481917de3e0fad58ab1d0c5595a47c6b6df25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-11 04:07:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07c2749839c6416daef79db94ca255b73ac209d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 04:21:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd3f49257db4b60a6c56e0c6f02e85b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8b8c1e0023c47aabfab6a20fba95384 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4235adec186145b58401950ec3aea805 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb018ae9ff64b1da1a7ce02903fea6c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8f81574bf964324a468e83d0e5fa8fd : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9782f43c23c413b8e788e80d7f2d906 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b236e8278dc94b5c9ec044a6ab101d38 : SUCCESS in 11m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0a3c4bf8c6b74f07868903ae25e0cb73 : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5bae9515e08a4edab4717002e06d6d17 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"44c65b818d5a50c74ba47880c0d2d6adb3cec707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-11 10:17:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5c5ac98db8f498f2a3799c355c02af22546fbba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 10:32:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6b25cd9276d43c6b6a6fe6e9dad7ed3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03357c179475416993f0c2018200673c : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ca297af2ac4b58bbd04e5dcaf7a3ac : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9222e7caf3147e292cb561108ca2bdb : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c65a30ae5d143cf8e7acf290c182ed0 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbfb4aefea564c25a94ee88548b0b55a : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b658e808b8ce4b4a985fc15d017ca22e : SUCCESS in 12m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2dde37a7c7b44be5879d7ba1140f62a9 : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0f2d401ddf32439cad48f0516672724c : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":2},{"id":"48394b2d26cbee4763ffb5661d5fdcce83db9171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-23 10:49:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ff4852818bb4be36dfbab53f64726755b5fa4277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 11:08:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3de3715482421ba078fccf44428dd6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c232bdb9cc4bbf905ed67d4d3ac48a : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7094988564a24a1b9533c63ddf0c9b37 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3d1dba4f7924d07b84c67cf41f3d6d0 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed6ac311f89645558d134bc30b484ef4 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ab81d95de34fc5a81e4598ab24ef1f : SUCCESS in 12m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/db10850217c6461c95e030ad03a6f88b : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a23bc1cdb11b47f1af9210302c01dcdb : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e22d7a39e805cb953fd77293b957f2a81966d0fd","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-13 12:51:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bac68be86c89c93e459578941939a46bd310c8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-10-15 02:20:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"564081df5c995cb53fa62ea85de7ee385697a9b9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-10-15 02:21:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d47a9b3328ad94367f2d6fbf29d5b6892d43d369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 02:40:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364128b58a044bf59c2180873a16636c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41cfccb278e4ba6931303238ef2b818 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a1e9349b596457a8d20e37ab29dc4f3 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbab07b253d84297bba324bd3b09ee64 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/296c6328ed32429a8b4f4f1884b7292a : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ef87873d0b4011acb86482629e48dc : SUCCESS in 11m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3e6c3cf9294e416bb0be183672752311 : SUCCESS in 6m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/300f9c150d3c4c28811b7ca31a2e4be5 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7cc67024d6998d85c105ceb972d4357f9d976c33","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-17 08:40:36.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ae6e716433e08791498c4192947b0f819b13c0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-11-14 13:02:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"edfc3064354d60ceb50423960541dbd84e34a5cc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-11-14 13:11:12.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a97a145a293b44239354bdbc526b0e41dcee766b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 13:19:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6319d10c1964630914e8bd413bcb853\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4d59933d46a4485bf9f741a79ae31bf : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fdcda61f70445ab6e0776a4f2a4edb : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b11b11f83cf6432bb2027a119daff205 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12c7444ad999464d9c6ddc384240060f : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9450be0a34a9453a9af81c71930ebb96 : SUCCESS in 11m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/52a7d44ae8c340d1b229aa5e102e9bd3 : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/240247f80b6744db8379253f68e38a37 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b36e43f975a1fdb41cddc8761d16fe823fee9428","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-11-28 11:56:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f9c8ad9de5a0d7ed0a5c8c2c6d440c2a264f3a25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2023-02-05 22:34:05.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"eb31fe167c06148e132ca9f602398c1fabdcb805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 22:53:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99d5cebd7ab4961b756bb822c883428\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be417aee8bf442368e72437dad88c134 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0440c49f90ab4de5822ad588b9d014ca : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4794c8f964084d28b7122d9140478f49 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dbe0728d1144e73b06d000a6de5b5d6 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a631e25802407b892d9f4f10dd46f6 : SUCCESS in 12m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/18b0a8950ce1488e8d773325af2084c0 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6727fb9baee64e2186c06577454e0314 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b0db4e6749df4a867d2ab398862674fde8bdc6d2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-06 07:27:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1f5ca8bead162b3410d0a3427edb0a14febd10bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 07:28:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d1c431ddebc092ccdb5ad09b29a082afb9d07dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:19:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f250f83b01ac446a9d0eea5687468b59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad131a05e8e433db3e78bc350cc8754 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/068dc6ee46334c758454e78186f231c8 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eedd6f02960403782bf0856edf229c5 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e0fa728066456d9f77d91c28101bb4 : SUCCESS in 10m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/daafa52546cb44b0ab05dd74b0a3835b : SUCCESS in 1h 16m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba74110e832848d99c0385738bfcb293 : SUCCESS in 1h 50m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e052c52ea5d46b5a51ecb3f23ab7e0b : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c8c47cc152dc45d182e669e63ffe4e80 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":6},{"id":"02075efd0c9f072741a3d4ab457d1905608016ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:19:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a525add085296341990874b9c595129d262cc930","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:20:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f720754343474b9e5bd0b79fb0bc88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c73c35ad6ae4db09de462f479b0867d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c55839caf2a5d17c2da23f6bbdccf08624eac810","revisions":{"df40df2d242ac6ee38169a201bec60686ce07902":{"kind":"REWORK","_number":1,"created":"2022-09-11 04:07:48.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/1"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 04:01:44.000000000","tz":540},"subject":"Fix misuse of assert statements","message":"Fix misuse of assert statements\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df40df2d242ac6ee38169a201bec60686ce07902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df40df2d242ac6ee38169a201bec60686ce07902"}]},"branch":"refs/heads/master"},"d21fa64d94757ddb52b3616f2ff3e9c421d47452":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-11 10:17:37.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/2"}}},"commit":{"parents":[{"commit":"6ec76b12825a205b03cd566505024fec73ad8083","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ec76b12825a205b03cd566505024fec73ad8083"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 10:17:37.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d21fa64d94757ddb52b3616f2ff3e9c421d47452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d21fa64d94757ddb52b3616f2ff3e9c421d47452"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fb4bcf51d5824bab7e01a2d761b65ef4804e3d67":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-23 10:49:16.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/3"}}},"commit":{"parents":[{"commit":"409c4bca77287c2d9207001db2d7073d7e402fc5","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/409c4bca77287c2d9207001db2d7073d7e402fc5"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-23 10:49:16.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb4bcf51d5824bab7e01a2d761b65ef4804e3d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb4bcf51d5824bab7e01a2d761b65ef4804e3d67"}]},"branch":"refs/heads/master","description":"Rebase"},"c0012f872d5d9eae8d80380f4a90c86368ff16f2":{"kind":"REWORK","_number":4,"created":"2022-10-15 02:20:47.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/4"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-10-15 02:20:44.000000000","tz":540},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0012f872d5d9eae8d80380f4a90c86368ff16f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0012f872d5d9eae8d80380f4a90c86368ff16f2"}]},"branch":"refs/heads/master"},"aafdcefddcaec0ab73229ab75514abb3b4aee7b3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-14 13:02:57.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/5"}}},"commit":{"parents":[{"commit":"529f8e2f52010ede293bebdaf7cd20a61758fe58","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/529f8e2f52010ede293bebdaf7cd20a61758fe58"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-11-14 13:02:57.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aafdcefddcaec0ab73229ab75514abb3b4aee7b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aafdcefddcaec0ab73229ab75514abb3b4aee7b3"}]},"branch":"refs/heads/master","description":"Rebase"},"c55839caf2a5d17c2da23f6bbdccf08624eac810":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-05 22:34:05.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/16/856916/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/856916/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/856916/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/856916/6"}}},"commit":{"parents":[{"commit":"f78da62e4a1707d1c872ffddf95dc78e0cc0f6fb","subject":"Merge \"Remove support for split listener configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f78da62e4a1707d1c872ffddf95dc78e0cc0f6fb"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-11 03:33:29.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2023-02-05 22:34:05.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027has_calls\u0027 with \u0027assert_has_calls\u0027.\n\nChange-Id: I7fed37a9cd8cc0c3187f57a2551ab7a9d87d33c1\nStory: 2010284\nTask: 46255\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55839caf2a5d17c2da23f6bbdccf08624eac810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c55839caf2a5d17c2da23f6bbdccf08624eac810"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
