)]}'
{"id":"openstack%2Foctavia~856991","triplet_id":"openstack%2Foctavia~stable%2Fyoga~If0b211b3ef4fce3b7a58823fc8d1e48448e4d346","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-16 19:30:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0b211b3ef4fce3b7a58823fc8d1e48448e4d346","subject":"Reconfigure amphora network interfaces seamlessly","status":"MERGED","created":"2022-09-12 06:56:11.000000000","updated":"2022-11-16 19:31:31.000000000","submitted":"2022-11-16 19:30:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856991","meta_rev_id":"73555e37b1f45afb31e107e970a66add6125caa8","_number":856991,"virtual_id_number":856991,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 19:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-09-19 12:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-11-14 17:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-11-16 17:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2022-11-16 17:26:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 08:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 12:11:53.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-11-14 17:53:03.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-11-16 17:26:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9caf0668c7f4563b9d488a0b508ad401fbaf263a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-12 06:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec308ceb9c6a909230d83ca1d7711dc8f7ac0f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 08:51:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a52a0c8ce3049328b89924dae5c75d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14b61c8626c448ccaef4c9ba4f7a8629 : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13e44b79856b40b9b2ccdef012ee8912 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badcef7e24ae447aa99ebdc38c21f7ed : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636eb0167fe049dba0f712ed0bfbc11e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca1d354939446ada27ecd786ee12bb6 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbaf17964fd74b87897b23f038d5e697 : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0afc89a7dfa84821a484682bc7425a90 : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8d8b8d4e3ad746da894dfdbb457bd0cc : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64c1a295e4c445418e83af731bc4626c : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85ccb02a65014830a25439cf5581a368 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55decc0d38e346b797c149b2a24e247c : SUCCESS in 1h 50m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9a4fea4fba34408906f05c4e9842329 : SUCCESS in 1h 39m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f57117d7db3342768c9e8e369b7521f9 : SUCCESS in 44m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b56e3f3da61d4ff581f12b22340b07e7 : SUCCESS in 52m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a4b04ae08b044a739360f76e31a3acd3 : SUCCESS in 1h 32m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8c75717f0db4edbaa949c48b2442880 : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/74d20b53817447968a78a8ed8ea03afb : SUCCESS in 1h 49m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc2526639dbb449bb9e421617d014579 : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2a797becf0427daa7003c2b63f42e6 : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c88c003b9e746f2b880d68cb017c11e : SUCCESS in 1h 34m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1c0e7db0742c45e0a50a9c7fe8197ba2 : SUCCESS in 1h 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebcfc666632141efb2292d5e76fabb0a : FAILURE in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e50670e468f3ba8d7bd09839e4dac1b6fe8a8adf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-09-19 12:11:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d895bd754c2a7c3df4d3cb822d4597c4fbf280d1","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-11-14 17:53:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5d10199164215680092f39a3180d2937d16647bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:26:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9207de8a303c272a733330732158c1e79ada3b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:27:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f91ea9dd2ad60872a92dc98ac0f390724bfe1617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:30:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d707be5e4b2240ab865ce299513202f0\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad228b1365f9410f9c8066155027c457 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ba61c8b68343caab29f1d997f47adf : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/367bad63238444f4b6c5312f1055fd43 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ca19deaebc6402e951a0c5c1c6bc6b4 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d2c1149acc434a8ea734bcedd021a5 : SUCCESS in 10m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10b6112042924d9c897548f855d41a04 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8cce08b3f7464d19ac1a1207ea5479d7 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0aba85a604bd477ea6335978da7b60c2 : SUCCESS in 1h 11m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1455537c63d841d5b2a67012ef01c9fc : SUCCESS in 2h 01m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ae013e8665842d98e2d6718e66a3f89 : SUCCESS in 43m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b7cbbb2c7404cbeb8efffe5ed8a2a85 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d390d5386a93831299266d5759001a8f4ff3ebab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:30:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73555e37b1f45afb31e107e970a66add6125caa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 19:31:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/204cc559299449858128cd04a3434d3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca095e8c2c941e880583f690add8cd8 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d75613fab2dd15e05cf002142c43797e9d31ffdd","revisions":{"d75613fab2dd15e05cf002142c43797e9d31ffdd":{"kind":"REWORK","_number":1,"created":"2022-09-12 06:56:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/91/856991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/856991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/856991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/856991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/856991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/856991/1"}}},"commit":{"parents":[{"commit":"4818f323abe60a88360e178b36a49b534287aa68","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4818f323abe60a88360e178b36a49b534287aa68"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-04 15:31:04.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-12 06:38:38.000000000","tz":120},"subject":"Reconfigure amphora network interfaces seamlessly","message":"Reconfigure amphora network interfaces seamlessly\n\nUpdate amphora-interface app to allow reconfiguration of the network\ninterfaces without having to set them down first.\n\nIP addresses/routes maintained by keepalived in Active/Standby topology\nare now explicitly defined and marked as \u0027ignored\u0027 in the interface\nfiles, allowing amphora-interface to not consider them as unused\naddresses.\n\nAdd fixed_ips support on the VIP port, enabling the possibility to set\nup many ip addresses on the VIP port\n\nStory: 2005608\nTask: 30847\n\nChange-Id: If0b211b3ef4fce3b7a58823fc8d1e48448e4d346\n(cherry picked from commit 7118e8d56f76a99f19fd0696be3f09b6fb258cb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d75613fab2dd15e05cf002142c43797e9d31ffdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d75613fab2dd15e05cf002142c43797e9d31ffdd"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
