)]}'
{"id":"openstack%2Foctavia~856995","triplet_id":"openstack%2Foctavia~stable%2Fxena~I1384c6f52eec99e6573a8e83fe5a80a632804083","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-12-07 13:58:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1384c6f52eec99e6573a8e83fe5a80a632804083","subject":"Fix plugging member subnets on existing networks","status":"MERGED","created":"2022-09-12 07:18:18.000000000","updated":"2022-12-07 14:00:58.000000000","submitted":"2022-12-07 13:58:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856995","meta_rev_id":"d3a0c3c853d21642f76f51829a52c46889f995da","_number":856995,"virtual_id_number":856995,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 13:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-11-16 20:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-12-07 11:40:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-12-07 11:40:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 07:18:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"CC"},{"updated":"2022-09-12 10:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-16 20:46:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-12-07 11:40:33.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"a42f87874b015f56d8c6dda19293bda186754ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-12 07:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8763bd0f4cd92db3ae1616552fb3ff356db5621b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 10:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ada87c3b7d49fbb03e539152f467e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dcc2d2949e142068b1cdfe90b7b8f0e : SUCCESS in 11m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3101ed5c6f60465f8b8852f7ffbc1124 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61e1e1e16c7426db532bf1f6dc32bdc : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d53fb804804bf9af199a33de619918 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a14847e78045452b9b195e4bcaa1c091 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eee0a066f814aae8d36a76aa6239df6 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a004c248002547a6a9e80f62e0410bd9 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb338e2d15d24488bd1560ce85ea209b : SUCCESS in 5m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3e071eea72745c494ef00d487f5b8a0 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/595452f2fce34b32b07330f64d389e43 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d67dab408cf54986a3faf9884b5f7eca : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8d62dd4ffee4b61a5ae53a3503c4efd : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4ee508ebdf948f9bc0f9060721167bd : SUCCESS in 2h 16m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8c9af4f3cd64757a8066e3ec6de7cef : SUCCESS in 2h 33m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5125e27eca242d7b494f8785f12f194 : SUCCESS in 53m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8537937b3f6243729828430ee09e89bb : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/666c1fa630bd4b8d82c146a94ad73fa6 : RETRY_LIMIT in 14m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c02c321026e442c3b99b9a9b2795d102 : SUCCESS in 49m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5dbeb8fa64504c8d955c5f6b901f1062 : SUCCESS in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb5758aaaabb4814819a277f57be7ebc : SUCCESS in 2h 53m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4864e09be320467ebf37a6269cc04fb9 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fdd3b127de3248a3915b742dad23b8a1 : SUCCESS in 1h 44m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9934888ec86e4a2aa498e0c1a0d9068a : SUCCESS in 1h 29m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3af780b1ed634ea6abc39f933b4a2b91 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f8af07ebe748d6f5d8e78c7741c036f483d898d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-12 14:05:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"561b18dd1429f8091c2a02c2f1bdca0dcd82a03d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-13 06:39:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9c6486a8bb2414b9db927c0219cb501f428a13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 09:05:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da37c8ae914645419ed55f7d5c25017f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/514bda5c49274c738402861901cecdb7 : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c221e267b1234fc5b3c5437452e664cf : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5035759856d741f19367119baacc018e : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c22ca8b3b44df3a15d9cc32903161b : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae75a9397c7444e9f063e8fbbfcc8ef : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e57ff56e6bf340438cbfc5e4992199fb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43d1cc092054c8c97ac58c72f12a420 : SUCCESS in 15m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3570587ffb1d4715b3243bc5a366edd1 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32bf581740064209a00e4ed703eb8e67 : SUCCESS in 10m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ed3f94b6013b460581463c10b3284c99 : SUCCESS in 9m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f80e6bad07d427ea51ef35f49a3fe89 : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bef3788e80543ce8d727377f4d16044 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/75a26aa2b99b4213b5eb640575f2c451 : SUCCESS in 1h 34m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a9292620a374e27a832748d596774be : SUCCESS in 2h 12m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e10fbb2957b5469099bb39e2403080c2 : SUCCESS in 51m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/516f8ff704c34d5ba8c4ff9263373677 : SUCCESS in 53m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d8922cbcbb1245cf9b161fe4029d867e : RETRY_LIMIT in 23m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e69f60aaf9147f69572a73f66748f0c : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e6b2ce48e1f40a0816019b244b241ac : SUCCESS in 1h 50m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1dcbd0a85304487bff27096e355b2a1 : SUCCESS in 2h 22m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3631af6d3741a0b855a6f5be6fd55a : SUCCESS in 2h 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72194919ad624797a457c94378e6c6a9 : SUCCESS in 1h 50m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/67e7074ff52441dfb4f56cefca3584b2 : SUCCESS in 1h 44m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e456b96e7ed24e2490549ec6f204df6e : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49b7815679188389c7a945ef3107314a7f3b2818","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 20:46:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54d97265b1c9494590cdef1630999b1cf0cc5d18","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-12-07 11:40:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03f54c63d3f7b70ea54b42b1c4087fc28739da3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 11:41:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e8b6c7bd3e9e91e4698ef44d3560a67d844d7f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bdf04b5f82c4fac8bd3f7d6b5a5a78d\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f12f6076ec442b6b1cccbbc86cf8ebe : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0fdbae1cb24a3c908c51e5136398d6 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d03aa3b06d1426cabf1b3511425cdc8 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cfc71b216374e4b826acdeb87452a13 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161d793e6f554c27bb2189ada953d336 : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c3788d262404c1892ac66cf5d41d4e8 : SUCCESS in 4m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9c62e54d17bc47dca058f5ef26989c84 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b10ebff642f48fba4777edd73422952 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e78556b101546b086f66330e7b0dd9d : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc38b2a575d6433389a854b49e70d617 : SUCCESS in 1h 32m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14a8d87729494cdf8edfa6c84f875229 : SUCCESS in 34m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8daa66d60902428ab29a75ab7287106d : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3aeadfe592cd61ccf1be6cdd2a3d63229e53dcfa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:58:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3a0c3c853d21642f76f51829a52c46889f995da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:00:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092856817f024697a2912d1535a6d5c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de05bfa2650446287c38a56ef3d4653 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84fc895a5c5e49c3bb31a63a4f3f23fc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16fa7f8eb9df7e5293260e4acef0e44015c3cdfb","revisions":{"b9357d75467ff426ab3b53148612bc2b6163fe2e":{"kind":"REWORK","_number":1,"created":"2022-09-12 07:18:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/95/856995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/856995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/856995/1"}}},"commit":{"parents":[{"commit":"c513c9e77a13994a91c30abaeb7bb9b56351a295","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c513c9e77a13994a91c30abaeb7bb9b56351a295"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-12 06:56:45.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n(cherry picked from commit 4b0d4c1b591e41d42fba87ce4a50a899abee6fcf)\n(cherry picked from commit cade09e72af8361479f883fc08bed9dc105024dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9357d75467ff426ab3b53148612bc2b6163fe2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9357d75467ff426ab3b53148612bc2b6163fe2e"}]},"branch":"refs/heads/stable/xena"},"16fa7f8eb9df7e5293260e4acef0e44015c3cdfb":{"kind":"REWORK","_number":2,"created":"2022-09-13 06:39:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/95/856995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/856995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/856995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/856995/2"}}},"commit":{"parents":[{"commit":"c513c9e77a13994a91c30abaeb7bb9b56351a295","subject":"Reconfigure amphora network interfaces seamlessly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c513c9e77a13994a91c30abaeb7bb9b56351a295"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-14 14:13:02.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-13 06:38:40.000000000","tz":120},"subject":"Fix plugging member subnets on existing networks","message":"Fix plugging member subnets on existing networks\n\nCalculateAmphoraDelta uses now subnets instead of networks.\nHandleNetworkDelta plugs subnets on existing ports, or removes subnet\nfrom ports\nThe Amphora server supports multiple \u0027plug\u0027 calls to the same port,\nallowing to update the settings of an interface.\n\nStable branches: Also includes a follow up patch for amphorav1 [0]\n\n[0] Ibbfca939ff580fd13c094c86721f703a711ffc46\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nStory: 2004112\nTask: 27533\n\nChange-Id: I1384c6f52eec99e6573a8e83fe5a80a632804083\n(cherry picked from commit 4b0d4c1b591e41d42fba87ce4a50a899abee6fcf)\n(cherry picked from commit cade09e72af8361479f883fc08bed9dc105024dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16fa7f8eb9df7e5293260e4acef0e44015c3cdfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16fa7f8eb9df7e5293260e4acef0e44015c3cdfb"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
