)]}'
{"id":"openstack%2Foctavia~856996","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~If0b211b3ef4fce3b7a58823fc8d1e48448e4d346","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-11-16 21:37:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0b211b3ef4fce3b7a58823fc8d1e48448e4d346","subject":"Reconfigure amphora network interfaces seamlessly","status":"MERGED","created":"2022-09-12 07:23:19.000000000","updated":"2022-11-16 21:38:57.000000000","submitted":"2022-11-16 21:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856996","meta_rev_id":"eddf0cb99fb69722305db67d1f5a34723029d6d4","_number":856996,"virtual_id_number":856996,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 21:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-09-19 12:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-11-16 17:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-11-16 17:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 09:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 12:11:41.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-11-16 17:49:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"82d0921f20173037d4ae99ccae503b312505ff42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-12 07:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"411adac04355dd943ab627893d51ece271de637d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:39:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14ae27c7cd543f688bff0b2b12eca6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a11d8318a9455ea70972102802f4dd : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2fe96921fa84c928734ead4809bb72e : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0998fb0093804c278fbe02398fab8d7d : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b385b695ff34bbab95ec6a2b822be02 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef63ce22ad0f418bb8025e578e943d80 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6831c9d88a6949efa5206cbe10826382 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5059e24edf4fb7bab5494516640fcd : SUCCESS in 15m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f3021baadf54250893bd33022885031 : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce148a5828e34847b9685671155c54d4 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58ff5c319c3049f8b9a3a4dcc6f67d88 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/582b3ff0477244a1baf1cd024cec9ccd : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ceeb3fd25664da385583eca987f2c6a : SUCCESS in 1h 50m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4d5e515c8874af59a74966eb4e542ea : SUCCESS in 1h 16m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4deb0f72e9d49528f55bc36f9123d22 : SUCCESS in 38m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9f7d2a55fb3747679731d2b951eefe2b : SUCCESS in 34m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d488ece463b24358b94e51f037d89385 : SUCCESS in 36m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f8166d535b144f04bbf3eef4c0ed1822 : RETRY_LIMIT in 24m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3df6de0455d44bda838ce09b46bbc0f : SUCCESS in 1h 35m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/458367551d8d4710b00a926def8ddb3e : FAILURE in 49m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c547050dcaeb49ed89fab306038aed93 : SUCCESS in 1h 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f301c88e1fea46da9c649a91489dc91b : SUCCESS in 2h 03m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/51e3fb3b031241cd89b8d220c2d637d8 : SUCCESS in 2h 11m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dfbe1db18f7b49d1978f370be03ac4ec : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9332acbe1429eb3c0a5cf813dc48e4a886ccb1fd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-09-19 12:11:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22fb1317a0d163c8feda1497a41f1a49a735ca45","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:49:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7175e6c458dfe8c82a25578089e9bbbf0bcd6020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 17:50:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63d032f643b9b3f6ff9408a459bcceb1ea547aef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 21:37:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/269a7f775a504525aecba4e839a7c902\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b79821d5d82c40a79584d05d3074a3cb : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52a579c62374e00bb4bedd9bb03d4c2 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26b7d339ab344aaa92460ec77c0fcdbe : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb4e2dc8f1c46779d72734fcfe4c4e5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc1c9c63d9348cbb4b0f34021b1b45a : SUCCESS in 11m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9dbb919560ec405a87573347184caa01 : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80afa291cb674dea8f0620d1410f04d6 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1441cd902324298adc3d386719821f8 : SUCCESS in 57m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dc60637cf4a449098066d209f2685d5 : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d83ba38c26ea4d049355f0ffb4f5a13f : SUCCESS in 34m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f23420d6f37f4461a284c89d70cc9abf : SUCCESS in 31m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cab80cdcf9349edb7f61e7f13d66607 : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"448ad77abb64933fc2f54e004f1e94152f27944d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 21:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eddf0cb99fb69722305db67d1f5a34723029d6d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 21:38:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3af5b952b8614d328539acfc1a41889a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81961ac83e6043dab49027d552ad396f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab62dd9d44fa751f7f5c5b2611f7f2a9a5137bb3","revisions":{"ab62dd9d44fa751f7f5c5b2611f7f2a9a5137bb3":{"kind":"REWORK","_number":1,"created":"2022-09-12 07:23:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/96/856996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/856996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/856996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/856996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/856996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/856996/1"}}},"commit":{"parents":[{"commit":"d1877698badc296654631c256bcf109c31086eb4","subject":"Merge \"Apply openstack-selinux policies in Centos amphorae\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1877698badc296654631c256bcf109c31086eb4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-04 15:31:04.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-09-12 07:18:44.000000000","tz":120},"subject":"Reconfigure amphora network interfaces seamlessly","message":"Reconfigure amphora network interfaces seamlessly\n\nUpdate amphora-interface app to allow reconfiguration of the network\ninterfaces without having to set them down first.\n\nIP addresses/routes maintained by keepalived in Active/Standby topology\nare now explicitly defined and marked as \u0027ignored\u0027 in the interface\nfiles, allowing amphora-interface to not consider them as unused\naddresses.\n\nAdd fixed_ips support on the VIP port, enabling the possibility to set\nup many ip addresses on the VIP port\n\nStory: 2005608\nTask: 30847\n\nChange-Id: If0b211b3ef4fce3b7a58823fc8d1e48448e4d346\n(cherry picked from commit 7118e8d56f76a99f19fd0696be3f09b6fb258cb2)\n(cherry picked from commit d75613fab2dd15e05cf002142c43797e9d31ffdd)\n(cherry picked from commit c513c9e77a13994a91c30abaeb7bb9b56351a295)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab62dd9d44fa751f7f5c5b2611f7f2a9a5137bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab62dd9d44fa751f7f5c5b2611f7f2a9a5137bb3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
