)]}'
{"id":"openstack%2Foctavia~858494","triplet_id":"openstack%2Foctavia~stable%2Fzed~Icc7dc9bb22187f908852c20415227574ded347aa","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-17 13:02:31.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-11-17 13:02:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc7dc9bb22187f908852c20415227574ded347aa","subject":"Handle feature compatibility of HAProxy server-state-file option","status":"MERGED","created":"2022-09-20 08:26:19.000000000","updated":"2022-11-17 13:04:01.000000000","submitted":"2022-11-17 13:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858494","meta_rev_id":"056fb6116e7a83a6db9b773e03505a9703b451bf","_number":858494,"virtual_id_number":858494,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 13:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-11-17 10:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-16 17:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-11-17 10:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 08:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 10:51:20.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-09-22 12:25:16.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-11-16 17:38:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"500a6c9b69d31de684db19018d03f4931e6bdfe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-20 08:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7f5aed1c73079e09b3b103b01c9f1796a03e776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 08:44:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c04ccc30562b4080af292a39a39a5980\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e37fdb118546aea4ec2c600e3c4281 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d308f1959084f97bc445e9c5768a93c : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/998266148498407b8f52be651d7f9832 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a70523dd1aa640a28a314cbdcafa6ab6 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05d4452574df4c1e8bb04c17c5a70669 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1236365fd9407583a9f9150e633244 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3a1c174e09e409cb8c6b0cc018aa62f : SUCCESS in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4212bc62ee094145bd241c48270d192b : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3ecb6099e9ac4825b418e5ca75974ba8 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f9ffcd0d84f4b1787c420bb930a45b0 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5e4d0eb814f4c6886ab9c06a18af1c9 : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c61df6bf987d43d08574bc9887e50a89 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c63f963a0a0c4c3c97d5ebc7eaf10064 : SUCCESS in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74d90ade5ebd9e1cfc347e04fe8348a3f9117ddb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-20 10:51:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18bf6b2a8c11e90e4f28050afb6ed121f1251ee7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-22 12:25:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1fbb3d0a10cfd4002e6acfee4f4c82aa0626c1a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-27 08:40:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b63b89cc4ef8045dae7f50ab4443a846bc8dfb5","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-27 08:40:40.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_34429\u003e\n","accounts_in_message":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"_revision_number":1},{"id":"99b822969f139d016aa669a946f83b842ec4d6f6","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-27 08:40:44.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_34429\u003e\n","accounts_in_message":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"_revision_number":1},{"id":"990c088ade85c7583bd605b8ac4e96cc52650ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 08:40:53.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"74d54a16163a7bcef35ce85c1f161068536909e1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-27 11:46:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afe144e52caf70213616535d0f89b4214155e2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 14:16:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ea10036c00941da87b6c2705c46f0e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95149d37b2e34c79bcf601607e0111c2 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4adb772eac664270adbb94305acb1d1b : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bc80b47b3864818a4fd5a1a45623c2a : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c50ead15fc084d8198bc0fb5f4e58c08 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68c79e9ec04943c3b53a48e17dea5920 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c118b2436cce458885d4f985114f331a : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6ba2c435a24492a1201426a2ba811b : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bfe695d1c9ed4e129a1cd1650969af53 : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9b4bc5a3ebdf4d4a99a2790fc4fb927f : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0d87fc5cd20b4bcd96eda6cd714b6889 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6baa9d5b70bb473aa18c084f3086c164 : SUCCESS in 9m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6eae238c9aa442d48dac6e1e44955f48 : SUCCESS in 9m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df5b98ed972f4853bcfbd8710c1a1608 : SUCCESS in 7m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2868449b5f504398aba410891353b777 : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4aef95597bc4b0795ec2beb1b503532 : SUCCESS in 1h 43m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/faa903eda7bd430ca73b0c0ba70c47e6 : SUCCESS in 43m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcbf781dbe824011a671cb403a8960ea : SUCCESS in 1h 12m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7d4d75fbe45948c2a46b6deed977fc73 : SUCCESS in 1h 24m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8678b4c91ef4fc4afaea4822ad8c103 : SUCCESS in 39m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6212b961e5e84b81bf2fd056abe42a18 : SUCCESS in 1h 54m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8cc5cadf04c416c83e34f7f71fdce40 : FAILURE in 2h 10m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d8c5f785aba4c828b0ab3ba1966d77f : SUCCESS in 2h 00m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f72970001474704afbe55d5f011caf4 : SUCCESS in 2h 14m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e92cbc01f9743bda23252e5c6aa14c5 : SUCCESS in 2h 20m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7198d6d16e84fef9cf84fc7828e630a : SUCCESS in 2h 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73de95195448d0233602e46b971191de41d67e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:37:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"43218c6929d1101c4f58aa45337ceda726b20f68","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-16 17:38:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe2efe2bfb6223b2d3c8fa93abbc5ba448f21da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 20:21:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2ac0dc502a49b58384cf4200a24ed7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffe1deb39cd2430cb4a1bfa9a05cd961 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312cae052d9c4f65a65a33a07769cc38 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39ee66b012af4c73974404efe70456ef : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74036ab7424442ed9364e7b15da004ec : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b753b8666d24eaf924a8885a1a9665e : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52584f401490469badbe486e6097b93d : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d6b8f174c19441ab07716711f97f262 : SUCCESS in 4m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d63da01441ca4e07acbd48d5f513d16e : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/56472ddae27f42b4b23125dba447c75b : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0ea61131e9b4a98a4d06154cf022291 : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd44e2441d0f44e284f679f5126f30de : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6330a590fe4a4b78a26c4e75184f33d4 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f29005e5274311b6c01ed1842f46eb : SUCCESS in 7m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c7099c0666c497f9c8f9df6a8dc4e28 : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b98b04b336040c88017f4c4889f2c13 : SUCCESS in 1h 52m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/455e3dcc40ab47e29398e13fd1056e8e : SUCCESS in 38m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3ce49d5e3c64cd4b767e29c03f0ced2 : SUCCESS in 1h 07m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c4ab3186e30d426694d32b1ffd83da4d : SUCCESS in 1h 18m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13a4a35fb91b452eafeeb0e1351d10b4 : SUCCESS in 38m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c16c43d3c353440e804f2e915c346f06 : SUCCESS in 2h 08m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4180879b9d964b999f3edecceece8348 : SUCCESS in 2h 40m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1db85a8996c4cd5bd1568f109fbb844 : SUCCESS in 1h 49m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a63e9291371b41939f7ddeb99352bca5 : SUCCESS in 1h 50m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/70f0c1ff32554e898db782587cf74339 : SUCCESS in 1h 53m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f5f120a764b4020aca01b13c31b2416 : SUCCESS in 2h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"723a9b68897ff2cbc33098902f8c0a73513f9b03","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-11-17 10:02:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a2e46447fd2abe0b430166704ef9817b67ad4e92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:03:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61fbd0dfd285674e5936735d4711e534419848e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6f538613a9242d7ba4cf0b24bf068ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18e17d2770241ec9b457ad199f137fc : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e639cb7de074e9cb73a6794d511000a : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc26a29a9c7c48d2902bdddd1d36e246 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa661a33f5641c68bc9f9e374194bd2 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2a4cef99d324ccb916e455f50d1a8b4 : SUCCESS in 5m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a62db48afc0d4640930a621aa408cfed : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d90d67a9666147529b0e9ef2e5af6a69 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff15d5a778dc4b25a5d875ccc0bebfdc : SUCCESS in 1h 43m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35c2c241539d4cc382bb09c80f17b9ba : SUCCESS in 2h 24m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a97952b226ba459680d69230421ddc5f : SUCCESS in 41m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9fc824dfaa742809e3ce8f4b41a4a0c : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":2},{"id":"195d101483d8e9cde2624ac177185f6450ee705d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:02:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"056fb6116e7a83a6db9b773e03505a9703b451bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:04:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd89ab6c2cb44d54be2890999a5666cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d4fb8511cb484cac5f37ce6121821e : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/024667b20dbf4ba8a2c212b92c09ea64 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb2e1d2e770a400d921348ede035dcadc79c01f6","revisions":{"7f220d41adaab75471e551f89b0c2b97e830bab7":{"kind":"REWORK","_number":1,"created":"2022-09-20 08:26:19.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/94/858494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/858494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/858494/1"}}},"commit":{"parents":[{"commit":"d8ec4499a351a69cb9c7ed7e7605e538e968c97d","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8ec4499a351a69cb9c7ed7e7605e538e968c97d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-20 08:25:57.000000000","tz":120},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f220d41adaab75471e551f89b0c2b97e830bab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f220d41adaab75471e551f89b0c2b97e830bab7"}]},"branch":"refs/heads/stable/zed"},"fb2e1d2e770a400d921348ede035dcadc79c01f6":{"kind":"REWORK","_number":2,"created":"2022-11-16 17:37:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/858494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/858494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/858494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/858494/2"}}},"commit":{"parents":[{"commit":"d8ec4499a351a69cb9c7ed7e7605e538e968c97d","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8ec4499a351a69cb9c7ed7e7605e538e968c97d"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-08-25 10:47:38.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-16 17:37:56.000000000","tz":0},"subject":"Handle feature compatibility of HAProxy server-state-file option","message":"Handle feature compatibility of HAProxy server-state-file option\n\nhttps://review.opendev.org/c/openstack/octavia/+/805955 introduced\na new feature that requires HAProxy 1.6 or higher to work. This change\nensures that with older HAProxy version this feature does not get\nenabled to avoid errors.\n\nStory: 2010254\nTask: 46145\nChange-Id: Icc7dc9bb22187f908852c20415227574ded347aa\n(cherry picked from commit 69ecdb190880d0c49e9f0ef92fd53de14cf66720)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb2e1d2e770a400d921348ede035dcadc79c01f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb2e1d2e770a400d921348ede035dcadc79c01f6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
